)]}'
{"id":"openstack%2Fkayobe~964862","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~If77366a0f8bb5f6ac3c9951533a2d75c9b0780a9","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-10-29 01:25:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If77366a0f8bb5f6ac3c9951533a2d75c9b0780a9","subject":"neutron: Remove support for linux bridge mechanism driver","status":"MERGED","created":"2025-10-27 08:30:49.000000000","updated":"2025-10-29 01:26:28.000000000","submitted":"2025-10-29 01:25:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"964862","cherry_pick_of_change":964789,"cherry_pick_of_patch_set":1,"meta_rev_id":"e8efeb98db0b8237d689ed3123562cb856d8405d","_number":964862,"virtual_id_number":964862,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-29 01:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-10-27 09:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-10-27 14:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-27 10:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-10-27 09:20:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-27 13:10:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-27 09:20:27.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-10-27 10:24:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-10-27 10:58:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 14:00:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"640df91a46cc38cc3437586886c08c1cda708501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-27 08:30:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"de2c04e7528859cbe066815e8d03fce5cc728f75","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-10-27 09:20:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dcbf0e7b9cb2a94e01cbb29508f0b985327cdaea","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-27 10:24:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4a01a46cc584b6765818f925d247eef07408ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 10:58:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efd112d7954f464aa03b04caa90422a2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0185d7923064818a1403941dc9dd559 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61322fed74864244b23503043f7660dc : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e41ab68509f64bada661d5b9ef1588c8 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e1b3b730ef4a52935aa95baa044f4f : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47f4705236fc42fbaa730808cf0e3c50 : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/149a84fc2bbc46af8d262229ce122905 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aefb320122064061951da75150c6f0e4 : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cbfb28a4179c431b92c99f043de6377c : SUCCESS in 10m 07s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/591494dacd014d908613e4c8ab3b6367 : SUCCESS in 15m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ffa24a018b104370ad05b2c371560420 : SUCCESS in 13m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcc1c714286b441e88b73c03146c7ece : SUCCESS in 10m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f255d43901bc470da66b93f41b02c4f1 : SUCCESS in 10m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3f2b3a4736154d8aa3b76f7a300bf5e5 : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1dc8101f7b445af9cbc6ab02b26ebde : SUCCESS in 13m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/71501e5093a944d68ad2aca80a20a8d1 : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/b45afb787ca24f9da48a25d43c5cd62b : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6b7b475b3d784000aab8a408fae54272 : SUCCESS in 1h 09m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92fd42d3a6f24e3792b665d138407d17 : SUCCESS in 55m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ffa4aaa4ec74d68a81757e5b22517d4 : SUCCESS in 58m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fe9009edd6b745e8bb132c4f14dc67a3 : SUCCESS in 2h 22m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1cec32b213844e21b4963929b1a9c8fd : SUCCESS in 2h 21m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e31f04e786b4b4883d1e072dbf4d2d3 : SUCCESS in 2h 25m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fd5658a180641b7b1ae579314948e3a : SUCCESS in 2h 14m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f7e11c629cd34da6ad2a077ce92b2652 : SUCCESS in 16m 43s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/bad95d5850164a6bb8708c531c2fc3eb : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d86777b8d4bb4f6bb38ca310a186523c : SUCCESS in 15m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1320801f7ee476b82fd2796a46e66d5 : FAILURE in 11m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/08468029cb5248d38963c9872a9cb349 : SUCCESS in 33m 36s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e121dac3fad74df2ac8811e334f549ef : SUCCESS in 29m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a699aaa5c5be4efe8b0389ffb8e696d9 : SUCCESS in 19m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/963c1661db404d93aec567cfd35ab6b8 : SUCCESS in 16m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d754d60b831438fafe31493b23dc85d : SUCCESS in 7m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0701764a21574f9288b2b3ce12895169 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3443bbdc161b46e74da8d196f9cfd94166bbe3a3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-27 13:10:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f8935c7ae18d3d9dbb875348e012f1bd493f48f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-27 13:10:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"69ea55e7e2845d05a1cacb5c85f802874b7e367d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-27 14:00:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5e33aadd83e223f355e109ee15072ec0db170b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:30:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41e4ee97d7324067899b413e19f4d1a7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c7dafa439a9490aada3def7e18299db : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9df5f53d0d34cdca2c37c63fb5fda2d : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8d28a69fcef4ab4967d0ee725cc2809 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e39f827df3415c900b5e3062ce4851 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d882e0128077465a840e590f98a5f93e : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f388b59668e411ab99dd3ecd29e582e : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a57942c9ad340ee861bc3a57e191b2b : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2d2c41c990245f7966a503c54c062b7 : SUCCESS in 5m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/929fad193ce649428fb9e79d9fb49fe5 : SUCCESS in 12m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9dc216ec37e6412489bbab77d1c45b6f : SUCCESS in 13m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74716ee167874709a6d64024fafc5805 : SUCCESS in 5m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3410ede06f3e4319a53983f38386b062 : SUCCESS in 16m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f23df1e62ee249a088c5d53baae98796 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2834c2c6159947038f9b725741a93b65 : SUCCESS in 10m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cf57f40c9c57466e916b4ac0cb513697 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/067547691a6d4bd6b80735454a9c79f6 : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/40eb3014be9f434d9055fe8fb51f5231 : SUCCESS in 1h 16m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54d9fa10a78b43209fd445c402045c6f : SUCCESS in 58m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c59a95964e2e47c98e862b60d5bee956 : SUCCESS in 1h 00m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e8025091ea1d468285de609c763f6c0a : SUCCESS in 1h 59m 41s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/dae17f85df054973bbcbfbb4414be67a : SUCCESS in 2h 15m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dae727634c34424a8e6f30bb44ccf5d : SUCCESS in 2h 08m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6e182633df0422ab0f5ccbf1265f735 : SUCCESS in 2h 01m 05s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ec38f06f4a2047879340ecc4ca5ca809 : SUCCESS in 8m 23s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/89e6c83753bb4d1cb18371818f0ee959 : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94b4f94cd66246d082966d332aeda335 : SUCCESS in 18m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ee4c0608dc74b2f995b10d71acfeadc : SUCCESS in 14m 34s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0ead1880132d46afb9517a24fef35ddd : SUCCESS in 33m 29s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/2c5e39371f244c588c4ec560147fe672 : SUCCESS in 43m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f111850e78374409b0660b9d13633509 : SUCCESS in 32m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/91255e424c8f4988ae808a25e9368308 : SUCCESS in 11m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/50f9a609f7ea4e45bbce5fe16e19f2dd : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6ebb49202ee4890826f26c19a2572f9 : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":1},{"id":"35e4d72c40e3a1ead856e2993c118918451ba350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 15:30:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14fc59151d33ff6a29f4bfbf69f467ea5eddb460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:56:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a79f9bd07a34d049c0f6e1c35dff737\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d92e7234f53b44399d4d63de488691da : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b07ba04ba0304b77b9bc2e3f8b8d82fa : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13f5080dc09f4e1eb1c2f1c51aaf40c7 : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bcb4a5e3c824c2ebe5182351691845a : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6519b86ac0cd4fe58fb047f3191c3d41 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/373b212f5f0d47b2aad2d60119ca868b : SUCCESS in 12m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fabd6ba66b834178848803d342815c36 : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f2373276cb044c481a6be2df5019efa : SUCCESS in 8m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5712c791cbf4b8db8eb2fbf5e16c3a0 : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63d3a5630b9e4167a0cbd16c7d7508b2 : SUCCESS in 11m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e80ea61de0e44f269499d0322952a65e : SUCCESS in 21m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/24bd42ac06674199828a450c4297b08c : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d364bc328d9c492ca562fedb3578b4c7 : SUCCESS in 13m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/979cb1d662da4b9a86c3877403dd53f6 : SUCCESS in 57m 40s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/cc17ae2fd8244a2f9f74cbb08419ba15 : FAILURE in 18m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e2b891049ba34ba48597bb3682ec8d28 : SUCCESS in 1h 10m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0beb07b3116245a5b1fb65e603b0276f : SUCCESS in 54m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/610208dd8f7141889a77726560aea9bc : SUCCESS in 54m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/94c2a8599a6f4db694221e6b57bec81e : SUCCESS in 2h 19m 35s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/acfc7ad32add4a94ad75e196bb6ba07a : SUCCESS in 2h 21m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04e8e3ecc41f436f8354fa28538c3d77 : SUCCESS in 2h 07m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c36cb31ce7614ac29246b253d44c070a : SUCCESS in 1h 10m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2d0c092a55da43da8c73a44f24e34097 : SUCCESS in 18m 09s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/87ef359be6e540c484027893136277a2 : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5c2ce3a6fae4c6b80f4c1a4eec7bfca : SUCCESS in 15m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/623a6fcd26f44d828a9b802d62e3949a : SUCCESS in 15m 56s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/46824088d3ce4d1692f1eec6e2b702df : SUCCESS in 40m 31s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8f7706ac06a74463a09a3394612ba16d : SUCCESS in 34m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fe339ab1c4041e3a1442275823ed28f : SUCCESS in 38m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/abe8743ac7e34cae95da8977a7045ed1 : SUCCESS in 8m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99a73e1fe3994a37ba95d4a5bb0fc329 : SUCCESS in 11m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f0e9ec01f4442f4b5f27cfd9719ea2a : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b81f2bedc59cf58afa59b7786c628b5a25d468f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-28 11:03:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6af3a2e5bf58f9f0c8d15a48c9d60c16b28a2c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 13:00:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/666a4ffcefff4efe91d6860f16d35a5e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e7f29c9fe8b4f329150031385c04142 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fec332d9b294286b6bf1b4d635f2e98 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/836e7af6acb548c98351a8698f65497b : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdeadd56d339489a85d7fe6c1aaea34a : SUCCESS in 9m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b354f8065ecc449588447be0da1f3222 : SUCCESS in 2m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49af07b46ee045a988c49c9a9d246153 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7dacf4f7cbbc40b18a98e33f07fc9b13 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97c104d6ef3f4f92bd09c4e2657f5c85 : SUCCESS in 12m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/20e2ef5b72684d609dbc8bd48e753947 : SUCCESS in 15m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7b5e66509cfe449abe6b6a773d5ac5c3 : SUCCESS in 12m 39s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9645c26d368c48d6925842c24dafa08c : SUCCESS in 5m 53s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/053d22cf0f9c430d91152ff1ba9cf18e : SUCCESS in 17m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1a9451a1b9c2429b89f352a29657a7b3 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2f9f80c261341999e0cd2ce492e00c7 : SUCCESS in 13m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2863960d508242799262fff9aca94686 : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/c8976e4ddde345f196b8979f78b8375d : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6f826449950943adb9f715c3d9667f8f : SUCCESS in 33m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e1a4f84e649488ab323afa336811562 : SUCCESS in 56m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c57d9f34cdd642bc9b81a453e386efb7 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7e01562095ca46f998d4affc56d5c34f : SUCCESS in 1h 52m 00s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/16dd763864e64bbf80316b5879b0f94d : SUCCESS in 1h 56m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a81dc6d7f5c43d4b22530aa24c45858 : SUCCESS in 1h 44m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b398fe4f7c2d4595ae32afa8a25f7d95 : NODE_FAILURE Node(set) request a334ca2caff04931bcd620f762e8ea08 failed in 0s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3bfe84648a4b4c708eb35c5b58575b0a : SUCCESS in 17m 34s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/8becbb770b7e4ae4ab6ae026d3be4e18 : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/925a88601f974e1b997a8b0f0ca33b2b : NODE_FAILURE Node(set) request 2211b30e55394d2d80a274bb75d61b31 failed in 0s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd431d8d706f4afdb3d36123bd12be32 : SUCCESS in 13m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4c1e44bcf6fe44b4a24fa32fc7131338 : SUCCESS in 36m 55s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ff5b4b45390f4c8992b262e98f62ece6 : SUCCESS in 20m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9ede62be5f74d0083226176472de384 : NODE_FAILURE Node(set) request d8d5e760fa01420a8425f7ec8a5c6af3 failed in 0s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6b3e82ddb01e41a792e2ef19ff881992 : SUCCESS in 11m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7584a1a0032d4e809ced89f368205a7e : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93fb04061a3d48b6a1947b28fc6a4b1c : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"412ee149ebabdff50936a7245baa822fde92c0e9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-28 13:37:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bcfa39ec6337879343492a7e3746f16e7fb987d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 16:17:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe29d0afe40546898ed3a7b07cb7733d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43aad5b7b7d54aabbd3d75a14dc0f1a3 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3896e7432b5e45a196e64929d5ddcb77 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d5c9b2f50cb4a9d86198a50b047690b : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1052f0ce5e4bc88e87febae9fb82f3 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c236807bd4b439c8c5d51180076fa62 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c7cda22713c419fb333982b68b4d7f7 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a54496d3fa064ea6b6b031603b815fa4 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ddde872c4a84055b2d12f1a81dfe8b8 : SUCCESS in 8m 59s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4bca8f0d7cf147aca9bf1525a5280f2b : SUCCESS in 7m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/369a516e77f64ca881f4ab9a228a728b : SUCCESS in 10m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7df063757a34205bc5d7913de822e3b : SUCCESS in 6m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/caf805f35dac46578f3b2eb4b05b3d57 : SUCCESS in 18m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5b549ca3bc004c48b7d73d537592b850 : SUCCESS in 17m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0f1c307638a4ef185b10eddb3e3c89f : SUCCESS in 12m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8533305078ca42b689c9e0b2a68c0372 : SUCCESS in 30m 29s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/97d6fbf92c2842449ac682d3db78c16c : SUCCESS in 47m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5ec4de3b57c14ae0b031aa24f9f5bda0 : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14a2beaeca0c4fb4918dc44025d272c1 : FAILURE in 7m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/db57631e9ba64cdb8fd4ba7e6aad9365 : SUCCESS in 50m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4097ccac00194a8da99a842089d132d7 : SUCCESS in 2h 09m 42s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/df7096bd6ad94b668076e61a4da3ec1f : SUCCESS in 2h 14m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d730003e5357458fbfc9a36f9dd910be : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8fb181a85fa4b7fb6be83ffa278cfc7 : SUCCESS in 2h 30m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c3bb7813ae5748bbb4725a2d133c3e59 : SUCCESS in 8m 42s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/5423ea1e53654ef3ad695a4d9647cbf6 : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1652082cab3c44c3818046a0d89c6993 : SUCCESS in 7m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/91572eefd38c40a18a828558c9032077 : SUCCESS in 14m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/337510221e1a44be876c1d58aa1cfbac : SUCCESS in 39m 19s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3e3c72c740164de182be36791df8da98 : SUCCESS in 34m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb55e5aac4df40198779db81d165b4e1 : SUCCESS in 25m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dde50a3fb1274d32a6fad6cb72805af7 : SUCCESS in 11m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/99501bfe40f242acaf57f626f0ccab2c : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d05911b15d304fb48f2e17b0a01743c4 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9bb200f719e777509bdcc6fea9e36919d617a0cf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-28 20:27:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b82a20d7a6d3304876897a2f870c46d356e0d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:49:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a97cf0edf80b41e797b1761b5ac3a4e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6117d2babe0e4f89b956ce950cfacd84 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0307a5e2e54a1a89c2d65b60e911d0 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/871f5a7056ba4b61a3e21334fe2a79d6 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004722d32140421db290d3a0cfa6b52b : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac1064daae874e2da9553cf54d081bc2 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/70b82df089b04751838ca998ffe9a371 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/490bf949b14c40e892a3fb5bbbc45b0c : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afbecd96b99648b180e9c60846d8395e : SUCCESS in 6m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/34d52caf3a294cd59fc146bfcdb0996c : SUCCESS in 13m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b61c7b90eff14883b4b115fe7ea79d00 : SUCCESS in 14m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42058d166076422c9731a8bde8f59da9 : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d85c19614f334617baa0551697488f10 : SUCCESS in 10m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f20b3b18c41147888bc9d85d23efae68 : SUCCESS in 19m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f1b368cdc4f4f5aaa43398ce8e53642 : SUCCESS in 14m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b31d03ecdf534e9c8f926bca0cfde838 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/468e9c2677964b02a5c6fdfe251ecaf8 : SUCCESS in 31m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/43209e4bb3dc4f9b93be295e4d8b84b6 : SUCCESS in 34m 12s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0959830a37346cfa9701370fa36d35c : SUCCESS in 1h 04m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b6178acddf84a48a4e012a675540337 : SUCCESS in 1h 06m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/17eedddba9c54fbeb5a140d32eae0514 : SUCCESS in 2h 03m 51s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/abad8ed787854e0aa8d87cd5ea74b6f4 : SUCCESS in 1h 14m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ed8cb099ff548a68ae88665d0ec3d22 : SUCCESS in 2h 09m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3687e8be40840288528a4d548b12dc1 : SUCCESS in 2h 11m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6d3b1f53de1140908dbc3e610b4cd3fd : SUCCESS in 17m 48s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f83d8aed356041f0a7dfd5b9229017c1 : SUCCESS in 19m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0d93ee49c034fdbb25cfbe7feec1af5 : SUCCESS in 16m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/109dc85fc91b4fd4962876bc814ce0d4 : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/adf5825381a14b65b8b2c17c3df64a87 : SUCCESS in 29m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c54177e2af1d44c286c5ea38e4b8bbe6 : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04ff64dee7be4340bbdfa2f5b00c19de : SUCCESS in 32m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d391d6d34639448c9ea68ed875bf7792 : SUCCESS in 16m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9187cbb2c40844e0bcccc41bb41995e1 : SUCCESS in 12m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2f5bbcc313a427bb1f633f9f0ec5ce5 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f599a189ae92f4966ed48f3c1e21be0c870cefeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 22:50:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82456c95800bdef12cb63af9ae6ab834269da2cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 01:25:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f77f90c11b024bb284a6ab0d7aa1c8a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c4b7a8139c4a3ba4e9f3d3e12346a5 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ef903af7f54480582053f6e09ea0c8e : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cae66e50a214a8e98971e6a476f118a : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f549cf01e234dbcbe32e771cea91d99 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fdc787331c464bbfa239725318db2351 : SUCCESS in 3m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d76ce27f0df4aedae18191a01de31dd : SUCCESS in 6m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/814ca336298e4679937525fcc8b2cd8d : SUCCESS in 9m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26083547586b451da36d70398e7342b4 : SUCCESS in 7m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/af8c0a93fcd840dead6ff2337fa56e02 : SUCCESS in 6m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/88233059d5a143fbaaba4fbd6ed0c74a : SUCCESS in 11m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/61d0868f2e004e6dba549da23548d14b : SUCCESS in 9m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ff952b9c93f34e0fb0abd6a773c3bb8f : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc31db9d9fa7482b9ec3e0a4ba3f5d00 : SUCCESS in 16m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/355e976636e84962b5b325b5fee5f663 : SUCCESS in 46m 54s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/087770795d7c411f9149788cc0556ede : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/984909f9725b4c558ebfa9d215733159 : SUCCESS in 1h 11m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c0d5dfa9a7d4b12b4617ffd28a802c0 : SUCCESS in 59m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f90456342d824f8eb1964b023b29509b : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1f6c45b9c8eb48829d52fd7ca8b5611f : SUCCESS in 2h 30m 32s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/cdef3af6238740ea9cd6a08b455a0bca : SUCCESS in 2h 21m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cad9ccc4f7d441a8aa8907374c7bc460 : SUCCESS in 2h 22m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e2022ae0765491f9584dc6e9ec8057d : SUCCESS in 2h 17m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5befabeccbec4ea984297594a50ba37c : SUCCESS in 18m 48s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/1c7ae61516eb4adc8fb1037ffa338da5 : SUCCESS in 18m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22727186d6a84a2f9226e3d7d009b766 : SUCCESS in 14m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d06a191965d8476db622b245ccb5e092 : SUCCESS in 7m 18s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0e1bd3421b9f4f7a802134b5d6037ac2 : SUCCESS in 35m 37s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c6b425aca7b24d14a6e3dab20d48fb40 : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44dee16b7e2f4f55b4e11153c0075dba : SUCCESS in 26m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0389ef6629d143a4841fe4727df6ab42 : SUCCESS in 14m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e435ea8418be4f83a7cd40e726af72e5 : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82ffa8b178fb4bd69bd901f1b1a97dbf : SUCCESS in 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"aeff8e03394e862f071c1bd035cf90851be1ff0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 01:25:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8efeb98db0b8237d689ed3123562cb856d8405d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 01:26:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d0a1ab03424a5e8d418e349ccfdd91\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79294d6cf4e944ec9bdd79cf167b2b01 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/975377d3a04a440aace1df63f14a8e28 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a384a09261b2553a6f78f647047f3b37d9997482","revisions":{"a384a09261b2553a6f78f647047f3b37d9997482":{"kind":"REWORK","_number":1,"created":"2025-10-27 08:30:49.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/62/964862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/964862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/964862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/964862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/964862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/964862/1"}}},"commit":{"parents":[{"commit":"6a5514edd00990795ff435f4f686fb43d8642bea","subject":"Merge \"Bump Ceph package repository to Squid\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a5514edd00990795ff435f4f686fb43d8642bea"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-24 14:44:04.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-10-27 08:30:49.000000000","tz":0},"subject":"neutron: Remove support for linux bridge mechanism driver","message":"neutron: Remove support for linux bridge mechanism driver\n\nNeutron and kolla already removed it [1][2].\n\n[1] https://review.opendev.org/c/openstack/neutron/+/927216\n[2] https://review.opendev.org/c/openstack/kolla-ansible/+/962278\n\nChange-Id: If77366a0f8bb5f6ac3c9951533a2d75c9b0780a9\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit 2b875095344b9c029becce3d8f3fb720633fab35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a384a09261b2553a6f78f647047f3b37d9997482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a384a09261b2553a6f78f647047f3b37d9997482"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
