)]}'
{"id":"openstack%2Fkayobe~965243","triplet_id":"openstack%2Fkayobe~master~Icfbc3531d998234cbb90d9fb964ccb887b9368cd","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Icfbc3531d998234cbb90d9fb964ccb887b9368cd","subject":"Switch to CentOS Stream 10 IPA images","status":"NEW","created":"2025-10-29 14:41:56.000000000","updated":"2026-10-09 10:23:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"8548352480743c9e78f88a0cae17c789526c494f","_number":965243,"virtual_id_number":965243,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-10-29 16:18:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c709a00d9e44f1be9e321a43a54c8ff70358071","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-10-29 14:41:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b8e424f2465e1855cc7d527f81a3a6bc3fae48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-29 16:18:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b956f5d3554cfda37a6fad7e1bc161\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3fd67e7300b4548902c29ee00570572 : SUCCESS in 4m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7d1b18b24a22427ca6777b097452d47a : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68a5092cc6874f0682c961036c842370 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/079c418f623f4128ab790101077a2231 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84bd0f19c2c142469c62f60d217bf9fb : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/755cd399e75a415c9914be12db74d429 : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4dcac8094d044889a4d238c5f04d5413 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ead8f426c65247ab839f759d512708f4 : SUCCESS in 14m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/963271ab39e4494588c1602b2d22a144 : SUCCESS in 9m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e0922d1851fc43ab839871f94322bbac : SUCCESS in 8m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/00c852a2986b4d24980227003e0a9591 : SUCCESS in 13m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6e2ef2cd2c94e009bec88ec88f09fe8 : SUCCESS in 11m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9dfc3a003b1d4788aabda6ee588d2945 : SUCCESS in 9m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/25b6af7557a64abea4b2a01c45729da0 : SUCCESS in 18m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf4eeff467a04f36b3b9e840bcec82fe : SUCCESS in 16m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/84ca88a039eb40c3a839a12c463155e3 : SUCCESS in 26m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e8426eb7b85b4172ad9c1434f20e7ff8 : SUCCESS in 39m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3aa92d55cd5641b3be2ff35a26336a75 : SUCCESS in 51m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb53c12db7be4047aaf57d650d39f236 : SUCCESS in 49m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1471660091974f068eba9dcb3d61d06b : FAILURE in 16m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a09b0649a2a547f98a7f949b4da22490 : SUCCESS in 1h 32m 12s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fef490aa576742a8848541ecc646613b : SUCCESS in 18m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/29f9ebf5be04435cb137bd30662a3d99 : SUCCESS in 12m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fef433e01b224cad8d24ea530ccc7f01 : SUCCESS in 15m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/13c2f18a5c9845cc94168a13ba8d8abc : SUCCESS in 15m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8e6fe6c1452490ebd0432c229ad83ed : SUCCESS in 27m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/84c9734dbd4a4f4591e9d97f27c926d2 : SUCCESS in 9m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d249bd16a6ac46a192d42f02c7b40733 : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60cd7d50a796485e8921f6ca6ee08371 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"35fa2faca7cefda77335fbd9c03f7bd582ef3548","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-27 16:44:58.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"adf0fe2c007ae86c948cb720581127146c15ffd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 19:14:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45c6dd7cf5f1444e8eaf59be47f03ea3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c499a882728459f9e6c0fcbd797e044 : SUCCESS in 8m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fc44be1f806441a8bb211fee9a33219d : SUCCESS in 6m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/895de258da714e698870d203489d4473 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d382540b53d449eb744047d96550384 : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d541bf877d9a41fa9713c3f757f510c1 : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a93b0854db924ccb8e6b195caf9a383f : SUCCESS in 3m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7904e619b4a74aa28f90b5c0961d6fd3 : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d9b3e0d764a42df89d07cb7ffd81e4a : SUCCESS in 10m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ffbc49f119f147a3bc7298f32a3c7dee : SUCCESS in 15m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52e300b8b4c4498fb49864a577d60466 : SUCCESS in 6m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9e2189faf7364fd8af13605e78f90fb9 : SUCCESS in 11m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ad7fd5014524cea878c596c3b76d578 : SUCCESS in 7m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e387e213f2d144b4a75c64f8301f73cb : FAILURE in 1h 49m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a8d4f153b13e46a8b77a7a6e4de56a7e : FAILURE in 1h 57m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b75ccc1dbf4e47d08636584b5a164a2a : FAILURE in 1h 57m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4052a969b6054ce192a5cd5be0d142ad : FAILURE in 1h 48m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2161c5e5498f4cb990cf83b2a7f770df : FAILURE in 1h 46m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ffa181cd36cb485686e350a2049e5cdd : FAILURE in 2h 28m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f554ad5b46e04dc2a68451aaf015d7f3 : FAILURE in 2h 23m 43s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbf918405ebe4e4286d06936a5e3103d : FAILURE in 2h 19m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f2eb38276be4c10bb54723c08c0c036 : FAILURE in 2h 27m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/74526a61ffbe4610b09ec260c9c00202 : SUCCESS in 15m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/eef797129a7749e686448baa62a1b862 : SUCCESS in 16m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af3cf2c56f4248f38cc7c5a3bc572794 : SUCCESS in 18m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0d6fbe2b03e41d894d3dcf604eeb0b3 : SUCCESS in 15m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f0bba9a1c0d547108a5ca52e33c6c9db : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ab836c668679419fab0edb8dcb5db13e : SUCCESS in 28m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5c364b0944d46918ec96c519e8a044c : SUCCESS in 30m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/614b3f31e8dc4b6bba629f481b4c2f1e : SUCCESS in 26m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/97283866fe7c48a3b78147588679ded5 : SUCCESS in 14m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/572847bd9ec2446d9a5ee8435c9ac0d9 : SUCCESS in 12m 58s","accounts_in_message":[],"_revision_number":2},{"id":"cac39b2c45a12895f7095a5869ca1fe456623ccd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-21 10:09:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"342433af49551ebefa911081640801bca1977bdd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-21 10:13:20.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c681540cdea8f9c8464f4b5b755357579fcd8d0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 11:03:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5d5f763eaf14eca8dbaa1a19838f97f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775bed908e52468bafb3e700a7a58447 : SUCCESS in 8m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f708ea52b4b94e6f83c7d92e6d4e9f14 : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81f996f2cd424c3691e6f13b7b79227e : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50edee29c7d540beb3c23e06f5cb5e76 : SUCCESS in 5m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ecaaf5c6874a4234b3e2e629f2e5460f : SUCCESS in 3m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dc8f03fe58cc43dea3d8d5da86a04a3d : SUCCESS in 8m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db33d304f4354926a594f81728f032e8 : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/59ff8cef6c284695a4a4ea246d99c321 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b52c66549d664efcacc3e99e597f8d82 : SUCCESS in 5m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83c0a32a1643476ca5c39ecac561c11b : SUCCESS in 9m 30s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3d632be0a5a04d5ea856b83c75812f3a : SUCCESS in 10m 27s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a2bbb5d7195848eeba9785616607d1b6 : SUCCESS in 13m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f2e084d566f4f1fb6fcd80c2d54e17e : SUCCESS in 15m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/853bd3f3f6c349a08b012e08261b7eb0 : SUCCESS in 13m 07s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d1a4019a05a84baa9731d347d0a1acf9 : SUCCESS in 10m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6005e3652e3b4c24ab9596c58f2bc83f : SUCCESS in 14m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c98f146ad8f84a18b48762b5cb203815 : FAILURE in 20m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e6e5c7be78d543f1a8b8e6684592b535 : FAILURE in 20m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bb55abf6b9b44d2ab2aa8b777e87eb36 : FAILURE in 25m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ccb5b8402e24aa9a0d87de4d56a3e29 : FAILURE in 21m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1cae463c1ad04c2fafdee2ecd33a6093 : FAILURE in 21m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/775d6ee2051a427a8df3cd0e29f55f3d : FAILURE in 52m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8d45f7cb40cb420ab872a792ac0ddc2e : FAILURE in 46m 35s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/959fb45b4df0457387d0ada10909e88f : FAILURE in 45m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e7530491a6742ceb99581a306ae92ab : FAILURE in 50m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4e4cfefc3bc7463597959945fa2734f4 : SUCCESS in 20m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/57833343aac8402e867fb5188d62c4e3 : SUCCESS in 21m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dc59f636ab5403fa677cc7965a4230e : SUCCESS in 21m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9f6001a3dd794b69a19183986aa46a40 : SUCCESS in 8m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6f908277a38d453693302487e8f0aa18 : SUCCESS in 15m 20s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6ab810fb54b74e90a823dcf9e456992e : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e59fd03f47784aeab6b416502f15fc1f : SUCCESS in 34m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a34d9a9b52184515a9290627728eaace : SUCCESS in 16m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca20ea1b0c5e442d95e120759cf3a21f : SUCCESS in 20m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09f5f5b27da54a62b424ff9e927d155e : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3fba2b07899612c15c813b55cda44cec6d09608e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-10-03 08:30:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"7300c9e211de4ae8639a7ed122dccef658bae156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-03 10:29:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7faef64f9348debbba68176ca6494f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8931505f33c244729f08d82abb9033f0 : SUCCESS in 9m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/84e189d355b54a7aa4855f22584f5aee : SUCCESS in 7m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/893728ed994649fcaf6292078c3791bf : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ab821bd24e04a2da579d20beffa3d4d : SUCCESS in 5m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5b8af7e2669c45b28f582bec80483df4 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c39118c6bdb342509c39c0d72834e0d3 : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2e2ea05b98642649d6ad1e566b793dc : SUCCESS in 13m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87b67f3bd7aa4d9690c95ceef37df7cb : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/80277ff009114080841ab6351e172d8a : SUCCESS in 16m 33s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/83b30c2f0128474db4b9aa20743d8d10 : SUCCESS in 21m 56s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/977aa936f97d4ea3855a32d71d744282 : SUCCESS in 22m 13s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da10decb07b84219b815b5fcb9adb355 : SUCCESS in 10m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ceebc6e87b2645e1976a2fbff7254327 : SUCCESS in 19m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/631240626490450a8de8ddc75fe3efa2 : SUCCESS in 24m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/861f7768e0654370acf0a3b8b643c5bc : SUCCESS in 16m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4cda1bf55bb247ca960bafa3fba7739e : SUCCESS in 34m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/313c5780156f445fae05b131b94d2792 : SUCCESS in 27m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/acfd040a08d7441c9a0f1c49f249eb48 : SUCCESS in 58m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9bd44869f3db40c6b283a64429d581bc : SUCCESS in 58m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a5b0d7267c2f4653ab459f28773f73ce : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c09ab75f3e4446a990ff88234629a45 : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0ebebb79cb2248ec845bbbd498110b14 : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8d3e1d5fb1ab4e6397af2eff99ce1444 : SUCCESS in 1h 50m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aac5ed8f8655484a83188d532a5252c6 : SUCCESS in 1h 57m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/844f3d926046416a9b512b399d309b6d : SUCCESS in 22m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2f7b51d7dd4e41f2aca606f1d80a8ba9 : SUCCESS in 14m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3fe98eec953f40569f640020bc8b4ee6 : SUCCESS in 23m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3cbe82ada0b140938a1be77d8027fcdd : SUCCESS in 22m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c4f4c454a7d9455c889ab4146ed971ea : SUCCESS in 40m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28f355b088744966ab014fde3cc8d951 : SUCCESS in 36m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/97d1c4e9a59d4c13b28dcc4d4c0720a0 : SUCCESS in 21m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f759efcb6b64ec49dbe383f2af6b761 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":4},{"id":"8548352480743c9e78f88a0cae17c789526c494f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-10-09 10:23:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5a4675139d926797a730a889d4f8c50939b79b53","revisions":{"d69178a82ae54d4f3ddc1c20185cb8929e870b5b":{"kind":"REWORK","_number":1,"created":"2025-10-29 14:41:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/965243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/965243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/965243/1"}}},"commit":{"parents":[{"commit":"ac156f45c309cd15dffaa00735c0cdbd6b40b93a","subject":"Merge \"Account for encoding in _get_direct_url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac156f45c309cd15dffaa00735c0cdbd6b40b93a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-29 14:41:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-10-29 14:41:14.000000000","tz":60},"subject":"Switch to CentOS Stream 10 IPA images","message":"Switch to CentOS Stream 10 IPA images\n\nChange-Id: Icfbc3531d998234cbb90d9fb964ccb887b9368cd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d69178a82ae54d4f3ddc1c20185cb8929e870b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d69178a82ae54d4f3ddc1c20185cb8929e870b5b"}]},"branch":"refs/heads/master"},"0dcb1eea417446c52d3baa55eeab57c242aca207":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-27 16:44:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/965243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/965243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/965243/2"}}},"commit":{"parents":[{"commit":"b622c92b37b16148505821f8eeec8f9ac6d78422","subject":"Merge \"Add support for url sources in kolla_sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b622c92b37b16148505821f8eeec8f9ac6d78422"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 16:44:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 16:44:29.000000000","tz":60},"subject":"Switch to CentOS Stream 10 IPA images","message":"Switch to CentOS Stream 10 IPA images\n\nChange-Id: Icfbc3531d998234cbb90d9fb964ccb887b9368cd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0dcb1eea417446c52d3baa55eeab57c242aca207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0dcb1eea417446c52d3baa55eeab57c242aca207"}]},"branch":"refs/heads/master"},"1a391265ac7d2792d93a857faba62eb78244a4ce":{"kind":"REWORK","_number":3,"created":"2026-05-21 10:09:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/965243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/965243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/965243/3"}}},"commit":{"parents":[{"commit":"70552faa0b9f1ce82380115b2d7bf83ea298a63b","subject":"Merge \"Fix broken conditional release note formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70552faa0b9f1ce82380115b2d7bf83ea298a63b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-21 10:09:33.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-21 10:09:33.000000000","tz":120},"subject":"Switch to CentOS Stream 10 IPA images","message":"Switch to CentOS Stream 10 IPA images\n\nChange-Id: Icfbc3531d998234cbb90d9fb964ccb887b9368cd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a391265ac7d2792d93a857faba62eb78244a4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a391265ac7d2792d93a857faba62eb78244a4ce"}]},"branch":"refs/heads/master"},"6d158f85807fb40ca2fbbecb0f9dad062a9431a7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-10-03 08:30:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/965243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/965243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/965243/4"}}},"commit":{"parents":[{"commit":"1bb276b3d16a5807c1dcfc76c1a62879a872b737","subject":"Merge \"Remove ALLOW_BROKEN_CONDITIONALS workaround\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1bb276b3d16a5807c1dcfc76c1a62879a872b737"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-21 10:09:33.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-10-03 08:30:56.000000000","tz":0},"subject":"Switch to CentOS Stream 10 IPA images","message":"Switch to CentOS Stream 10 IPA images\n\nChange-Id: Icfbc3531d998234cbb90d9fb964ccb887b9368cd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d158f85807fb40ca2fbbecb0f9dad062a9431a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d158f85807fb40ca2fbbecb0f9dad062a9431a7"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"70552faa0b9f1ce82380115b2d7bf83ea298a63b","ours":"1a391265ac7d2792d93a857faba62eb78244a4ce","theirs":"1bb276b3d16a5807c1dcfc76c1a62879a872b737","merge_strategy":"recursive","contains_conflicts":false}},"5a4675139d926797a730a889d4f8c50939b79b53":{"kind":"REWORK","_number":5,"created":"2026-10-09 10:23:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/965243/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/965243/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/965243/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/965243/5"}}},"commit":{"parents":[{"commit":"ba1598b4f0166e877c87be110979fdd14e51487b","subject":"Merge \"Fix merge_configs/yaml and concat behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba1598b4f0166e877c87be110979fdd14e51487b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-10-09 10:23:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-10-09 10:23:36.000000000","tz":120},"subject":"Switch to CentOS Stream 10 IPA images","message":"Switch to CentOS Stream 10 IPA images\n\nChange-Id: Icfbc3531d998234cbb90d9fb964ccb887b9368cd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a4675139d926797a730a889d4f8c50939b79b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a4675139d926797a730a889d4f8c50939b79b53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY"},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
