)]}'
{"id":"openstack%2Fkayobe~965875","triplet_id":"openstack%2Fkayobe~master~I223c5f4d1a1ec8629ef841950f0ea78c5238d1d1","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I223c5f4d1a1ec8629ef841950f0ea78c5238d1d1","subject":"WIP: Support baremetal port configuration","status":"NEW","created":"2025-11-01 17:14:02.000000000","updated":"2025-11-01 22:18:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"34326a2d7f472bb43a91689038c18dd71a1c5f04","_number":965875,"virtual_id_number":965875,"owner":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-01 22:18:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-11-01 18:39:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b0fc7c7c2a3fa3ae9a32098c1b60ee7b193209e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 17:14:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5c9accbb1696d39c9b9ef5177f1bc90e1a77e00","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 17:14:21.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9b1c256b08855949ee7a6912e0c64d3cec066be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 18:39:13.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8708d0b0a8b9495f9e8b9a389ff89087\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/63adbc2bbfc4493ea1c73f4d364aa362 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7576b0969f4fcea9397ee3b3c8aa7b : SUCCESS in 8m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/599100f701bd4e53bcfbd77100f2e920 : SUCCESS in 8m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ed87fa936974ac7ae73d8813f09e465 : FAILURE in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a716be79264942a99d6b8e44d963faa9 : FAILURE in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f70e028429d4d1e847fde7ce1093780 : FAILURE in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/003c1f08d27b46c9bf291c1b46117130 : FAILURE in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4460284ae0f46f7929e11219fc9f5e5 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2cb7eeeb691141d09477f10a2cea3d8c : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a3aa83ee720b4c92bf286fd3befb527f : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7f7e73b05b3647bd9b2ab25d5749fba0 : SUCCESS in 12m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b7ca49d16ecb4719bf3ce469cc1dd26a : SUCCESS in 13m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3fbc6fc100145c4905768e1009e1d9e : SUCCESS in 9m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e44f5edfb41946629e6c5710417b7116 : SUCCESS in 15m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fb2271bf118c40dc9617d4ad2b6a76fa : SUCCESS in 17m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7c669c4db0a4d6a87f1ef3496ceb9dc : SUCCESS in 13m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dba6cc38bdc0412bbb2bf49db2572406 : SUCCESS in 46m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/718f0fe3e74f42289c1c4012fc67efb3 : SUCCESS in 47m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ca1a808a3d294b819c925e7031f92907 : SUCCESS in 47m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c93f94b21ede43a0bf3626372fa12874 : SUCCESS in 49m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1d706e4a91c44e529b1c6d33c7174cfb : SUCCESS in 51m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d610722e9b834d2b9add80feb6f2b3c5 : FAILURE in 1h 20m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6b3ffa67085f4e84982adc6dfc89e334 : SUCCESS in 11m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/49938fc0964f401b9466dccc492885d4 : SUCCESS in 12m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28cb26b48bcc4ec2969bb78867c21692 : SUCCESS in 14m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/da55c38abab54c959a34cbe60bbf531e : SUCCESS in 13m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f7fac24929f43b0941d226872e6a545 : SUCCESS in 28m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1282c977094a426b9d1f899b101de574 : SUCCESS in 12m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cf70a1b51fd3487b805f688cbab7d7f9 : SUCCESS in 16m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84b868a1fea244fab6821470e2445f96 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d33da72f2705c4c7306cf3ceebd4857715a67654","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 20:02:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dad588fe6e02c75a0b7859156913fa260b75658b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 20:04:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d5752e0b5c6d1466ded70043f8761537edd741f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"date":"2025-11-01 21:03:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"34326a2d7f472bb43a91689038c18dd71a1c5f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-01 22:18:58.000000000","message":"Patch Set 4: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16beac9b3c80467fa12fbcc874298aad\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dbeebca3a20b4cf695c772c5f1437dac : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cdbb89e004b4c5eb104f90290ddd437 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/091acece7257499abc51b92df26cf6ed : FAILURE in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94d12304ce5f4e489bb713cfa3c276cd : FAILURE in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b23b6612e6ba4f5190eb6ff0b352a45e : FAILURE in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/598a6f3cc1834ff4b0a57233285dad15 : FAILURE in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3dbd2ea0799d47449bc6e76b97245f54 : FAILURE in 7m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e91e964111ad4805b52fe03f8562aea3 : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/937d027dda1c4f779f6b71923c37ec37 : SUCCESS in 14m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/333e9792fa48428e8ebdd4cc33cdeacf : SUCCESS in 12m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4cc4a5e51caa49338a93d8c631c34409 : SUCCESS in 16m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad521b17298c459cbed51500015b2a70 : SUCCESS in 12m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f1f3c6ca049412e98d3920fa4463735 : SUCCESS in 11m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7757d1a1e94d4116a967b49935131b4b : SUCCESS in 20m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9c5dbb69548743bcbfe519c81a8e9081 : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13692a5af027408f91d3e19d726ed701 : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/520e2eade4194c01974d4dc71332213d : SUCCESS in 34m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1dac7c5be97447b8b771bf5c424fcae3 : SUCCESS in 58m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/34100de0f1e442fab4d4ed74f19603ea : SUCCESS in 38m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fab1a74934984e2782b7e45e74cc7878 : SUCCESS in 50m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3ce1173a0bc4bf99c22d2a4f6a160f3 : SUCCESS in 24m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10dd5c86c822493db45228b3069ff38e : FAILURE in 1h 11m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0bbc8a6452e44457ac34a99d3a7a0922 : SUCCESS in 10m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ac91f173b6a644bdb97b2601aadd58de : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e40c0b31104b467fbbab9e8cdca77e82 : SUCCESS in 12m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b29872b9aacb4114bbfd7a0bc541e907 : SUCCESS in 16m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b68fb205d2b4dd2a9c32de64af9bf2f : SUCCESS in 26m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4f3734496b63405993c9e5ebbfce7250 : SUCCESS in 7m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f39e88644c84be78a2c8b187244cf69 : SUCCESS in 6m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97a8a4ba47244171b7137eda75550a8e : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2cff31eb6ca57a97dce0754b8157dd2b1e475d14","revisions":{"fa2748043d7278837a793979fc44bcd2347e1fb2":{"kind":"REWORK","_number":1,"created":"2025-11-01 17:14:02.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/75/965875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/965875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/965875/1"}}},"commit":{"parents":[{"commit":"84d743689a97dfccab9f83f0c2f3fff85a9b8592","subject":"Refactor baremetal node registration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84d743689a97dfccab9f83f0c2f3fff85a9b8592"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 17:12:39.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 17:12:39.000000000","tz":0},"subject":"WIP: Support baremetal port configuration","message":"WIP: Support baremetal port configuration\n\nAllow baremetal port configuration to set attributes like physnets,\nPXE support and switch link information.\n\nChange-Id: I223c5f4d1a1ec8629ef841950f0ea78c5238d1d1\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa2748043d7278837a793979fc44bcd2347e1fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa2748043d7278837a793979fc44bcd2347e1fb2"}]},"branch":"refs/heads/master"},"743dca0a13989b6a609d0f9311669c2799e59012":{"kind":"REWORK","_number":2,"created":"2025-11-01 20:02:19.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/75/965875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/965875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/965875/2"}}},"commit":{"parents":[{"commit":"84d743689a97dfccab9f83f0c2f3fff85a9b8592","subject":"Refactor baremetal node registration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84d743689a97dfccab9f83f0c2f3fff85a9b8592"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 17:12:39.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 20:02:12.000000000","tz":0},"subject":"WIP: Support baremetal port configuration","message":"WIP: Support baremetal port configuration\n\nAllow baremetal port configuration to set attributes like physnets,\nPXE support and switch link information.\n\nChange-Id: I223c5f4d1a1ec8629ef841950f0ea78c5238d1d1\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/743dca0a13989b6a609d0f9311669c2799e59012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/743dca0a13989b6a609d0f9311669c2799e59012"}]},"branch":"refs/heads/master"},"2525011009a455e60537119aa2eec8d640531b64":{"kind":"REWORK","_number":3,"created":"2025-11-01 20:04:07.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/75/965875/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/965875/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/965875/3"}}},"commit":{"parents":[{"commit":"84d743689a97dfccab9f83f0c2f3fff85a9b8592","subject":"Refactor baremetal node registration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84d743689a97dfccab9f83f0c2f3fff85a9b8592"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 17:12:39.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 20:04:00.000000000","tz":0},"subject":"WIP: Support baremetal port configuration","message":"WIP: Support baremetal port configuration\n\nAllow baremetal port configuration to set attributes like physnets,\nPXE support and switch link information.\n\nChange-Id: I223c5f4d1a1ec8629ef841950f0ea78c5238d1d1\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2525011009a455e60537119aa2eec8d640531b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2525011009a455e60537119aa2eec8d640531b64"}]},"branch":"refs/heads/master"},"2cff31eb6ca57a97dce0754b8157dd2b1e475d14":{"kind":"NO_CHANGE","_number":4,"created":"2025-11-01 21:03:00.000000000","uploader":{"_account_id":36238,"name":"Hollie Hutchinson","display_name":"Hollie Hutchinson","email":"hollie@stackhpc.com","username":"assumptions"},"ref":"refs/changes/75/965875/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/965875/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/965875/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/965875/4"}}},"commit":{"parents":[{"commit":"a62811949217d463b661acbd71f644f18c3ed70b","subject":"Refactor baremetal node registration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a62811949217d463b661acbd71f644f18c3ed70b"}]}],"author":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 17:12:39.000000000","tz":0},"committer":{"name":"Hollie Hutchinson","email":"hollie@stackhpc.com","date":"2025-11-01 21:03:00.000000000","tz":0},"subject":"WIP: Support baremetal port configuration","message":"WIP: Support baremetal port configuration\n\nAllow baremetal port configuration to set attributes like physnets,\nPXE support and switch link information.\n\nChange-Id: I223c5f4d1a1ec8629ef841950f0ea78c5238d1d1\nSigned-off-by: Hollie Hutchinson \u003chollie@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cff31eb6ca57a97dce0754b8157dd2b1e475d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cff31eb6ca57a97dce0754b8157dd2b1e475d14"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
