)]}'
{"id":"openstack%2Fkayobe~966093","triplet_id":"openstack%2Fkayobe~master~I39eec23d9ca3f8fe9b8d7e9693082ef5841dcffa","project":"openstack/kayobe","branch":"master","topic":"fix_venv_setup","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-07 18:29:13.000000000","reason":"Change was submitted"},"38563":{"account":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"last_update":"2025-11-07 18:29:13.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-07 18:29:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I39eec23d9ca3f8fe9b8d7e9693082ef5841dcffa","subject":"Always call virtualenv as a Python module during setup","status":"MERGED","created":"2025-11-04 17:27:27.000000000","updated":"2025-11-07 18:30:06.000000000","submitted":"2025-11-07 18:29:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"966093-fix_venv_setup","meta_rev_id":"9a3c67a563308ffb7536b1c118f3e77d1242580c","_number":966093,"virtual_id_number":966093,"owner":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-07 18:29:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2025-11-07 14:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2025-11-07 14:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-07 14:43:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-07 14:43:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-04 18:48:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-05 07:37:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-11-07 10:11:20.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-07 12:54:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-11-07 14:44:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"0614821cb88039e8fc568d0258a7ea1560fee2c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"date":"2025-11-04 17:27:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a7a86601f1628d5e3ffa702c24767af0a5d434d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"date":"2025-11-04 17:27:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d30ba5fb93d84d1ca4871aea8e82df929e803fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 18:48:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c106f486f2443d912476f1a7ca962f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0132e739c1034716b6c293224ab92f4d : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/78d9489390e442b9965c5a62c599f28c : SUCCESS in 6m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dea2908898d4d1da560b128d035b3e8 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114e8f00b3a944e48bef5d171a410e8b : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8924c9367a2b4611b278207799bb4ee8 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9460801ad4c94ea9a36ecaa2db90e970 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2f9aa4706dc4903aa0b061378e54b2d : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19f1ac7a85b84963bec50e8c1a378671 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1e8fdd720c74e2b970754525412fb93 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c174670430394a2383f219ab4df2b9da : SUCCESS in 6m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6755b467f6d84aa9ad1d0a5f7ddd0389 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4ce8e228faf944a6a3268aeb787f7964 : SUCCESS in 12m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4709206da11a4b11a1a889b89db4c948 : SUCCESS in 11m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3ee6a4c6de59451e9d6a8229bcbfd570 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b728ee448d04821bc53ece5b833eeb8 : SUCCESS in 17m 34s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14b214d682d64119b2787e4ad9cc2f23 : SUCCESS in 13m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ebf156c90b274e0d8fdc0f2a5ba328de : SUCCESS in 41m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/18cc2d69b2db45c28f5084a4b49c5a61 : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/776c41ad51834077980f7fd689df47da : SUCCESS in 51m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b1e7febe12164cc09b437b1b05ade4ed : SUCCESS in 55m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/429056e68efe4afd8b04571115478ca5 : SUCCESS in 48m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23bb66d50cc54760b7a2cd618d587ed8 : SUCCESS in 1h 19m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/bb2b2642bf9147448a34e2b991636b98 : SUCCESS in 10m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bbf4b63cfed34cd59dcd53207cda0f1a : SUCCESS in 9m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d84445797eb847cfa36ecbaaa034dab2 : SUCCESS in 15m 17s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5a9af667c75045d686b2cd5de56ffece : SUCCESS in 13m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b51ef491849440e88590bc481eeb5ed3 : SUCCESS in 27m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bfefdaeaf0d44750a98c684ec4005b65 : SUCCESS in 9m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b7f981f9b0ee4f748c7734929b26248b : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3f5f0a62d4d14caba55364b579ed0ee1 : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5b6c389b5491a2ba909d1b4845807881f113ec9f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-05 07:37:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41040e3636b88646544137ba29806b95933503d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"date":"2025-11-06 07:48:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d1a1bf885ac1aaa8a88eb8193664d812eca62dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 09:28:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/717119e9260b4c61a76b190e5b2dcd3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4cb8437d43478fb980235204d45842 : SUCCESS in 7m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b534e149cff243fb86e38e1c075b3e0b : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75f638f5ab7645888c17803b86aa2047 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e412f8622364115b78c47d5e422511b : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2a3bd81759c4522a01807c4ad0b854c : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f52942770cf64389ad816dd39497b5a8 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bba024cee5ba4322979883c153bb9207 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9cc105b099ca4f65bd4ea9d7420db7b9 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/939a0685da8f40208cc468ebf50c4751 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/07bd7dfb673d42b3892b732f800d79e2 : SUCCESS in 9m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d93294d4b8b24263a0f056ea6098cec6 : SUCCESS in 9m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6086c0ce216f46a6a0e5dc22490f8945 : SUCCESS in 12m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b9d11df8c324caf883799d7d1b34cd6 : SUCCESS in 11m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/34367d190f8c437aac7926c53a40a992 : SUCCESS in 16m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4a6d4c1acb3b4aaab50c03c0165381ec : SUCCESS in 17m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31d4b6bb6db747848335eacfafe217b6 : SUCCESS in 13m 09s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/873a74b626f7445bbd846e184c67c1a8 : SUCCESS in 48m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c1c4934efa9c4ff3bea2ef8bccad9988 : SUCCESS in 1h 06m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/da60e9fd90f44492a396c7636d9e3d90 : SUCCESS in 49m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daf80f3664d84997a2acf10d055bfd15 : SUCCESS in 51m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f4f8d2dfa7e43da9289728d888b2663 : SUCCESS in 49m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5e3c951ee7c49d6809f9f39454ed9cd : SUCCESS in 1h 34m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6a628f626c2b4f11abc9fa228d8103ef : SUCCESS in 20m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2c249d96b55a461b807264634aeeeb83 : SUCCESS in 13m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a158ea65c6684b4d910aee6e25a69fbf : SUCCESS in 14m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a39a4374da8140e2a86c057f9967568f : SUCCESS in 12m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dcf6be84141490581a847d002fa5eca : SUCCESS in 22m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9e2e4e16de0342d0bbacf4f6c509ef03 : SUCCESS in 13m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3aa11843ee5448ae9dde86af893bb287 : SUCCESS in 16m 42s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95b0ee66d8434f048ed51c14508a4d1e : SUCCESS in 12m 29s","accounts_in_message":[],"_revision_number":3},{"id":"68c12db9761c3ee9c11a68717adf61eb1199b073","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-07 10:11:20.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ed7bab3b7069d6ecfff7af0d29c666e8fb40ba67","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-07 12:54:55.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"06a1b17ddaab70d7aa59ae145d0907194b93f0da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 12:55:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0fdcd02266d4c1ea5aa4f6c8dbf38a9f91a37549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 14:28:16.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4d286debce84c8fbbe5ea7e140c6494\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e6f940f0704ff0b3a100fcc869e4cf : SUCCESS in 9m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/33e442e9fb254d4084af7b19a41e9c6a : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3604c0d953742a58cfd80d32facdf32 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c45ddad461845d48f62f698206d739a : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4f2f28ebb7aa49b4b11a40120159a324 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fc125809fa946c9a561f89a28729ea1 : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76ba134e019a4ac8bfc01c012cc9e27c : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce97635c7d4f4c8eaf5982f8b536562d : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4cde9a3e04b14277986b825523ccf3e9 : SUCCESS in 10m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fa3d6f1e393b41909d4f59f9f971c21e : SUCCESS in 14m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/97c11a03d2b945c79cea8d37d2a56b41 : SUCCESS in 13m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c66c7cbb516c4a26a621b2cca5ddcef2 : SUCCESS in 7m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eeee751c7d1445d8b4c8bab304aa6883 : SUCCESS in 22m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3dca42225c9743d4b7955ac995ccf20b : SUCCESS in 11m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f0713a1454c4cb58939aea2c3079be8 : SUCCESS in 14m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/52a52bf657c24dcc825b7c4a0d13c655 : FAILURE in 25m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cd45dc17c45c4c2f96642448f9c07825 : SUCCESS in 1h 03m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3515a8a369d34658bdfc0b9f4bd89ed2 : SUCCESS in 32m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0893c3c13c7b4efa80776cd9e193eda7 : SUCCESS in 58m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dfd31ee8be334bc282a685ea63759301 : SUCCESS in 51m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99814510fcce44bcb4cc85d788d808bf : SUCCESS in 1h 31m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/32a2f248943d4c74a58d31a1566f24de : SUCCESS in 10m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ff5f0c457d5546c9a1f4345b80f9f3df : SUCCESS in 16m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c796b5fc6a49429d98db66f848c64832 : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0ed29b1ca51431bb7f2370ee178fa17 : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b57d0056d1448c8888710844b3fb744 : SUCCESS in 33m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/39b385343e50411d871b06e09493eb42 : SUCCESS in 18m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2112b55b033140f28360f266f6914005 : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fdd0ac85f60458ea7d25f87e0b80355 : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":3},{"id":"9c0e934d78b7a7b6826b83a4d6eab523ebf535a7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-07 14:31:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e98ac83d10471acd45be053426b7c419126fd638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-07 14:43:02.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e03b3f2107b24fe7f37396147939b168b60cb4cb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-07 14:44:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"daf13f0f856b5418f73c088f5f6ad65b54f9fe95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 16:38:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8014ab8888ce4e0db0ea671bcb2459e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e065ead333b147d893eb0d12e8d45cbf : SUCCESS in 8m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/891a0d8b361c405b955231ad10420326 : SUCCESS in 6m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cacb9767e2af47b39697911ae298a3b2 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0dcb59cb92b445e835977c87257d31a : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bbd3015135c4f39a62f9c4fd3b0c625 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ffa197b00f444825b9347f94704d0666 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99d8e6b903484119b931f19bd7d16aad : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77e5d33fca864ec0a8381c364fac1c11 : SUCCESS in 3m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39a85b21050f482cacbce9e9dbb7ee25 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a178dfc6274f4d729568123ac40f857a : SUCCESS in 10m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/001ceaa7f4d24c3db720e8e405c8347d : SUCCESS in 13m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/506a6aba01074344bd0a825aca2c18fe : SUCCESS in 12m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f31962f7fb9f4f75aad2b99c4bbb1bb0 : SUCCESS in 9m 21s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a6069e97df6e43939c4144fe744eb6ff : SUCCESS in 24m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c366b29fb1d042a48b86b551795b0088 : SUCCESS in 11m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37370cbf85954a17a23a4df092bd878a : SUCCESS in 14m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/34c79e3986c24f9f85a6f23fe4702b05 : SUCCESS in 32m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/60bf92f106f546468cb0150ff69d3233 : SUCCESS in 45m 48s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5022a9c31fa14f8bab88c6450bb9cf37 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e4354d70feb47e680959ce7b3ad4447 : SUCCESS in 51m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7a12eab480004be9bd3266903e4c5b20 : SUCCESS in 51m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0563071d36f14e22a166e409d535c9aa : SUCCESS in 1h 50m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d4097382dc7c41669d9c40562bdb8ab2 : SUCCESS in 13m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aaf22875c1034885ab851ca70f3823d9 : SUCCESS in 20m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b19730d4c2447fd8480f0d56c92d563 : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4b2b09426ef7475fa0bea028cb193f54 : SUCCESS in 12m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a052ab355454444e90ec54d4c3a76ec8 : SUCCESS in 27m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a059496a8f564de8a9ce60d8b5c51cba : SUCCESS in 11m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1ad4ca3e14a4aa7bc5a3d72904eb7b3 : SUCCESS in 15m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/758e66b4730f46a88f5577b38d575bc3 : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":4},{"id":"176cd93af7ae6cd03d20b6b46b22fd30cef4439c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 16:39:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d12e4183bf51837179b74900f171444c6152d566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 18:29:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f97e10e555a343189958390ab29854e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467f908c302243979b16e0e3962bb7a9 : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/05cbaad475264211ba0c523b72c99cc0 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02657c8685944ca8bb3158348fdb5b3 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8663d55639384b20a233555adcb2479a : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b414cc89291742daa3a0460f14923884 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76dbef14396f442e9548a45799ad0752 : SUCCESS in 2m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a104e71ddf8434893677970e310c86b : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aba75a0bd67e413cbbfae02b6bd04ec7 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c01c7e24949045318ec8391146d51576 : SUCCESS in 10m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9afe9cf268c54d849a933e25fb82cff0 : SUCCESS in 16m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9dfa743fae984439b2a0f8dfe9a241fc : SUCCESS in 16m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cb48683ea7e4be6845badc5cf260c62 : SUCCESS in 11m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5062059eaa284b7b9d64651c029b2a66 : SUCCESS in 22m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f99b44a7db3e4acdbab52a77a7ebbbbb : SUCCESS in 19m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed987e397d0949c6a77db132e91a6169 : SUCCESS in 17m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/412bbb3235c94e48b964c580f51217e1 : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7d9c10e20c3142209f42ddb2e4dc5cf1 : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e2a00a72a7704638860bcf2832b253b3 : SUCCESS in 33m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9248fb2dac8140aa971a69532f16ce25 : SUCCESS in 54m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1e43f23b5df34f63a5df47ee8433b740 : SUCCESS in 49m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9d5f1df2e1c4cc9a20794968ea4fbc0 : SUCCESS in 1h 35m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b4eb727d64f54457b9834c581d6dab67 : SUCCESS in 12m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d3a2fc3de8c847af9f888ca37958074a : SUCCESS in 13m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71c6493965374effaaa72ec74d2971f0 : SUCCESS in 16m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a2fa72f9bbb46c292ad4461f536e673 : SUCCESS in 13m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4c22d763a604744b9a1457f7a0f6e76 : SUCCESS in 28m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/71b048a009934d13908739be3cdf459f : SUCCESS in 8m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4240fa7635364743a100f038c17d81ed : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49573fd5f1b34f40b7118f39aca0d033 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":4},{"id":"142c4d331ff00ae5eb0df6c2a81d4b08ec79b74f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 18:29:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9a3c67a563308ffb7536b1c118f3e77d1242580c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 18:30:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/808c52a17ab4434a950d92d00235aa9d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3186b528e0d84e0abe4b761340bd0767 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0fde769dbd649b59bbcfca39636401c : SUCCESS in 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"15353c00b3399cb6b94874d0d508cf0874addbef","revisions":{"cced8d03e747bce69b3c0745abe959c9f26afe8e":{"kind":"REWORK","_number":1,"created":"2025-11-04 17:27:27.000000000","uploader":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"ref":"refs/changes/93/966093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/966093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/966093/1"}}},"commit":{"parents":[{"commit":"ac156f45c309cd15dffaa00735c0cdbd6b40b93a","subject":"Merge \"Account for encoding in _get_direct_url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac156f45c309cd15dffaa00735c0cdbd6b40b93a"}]}],"author":{"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","date":"2025-11-04 17:06:45.000000000","tz":60},"committer":{"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","date":"2025-11-04 17:25:46.000000000","tz":60},"subject":"ALways call virtualenv as a Python module during setup","message":"ALways call virtualenv as a Python module during setup\n\nCloses-Bug: #2130643\nChange-Id: I39eec23d9ca3f8fe9b8d7e9693082ef5841dcffa\nSigned-off-by: Thomas Sell \u003cthomas.sell@bih-charite.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cced8d03e747bce69b3c0745abe959c9f26afe8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cced8d03e747bce69b3c0745abe959c9f26afe8e"}]},"branch":"refs/heads/master"},"2ed7f0c248b5868763887dae81151e3cae1b9c9e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-04 17:27:58.000000000","uploader":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"ref":"refs/changes/93/966093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/966093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/966093/2"}}},"commit":{"parents":[{"commit":"ac156f45c309cd15dffaa00735c0cdbd6b40b93a","subject":"Merge \"Account for encoding in _get_direct_url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac156f45c309cd15dffaa00735c0cdbd6b40b93a"}]}],"author":{"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","date":"2025-11-04 17:06:45.000000000","tz":60},"committer":{"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","date":"2025-11-04 17:27:46.000000000","tz":60},"subject":"Always call virtualenv as a Python module during setup","message":"Always call virtualenv as a Python module during setup\n\nCloses-Bug: #2130643\nChange-Id: I39eec23d9ca3f8fe9b8d7e9693082ef5841dcffa\nSigned-off-by: Thomas Sell \u003cthomas.sell@bih-charite.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ed7f0c248b5868763887dae81151e3cae1b9c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ed7f0c248b5868763887dae81151e3cae1b9c9e"}]},"branch":"refs/heads/master"},"0bae5273c5a24ba95f3ff1c1e198bbd6b3e1f7dc":{"kind":"REWORK","_number":3,"created":"2025-11-06 07:48:49.000000000","uploader":{"_account_id":38563,"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","username":"sellth"},"ref":"refs/changes/93/966093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/966093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/966093/3"}}},"commit":{"parents":[{"commit":"ac156f45c309cd15dffaa00735c0cdbd6b40b93a","subject":"Merge \"Account for encoding in _get_direct_url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac156f45c309cd15dffaa00735c0cdbd6b40b93a"}]}],"author":{"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","date":"2025-11-04 17:06:45.000000000","tz":60},"committer":{"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","date":"2025-11-06 07:48:33.000000000","tz":60},"subject":"Always call virtualenv as a Python module during setup","message":"Always call virtualenv as a Python module during setup\n\nCloses-Bug: #2130643\nChange-Id: I39eec23d9ca3f8fe9b8d7e9693082ef5841dcffa\nSigned-off-by: Thomas Sell \u003cthomas.sell@bih-charite.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bae5273c5a24ba95f3ff1c1e198bbd6b3e1f7dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bae5273c5a24ba95f3ff1c1e198bbd6b3e1f7dc"}]},"branch":"refs/heads/master"},"15353c00b3399cb6b94874d0d508cf0874addbef":{"kind":"REWORK","_number":4,"created":"2025-11-07 14:43:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/966093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/966093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/966093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/966093/4"}}},"commit":{"parents":[{"commit":"22fc03f15ef2930fe53442b53ac0003216e6ca8b","subject":"CI: Fix CentOS Stream 10 host configure job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22fc03f15ef2930fe53442b53ac0003216e6ca8b"}]}],"author":{"name":"Thomas Sell","email":"thomas.sell@bih-charite.de","date":"2025-11-04 17:06:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-07 14:42:54.000000000","tz":60},"subject":"Always call virtualenv as a Python module during setup","message":"Always call virtualenv as a Python module during setup\n\nCloses-Bug: #2130643\nChange-Id: I39eec23d9ca3f8fe9b8d7e9693082ef5841dcffa\nSigned-off-by: Thomas Sell \u003cthomas.sell@bih-charite.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15353c00b3399cb6b94874d0d508cf0874addbef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/15353c00b3399cb6b94874d0d508cf0874addbef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
