)]}'
{"id":"openstack%2Fkayobe~967668","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I8699acef0e88618f86fe854be170d885fc1ab8c8","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-21 18:56:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8699acef0e88618f86fe854be170d885fc1ab8c8","subject":"Bump minimum Python version to 3.12","status":"MERGED","created":"2025-11-19 09:53:23.000000000","updated":"2025-11-21 18:57:02.000000000","submitted":"2025-11-21 18:56:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967668","cherry_pick_of_change":967297,"cherry_pick_of_patch_set":2,"meta_rev_id":"783de5ec66bd39ec0ecbcefd2b849d6ed01e5839","_number":967668,"virtual_id_number":967668,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-21 18:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-21 18:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-19 10:06:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":1,"date":"2025-11-19 10:22:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-19 14:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-21 18:43:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 10:06:57.000000000","updated_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"reviewer":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"state":"REVIEWER"},{"updated":"2025-11-19 10:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-19 10:22:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-19 14:43:47.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-11-21 18:43:52.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"2738fb68af2666f65d0e13ec697ad75212888e65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-19 09:53:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a7198d0bd9a2098c06bbdd8bf5694095af970b14","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-11-19 10:06:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb0adeab263778517eefe12dc3bc8a44fb7c4f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 10:08:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f829c20740fe43d8908b47e97ae55a79\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8eec3ba5d9c549f8b617d19edb7dcd3b : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaa30150e85140598814555c2a864864 : SUCCESS in 9m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c807d440351d499bb0ed6d459d07ceb1 : SUCCESS in 7m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f0a0a937c094389b735d533de5e9f36 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b69d52c54b4343c4b51407e3c9fe77c3 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1730e677872a4abd9289c1733dc2f001 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3eef246f12a34361bf698b7254250bde : SUCCESS in 4m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b207126af722453d8401779ecb01d900 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f357d9871b674a387daa362e50522d05d40ac027","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-19 10:22:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d825c650e3773371db5828975f11608a8a359c15","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-19 14:43:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5aa1f6ea685012dbf9ff763ce80c6f404547c738","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-21 18:43:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99aa6c9e345db566167ea5ead2e2beb4350f5f65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 18:44:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"861a43767d5b75e676ae307608cdab16efc7dda7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 18:56:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe20e69fd984016af6916fe9015759f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b109283846834721bcaa77a916212847 : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caeba8d385547238fbce2fcc4b69700 : SUCCESS in 10m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db0f335c584e4ce68065fed0a4c7dc2a : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/074d7cf3b0304dccaa55e36794456571 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3725cac6bcd84c0d86923719a0dc956f : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c06c3e9bc6d34907a729b953b424e048 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"90366a57cf6ccbc6e7b527667cb9fb737a0027c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 18:56:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"783de5ec66bd39ec0ecbcefd2b849d6ed01e5839","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 18:57:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83b737ca206f4a90b9bea8fdfd69c7b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857377c54e8b4baa8ade20a4e8d618e5 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64e0dff6140943c793c3360d36eff6c0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91f4c8ef9c882592086bb67fd87202b05917d63e","revisions":{"91f4c8ef9c882592086bb67fd87202b05917d63e":{"kind":"REWORK","_number":1,"created":"2025-11-19 09:53:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/68/967668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/967668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/967668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/967668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/967668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/967668/1"}}},"commit":{"parents":[{"commit":"8bb8a44c989ed6e5d1aa2d7f261adc1646c1dd60","subject":"Merge \"Fix comment in globals.yml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bb8a44c989ed6e5d1aa2d7f261adc1646c1dd60"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-17 11:06:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-19 09:53:23.000000000","tz":0},"subject":"Bump minimum Python version to 3.12","message":"Bump minimum Python version to 3.12\n\nWe require ansible\u003e\u003d11 which needs Python \u003e\u003d3.11. Since all our\nsupported distributions have Python 3.12 as default, make this the\nminimum.\n\nChange-Id: I8699acef0e88618f86fe854be170d885fc1ab8c8\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 191e4d9899105bb49a2298e7ffb4cf60b5a80743)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91f4c8ef9c882592086bb67fd87202b05917d63e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91f4c8ef9c882592086bb67fd87202b05917d63e"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
