)]}'
{"id":"openstack%2Fkayobe~968188","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I552f21c73a69340132244ea15089f01a2f9d20d0","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-24 16:58:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I552f21c73a69340132244ea15089f01a2f9d20d0","subject":"Fix issues with inspection rules","status":"MERGED","created":"2025-11-24 12:55:41.000000000","updated":"2025-11-24 16:59:13.000000000","submitted":"2025-11-24 16:58:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968188","cherry_pick_of_change":967256,"cherry_pick_of_patch_set":8,"meta_rev_id":"ebec7a1a944fa4137e3b5f0e515b1c8dee157594","_number":968188,"virtual_id_number":968188,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-24 16:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-24 12:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-11-24 12:58:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-11-24 15:28:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-11-24 12:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-24 12:59:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 12:57:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-11-24 12:58:12.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-11-24 12:59:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-24 14:58:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-24 15:28:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"1c6a0d654ee3856cd82fde640499c6badcde064b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-24 12:55:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5c4f2df5a5926381cbfdf1c28dbdb966a85bb683","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-24 12:57:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1b6722ee05d0d7b50226db6ebad89be4241d75c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-11-24 12:58:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d492ee4acecf901e76f23ecbaf729be3cd48800","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-24 12:59:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d4605803eac73c79a65a2a3833665c6d93c1297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:58:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a46a1fcd204307a57d846fa6376acb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6349b8b24d9d483dae384088be2deb08 : SUCCESS in 10m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e26210b2b754a989b1ed13177d1a766 : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/134ba7afc0eb419f9c64eb51d2248660 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e82f02d2ceb48dcb2bdf593a5546831 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4faf68b4a6df42abaa46f306e4a2eff2 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33072ed58690415cb478a405463f0f3a : SUCCESS in 7m 02s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/95250084521042a8bc6194566748b406 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b217d0e5fa74e6c83bd14e4445243e1 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4ce153031000411e893252b728b2da17 : SUCCESS in 9m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c9966881fd164017b548da93b1c205e4 : SUCCESS in 16m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db3888d73c4640ad825d7bd08f5eafce : SUCCESS in 12m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/29607fdea4b44c84abf9f2ee9f209943 : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a25e6b85ae5409094cf6bebe362b6ad : SUCCESS in 18m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9d961dd7efc54ecbacd1716644def294 : SUCCESS in 1h 11m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4cf13bbc8c5c452bb65034610a97716c : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0122c08d1bbf42a78f83bb50267f41d2 : SUCCESS in 1h 20m 49s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97be25b4800e48a297ea2f8d9e66d2ad : SUCCESS in 1h 10m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e3d39597095843b296c45a26f1d359b4 : SUCCESS in 1h 01m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59439dc8f6e34340a890355953354391 : SUCCESS in 2h 00m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/83c7fb5594d9448a83c6dc28b8676ad5 : SUCCESS in 21m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dd01e241b4674c48aa5ca1ff741302f6 : SUCCESS in 14m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba0c08686ad342c294ac5ebaeb49f80c : SUCCESS in 8m 14s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/856990049f10441895089e09808adc83 : SUCCESS in 7m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a4714d880ff4d79be2b3140f39c83cf : SUCCESS in 27m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3ff6b3d2313b44029a01472fc09f3894 : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea441b71706d438f98abc78047068627 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"69dc025df661e3d4fce74d2cfa6e7caf3b1c5acd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:01:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5b2032e3a9b03388f376cd7c8fd815b6f1e8c88","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-24 15:28:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c5112b9e58036699a8a1ca8e9b043b524dab8b15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 16:58:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7a5b6b5702145acb026b7d477e9ac11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14cbc2d30f84465589372ceee2413079 : SUCCESS in 10m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6e4aba250bb44e2abb532949ae7aa1bf : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6ec6ab8de140bf8c263e97b3a93be7 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6067cc1f17194c40abc8d6ddcc32f010 : SUCCESS in 5m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2755a311514b4fcb85b4e0cea80fee8f : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ded159850df4a62a09b157185d89e1a : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/094f5aed609341b8b9fff77a6c365b96 : SUCCESS in 11m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/58be62809a0a447da0120ef366640c49 : SUCCESS in 12m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4b29e54dfdb46a2a52ae1f338dc7099 : SUCCESS in 10m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3fa734bca4fe42d9a2e70912c6075670 : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cabc30919d7b4c2ab6357ebcf6518cb8 : SUCCESS in 14m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6f0b688ff1ea41dc9e4ae5577ad5f198 : SUCCESS in 33m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0396705862e948d69317c7d6f6d4db66 : SUCCESS in 34m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1784288ea3514743ba42509adf2a3768 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d00640dd49b84b11bd34e1e0c6912685 : SUCCESS in 51m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/988b9e92bd5f4c26a5c1917071406347 : SUCCESS in 58m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cf053dbc41045b596d51d91b2b30f03 : SUCCESS in 1h 40m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2a953f10cdf24010894a608f6b06dd0b : SUCCESS in 10m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ddbfa94b2ca6498a88b60843548b0140 : SUCCESS in 9m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18175ab1e0b8400a834b22e7d3389090 : SUCCESS in 18m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/716f3757197d42d1bf3ad958b33a8f69 : SUCCESS in 13m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e07bb5fae29c4ced9a26b10700290293 : SUCCESS in 28m 48s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0ef706dbaa65459c93d33ef71cf10999 : SUCCESS in 8m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/282afc7cf678414e9a874c65cefbfe85 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"941e550c8942e5f75c3afb882f4323cf4c8a4678","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 16:58:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ebec7a1a944fa4137e3b5f0e515b1c8dee157594","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 16:59:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/985e9f8802994bbe8ef4d1241eae7d6e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa62f934cbe1418a858d5fd63cef09d4 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"823b020910d470fcd73ae7459b4e319cdc618879","revisions":{"823b020910d470fcd73ae7459b4e319cdc618879":{"kind":"REWORK","_number":1,"created":"2025-11-24 12:55:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/968188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/968188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/968188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/968188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/968188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/968188/1"}}},"commit":{"parents":[{"commit":"76e8382981b0023b40279b2cb74b2b26b3ddf1b2","subject":"Merge \"Bump minimum Python version to 3.12\" into stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76e8382981b0023b40279b2cb74b2b26b3ddf1b2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-21 10:35:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-24 12:55:41.000000000","tz":0},"subject":"Fix issues with inspection rules","message":"Fix issues with inspection rules\n\nThe by_path attribute from plugin data was being used with the\nroot_device/name property, which failed to match. Revert to using\nserial, which was the original intent anyway. Check if root disk serial\nis not empty to avoid setting root_device to {\u0027serial\u0027: \u0027None\u0027} which\nwould cause deploy failures.\n\nFix syntax for accessing node data inside dictionaries: interpolation of\n\"{node.driver_info.ipmi_username}\" does not work, we need to use\n\"{node.driver_info[ipmi_username]}\" instead.\n\nFix syntax for accessing inventory data: inventory is a ShallowMaskDict\nobject, so we need to use \"{inventory[system_vendor]}\" instead of\n\"{inventory.system_vendor}\".\n\nUpdate all regular expressions to match and remove unneeded escaping of\ncurly braces.\n\nFix syntax for setting properties: the path needs to be\n\"/driver_info/ipmi_username\" instead of \"driver_info/ipmi_username\".\n\nChange-Id: I552f21c73a69340132244ea15089f01a2f9d20d0\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 4275c90eb511930dd9b327fcd3970824b70399b4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/823b020910d470fcd73ae7459b4e319cdc618879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/823b020910d470fcd73ae7459b4e319cdc618879"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
