)]}'
{"id":"openstack%2Fkayobe~968231","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~I4f6af042eb38d98429075d845b2a1c9ecebebf88","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-25 08:02:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4f6af042eb38d98429075d845b2a1c9ecebebf88","subject":"Rename stable/2024.1 to unmaintained/2024.1","status":"ABANDONED","created":"2025-11-24 17:23:26.000000000","updated":"2025-11-25 08:30:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"387b2d0f14a0916fc2a7ad09e56f89f7fe172da7","_number":968231,"virtual_id_number":968231,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 17:32:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1dbd2428538c29e544fdc275fb3c0e24c562f947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-24 17:23:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"893b2ee3dd2fba832ddaec78632e23a7eb0e3385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:32:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea62150a22354d43a7819f5fdad5384c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b427129476e4f348bdba7d5969cf4af : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d15456093147d3a7df287d7501d843 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c65056a1a4db46f7a54907bdd821b443 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e50b3f70b5864f0cb7211f305edbd2b3 : FAILURE in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3d9abab670d4fb3930d41ba156ee1da : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95c362af3bcb443fb3d76d453dec565a : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2379cff094ba24a073cf5ccb0bf8e4af70a6589c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-25 08:01:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"942fc41c5d0c77a72f9380ab59a28ce32244ba5f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-25 08:02:51.000000000","message":"Abandoned\n\nMerged into https://review.opendev.org/c/openstack/kayobe/+/968240","accounts_in_message":[],"_revision_number":2},{"id":"a5cfda38ef68a344ac83ecc58cd9144b9e7f03c1","tag":"autogenerated:gerrit:restore","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-25 08:29:56.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"387b2d0f14a0916fc2a7ad09e56f89f7fe172da7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-25 08:30:24.000000000","message":"Abandoned\n\nWrong branch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8bebacfda8ed22e8aa877dfa5d5989bca201862c","revisions":{"602dfab72fed5a1bb645b18dec332d6ca1b02d0f":{"kind":"REWORK","_number":1,"created":"2025-11-24 17:23:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/968231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/968231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/968231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/968231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/968231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/968231/1"}}},"commit":{"parents":[{"commit":"40c35d994fe40b8d6d7309dd5b997b5086fd790c","subject":"Remove default(omit) from Bifrost inventory template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40c35d994fe40b8d6d7309dd5b997b5086fd790c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-24 17:23:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-24 17:23:13.000000000","tz":60},"subject":"CI: Remove upgrade jobs following 2023.1 Antelope EOL","message":"CI: Remove upgrade jobs following 2023.1 Antelope EOL\n\nThe unmaintained/2023.1 branch was deleted with the end of life of the\n2023.1 Antelope release [1], which will cause upgrade jobs to fail.\n\n[1] https://review.opendev.org/c/openstack/releases/+/967217\n\nChange-Id: I4f6af042eb38d98429075d845b2a1c9ecebebf88\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/602dfab72fed5a1bb645b18dec332d6ca1b02d0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/602dfab72fed5a1bb645b18dec332d6ca1b02d0f"}]},"branch":"refs/heads/stable/2024.1"},"8bebacfda8ed22e8aa877dfa5d5989bca201862c":{"kind":"REWORK","_number":2,"created":"2025-11-25 08:01:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/968231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/968231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/968231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/968231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/968231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/968231/2"}}},"commit":{"parents":[{"commit":"40c35d994fe40b8d6d7309dd5b997b5086fd790c","subject":"Remove default(omit) from Bifrost inventory template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40c35d994fe40b8d6d7309dd5b997b5086fd790c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-25 08:01:22.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-25 08:01:22.000000000","tz":60},"subject":"Rename stable/2024.1 to unmaintained/2024.1","message":"Rename stable/2024.1 to unmaintained/2024.1\n\nThe 2024.1 Caracal release has transitioned to unmaintained [1].\n\nAlso remove upgrade jobs following 2023.1 Antelope EOL: the\nunmaintained/2023.1 branch was deleted with the end of life of the\n2023.1 Antelope release [2], which will cause upgrade jobs to fail.\n\n[1] https://review.opendev.org/c/openstack/releases/+/967694\n[2] https://review.opendev.org/c/openstack/releases/+/967217\n\nChange-Id: Iec29696248c9271c35c8365d79b622b6a5c05ab9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n\nChange-Id: I4f6af042eb38d98429075d845b2a1c9ecebebf88\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bebacfda8ed22e8aa877dfa5d5989bca201862c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8bebacfda8ed22e8aa877dfa5d5989bca201862c"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
