)]}'
{"id":"openstack%2Fkayobe~968240","triplet_id":"openstack%2Fkayobe~stable%2F2024.1~Iec29696248c9271c35c8365d79b622b6a5c05ab9","project":"openstack/kayobe","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-25 08:04:43.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-25 08:04:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iec29696248c9271c35c8365d79b622b6a5c05ab9","subject":"Rename stable/2024.1 to unmaintained/2024.1","status":"ABANDONED","created":"2025-11-24 18:04:48.000000000","updated":"2025-11-25 08:04:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"19c5bc6df36f411f8527f08822e67cdcabd08a7b","_number":968240,"virtual_id_number":968240,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-24 18:05:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-24 18:44:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e135d3d39082a71a37ac9b2c6754e5236660442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-24 18:04:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65a6b9e4d93fa05658603b8133932596cb21d02c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-24 18:05:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"87c8ce29eabee9a376c814ebd4509336d76adaa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-24 18:06:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"61bd167d0d30ff848b7a0105ca34879c3bda4b17","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-24 18:08:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"91ce5ccec099ab954078914aece97476d307bb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 18:44:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e7e3cef726e4b28b48f0897615563a6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32a828c0cba141159e55d69855f73fc0 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e85a760abd24336b9901cd695044cc9 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1cd61fa85184c9da25cabd846ae4ab2 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d0685150a324c169640b19ba6c20e08 : SUCCESS in 5m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/821cf73ea4de41e2a24625c022423d89 : SUCCESS in 10m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/78da5370aad3486cb5e034fc2d32bf41 : FAILURE in 11m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6b1349dfbccc42f781ad5e9ab77979fc : FAILURE in 8m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f181890f46d4ea88bc418637b775e66 : SUCCESS in 31m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/909b175c11e24ad68d60fb5c55c06c9f : SUCCESS in 31m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b9556d4fd0b946b0806231875d5bed1b : FAILURE in 11m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/06c4e4e91d6d4f6897215c568ba30adc : FAILURE in 5m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ef501992f10d47cc8c7c0550b6071b5a : FAILURE in 8m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c209d8a48e648b8801e5e8a773e0b8a : SUCCESS in 14m 37s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27348423f7c742a5b21f9d8a09e2be55 : SUCCESS in 8m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2b35a514430145339bdb9717cec431be : FAILURE in 6m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fea019d81564e39bd43d73127b1dc24 : SUCCESS in 11m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21aab0f5f1bd4e2584c86c1dc38e7445 : SUCCESS in 11m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/45fd98545c854a65b6c33ec03209008e : FAILURE in 10m 33s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/4ad2f3783d814f96beb0bdb3efe9bf2b : FAILURE in 5m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/550e6abfeba94a1fb6a443e20c874d78 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-jammy-efi https://zuul.opendev.org/t/openstack/build/277e5bc0c7154bfd9abdbcdab139d728 : SUCCESS in 10m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf58aca13cc949a4aa660fd78927ab61 : SUCCESS in 10m 58s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/22e057b93b6847568ee7cca0e264c883 : SUCCESS in 16m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b09faeff5bb444bae537c3b8adbd457 : FAILURE in 10m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22355810d5934f599ba99a83f9aa8087 : SUCCESS in 9m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53ff5357f06941db81c417f8b34f9763 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0bb72f798bf43458f163a5620c39ead : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74467954024f4361b90d3114849ad4b3 : FAILURE in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/450edb53efd14a1abe34968daeee5eca : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/309b1b8d96a74715941d83cea56ff149 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8dd58ba9c6f14238b017cd8d9cac5734ee5f5615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-25 08:02:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"19c5bc6df36f411f8527f08822e67cdcabd08a7b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-25 08:04:43.000000000","message":"Abandoned\n\nWrong branch","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f5a4173eafa7ea379081c5bb43b05223d2cf8ef4","revisions":{"083b67d39ab907a3235dbd49e1c3746278355875":{"kind":"REWORK","_number":1,"created":"2025-11-24 18:04:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/968240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/968240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/968240/1"}}},"commit":{"parents":[{"commit":"602dfab72fed5a1bb645b18dec332d6ca1b02d0f","subject":"CI: Remove upgrade jobs following 2023.1 Antelope EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/602dfab72fed5a1bb645b18dec332d6ca1b02d0f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-24 18:04:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-24 18:04:13.000000000","tz":60},"subject":"Rename stable/2024.1 to unmaintained/2024.1","message":"Rename stable/2024.1 to unmaintained/2024.1\n\nThe 2024.1 Caracal release has transitioned to unmaintained [1].\n\n[1] https://review.opendev.org/c/openstack/releases/+/967694\n\nChange-Id: Iec29696248c9271c35c8365d79b622b6a5c05ab9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/083b67d39ab907a3235dbd49e1c3746278355875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/083b67d39ab907a3235dbd49e1c3746278355875"}]},"branch":"refs/heads/stable/2024.1"},"fcbd063165c1bd9eb0316db22815bda41440e219":{"kind":"REWORK","_number":2,"created":"2025-11-24 18:06:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/968240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/968240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/968240/2"}}},"commit":{"parents":[{"commit":"602dfab72fed5a1bb645b18dec332d6ca1b02d0f","subject":"CI: Remove upgrade jobs following 2023.1 Antelope EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/602dfab72fed5a1bb645b18dec332d6ca1b02d0f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-24 18:06:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-24 18:06:24.000000000","tz":60},"subject":"Rename stable/2024.1 to unmaintained/2024.1","message":"Rename stable/2024.1 to unmaintained/2024.1\n\nThe 2024.1 Caracal release has transitioned to unmaintained [1].\n\n[1] https://review.opendev.org/c/openstack/releases/+/967694\n\nChange-Id: Iec29696248c9271c35c8365d79b622b6a5c05ab9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcbd063165c1bd9eb0316db22815bda41440e219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcbd063165c1bd9eb0316db22815bda41440e219"}]},"branch":"refs/heads/stable/2024.1"},"f5a4173eafa7ea379081c5bb43b05223d2cf8ef4":{"kind":"REWORK","_number":3,"created":"2025-11-25 08:02:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/968240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/968240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/968240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/968240/3"}}},"commit":{"parents":[{"commit":"40c35d994fe40b8d6d7309dd5b997b5086fd790c","subject":"Remove default(omit) from Bifrost inventory template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40c35d994fe40b8d6d7309dd5b997b5086fd790c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-25 08:02:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-25 08:02:06.000000000","tz":60},"subject":"Rename stable/2024.1 to unmaintained/2024.1","message":"Rename stable/2024.1 to unmaintained/2024.1\n\nThe 2024.1 Caracal release has transitioned to unmaintained [1].\n\nAlso remove upgrade jobs following 2023.1 Antelope EOL: the\nunmaintained/2023.1 branch was deleted with the end of life of the\n2023.1 Antelope release [2], which will cause upgrade jobs to fail.\n\n[1] https://review.opendev.org/c/openstack/releases/+/967694\n[2] https://review.opendev.org/c/openstack/releases/+/967217\n\nChange-Id: Iec29696248c9271c35c8365d79b622b6a5c05ab9\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5a4173eafa7ea379081c5bb43b05223d2cf8ef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5a4173eafa7ea379081c5bb43b05223d2cf8ef4"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
