)]}'
{"id":"openstack%2Fkayobe~968479","triplet_id":"openstack%2Fkayobe~master~Iffc8de818e80ee42f5946e2ef04b997d8a9a2f67","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-12-15 09:57:34.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-30 10:12:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-12-15 11:49:40.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-15 11:49:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iffc8de818e80ee42f5946e2ef04b997d8a9a2f67","subject":"Bifrost: Add inspector_keep_ports option","status":"MERGED","created":"2025-11-26 13:39:10.000000000","updated":"2025-12-15 11:50:36.000000000","submitted":"2025-12-15 11:49:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"968479","meta_rev_id":"093fe8ac12a2f9ed51b693c048f2992bb5c08cc3","_number":968479,"virtual_id_number":968479,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-15 11:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-15 09:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-30 10:12:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2025-12-03 09:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-15 09:57:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2025-12-15 09:57:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2025-12-05 11:05:59.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-15 09:57:34.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 14:49:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-26 16:40:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-28 11:08:57.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-03 09:23:28.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"6ca790685bb7c2ae8720353443162e736286d2f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 13:39:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24dbc024107444ff172e6e74342a195d24af3192","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 13:41:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"db2e5066c20a400e5118bfbdddd6d4f7058b7204","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 13:48:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4eea0c5ba19bd4bb5678340a365047e0aa116dbb","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 13:48:16.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"90e3822e270739a778166509e7271fe7dfea12b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 14:49:10.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6ba3d023e374b4116db3e809541711f15a53582c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 14:49:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"74f2f6cc5453ab6b069d824c36395f88e0bf140c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 14:52:09.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"38ec50566e34846fb0809ef5ae3de34b7f1b5078","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 14:54:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"76387c8ae9857b35ea6c79b1fe6f217283ead0eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 16:40:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe97f39bb27141cea430193d2d0f296b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d732587019424896a25144461721bd : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c7d156659de646158821327226c66a7e : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5f2c9f70764f28bf4ad6d4c8b71c74 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7557c2ca7c34a7796042899c57465fa : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96c9b4af20024f16adf26258ae802967 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9595c23492a4eef95155b5e4d27d958 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ddc9e0a86f54faf9015dfd8b70b93fa : SUCCESS in 3m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/701cb7c43d22407fa3058d4e97ceb867 : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea09a72417d744bbac77d79568eab088 : SUCCESS in 11m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/77cafb36dde547b4900846e065a8192a : SUCCESS in 12m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52d43f920a0c475abd99b9ed61d4a996 : SUCCESS in 9m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/aa80b975ad3545108acb0dec6f676bd1 : SUCCESS in 20m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79901fa9bec9443685b72e7c8b7b412a : SUCCESS in 9m 49s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/24f566be3fd344458cf6ccd491fdef5a : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d1c8d3fdd5104145a96972c450ff236a : SUCCESS in 35m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1e4c9dfac1334747adf164cf8c96c23a : SUCCESS in 57m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e287fe6e84ba40e3a7c15713b44b36c0 : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/05a0b45e3e614cd281da94cfd8b70a31 : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed638dd4a8f14962b3f4b898f891d8bd : SUCCESS in 1h 45m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9c7f8347b3c9406cbef3e2c363f20126 : SUCCESS in 10m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bfc37301ea434915b420975056c51938 : SUCCESS in 15m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd9909f5f38044e7912a734e4e18e96b : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e0c8cd317c74c7d884a6a709c7e74ee : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1df5d9d1c18d4c33ab1a4c6572a86c40 : SUCCESS in 28m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/38bfd5c6523b4f27a50b3276e76987d6 : SUCCESS in 11m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e67a14398944c93b8b0d6c2bb3d5ed5 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"70d16f0329f82edfb0316cc4f4c831c323ea3be6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 11:08:57.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"247bc4fad6bc914f0ae5e7039cfde7fcc7ba8322","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-28 11:12:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a4273eb1cec66d742cd56b9ae9082faeae6495a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 11:13:55.000000000","message":"Patch Set 4: -Review-Priority Backport-Candidate-1","accounts_in_message":[],"_revision_number":4},{"id":"147441614b559085ab7f705f168bf1cd0eab8bb6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 11:14:26.000000000","message":"Patch Set 4: Backport-Candidate+1 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"2692426a515ed4955848d0685d873c91b5bcda84","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 11:19:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab02542c87cb726a54dbdadd50b71b2974a36586","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-28 11:23:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ddf277e2c508473d728ef2be091609e86bb3bb16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 21:36:01.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ca8da03332d63ff09033df192b53b61bb9f81628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 23:06:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e2a924dcd14157bd693eb8b16d23b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/449747ce06c943b9a07413da22893f48 : SUCCESS in 5m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/87e61984600a40f5a0f9fe6a63a581a1 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a949187c2e534528b9858ae65ba4fb70 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d158ab1959d445bdb8efd58e5499c9f8 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca9eff61d19b46e3af42e6497977983c : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06cde1d3968e4bf0975094c69de91811 : SUCCESS in 2m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f9769d7f6a384f23baa549e57da3498c : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84760de6a3464f7e85da260ee3b48997 : SUCCESS in 6m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b5a68790bef45c38710e7d02a72a6c6 : SUCCESS in 11m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8ea0cd8462d7429aac7d7ef281297907 : SUCCESS in 9m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf120579b6dc451e8cfaff2910eb4e10 : SUCCESS in 13m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9a4c80cb3e3a4ecf8e27c01d2e79c586 : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2e1ecd815ef44eba2e6c5849c633b7a : SUCCESS in 14m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9ac8a926289d490cb4b01c1076c51b5b : SUCCESS in 38m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2371007132034e009344b3d356444eb8 : SUCCESS in 40m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5026dfa82c7c47148c2deea61ebc98cc : SUCCESS in 37m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d43e7de64ee459095bf17d7caa0b5fc : SUCCESS in 1h 10m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06b522e26ce54a2eb5c8042e211eefab : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/577243531e8e4e78932a03e610e2ff79 : SUCCESS in 1h 22m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/548dc0bcdf104b19b5890ac509ed7af7 : SUCCESS in 9m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4e2920233f2d4fef8e7952d5cd1c0230 : SUCCESS in 9m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96faded5059444e48a048cf9c615acc6 : SUCCESS in 17m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2ca344a9b73446c9000e9fa1c7d1bc7 : SUCCESS in 15m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24e732dd434b451bafed2238bcd2591d : SUCCESS in 28m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f72a560fcc334816bcc9dfda2169a279 : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da0a2b17a6c74595b809db9313ec6ff2 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":5},{"id":"99f0427a659de8a64ea10154ed5f9d35d5bc9b1b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-30 10:12:16.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"75c60dfe098118e09d3918eb9cbfdb89e18e2ca6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-03 09:23:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"bb6db1e62d22e22da529291d5bc0f24c1f34c212","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-05 11:05:44.000000000","message":"Patch Set 5: Backport-Candidate-1","accounts_in_message":[],"_revision_number":5},{"id":"a9c7aa365f179f51e3b885329a19794018dbe7d9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-05 11:05:59.000000000","message":"Patch Set 5: -Backport-Candidate","accounts_in_message":[],"_revision_number":5},{"id":"24e928cff51625b2bf8fa6bcef6ef4f1fee17861","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 09:57:34.000000000","message":"Patch Set 5: -Backport-Candidate Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d0a6af9bb64d0bbf7d2cb6d5df20a9b3f309c3d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 09:57:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"998167a54a8cb11caf862e4dd88d3f491c873ad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:49:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90b9c5206eb643bb8ddbd038a6067b7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab0d9f20a7f44a808606f31a08f42d9e : SUCCESS in 10m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3314a1a01764ff4b87c8897d644d5a7 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97fda086ba054990a9c6beb0c1b8906b : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39e0cc31496a4b87a7afee766c5853a5 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9878be07c55d465888ce626a996a2db9 : SUCCESS in 2m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f95c563aa58447bc8717f20165465baf : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ceb339c0f724238901c36dc19ad1ee0 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4c3654370d44da09951a9debc7b417c : SUCCESS in 7m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/12516b5fd7724d409dbefae93cf2402c : SUCCESS in 10m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8e17a5afecb46ee8858266d3f15ce5b : SUCCESS in 11m 32s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/09bf53e5f93644638851b135e7404719 : SUCCESS in 12m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0435cf1086b44c5bbf664a071831ec43 : SUCCESS in 10m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/81e6ffd6b59e4e49b6e485cebd54a7e8 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9ba24897ab8e46f784ab11f595afcc42 : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/37c02686641a4c61b9b0d3bc92a46946 : SUCCESS in 1h 22m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/892beb6db96c46dcae9c48aa574e6b78 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/853318dd4f9d44cabf532738ab3726d6 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/093e6fdd861b4de994e02eb3deb552e9 : SUCCESS in 1h 46m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a15ca6d8b60435bafb9eb245bcb8454 : SUCCESS in 11m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a6a51c7d6e24e7898e0bb5c2469de52 : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06ef8cc7647144be9c392f2281dd94e9 : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b8fa29694434a9aa86cfb7933439102 : SUCCESS in 10m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b05a3e0ca80044dea03d942fb1253e55 : SUCCESS in 27m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dfb4975c983f415fa4053a81dda4a82c : SUCCESS in 18m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d240f6431064301a562c55fd261d7be : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":5},{"id":"a7123cc4abcd351c7b07e4be27f6c7f4d278b783","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:49:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"093fe8ac12a2f9ed51b693c048f2992bb5c08cc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:50:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d173550bc247ee85bc2b035b4c1853\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c1c89932b84beaa95c762ee1d46a51 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ebbf60607bc4716a443bf5f4a4babcc : SUCCESS in 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6194b950053c37295213323c95fc7220466367f3","revisions":{"9281ce222cad6735e543b6a75e06bfb7c34513aa":{"kind":"REWORK","_number":1,"created":"2025-11-26 13:39:10.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/79/968479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/968479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/968479/1"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 13:36:51.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 13:39:09.000000000","tz":60},"subject":"Bifrost: Add inspector_keep_ports option","message":"Bifrost: Add inspector_keep_ports option\n\nChange-Id: Iffc8de818e80ee42f5946e2ef04b997d8a9a2f67\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9281ce222cad6735e543b6a75e06bfb7c34513aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9281ce222cad6735e543b6a75e06bfb7c34513aa"}]},"branch":"refs/heads/master"},"c200b43b7338bed7f0786f660d0e48be4f4dad50":{"kind":"REWORK","_number":2,"created":"2025-11-26 13:48:10.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/79/968479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/968479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/968479/2"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 13:36:51.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 13:46:55.000000000","tz":60},"subject":"Bifrost: Add inspector_keep_ports option","message":"Bifrost: Add inspector_keep_ports option\n\nDefault stays `present` to match Bifrost,\n(overcloud inspector default remains added).\n\nChange-Id: Iffc8de818e80ee42f5946e2ef04b997d8a9a2f67\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c200b43b7338bed7f0786f660d0e48be4f4dad50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c200b43b7338bed7f0786f660d0e48be4f4dad50"}]},"branch":"refs/heads/master"},"bb0f9ae6766a7ecdbf22f5212b8f15b8b3d033b0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-26 14:49:10.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/79/968479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/968479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/968479/3"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 13:36:51.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 14:49:10.000000000","tz":0},"subject":"Bifrost: Add inspector_keep_ports option","message":"Bifrost: Add inspector_keep_ports option\n\nDefault stays `present` to match Bifrost,\n(overcloud inspector default remains `added`).\n\nChange-Id: Iffc8de818e80ee42f5946e2ef04b997d8a9a2f67\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb0f9ae6766a7ecdbf22f5212b8f15b8b3d033b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb0f9ae6766a7ecdbf22f5212b8f15b8b3d033b0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b28e982d3c0a5d6aaf6816d456b5edb31c5e9a9a":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-11-26 14:52:09.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/79/968479/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/968479/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/968479/4"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 13:36:51.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 14:52:09.000000000","tz":0},"subject":"Bifrost: Add inspector_keep_ports option","message":"Bifrost: Add inspector_keep_ports option\n\nDefault stays `present` to match Bifrost,\n(overcloud ironic default remains `added`).\n\nChange-Id: Iffc8de818e80ee42f5946e2ef04b997d8a9a2f67\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b28e982d3c0a5d6aaf6816d456b5edb31c5e9a9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b28e982d3c0a5d6aaf6816d456b5edb31c5e9a9a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6194b950053c37295213323c95fc7220466367f3":{"kind":"REWORK","_number":5,"created":"2025-11-28 21:36:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/79/968479/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/968479/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/968479/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/968479/5"}}},"commit":{"parents":[{"commit":"1b7b9f292e54bc307b6ecfe0b68a5c414317f35a","subject":"Merge \"Handle empty remote_hosts in kayobe connectivity check\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b7b9f292e54bc307b6ecfe0b68a5c414317f35a"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 13:36:51.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-28 21:33:58.000000000","tz":60},"subject":"Bifrost: Add inspector_keep_ports option","message":"Bifrost: Add inspector_keep_ports option\n\nDefault value stays `present` to match Bifrost (overcloud ironic default\nremains `added`).\n\nChange-Id: Iffc8de818e80ee42f5946e2ef04b997d8a9a2f67\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6194b950053c37295213323c95fc7220466367f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6194b950053c37295213323c95fc7220466367f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
