)]}'
{"id":"openstack%2Fkayobe~968535","triplet_id":"openstack%2Fkayobe~master~Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-28 13:22:10.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-28 11:23:12.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-27 12:15:31.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-28 13:22:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9","subject":"Handle empty remote_hosts in kayobe connectivity check","status":"MERGED","created":"2025-11-26 17:45:22.000000000","updated":"2025-11-28 13:23:45.000000000","submitted":"2025-11-28 13:22:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968535","meta_rev_id":"c57b2411117fb7f18b70f2bf85787d82bc2a5fb1","_number":968535,"virtual_id_number":968535,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 13:22:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-28 11:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-11-28 11:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-11-27 12:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-28 11:23:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-28 11:02:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 19:26:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-26 19:42:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-26 19:51:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-11-27 10:55:42.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2025-11-28 11:02:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-28 11:23:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-28 11:49:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"073d02842f7b81f11d53a9b6f95151b8f1016035","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 17:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9351d27b00f27ad90cdc354c011b3fd53555dd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 19:26:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f4fdafd18154662aabd61c4d6f40b73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e83ebb923434fa2850e7048a45e71e1 : SUCCESS in 9m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d679f0f1b8934bc8b81c285a485eed3b : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3116840c270040bf8e6be964e3347b8a : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e0964732c747248a2fa009a585d7e5 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd6b9843a89d4cf5a4094beee0883f63 : SUCCESS in 6m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33fd3467b3d14a5c8039ddf1dba80f9d : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36a6da8f523045beb66e749052408a3e : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae0da345be694e798f1a2b94825eb2ae : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a883ec8cab994046b73971adc352d4ef : SUCCESS in 16m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ea8273ca45ba49d7925071f5922885ce : SUCCESS in 12m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1e965c76b49e42bd9c1c416269429d4e : SUCCESS in 21m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fd98633d8e249b0b0560b7203b625e9 : SUCCESS in 15m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/87b22beda2544cc29f5a53ffc13375c8 : SUCCESS in 47m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/44007dca454f45f7b6979ee5e64313c0 : SUCCESS in 58m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5b43e473bb46450b86888fee4ca95840 : SUCCESS in 37m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e316f5d2a54948b6a96c5c772856b52b : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4acbcc677ba04d5daa89ef8204368ce6 : SUCCESS in 59m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ece01558f7eb42048374ea36f0877538 : SUCCESS in 1h 36m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f68430fc16d44f1fbb6141858b0581c7 : SUCCESS in 16m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dab56c39b10b41d0a32e01ef7ef0f662 : SUCCESS in 9m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3cc70c77ea5e4b55858bd7b46221383a : SUCCESS in 14m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89961bef078b43a9b57a216be1b1a146 : SUCCESS in 14m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89a63344dcb34586abf4283edd05b992 : SUCCESS in 31m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f8c18ca12d9c432d9120044beb7d22c1 : SUCCESS in 11m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2369e036e064245b5974391966e1822 : SUCCESS in 9m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4f5808e586b65cd6213e707a5aea61e6a6744a3c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 19:42:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07ca33384413fd41158ed0ec88f81ed912f4eb8a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-26 19:51:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc497a1db35c9a8a4f436ab3d48327fc7e8b3174","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 20:08:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c81570df24b6ea5d0c41490be3e8df0f872ad372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 20:10:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b55b84046601457f91f1f58fcea4f119cf2a4d1a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-26 20:11:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74d8c34adc538b31d046bf0ec399fc6241d67479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 22:04:51.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e922f88ff0eb44b69b246a996c9c7d87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe759a7fb8445fb82aa5ad1e83e92c8 : SUCCESS in 9m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a665fe53d744c78bb37f16ac79c466d : FAILURE in 7m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bc0bcf8d14a49e881098641ae4c0176 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aefe4fc817744a4f8c038c56da3c3336 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/374bc472759549dbb279eb13efc23252 : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64775cc99fb446168de89b407a70cdb8 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb89ed54e3674942ade86c65ebadf8a5 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b8e8c51ef72469a85d364f3b0688fbb : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2a4a1d35ee6a4c8a8438c2da0155acbc : SUCCESS in 12m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9167694b79004db98d8668cfc564a392 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a968ef739a3448d4b559a171ffcd1a4d : SUCCESS in 10m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b754b6f2ef824913a5d01346dd491980 : SUCCESS in 15m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f74d9ace59f74c058a5d99184fa0525c : SUCCESS in 33m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/939762fa77454689879ff1ef265bf519 : SUCCESS in 34m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2d0825eb3eb54955afd26b04a3dec966 : SUCCESS in 1h 25m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/baa9b24c01984374a5c0933a38ba59bd : SUCCESS in 1h 13m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e48e44692804e4ab6a1a6f7faccc696 : SUCCESS in 30m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe283a26c3c24a70bc487e0300e5be5a : SUCCESS in 1h 49m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fbe7c8654ad844e585225af86796001b : SUCCESS in 14m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7b31ca725c744421a116e61a8cc5de1b : SUCCESS in 22m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d551e8ddba8f4b6cbebcf2be4e199c95 : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5a45112e221442268916afe14f9172c2 : SUCCESS in 7m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8195cb8879554d638e99e8b7806b13c1 : SUCCESS in 30m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ff858c5f91e24908b2d77da62c4a8c9b : SUCCESS in 7m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4492ac971752411bb7304216f12eab63 : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ddb815a280a4adbc3f63af5dc3fe2cf79d3bc183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-27 08:43:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e3d3d1521b4ad7c2cefde6cb056af042e6e0fa5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-27 09:04:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ebbc6d3eeb859652501249f411a2e950720bacda","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-27 10:55:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a6b64ef253fcff1ed9a1dac0cf8d3c3a3b114d4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-27 11:02:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"09453f1949646f774e6e31c27ccff489d04441df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 11:17:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a419efd391f4e17b6342d727f543579\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55d476f299e41b2bfd01169162f8950 : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/58095950d39d42e4ae673e57a473d048 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b4040546ca246bfa11c563c8b40e3c1 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33a8945aec442e39a5f8649d2e5fd35 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c23d4dcb37b94f97bd7f8c2448e240bd : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49691f9861aa49419b1c74f4bf44b9cf : SUCCESS in 3m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e19c7e7b1af4d6b87d57628d7cb2af7 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7109e624e5e244009f2796e22f96c4e8 : SUCCESS in 10m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b344eda57b2b42bea6aa38df76e907e4 : SUCCESS in 7m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a11704a27624405a9e9e49051eecb2fc : SUCCESS in 15m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8e564b4b55bc4707becc7e129a10f279 : SUCCESS in 10m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a45f31ca24834975a708648aee396f03 : SUCCESS in 18m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/58079271a0304ff28a271652c79eda44 : SUCCESS in 37m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7de2043a7a3e456cac66e7cb8c7fb074 : SUCCESS in 1h 19m 37s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1820c38ab021459ea42109aa1d465588 : SUCCESS in 41m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/144dee2e37984d8b90ba3ddf2e062834 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e0bb599bed104994872a6a5f4c55f3bc : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85d1c787dc33416e87ed369e5e9a51a1 : SUCCESS in 2h 07m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b48f0279a500442caa3e342a8784adb5 : SUCCESS in 14m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ed716354794244d7b88c4dd13cdd5f97 : SUCCESS in 14m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5211a52229024a96b8fd060740fc7542 : SUCCESS in 15m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2ae0543b00e141f7be2a3af4c71a2f5d : SUCCESS in 15m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b52e27e8da8e4e4a90d9df292956f3e6 : SUCCESS in 27m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e9265468661949938f30db14b17c7ef3 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b318bc8fa8474322b81e637394e7576b : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":4},{"id":"b6e761fe833dd226251c60258697c78e8ba5408e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-27 12:15:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"25960821454528bcfd38481747a0db083ec45493","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 11:02:03.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1a1931742d26a0951bc616a302581fde90f3221d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-28 11:23:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3c52ea1c0b71c7e8967561e7119efafa801404e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:23:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3f3141956238f194a475346c948edf7881a5d4a3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-28 11:26:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"73eb4b4391a8f056a5bc59051d73f0eff7aaa448","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 11:49:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d8eca03f8dd695bf7782fb29742217841e79ac50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f57e101db74fbf845d171be0f34619\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d48bdeda404c309462c32f0a4854e2 : SUCCESS in 9m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/43a445a03a9d4f9498af449d79e233e9 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cdf48eb864b43c59d05683742ccdef3 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/866d5a5764854adb89668ba2eb2bdf89 : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a7512053b1d4413b2fa3272deda1676 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5798b4689cdc48249c01e5cb7398188a : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6792b0d5bb8046839c5984ae1538ffcf : SUCCESS in 9m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a8f4b30f316544f2bca629b0541fcf16 : SUCCESS in 8m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aba99da66b654c9b915e4282f1e8c861 : SUCCESS in 12m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/69f6ba263b384bb685587e15caabcce5 : SUCCESS in 12m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f3192f03efa4a53a6eb3903131c2795 : SUCCESS in 18m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2a6b7ba8ea2f4624b39ff21b9b06e658 : SUCCESS in 55m 45s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0146817d2e084e279c11afe7e00300bb : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6d4fd426558f42ff873e75997c8031a3 : SUCCESS in 1h 21m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09417133590a4902bc2cdb989373b40d : SUCCESS in 36m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9c338ae8378b4b25a5764a142460342b : SUCCESS in 1h 09m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71e8347a5d8141038f925f8d902bf05a : SUCCESS in 1h 41m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/420b9462b76b4a0e9ba59884e022ca5c : SUCCESS in 14m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a7f55a0a8786429b95239cf8e0fbef9a : SUCCESS in 14m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/882735a149a34659b673322d91c17950 : SUCCESS in 15m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d1e11d186b4c47e9850141b9a1e4adb9 : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c0d9a7b8ddb42c591ca947f661558c7 : SUCCESS in 27m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dce14020e1ca495a930b933227bb9e11 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa1a0dbce196474aaf9dda002f6e17ba : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":4},{"id":"92f6d0c1340042301e242ba98dcca6333c6a778e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c57b2411117fb7f18b70f2bf85787d82bc2a5fb1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:23:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9c612cedb14197af379b700cb53dee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1566244feef4ceb9b68f25172d77b29 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c59e26189f8eb94c55f46aab30208204d0d5bbd0","revisions":{"e3f9d518f938c94375a922b0f549f5bb87eca22d":{"kind":"REWORK","_number":1,"created":"2025-11-26 17:45:22.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/35/968535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/968535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/968535/1"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 17:43:11.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 17:45:01.000000000","tz":60},"subject":"Handle empty remote_hosts in kayobe connectivity check","message":"Handle empty remote_hosts in kayobe connectivity check\n\nCloses-Bug: #2133047\n\nChange-Id: Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3f9d518f938c94375a922b0f549f5bb87eca22d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3f9d518f938c94375a922b0f549f5bb87eca22d"}]},"branch":"refs/heads/master"},"72ed9b0b8b7bb2e505d457b77d164c27904fccf0":{"kind":"REWORK","_number":2,"created":"2025-11-26 20:10:30.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/35/968535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/968535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/968535/2"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 17:43:11.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 20:10:25.000000000","tz":60},"subject":"Handle empty remote_hosts in kayobe connectivity check","message":"Handle empty remote_hosts in kayobe connectivity check\n\nCloses-Bug: #2133047\n\nChange-Id: Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72ed9b0b8b7bb2e505d457b77d164c27904fccf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72ed9b0b8b7bb2e505d457b77d164c27904fccf0"}]},"branch":"refs/heads/master"},"4de27d7664a8e6c48bbcf51b7636df4c07fd3f6a":{"kind":"REWORK","_number":3,"created":"2025-11-27 08:43:12.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/35/968535/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/968535/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/968535/3"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 17:43:11.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-27 08:43:08.000000000","tz":60},"subject":"Handle empty remote_hosts in kayobe connectivity check","message":"Handle empty remote_hosts in kayobe connectivity check\n\nCloses-Bug: #2133047\n\nChange-Id: Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4de27d7664a8e6c48bbcf51b7636df4c07fd3f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4de27d7664a8e6c48bbcf51b7636df4c07fd3f6a"}]},"branch":"refs/heads/master"},"c59e26189f8eb94c55f46aab30208204d0d5bbd0":{"kind":"REWORK","_number":4,"created":"2025-11-27 09:04:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/35/968535/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/968535/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/968535/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/968535/4"}}},"commit":{"parents":[{"commit":"2858b253b147c6198b112c3437ac23bb364d6192","subject":"Document moving releases to unmaintained and EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2858b253b147c6198b112c3437ac23bb364d6192"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 17:43:11.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-27 09:04:23.000000000","tz":60},"subject":"Handle empty remote_hosts in kayobe connectivity check","message":"Handle empty remote_hosts in kayobe connectivity check\n\nCloses-Bug: #2133047\n\nChange-Id: Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c59e26189f8eb94c55f46aab30208204d0d5bbd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c59e26189f8eb94c55f46aab30208204d0d5bbd0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
