)]}'
{"id":"openstack%2Fkayobe~968571","triplet_id":"openstack%2Fkayobe~master~I066ef977825df4c4c34fcdc92523794656ece0d7","project":"openstack/kayobe","branch":"master","topic":"fix-typo","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-27 16:52:28.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-27 16:52:28.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-27 13:13:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-27 16:52:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I066ef977825df4c4c34fcdc92523794656ece0d7","subject":"Fix typo of the task name","status":"MERGED","created":"2025-11-26 21:01:51.000000000","updated":"2025-11-27 16:53:26.000000000","submitted":"2025-11-27 16:52:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968571-fix-typo","meta_rev_id":"d864cf422065adc064a25fec9af1216a65d0da77","_number":968571,"virtual_id_number":968571,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-27 16:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-27 06:57:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2025-11-27 09:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-11-27 09:00:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 21:06:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-26 21:06:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-11-26 21:06:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-26 22:59:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-27 06:57:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-11-27 09:00:19.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"33e867f3d7d635d947557daaf6312a558307195c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 21:01:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d98644383609ac3bb49c5777c7e2959c7a959aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 22:59:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a19be11abd4046cba88edc1a822ce5a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca880cff326416390da11fa195cba0f : SUCCESS in 9m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6aa3d3e9a0c49d2bce90f0970bc533d : SUCCESS in 6m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/049e80bd53c643f4bdb689854c1fac72 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b7314ed00304a14a2d13f7310f5fc75 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32ed3ec75dad4eb88ee04117faa27466 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41725a716ff840a19e0b700a83a4fd81 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f87c096ae55345d38ceb1911e76c82b0 : SUCCESS in 6m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b0650b2ec97740a8aeb45eb12ead0b72 : SUCCESS in 5m 58s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab5bcc83194649679adafd001b8c36c8 : SUCCESS in 8m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ec4f2317d4443f9b53e2c0bc31c030f : SUCCESS in 12m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fd7c64dbfdb74e0c952e3927baf6e191 : SUCCESS in 24m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e05faf511bc646c3a0fa4e3c9e907e6f : SUCCESS in 20m 28s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0be2dd7dbc154a82be1c0cb1cd440244 : SUCCESS in 42m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5c4e294ccc8a4248b19ab3e1cee1e730 : SUCCESS in 1h 23m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b001979c92d74c7cbfbdb190c4387950 : SUCCESS in 45m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10f156e8a8a8463f8d7de2104b8c216e : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d97ec9b753b24da8a967b0cc34565ae5 : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c60204c55671470d9f42e811989ae5b9 : SUCCESS in 1h 51m 47s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f67f19ff27404984890afc87362f2863 : SUCCESS in 17m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/36ba7e3ca9304117841840a79b936336 : SUCCESS in 16m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1d4d7228692456fb8dd87483d3beeac : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/68263ba219ee4da2b88158f53de6408b : SUCCESS in 15m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0899a769cb024a9ebd485eee1b6d2c9e : SUCCESS in 29m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/48ddc3ec57e7448993d326569adb50ce : SUCCESS in 17m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/344472c729524095842379213cdb595c : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"776a06b9ef42c27dcc9c3c2809177fbfb9f01156","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 06:57:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9865d093bd517066215b23889627ca9360d6688","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-27 09:00:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b461174b6b17e23fdc5498dd7b28cc4330ed42e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 09:00:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adda499ac4e421e25016851132f415dd0ff39603","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 10:58:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13284d7a04064c099154b8b0e5428b26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f2a2f3cefff4c3bbe389485d0a83b50 : SUCCESS in 4m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39c665d2a4db47bb96840ef2a0adae12 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34affcf750af4d538034fadb1f20ffde : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b6ae75975b84b05a6b52238bd3a274b : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7141a691c01240ef8cdc36ee08d66f9e : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/559768ca2bcc45108a42c14086ebec57 : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c45dbbaa1cf44a539666e40ebe89008d : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4942a563b0fd435780b28b3c382e0823 : SUCCESS in 8m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f0292dc3461400aa92386e135f9319a : SUCCESS in 12m 18s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dab43d9165994f0581d821a20a8a9f38 : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fd0add78acd43129100d2554f600a65 : SUCCESS in 13m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7887318999fa4756a5b78657a5f1c03c : SUCCESS in 39m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6aecc353334c4f1a967937fe470dedf4 : SUCCESS in 56m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/25ab213bad524126802d50c1fd994a59 : FAILURE in 1h 33m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a451bc3f3694504b9a07e50183a47fc : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/71904881d3614c17a402112642402e21 : SUCCESS in 1h 06m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6daaac41c254239835c355a47d4e4f0 : SUCCESS in 1h 39m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4054effde2034f4a88cb63d63f98eaad : SUCCESS in 11m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e8a9c08b1d1b4c5abd6f7b55bbcf799c : SUCCESS in 9m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d303ac161d6640c49dc57d573460d619 : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1e6977e47d994672be32d673163c6e00 : SUCCESS in 14m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/049c750c406048bab64a28e373f72322 : SUCCESS in 28m 18s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c8f2a66d6a3d4fc28c8ddbb0ffc39b6a : SUCCESS in 11m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8d8f91c0bc143018658749851c996af : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"aa4d89228b701a8607fc965d585a800d35c87951","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-11-27 10:59:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e7f61cd132f51de1953c963390eec68f6406a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 12:59:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd9e7480ed204b5b8d85dd1bdd97c1f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b4b95af7184efd8700bbe12ecb0131 : SUCCESS in 6m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b3ee92f03a7f4238bb8c08bd6e7d1ca7 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9581952fe2c04273b32d23063563ff96 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb66184e475440ea5cb553c94eb2919 : SUCCESS in 2m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5b4dbde6f784ae38985539ebcb806b9 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d2279eda6ea491288d917b1e29ab9f4 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d6f20df18be492d924a10dcb7c45165 : SUCCESS in 14m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eee25270d81b4810891cf8804d2cd269 : SUCCESS in 7m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c2a6676f13704a27a05699841351711c : SUCCESS in 16m 09s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/181a3e72a096454eb0c5964582f58a57 : SUCCESS in 12m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/20db4ce258e34925bf7263c813ccc4ff : SUCCESS in 24m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a20947ed2df40b1a119fe974bdcd9fa : SUCCESS in 14m 32s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/013272fdd7074b898d019a56ef9bf1e1 : SUCCESS in 1h 18m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/685d8a3609cd4c959987462633165f82 : SUCCESS in 1h 18m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7c8ab0da39bf46529bb4e06e37e0b9ef : FAILURE in 1h 44m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/178e28bc2ee746c3b193d30a04fb3d02 : SUCCESS in 34m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7772b69a60c84b53aad81d3e13ba8c85 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40ef35756664452eb4a09d253725e83f : SUCCESS in 1h 57m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a21a30e1e63547e7b662d6534feb5073 : SUCCESS in 11m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/468bf3e915f44fccb8f9921e5f4a92fd : SUCCESS in 12m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82ca58c950be4cff979ea5f9ab41f7c3 : SUCCESS in 19m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ba05c924bc242af9ec27eab35953730 : SUCCESS in 15m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f20bbdd260d94b0a90269809ebb4b2ff : SUCCESS in 26m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ac012c7eb3744f9b9d0e480d0b120b69 : SUCCESS in 8m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e626328285474eb0848a21d489ac721b : SUCCESS in 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0609bbb6e435028f516c4e9439161f536408f27d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-27 13:13:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae1852a63767cb8946dfb688d73dc4818b22c09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:02:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780a85ba2b554b5b9f9a63cedb7dd093\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd07e3ed7acf48acb48f19dcfe2c07ae : SUCCESS in 5m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4854f82beea04afca7ef969aa9e76b14 : SUCCESS in 6m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/666c98e9ab7b4718a7ab206518b4987b : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/565e4a5cc2e44761907c2917becfc279 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0214e6164fe4375bbc2f934170f3dba : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb8436413820427093f7b9d0734307ba : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d305685a6fc14ef49b9bfbc4d2fcbfa2 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66aafbe5dc164c7daddb5a4d9cb5863b : SUCCESS in 8m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3506cf91439741b9a0475c5b51bdecd2 : SUCCESS in 7m 35s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1c8bdd41e2b46889c867e0b5e315adf : SUCCESS in 11m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1c4d566c79134b15afb9b174c7ba5fe9 : SUCCESS in 11m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fd78fe8a8ba49aca71807459ee4b463 : SUCCESS in 18m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fe06dd7b6cc44db99a329fa032651b6a : SUCCESS in 37m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c0443349c91444bfa3ab7c418d47da13 : SUCCESS in 1h 17m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ae1592bacca240c5853f8fadd2ebcb1a : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adcfc97742ec4169b07957a1a8edb625 : SUCCESS in 54m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a477b2b6157a40028023a3cbec089eae : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ccd4ace39c794a4592892284466e6ec2 : SUCCESS in 1h 43m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5ccc88c0813f474ca3d07b850268d887 : SUCCESS in 22m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0e8adef810544e968f5567a220d4f784 : SUCCESS in 10m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/264741074a2946009fefffce021511ee : SUCCESS in 17m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bcd581f2abf44906a0bb1e6b36682cc2 : SUCCESS in 11m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eae6c6da816414798a011f4e9daa999 : SUCCESS in 27m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/19c8e348fa9a48619aa42f9dcab0f71d : SUCCESS in 16m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15a9e6f79a9a498ea77cb1f996a7a6a5 : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9cc10d3d70bfadb35236fde1a2007bcfd2833ddf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 15:03:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b68593665040e89d6150655c3c4836d8ae9aa6fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 16:52:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4da844652d194dd5a7e1c011de86fcd8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f3edfd4376c49af8d010444c1292cf7 : SUCCESS in 8m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d11d9594c8a4678930c95d289eada1c : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6450494148a41158d5e7b69d80b65b1 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7472018877234ba388e09d89ac24bd86 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a494429d0f7540ce80cae57ba7804c44 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35780fb8eccc4db789fc7241576a5831 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52eebe17dfe44127ba41f53c69918ae0 : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7bf8a8f0ca46416dabfc349541bef6f1 : SUCCESS in 7m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e540414cf764a7781d73ad8113dce18 : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/36e5f018c1d4467291b5c5562c015bd9 : SUCCESS in 18m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/295ff01d044a49f297cad7539bb5aa05 : SUCCESS in 8m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/871258bc9d9643b59f03e1d537ef0162 : SUCCESS in 35m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e24114dce82f4e8f8c4f8c8477cc72a0 : SUCCESS in 39m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fecbc09d2dd64c2c984029645b659a56 : SUCCESS in 38m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8074444668ff4615936a8128f0be925d : SUCCESS in 35m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/221271608a014fc99955e5aa713e0ed5 : SUCCESS in 1h 04m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d28c6a7e4bb4c1e98f5554389efeddd : SUCCESS in 1h 44m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/03c54d51aed34eef93e010cd508740dd : SUCCESS in 9m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b715846b59fd4a1c85be0fff91f0a181 : SUCCESS in 14m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/415dbfbfe0874a0387d93d49afe84dfc : SUCCESS in 7m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/23646c01594b422485bcd769a5f07da7 : SUCCESS in 14m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4ff363830444fc492e4f24c52e41d78 : SUCCESS in 14m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/56904044929f48898141b636c2344344 : SUCCESS in 7m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16b6058a8b8148eb9aae9d2bd92a0c49 : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":1},{"id":"94dcd6d404258dcba80ffee5263ea393211a388a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 16:52:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d864cf422065adc064a25fec9af1216a65d0da77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 16:53:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52025f207db74db0b60dab8f570ceb0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6927f518ade14ec0b307e95abf7a2923 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","revisions":{"645a3074063f9df8c958c2b241d62b3fa4aef9ea":{"kind":"REWORK","_number":1,"created":"2025-11-26 21:01:51.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/71/968571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/968571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/968571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/968571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/968571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/968571/1"}}},"commit":{"parents":[{"commit":"03c849c6cab0045a78257eb37db95c75cc1c42c2","subject":"Test bare metal node inspection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03c849c6cab0045a78257eb37db95c75cc1c42c2"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-11-26 21:01:16.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-11-26 21:01:16.000000000","tz":180},"subject":"Fix typo of the task name","message":"Fix typo of the task name\n\nTrivialFix\n\nChange-Id: I066ef977825df4c4c34fcdc92523794656ece0d7\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
