)]}'
{"id":"openstack%2Fkayobe~968574","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I066ef977825df4c4c34fcdc92523794656ece0d7","project":"openstack/kayobe","branch":"stable/2024.2","topic":"fix-typo","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-11-28 10:58:34.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-11-28 13:22:04.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-11-28 13:22:04.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-28 13:22:04.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-28 13:22:04.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-11-28 13:22:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I066ef977825df4c4c34fcdc92523794656ece0d7","subject":"Fix typo of the task name","status":"MERGED","created":"2025-11-26 21:04:47.000000000","updated":"2025-11-28 13:22:58.000000000","submitted":"2025-11-28 13:22:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968574-fix-typo","meta_rev_id":"548f0291fdedf3dffb795d225b158f68c5b49ec8","_number":968574,"virtual_id_number":968574,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 13:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-28 10:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-28 10:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-26 23:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-27 10:27:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-11-27 10:27:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-11-27 17:42:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-27 17:42:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-11-27 17:42:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"552c4ec04dbd43f2548f51022c47474eb13b4af2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 21:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e23f91c43c838de57613141feb983a6cda23bc77","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-26 21:07:16.000000000","message":"Topic set to fix-typo","accounts_in_message":[],"_revision_number":1},{"id":"e831e87b002396c4d73f61ade1169146574ee03e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 23:15:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47821cd567ce498cb314161360837193\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2af089fee4e431c9b607f298d2b5d54 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6a8fae39a646a6ae1d67aa0ed13e7f : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce30e952b2fa472b99d8bf1bf0b9239b : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f91d40ea6ba460da87a9405e2bb05df : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c77b5e36a004cd186539f286e36bce6 : SUCCESS in 11m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46ff466c1b534c8d95b811f4904eeb9a : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7bce45355fe44aab01e84c2e17e8596 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/69d98375752443cba8524203870a65dd : SUCCESS in 8m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7a5b0ef5440a43d4ba619e95f228c641 : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/520e32ec5b31483a9a21358e5979821c : FAILURE in 8m 53s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9e7312adbd24686b01a040cb1d9b9da : SUCCESS in 1h 00m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/110b08b7ce564022ab097f1334727592 : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d971995a08494b27a2c9eef6cb35ec18 : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f165a666b14147e79e8cc34382a996af : FAILURE in 8m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b93701a679c84176b6dc741b29bf6cf7 : SUCCESS in 13m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/145197256576449ab2869cdec8f9689b : SUCCESS in 2h 04m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6eb6ab48d1e742a5934c274df2dbb11b : SUCCESS in 1h 37m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1422cffd95644daeb612c8b31adf10f0 : SUCCESS in 24m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bf8638cff2f4443be4df4301e2b3c9c : SUCCESS in 19m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/241dfba238a34e9989a0843db5386ffa : SUCCESS in 23m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc5b895b7ea742e5ad798e63c2f8298a : SUCCESS in 26m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4e72e8a08ea41bf862ac39b3b5e3722 : SUCCESS in 19m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7dfb572875064457b6e0e7ada2d42a0a : SUCCESS in 11m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8cce22c12a1f49b9b3c95b33c538df65 : SUCCESS in 12m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3007938b1052434c848011f677e44314 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"72fcfc075f4571624b784943fd824a2b987b9fc4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-27 10:27:18.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"8998c300136efe6013339146ae0c65a8d38dd16f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 10:58:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cfc2c8f297b106a0808dff4e5efb99590ffde41b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 10:58:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc5e78bf1b24d89527458521cf7335d347ee3668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/895c7942b5a04658b2916b6779cc05ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c533da3ba0294bc488f5738fb1e51778 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a09175172eeb449886b0ead9692f5b9f : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6868929beb4e4a16a5f2c6df6f66523f : SUCCESS in 7m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f10126d0257a4e859416ec94f0f535f0 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/95fdea77170241d59b35460cc97fda86 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88cc5b5f83cb4482bd802615198ac640 : SUCCESS in 9m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/95576fcd6fef448584487d3e63af0353 : SUCCESS in 1h 10m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a3c2c39bd0c452a831323c63eb5c5ab : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/364a84166de9401aa4887da27ea76ebf : SUCCESS in 1h 16m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b205ca76eeb9421093962357f3edb55f : SUCCESS in 12m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/923dffee79844276ad27dff079212136 : SUCCESS in 13m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f7aed51d610e4b9f99a79b00682d4800 : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cec2aabde234b90af71bbcaf44348bf : SUCCESS in 1h 36m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7235d039f7b24e9fb3de7c0e63c2b7d8 : SUCCESS in 19m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f66d2d51cb1943fc9232a3124132ff2f : SUCCESS in 20m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/35367807913f4b7dae41b8fc73aafe07 : SUCCESS in 35m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f04d89bfb58c4e2ab9a567dbf1010c55 : SUCCESS in 30m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/086b3c2a5d354b2a9b63e7d843fefc39 : SUCCESS in 20m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3b2843ef1d74554a5f8d2258e89972a : SUCCESS in 15m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/395defdf5c9248e1b19e5368ca8a172d : SUCCESS in 16m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/103de6fe1acd49d3b8640b403eb3889b : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f3e94d5680b5b9c11a15f0889b5b55718161deda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"548f0291fdedf3dffb795d225b158f68c5b49ec8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 13:22:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52a21250a3a046549417c76b3834d962\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfea40323be44eb389894e2fc6b81a92 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5e617de721817b0496961bc6d378a6edb767fbf","revisions":{"b5e617de721817b0496961bc6d378a6edb767fbf":{"kind":"REWORK","_number":1,"created":"2025-11-26 21:04:47.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/74/968574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/968574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/968574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/968574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/968574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/968574/1"}}},"commit":{"parents":[{"commit":"1865865b7f78401edc6f8c6f0825f10f7541ba1d","subject":"Move nvidia.nvue to 1.2.9 for cumulus 5.13","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1865865b7f78401edc6f8c6f0825f10f7541ba1d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-11-26 21:01:16.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-11-26 21:04:36.000000000","tz":180},"subject":"Fix typo of the task name","message":"Fix typo of the task name\n\nTrivialFix\n\nChange-Id: I066ef977825df4c4c34fcdc92523794656ece0d7\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 645a3074063f9df8c958c2b241d62b3fa4aef9ea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5e617de721817b0496961bc6d378a6edb767fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5e617de721817b0496961bc6d378a6edb767fbf"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
