)]}'
{"id":"openstack%2Fkayobe~968725","triplet_id":"openstack%2Fkayobe~master~Icb74fd51735ee0f45fb7d4109f214bfbb7f50057","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-28 11:36:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb74fd51735ee0f45fb7d4109f214bfbb7f50057","subject":"CI: Skip baremetal inspect for UEFI boot mode","status":"MERGED","created":"2025-11-27 15:40:24.000000000","updated":"2025-11-28 11:36:57.000000000","submitted":"2025-11-28 11:36:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968725","meta_rev_id":"9b2093c362fb8594aa185e4046abd98f51271b39","_number":968725,"virtual_id_number":968725,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 11:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-28 09:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-28 09:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-28 09:32:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-27 17:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-28 09:29:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-28 09:32:52.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"de023ed27589c7ddf202d2ccf7ef89a50d47d75d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 15:40:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dcb8271ff96cff6ca114ca36c38ffdb281ba011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 17:31:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/355e66c24bc54cf781b0bbd5553ef8b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90837a0ab6854bb1b87bbf4090b74fb9 : SUCCESS in 4m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae39d815440e4b4ab18c8d37d24e63f6 : SUCCESS in 5m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e2143b8964c4d7681f5dc1691941ee1 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b96ec6deada4be0b307155d5614a19d : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3cbca4a00284fab9bbc17a755b637a0 : SUCCESS in 4m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ed30c7dffd3a4fbbb57d567cf64618d2 : SUCCESS in 7m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2626f822b09e4eec8b7aed2d1c3b3b23 : SUCCESS in 10m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4567738a766247b1bdd262d679f3cbc8 : SUCCESS in 11m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f77264703bc44f838b798a2c9e329026 : SUCCESS in 8m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e3aef140d65d44faab7f24716f80cfc3 : FAILURE in 1h 21m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5323269927224d83b1d96abbffc1ad30 : FAILURE in 1h 01m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abe3f269dc5a49daa1f6ee825984a8cd : FAILURE in 1h 25m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a5e247b38374604936c2c34174f6afd : FAILURE in 1h 25m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f91f9322f2e42e2961cd0bb11528a00 : SUCCESS in 1h 45m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/627683718c5b46da9d389c3c94e1cb20 : SUCCESS in 9m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0c1228726baf4a7ab4b36274886bf357 : SUCCESS in 10m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65ec3c2b9fd1441fa2dbc25c3be08a3d : SUCCESS in 16m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ccd1ecd958b14f6e8b8e6cdb060b9aa2 : SUCCESS in 7m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db6ef9ed4aa243c9a2ff5b5aa9321c5e : SUCCESS in 14m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3e8d572fe4f543e1956b32961bce0f7a : SUCCESS in 8m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/836aa906b5a743c5b62b75f9881a73fc : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6f5f355d9947cb17bd4830b5d083f18d0a18f590","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 18:13:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"df1ed689a80f8a33f09240dd199f9f55244ee0e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 18:13:42.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch stable/2025.1.  The\nproblem was:\n\n  Nodeset kayobe-rocky10 does not match existing definition in branch master\n\nThe problem appears in the the \"kayobe-rocky10\" nodeset stanza:\n\n  nodeset:\n      name: kayobe-rocky10\n      nodes:\n        - name: primary\n          label: rockylinux-10-8GB\n\n  in \"openstack/kayobe/zuul.d/nodesets.yaml@stable/2025.1\", line 20","accounts_in_message":[],"_revision_number":2},{"id":"527323a65128b9c7aed307c2a0ee29d882c4a0d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 19:55:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1f6c66424e4a11607a5939aa1cf2bf04723c6d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 19:56:03.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch stable/2025.1.  The\nproblem was:\n\n  Nodeset kayobe-rocky10 does not match existing definition in branch master\n\nThe problem appears in the the \"kayobe-rocky10\" nodeset stanza:\n\n  nodeset:\n      name: kayobe-rocky10\n      nodes:\n        - name: primary\n          label: rockylinux-10-8GB\n\n  in \"openstack/kayobe/zuul.d/nodesets.yaml@stable/2025.1\", line 20","accounts_in_message":[],"_revision_number":3},{"id":"478c1defb5ec8c75ddd10ce449a0e733b76e8357","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 19:57:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c0d54cb7ae96291415f71836aec490f4e6c6f152","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 19:57:20.000000000","message":"Patch Set 4: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch stable/2025.1.  The\nproblem was:\n\n  Nodeset kayobe-rocky10 does not match existing definition in branch master\n\nThe problem appears in the the \"kayobe-rocky10\" nodeset stanza:\n\n  nodeset:\n      name: kayobe-rocky10\n      nodes:\n        - name: primary\n          label: rockylinux-10-8GB\n\n  in \"openstack/kayobe/zuul.d/nodesets.yaml@stable/2025.1\", line 20","accounts_in_message":[],"_revision_number":4},{"id":"f719c561cbe1fa8b52b158135f03d3ad08ca0644","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 20:03:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a1b72baf295000287fce653d069ab7982e3a2fa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 21:21:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87da48078f1c4f8682a365a2b8074929\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4c440863fe4861ba149d82477751af : SUCCESS in 9m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/310902171e564c218d0a0a37e4d7224c : SUCCESS in 7m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8208e3cd620e4cf9896110081fcacd6e : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb2f2dc377f4774a62f6afb0bcd88ec : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/977904635bc84625a594c8d85310086b : SUCCESS in 3m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/206472a83e86452a8ad4f199850d430c : SUCCESS in 12m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab73bcb5474648d4b4357a907dcee24c : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6818a110a143458bbbd0c6797371ae85 : SUCCESS in 24m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a74f13c74ed84808a59387b71edd5051 : SUCCESS in 12m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5fb91618ed434a45a223e8409a6951e8 : FAILURE in 1h 11m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0aced370ec0d4f43963b2e0d86d1e54e : FAILURE in 1h 12m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7614080fd91349c8847c7878ac6903a6 : SUCCESS in 58m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39abc6a58e444839b2257d2fe9fb5273 : FAILURE in 55m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3cd55b1eda33422994c98ee6aac07808 : FAILURE in 1h 16m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8e923a03a9614eebbf65befcff0b4ddb : SUCCESS in 52m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/35d2779955db40d18b175022c946c081 : SUCCESS in 13m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dd78c71b73a54506be4881b1cf6698e5 : SUCCESS in 22m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/918fa36189cd4fca993b025d9073593b : SUCCESS in 19m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/016ca65a02674278970c7767b2431ded : SUCCESS in 13m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aba69dc71fd4cc1ab0656fba401053c : SUCCESS in 15m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c1779929608e4b359c780b5bb3e6b5df : SUCCESS in 10m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5a7049ded6547bf88feb0a90ae3aaf8 : SUCCESS in 17m 00s","accounts_in_message":[],"_revision_number":5},{"id":"e506cd028bfc1e9cdfa2954f8ff182fb1ba64297","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 21:30:19.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"099e4278e0d9b34bea37d53948eff68bd3224e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 21:30:37.000000000","message":"Patch Set 6: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch master.  The\nproblem was:\n\n  Nodeset kayobe-ubuntu-noble already defined\n\nThe problem appears in the the \"kayobe-ubuntu-noble\" nodeset stanza:\n\n  nodeset:\n      name: kayobe-ubuntu-noble\n      nodes:\n        - name: primary\n          label: ubuntu-noble-16GB\n\n  in \"openstack/kayobe/zuul.d/nodesets.yaml@master\", line 26","accounts_in_message":[],"_revision_number":6},{"id":"6792dd313f3ad8433f0e0064e943b9c6eba4a047","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-27 21:32:42.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"823051bf6754e07e2641f7f1c7df73332e4fab3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 23:22:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51ec3a70528e4355a700f63905bf5e63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a48ea3bf2034d7eb0a75e3d3609c2fb : SUCCESS in 8m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c90380a6e694bfbad3676442391f477 : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/196ddf58219c4c01a3b36743392ba51f : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05426b3ca80f4f56a1396d5170b8da44 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9041c17e56174dceb9cee617e365d1a2 : SUCCESS in 6m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2d30e44b8d1248fdbb505674881518a0 : SUCCESS in 7m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0b6d6d3ff0641748011126568712cb9 : SUCCESS in 6m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e5dbc1fd27dc4672b812a432c95995f5 : SUCCESS in 16m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f45db786a9384b7e9b48cc05983663a2 : SUCCESS in 8m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8cf2243d385d4e6cac1fb484c0dcc933 : FAILURE in 1h 11m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5e7589f35e0b42f69e85ff4a92eba546 : FAILURE in 1h 14m 52s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/06b1fcdab6194010974e5663f39de9a0 : SUCCESS in 58m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dd46a34223548969fdc2ff8b2f49df7 : FAILURE in 1h 07m 49s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9e4ec7b77dd54298afdaab80f3881a10 : FAILURE in 1h 08m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4cc2e9dda734f4c9fa66548ae2d9482 : SUCCESS in 1h 44m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b6c61287b80a406794164b7b4a2fb1b9 : SUCCESS in 10m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e0229c42283414587db3148cd3ef8a3 : SUCCESS in 13m 08s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7f8beba81124b7288d89859a356b784 : SUCCESS in 15m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/36990a86a8cd4f14825c7eaa96535a51 : SUCCESS in 17m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de2550c7d1794e2d847ce06baaf33d16 : SUCCESS in 25m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ba8c27485bbd47efaa8d2e6012f9e9b9 : SUCCESS in 7m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04fac5b87295456594d8aaa57d140d5c : SUCCESS in 12m 32s","accounts_in_message":[],"_revision_number":7},{"id":"93d4b3f564aedf1a563da0e5d5938bf2736e0334","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 07:46:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"46c802f23293421b487daf4548e51e7a2b716107","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 08:06:03.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"abecea2f5ee3e8f10b2f77a122f4f8f3f011ff31","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 08:58:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"bb9fb206116e809f29c2c7990abe2008fe9035f4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-28 09:29:29.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"bee13760802100f3953dc3899902285ba8d8e677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:31:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeb6e201b88f4dc8bde3419293947ceb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7781bf2c308c4ef2b3bec24ab58ea2d5 : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7255307e381146cc9499b09d5d159831 : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd32a74c741440f5ae6e38fdce9f1fd5 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1818a5dcbd749299a76d7a30051e936 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87f8822c7a9f491fb7fdebd6a04a90d7 : SUCCESS in 8m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef66db4a0d084080aa909f176c76bbe6 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4923af5db3e84c52b88c9b7cfe3b6f2d : SUCCESS in 11m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e767da5b3cf465d840441b4464434f8 : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7793a645097a45a0b64524af476ad820 : SUCCESS in 7m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f3c2b2acc8b4e589f528790d2f7ba59 : SUCCESS in 12m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/46dcab9300f7448d99275c3deb00dbec : SUCCESS in 11m 10s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d760ba58073426bbbbe1c94e0703c3a : SUCCESS in 17m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/de3442dd0cae497f8920f156981c8f13 : SUCCESS in 1h 06m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/955f1ceedf0249d2a4b00cc495a79953 : SUCCESS in 38m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e9803db9f68e446e8690414d78cd2808 : SUCCESS in 37m 05s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2b65ea4e7ee4a41a7c16eca537a34cd : SUCCESS in 1h 13m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8ec366c82ecc4029a2f846eb1e3288cc : SUCCESS in 39m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ac3ef2586234e8995b77dcf88a5476f : SUCCESS in 54m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/19a577aa38794491b95c34f1cfb57322 : SUCCESS in 10m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d3e7b50a3ad14558b0501210208c0554 : SUCCESS in 9m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c058954540294eed90a0f17111943280 : SUCCESS in 15m 43s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f8a2cadd31854656a3fb7eb71ba99722 : SUCCESS in 12m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/066a06ac0efd4875a13d80c35a18a926 : SUCCESS in 27m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c05883a5c3d4eb8bfb09e3d0983df08 : SUCCESS in 11m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f8208af51f84766976d6af5fed767d0 : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":8},{"id":"9e94a75997e1d311a31a523a3924a9f03bf8e64c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 09:32:52.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"96a59da62507fd2952552d601f020c9403316f19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 09:33:05.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"56638929d690e2821f27ede2565a4006821ea52c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:36:10.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/194347748125431398794eeb8f9da75a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a6f7667bdc4dff818722935509a29a : SUCCESS in 7m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/290a60061b844921b84acb8797185568 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3ebf77009c8431e860162376d7a426f : SUCCESS in 6m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a84b0d992db4c5e91f80189c420dcec : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fcec1f39cf3c4adca98437a9ea64e930 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5195d58f29274a5db3ac1c76d38b2c4f : SUCCESS in 12m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e876ca25ace84432aa615b29ca54ec04 : SUCCESS in 9m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/94f3689c0527446c8e1c74de4c654798 : SUCCESS in 9m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c932ece40d1b41138e85f9d84027fa98 : SUCCESS in 13m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a75754af77b84e718209a485e6edc9b5 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/991157857cc44b2cb07fb68dd31ddc0e : SUCCESS in 15m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/beeb42274e4d48df87994996626f40f3 : SUCCESS in 53m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/01267979da1b4da2aeae7d4ba60c46d5 : SUCCESS in 40m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/15aca71fc15a435fa160a12fd5279db7 : SUCCESS in 1h 22m 14s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7bc4a6d6ba34b1b95abcd4e9f718224 : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ab1e7b453aa4c358bc3d1a90d273ccd : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/219e56ee7c804e0bba0698f29e53548f : SUCCESS in 1h 57m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/eb2ee615093d4526b049c9b93c937548 : SUCCESS in 24m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/040b15edd4cd4f448733b1ade1e27fd7 : SUCCESS in 17m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c7cd963105f49968b4101e829ab538f : SUCCESS in 9m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/353eaee03b6e49d1a397c6663c5ee479 : SUCCESS in 15m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fac5530098949989d2eef34417e206e : SUCCESS in 28m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab3e9dbd6a5648dfb9a814cf29347485 : SUCCESS in 11m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/367cc5ade3ab47bebbc6ac1eb8d0fc1a : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":8},{"id":"2c6533c16d29203f5fe619e04eb6ce3cf1170826","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:36:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"9b2093c362fb8594aa185e4046abd98f51271b39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 11:36:57.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a077e4fe6fb40dd8242206c9e80e9e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a646519e264643dfbcdfe3990b8edc6b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"fe86f0af7d5bc835cd139006a59c4e6c3b63b50f","revisions":{"70666ca57e051f5f9135e43654dd6a533ef27306":{"kind":"REWORK","_number":1,"created":"2025-11-27 15:40:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/1"}}},"commit":{"parents":[{"commit":"03c849c6cab0045a78257eb37db95c75cc1c42c2","subject":"Test bare metal node inspection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03c849c6cab0045a78257eb37db95c75cc1c42c2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 15:40:19.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 15:40:19.000000000","tz":60},"subject":"CI: Switch Tenks nodes to UEFI","message":"CI: Switch Tenks nodes to UEFI\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70666ca57e051f5f9135e43654dd6a533ef27306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70666ca57e051f5f9135e43654dd6a533ef27306"}]},"branch":"refs/heads/master"},"ce10a184718a2652a5f3fe6eda0c2025d368abc2":{"kind":"REWORK","_number":2,"created":"2025-11-27 18:13:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/2"}}},"commit":{"parents":[{"commit":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 18:12:58.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 18:12:58.000000000","tz":60},"subject":"CI: Switch Tenks nodes to UEFI","message":"CI: Switch Tenks nodes to UEFI\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce10a184718a2652a5f3fe6eda0c2025d368abc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce10a184718a2652a5f3fe6eda0c2025d368abc2"}]},"branch":"refs/heads/master"},"dbf16db92af1aee6f2a701e775e5cafae76d891e":{"kind":"REWORK","_number":3,"created":"2025-11-27 19:55:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/3"}}},"commit":{"parents":[{"commit":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 19:55:31.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 19:55:31.000000000","tz":60},"subject":"[DNM] CI: Switch Tenks nodes to UEFI","message":"[DNM] CI: Switch Tenks nodes to UEFI\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbf16db92af1aee6f2a701e775e5cafae76d891e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbf16db92af1aee6f2a701e775e5cafae76d891e"}]},"branch":"refs/heads/master"},"54d0eca8801837e14eaf60f6903a1cfb4b568bda":{"kind":"REWORK","_number":4,"created":"2025-11-27 19:57:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/4"}}},"commit":{"parents":[{"commit":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 19:56:56.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 19:56:56.000000000","tz":60},"subject":"[DNM] CI: Switch Tenks nodes to UEFI","message":"[DNM] CI: Switch Tenks nodes to UEFI\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/54d0eca8801837e14eaf60f6903a1cfb4b568bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/54d0eca8801837e14eaf60f6903a1cfb4b568bda"}]},"branch":"refs/heads/master"},"aee872c51215b8de6ef388afbef4cbbcb09bf9fb":{"kind":"REWORK","_number":5,"created":"2025-11-27 20:03:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/5"}}},"commit":{"parents":[{"commit":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 20:03:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 20:03:37.000000000","tz":60},"subject":"CI: Switch Tenks nodes to UEFI","message":"CI: Switch Tenks nodes to UEFI\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aee872c51215b8de6ef388afbef4cbbcb09bf9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aee872c51215b8de6ef388afbef4cbbcb09bf9fb"}]},"branch":"refs/heads/master"},"77604c5e6740ffce9cd0bc33f01e988628c688cf":{"kind":"REWORK","_number":6,"created":"2025-11-27 21:30:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/6"}}},"commit":{"parents":[{"commit":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 21:11:01.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 21:11:01.000000000","tz":60},"subject":"CI: Switch Tenks nodes to UEFI","message":"CI: Switch Tenks nodes to UEFI\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77604c5e6740ffce9cd0bc33f01e988628c688cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77604c5e6740ffce9cd0bc33f01e988628c688cf"}]},"branch":"refs/heads/master"},"6c2c2f8a3b23890f31ba85a28373459464dc8331":{"kind":"REWORK","_number":7,"created":"2025-11-27 21:32:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/7"}}},"commit":{"parents":[{"commit":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 21:32:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-27 21:32:36.000000000","tz":60},"subject":"CI: Switch Tenks nodes to UEFI","message":"CI: Switch Tenks nodes to UEFI\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c2c2f8a3b23890f31ba85a28373459464dc8331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6c2c2f8a3b23890f31ba85a28373459464dc8331"}]},"branch":"refs/heads/master"},"fe86f0af7d5bc835cd139006a59c4e6c3b63b50f":{"kind":"REWORK","_number":8,"created":"2025-11-28 08:06:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/25/968725/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/968725/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/968725/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/968725/8"}}},"commit":{"parents":[{"commit":"645a3074063f9df8c958c2b241d62b3fa4aef9ea","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/645a3074063f9df8c958c2b241d62b3fa4aef9ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-28 08:05:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-28 08:05:23.000000000","tz":60},"subject":"CI: Skip baremetal inspect for UEFI boot mode","message":"CI: Skip baremetal inspect for UEFI boot mode\n\nThe kayobe-overcloud-tls-rocky10 job is frequently failing with:\n\n    Node tk0 failed to reach state manageable. It\u0027s in state inspect\n    failed, and has error: timeout reached while inspecting the node\n\nThis happens when ironic_boot_mode is set to uefi. Skip baremetal\ninspect for UEFI boot mode until the issue is fixed.\n\nChange-Id: Icb74fd51735ee0f45fb7d4109f214bfbb7f50057\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe86f0af7d5bc835cd139006a59c4e6c3b63b50f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe86f0af7d5bc835cd139006a59c4e6c3b63b50f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
