)]}'
{"id":"openstack%2Fkayobe~968796","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-11-28 19:06:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9","subject":"Handle empty remote_hosts in kayobe connectivity check","status":"MERGED","created":"2025-11-28 13:52:57.000000000","updated":"2025-11-28 19:07:07.000000000","submitted":"2025-11-28 19:06:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968796","cherry_pick_of_change":968535,"cherry_pick_of_patch_set":4,"meta_rev_id":"4975646a33d0c536606d2a557d484d86c07c3d16","_number":968796,"virtual_id_number":968796,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 19:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-28 17:20:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-28 14:56:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-28 13:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-28 17:20:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 13:52:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-11-28 14:56:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-28 15:39:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-28 15:50:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-28 15:50:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-11-28 15:50:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-11-28 15:50:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-11-28 15:50:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-11-28 15:50:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-11-28 15:50:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"696a6ece2a065c4ac81f1ef2e5c35e3f8f131c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 13:52:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0744e0db80e959f2700b82def5ba79d370bbfd23","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-11-28 13:55:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15f7710705fef7bb685302f6730fb4c10a3df6fd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-28 14:56:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"55b4346398f84e87311a21bb400f21bfbe1c77a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 15:39:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe1fa9de97e3477f919ad313851cb411\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bbbe916af7e406fa332156978587032 : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c53cc0b9c81245a8b908586b07a8f3e4 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24e35c7863234f51b9563379e8d4cf6b : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c145b374d1e14769a89d046ff08f7f66 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6351f2571e9463ca4c8c884fdf71291 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91e7a7f83fc9495987410aa4f1bec3d4 : SUCCESS in 7m 46s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67cd21c4256d45d98f7677492af5e33c : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1533787c9e57469e97e0a66e9695dca3 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9599b994241944c3b7e9e346ebb785d0 : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cdc0366565fd4831adcef36b0e6a85bd : SUCCESS in 10m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/645e56c39c204bebb629d2053f9d3c35 : SUCCESS in 12m 48s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b7cab9b3218546a2a37c4b3c943da91d : SUCCESS in 12m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5750e32338c4c89afa15a992ab7777c : SUCCESS in 14m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/04a772d7bb454bb4be12008f0a120182 : SUCCESS in 57m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6d534be71ca14d7fa4ab464e85c72d46 : SUCCESS in 1h 10m 40s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9437c8fb98bd4e1780573afcac931520 : SUCCESS in 1h 25m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6973e2ef21054da9bf49438ce03b1d93 : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8540007d6eac4ed28c469507c6b5374e : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb4c7ffb04334d35aa907dcd17a2b4ee : SUCCESS in 1h 38m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/41224d2fe58d4949a3a994553618357e : SUCCESS in 9m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9611108ec09b42cf82b298851ce85083 : SUCCESS in 20m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b56852f22fc14976b2ceba8951112e3d : SUCCESS in 15m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/86b477e2826a40039c803aa81df11b4a : SUCCESS in 14m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84c508d0477c44b884536d3115ec331b : SUCCESS in 31m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/74dd278b12ad437d9827b4d8deaa85b7 : SUCCESS in 18m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b6990c0dc054cb6982f97a4a6b83708 : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"95ecb341364a8c766c958e88c4d93b9cdb952fa3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 17:20:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a416b5a4ef2e8cbc1249c1091cdac764afac31ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 17:21:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aba6604b9a1ac6e0f39200ebf2d46682edf0063d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:06:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2412a285f1b49b59bf41dc8580d67e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e863d922fcb045d9901fc44f8805c1b8 : SUCCESS in 10m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82bb0a88d92241e1b603839deab1a135 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c819fbd0761f48bd8837243f52cb2385 : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/835d8162e42641fdad70dcd14cd42dc8 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d40c60261a847039db97ecc7a9b1267 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/425d3c17e3174e2584ff75f6d9cfc6b3 : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d174681474e4adb94e6e9281a0a194c : SUCCESS in 6m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ef8f10ce21ab4879a7ef11f4b770b723 : SUCCESS in 8m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f69d61874a64d97beb8f4a1554e938f : SUCCESS in 14m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6a1ce1c37ab44922a20d51fc3cedf8fa : SUCCESS in 11m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af8c780d34f84edcb8229cf683a7d7a1 : SUCCESS in 15m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1c7daf12315d4eb3bc3dfb8437948d99 : SUCCESS in 34m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/741eb86415e94ca680b849588a861ece : SUCCESS in 37m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d2750f6cf58546c7b12a874111e75ee6 : SUCCESS in 1h 23m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69b2ad25f681415fb261c1d372cdc0ab : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/374eb1d68ce24f5dbb899d9730dca07f : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67a6d46acb6140dca9af5eb2c349629a : SUCCESS in 1h 40m 41s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0ba48444787b4f2a818728bd60200b8a : SUCCESS in 10m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7cb5bbd016c84c318be3cf383e7812e6 : SUCCESS in 21m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4651f7c0d3b44375aa80d09fb34c3389 : SUCCESS in 18m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/99ff7b1ad3c242e2ad96116e21bd15e6 : SUCCESS in 17m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d92317f6f28457a99b051e49d5d2533 : SUCCESS in 31m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e5e056e21a964d64aa96581df7fbe412 : SUCCESS in 13m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc6c6b012d9543cc88b75b32ad987741 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"583e8d7cef60f45981ce28c5ac2611c4ef17c2f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:06:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4975646a33d0c536606d2a557d484d86c07c3d16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 19:07:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/536c534a1fe5498193eda210279841fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c024c3e538b40b0bfb279b259e0e60c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c013a3f84ad2569ad76a56e7b55fb960b0ceef04","revisions":{"c013a3f84ad2569ad76a56e7b55fb960b0ceef04":{"kind":"REWORK","_number":1,"created":"2025-11-28 13:52:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/96/968796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/968796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/968796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/968796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/968796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/968796/1"}}},"commit":{"parents":[{"commit":"046166125194d109aee7facf43bcd99355f92631","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/046166125194d109aee7facf43bcd99355f92631"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 17:43:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-11-28 13:52:57.000000000","tz":0},"subject":"Handle empty remote_hosts in kayobe connectivity check","message":"Handle empty remote_hosts in kayobe connectivity check\n\nCloses-Bug: #2133047\n\nChange-Id: Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit c59e26189f8eb94c55f46aab30208204d0d5bbd0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c013a3f84ad2569ad76a56e7b55fb960b0ceef04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c013a3f84ad2569ad76a56e7b55fb960b0ceef04"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
