)]}'
{"id":"openstack%2Fkayobe~968798","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-11-28 20:05:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9","subject":"Handle empty remote_hosts in kayobe connectivity check","status":"MERGED","created":"2025-11-28 14:08:00.000000000","updated":"2025-11-28 20:06:14.000000000","submitted":"2025-11-28 20:05:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968798","cherry_pick_of_change":968535,"cherry_pick_of_patch_set":4,"meta_rev_id":"e736fa0b679663be7b1acf4fb9fad4e1955dca0e","_number":968798,"virtual_id_number":968798,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 20:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-11-28 17:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-11-28 14:57:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-11-28 17:21:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 14:57:03.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-11-28 17:21:11.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-11-28 17:25:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"97eb41e1ccd3a57c3e4acee7afee79880442b14c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-28 14:08:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fa22a31541527910a61d01605081e495a628ebd8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-28 14:09:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"7cfa909787bb80093748c2a0d034317ea28b5bbe","tag":"autogenerated:gerrit:restore","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-11-28 14:43:08.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"cc57cbf7302f8b727085bf08f394d33b8fe35605","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-11-28 14:57:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c0b8c81f2b3b9db984a9942048b29a019bfe9908","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-11-28 17:21:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e54266da1e9d2f8d4878d204f53dc6ffac94808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 17:25:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc889ed73f540eea706502f73fc7935\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95104a9e47c0422e92740d08edd15bb2 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8d2d3fe2a2e4d0aa9ac2c7bbedf11ba : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aea86068c34c451c98285b7513f20686 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95e1684f5a24419dac18eae446b975a8 : SUCCESS in 5m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8014110c65224d829f37bb09ebded4ef : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b9ec4d3ee1e946e8ba4f42c19dccdc85 : SUCCESS in 8m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09d679f670b34a9c99a2034559492d9d : SUCCESS in 9m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c4012fd920274566a9131943c1613ac6 : SUCCESS in 8m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/554a7bda7a764e24b3bacad42fb158fb : SUCCESS in 7m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/430a610d131c4c29bc796f4424772c14 : SUCCESS in 11m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/07ec33443ada4b4dbdebbe8c152d35d4 : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2a0cc28e72a2471f959d901e176f5e5a : SUCCESS in 19m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bc282e91556417fb409b0b9dbbd796b : SUCCESS in 7m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/218524e272cd49019f93f4de9c3f159c : SUCCESS in 33m 51s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/08919ba2f7f44435a63778da4bc7cdff : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/841bf8482c844333a0d5c1d9bca5c4e2 : SUCCESS in 35m 42s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52f676fd57c544b5a181264d2853b7b1 : SUCCESS in 58m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9baa00a9c73541cf93257dce68beb7ba : SUCCESS in 51m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d8c3283f82e54584ae98325c036b5932 : SUCCESS in 2h 23m 01s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/5b8716683adf4ee19c3e88b571d6cdc5 : SUCCESS in 2h 37m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/adb82a1d4bba450dbb9415032ba8813a : SUCCESS in 2h 21m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd73a979a3054f389799d23f202a7496 : SUCCESS in 2h 10m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4f26c13f36cb4610bc759c21a50db91b : SUCCESS in 21m 48s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f8d2a4a6d1e4462781a369a47efcf270 : SUCCESS in 8m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9223b5e13b31478f80c05a895659b4db : SUCCESS in 11m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d975b2de579b4e97823194d272be42e5 : SUCCESS in 7m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dbea405b0e364df8a832eb76989d8c88 : SUCCESS in 33m 26s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/828ca30192c64b4f9c9ca7843500baf2 : SUCCESS in 21m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4964ab9bf6b547ea88065aa32b6e6427 : SUCCESS in 20m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0d622eae67e248148a25afcb186acd47 : SUCCESS in 14m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8ba29265ab2f41ac80988470b8ba5c60 : SUCCESS in 13m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91b0c7a1cbfc401ca7d4612486bc8e77 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ea29c973b076891753a729c6eb0ba87d34173adb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 17:26:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a84badd22f192bad7a36468053099fbfc6c7bae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 20:05:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d1dd515cbc343b7ad869b4876407a22\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191a6e4a84eb425583b648f8e4e47972 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45cf3cdc6b7a4b15a5d3283c90d0ce2f : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a593f5c01ed40e4bdbb0111cb3250ca : SUCCESS in 10m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d8addb0285a40e587691abd8f7ab490 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0530d80394d944df9d47d8cebd057e42 : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/796135fba4cc4171912d59d3cdf45789 : SUCCESS in 11m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2276f375d6e14e5dabbd632d8a35e0d8 : SUCCESS in 11m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/639060ac87cd47748d2c68baf8219dc2 : SUCCESS in 12m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed52f70dedf141618fac7d510eb7241a : SUCCESS in 5m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4c5c4d81fe3a49b4a9cd48cd33efb1fb : SUCCESS in 10m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c119b015708f4c1f973be924502fc160 : SUCCESS in 20m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecbe683f329c47b1af1bc5da7c954f8e : SUCCESS in 7m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e48e85ddbaaf4bfaa0ebc434591f1eff : SUCCESS in 56m 08s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/49f159bc867848abb5e76fb610540b11 : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/40420131f9454175a0549f53c3a751c2 : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ad4c26c7af543a2b090f2567037d2a2 : SUCCESS in 59m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e7141332825d4c63b3751b891ab93733 : SUCCESS in 56m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/49c2c44148a8438fb6dbdc759f4a8fe0 : SUCCESS in 1h 22m 37s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/07164354fed34ed2ad54093dd6faf336 : SUCCESS in 2h 33m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/663765c96cb0468aad689f195034288a : SUCCESS in 1h 13m 59s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e99517516644ec6ba297ca46a373094 : SUCCESS in 2h 21m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9c29a23f3b9744bf99880eb375e3951e : SUCCESS in 15m 32s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/dbf26f3272b344d3988b85a541c8bfc6 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb3817ae4be64bf5af9fe312bb95b029 : SUCCESS in 19m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/33743894f4b6451e92f3b37971504a23 : SUCCESS in 8m 03s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/565b2852326c41039bb27fae9c583b55 : SUCCESS in 21m 49s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/b8dba1a4fbe2480a9011d77799ddcb7d : SUCCESS in 29m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96ce081350d14e82873a3f5134d4696c : SUCCESS in 30m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/42a9e07893fc4f17b21b8c217d6d423b : SUCCESS in 8m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/02f4792288074a02acba7d44af26d227 : SUCCESS in 8m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec3b304891444a6ab72ee32d1f626dc8 : SUCCESS in 11m 59s","accounts_in_message":[],"_revision_number":1},{"id":"068b97d3ed50056942d95df8cdcad7e409275562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 20:05:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e736fa0b679663be7b1acf4fb9fad4e1955dca0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 20:06:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c0ab829a7646529a5bd1bfc86f9780\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/687b0fdaf93a4109804c7e0088cc867a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1785385012df43131ad48a40e4e5478552ff5a3b","revisions":{"1785385012df43131ad48a40e4e5478552ff5a3b":{"kind":"REWORK","_number":1,"created":"2025-11-28 14:08:00.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/98/968798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/968798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/968798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/968798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/968798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/968798/1"}}},"commit":{"parents":[{"commit":"c22bb5602bc7f4f5608cc8dccc8995e0eccd956d","subject":"Fix typo of the task name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c22bb5602bc7f4f5608cc8dccc8995e0eccd956d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-26 17:43:11.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-11-28 14:08:00.000000000","tz":0},"subject":"Handle empty remote_hosts in kayobe connectivity check","message":"Handle empty remote_hosts in kayobe connectivity check\n\nCloses-Bug: #2133047\n\nChange-Id: Ibecc0302a2a9620ef5c2bc93b73cb1a6ad71b4c9\nSigned-off-by: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n(cherry picked from commit c59e26189f8eb94c55f46aab30208204d0d5bbd0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1785385012df43131ad48a40e4e5478552ff5a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1785385012df43131ad48a40e4e5478552ff5a3b"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
