)]}'
{"id":"openstack%2Fkayobe~969042","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I988810541b4620ac196d216bf54896a774282fdd","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-12-12 09:50:53.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-12 12:05:34.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-12-12 12:05:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I988810541b4620ac196d216bf54896a774282fdd","subject":"Fix Bifrost host variable file generation","status":"MERGED","created":"2025-12-01 13:02:48.000000000","updated":"2025-12-12 12:06:28.000000000","submitted":"2025-12-12 12:05:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969042","cherry_pick_of_change":969022,"cherry_pick_of_patch_set":1,"meta_rev_id":"b8b63b4352bcd3f1222378bc1a133ba2298dcbbf","_number":969042,"virtual_id_number":969042,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 12:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-12 09:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-12-11 09:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-12-10 21:41:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-12 09:50:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 15:33:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-03 10:00:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-11 09:02:51.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-12-11 11:25:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-11 11:25:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"28671958cf8239ef7dc4f83974f82954cdf364c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-01 13:02:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dc572444a1e160b83ff798481c55aac307d175e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 15:33:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9617ab662d564565ae81c3b7c01c03fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1082c7f749e04bed88228b1f22e46365 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91625a7538d24f9b89d50b6232051a88 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d3482e6a8b5411987446f85116aa555 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f74c21c1d842b090abcaad48b52384 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bf5b52b533e448eb3c9be5210a475a5 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79dda929cbf142199b421a27cbfa3f30 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/204c67a7e155464aa9d7e1f78910fcaf : SUCCESS in 14m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b732643b7904f4cacafdfb783f694e6 : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/91da8cb6206940e2880c3b3376e0c1c9 : SUCCESS in 13m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/494ef6a7e5804bc3aa2e4cfe734f78ef : SUCCESS in 14m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9145e1d541d844b788592ae0afd88c56 : SUCCESS in 11m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1374905b20a74e6da3ea975fedcbecb9 : SUCCESS in 10m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5ee1207f8abd45a6973fd25548a8d98a : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d81204191fec4f1987db7a914f257e32 : SUCCESS in 12m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/989f252e333b4d3489e71cf5dc5302fc : SUCCESS in 59m 25s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/3297991bdc0b4af19ea80894cf88db1e : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cf11de69e9da4f2884a6e229a4b7582b : SUCCESS in 1h 15m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a3c0e4d41484615a1e23b65d6b6fd33 : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0557b6479f9d43138f90e2cbbe3a3b6b : SUCCESS in 34m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/24a9879f86a741a086fa752cde311a8e : SUCCESS in 1h 16m 42s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/622c1c28237e406d8ba099a8c2836f4e : RETRY_LIMIT in 3m 41s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a81176020f7446cbe9595ed0b35ee83 : RETRY_LIMIT in 2m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6fce969d28d482687039d0d83db326d : SUCCESS in 2h 14m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/eda46c9163a8460f9d84c86b799eeded : SUCCESS in 19m 27s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/01e85e5df13d410abd0d61c9bd0236ed : SUCCESS in 9m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e57eba11ccee4ba18ab7a5ff3f1284b7 : SUCCESS in 18m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8e3f2143d45e4b3a8385594ee67b5910 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/69e6c9605281458ab1f93218f2b77a9e : SUCCESS in 34m 42s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a97b695e81ad4fb8b0f964484f0ad9d1 : RETRY_LIMIT in 1m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8191eb7ab7944b24bb00163817d61791 : SUCCESS in 32m 59s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9449ec35d9824ee5828b43e5d19adfc5 : SUCCESS in 15m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/13724f81991647e9acb382092c0d0fb9 : SUCCESS in 14m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b759fa6cd76143348bc99a7cf1d0842f : SUCCESS in 8m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1bfb636b605c1a907ecdd3dbc38c931e3545c9a0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-03 08:36:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e81ec3858306910ef7020199bb4b0c7c8f72cdb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-03 10:00:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf0af1ea7a0064b02cd8af1c2cd589f412847a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 10:54:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96bf4587b34d49449f3c9c2e111ac628\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dacfdd86fcad467388b1c02a8462da19 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f32b9afd487c46248c2c82611bd9d01a : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c379163757f44bfcbe5375a1a0ab2b64 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a0ca42d007c4cb5af845d5eaa5da1f7 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fc6f7bae78149138058c91f2d968902 : SUCCESS in 2m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bbd63a8f76f4578bbad073cf85840f7 : SUCCESS in 3m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f61897fa8de64048bff5aea20420ea43 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eac556dede71484cabd58577dd97bdde : SUCCESS in 5m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0e1a72f829564e54928b4bda0336a6a5 : SUCCESS in 7m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d6130268a28e498da856fd956607a057 : SUCCESS in 7m 07s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f80af83b6364794a0092f592501fe3e : SUCCESS in 11m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/55348279793045e7ac18124d995953ef : SUCCESS in 9m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9282c45b8f86495ca36f6be9d7bdfa41 : SUCCESS in 19m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5dfa12d3ac2248a18949198b0f3aa250 : SUCCESS in 16m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/04e5156c944341e3b1fb1c4e0f93e852 : SUCCESS in 57m 58s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4abd301c09494c1b92be06c934d00bb8 : SUCCESS in 32m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4e23aa8e68884201acdb31a31b9b401c : SUCCESS in 35m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdc91e6aff364162a91d1c3eb7748f1e : SUCCESS in 27m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f8e52b7d813e4a238132c2079ae09a2b : SUCCESS in 29m 35s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/017de76a62144f669a522d03e1c74224 : SUCCESS in 1h 18m 15s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/7d4045802fce4b9885a4157049331985 : RETRY_LIMIT in 3m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3c172d136a54778b7d793c2bea5cf12 : RETRY_LIMIT in 1m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27529434aa61410a834aabb9b359b042 : SUCCESS in 2h 13m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c37d26847de043d89b496ea3ebf1fbf3 : SUCCESS in 20m 48s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/6495149d11e1403ca4ca54e72004cb52 : SUCCESS in 18m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b10f5d6f4674206af878f2d6b226c8b : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/643134ad7b104ddfb2466245cd7ba6f2 : SUCCESS in 12m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ede37daea0a94775b080b8e0a3dc8156 : SUCCESS in 35m 04s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/64d2e69d22ad4e56a92938b84cf2ffc5 : RETRY_LIMIT in 3m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c613d8fbafd49a881648bb32e060c7d : SUCCESS in 19m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2b71308ea6ba467986a10207686fc7f1 : SUCCESS in 12m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4edc87729d1c4954a8e79032c2b5d824 : SUCCESS in 7m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6e5541e056e425284db28758df7f2fb : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"419d14388c82b244abcdf1294ae022338448fdf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-10 09:48:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f95ad1a2918759606957d17549839f188791a83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 11:55:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68690669603f418a9c6f1f007830424a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdbb86c1ff9d4bd4974a0751ff8eb39b : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec39ad2b33848aa9fcaa539078086c4 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/365bfb980ec040b7815806230477a7a6 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57605fbef55f4b0ab54ca055bb81bcbd : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9969affef47f460aacf44f33a3c5d576 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c107322c1fb4f1bb8866bce0dd70921 : SUCCESS in 2m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ae5a019648c4774b6d48975f723ce95 : SUCCESS in 5m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/faade5089a6f472bbb53b30c5ca5396f : SUCCESS in 5m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9cb9d511a4424211bdf97f7eb2c29b05 : SUCCESS in 8m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/50a8518f7ef74c4c8b04b3c5bf4d92ea : SUCCESS in 11m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c077537934a94950b685280a6de0de25 : SUCCESS in 6m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b507f31895184bfab8b80d4bc4807b0d : FAILURE in 10m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4d2c3523d95e49139aeaa2bbf35dc22f : FAILURE in 5m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94faf540180c4c0dab8c74c55025aed7 : FAILURE in 4m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1376adea4329450a90ced84fb9768c36 : SUCCESS in 45m 41s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/94acef3062224928b69ae302b416eea2 : SUCCESS in 33m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a8296d05499c42098fe44be847f4ae35 : SUCCESS in 1h 04m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce8806d694cf4454b768bac502bc597b : SUCCESS in 28m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dfd4aa5d351d4fb78572b0dd1988615a : SUCCESS in 29m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f85cf9bc447744e891815787092c2a0b : SUCCESS in 2h 05m 11s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e5edcd8429e847218904fafb97ff377c : SUCCESS in 1h 49m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c85305b47f747359f642178386f24e5 : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3df5be177c2f486a9e75bf3ca3a157f0 : SUCCESS in 1h 59m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/648ab0397b4e4305a158cb716b9d141f : SUCCESS in 9m 26s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/613034707f8f49b2bf6cdd935b8e5176 : SUCCESS in 16m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5973156a33b1412d917691bc8ad7158c : SUCCESS in 13m 54s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e84ca451030f4fc0bbad5312607430ef : SUCCESS in 7m 20s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a8e5a4b0ecb448f689216639a9cb999b : SUCCESS in 26m 54s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1a5d0e004a4e4050a1bab0f6b243dba3 : SUCCESS in 36m 27s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a193f9c491145cb81a3cf63604f7346 : SUCCESS in 19m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d744855cc1d44435899b0b5b0b1e9817 : SUCCESS in 18m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/604a81a8a5794fc28c81837e0cd57872 : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd44492625f64b60b8f1ac23674f66db : SUCCESS in 12m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0aefdc78eb91eab9c791a5a381669744921d56f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-10 21:41:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"801556d41565e71788bccc603e11f9ef261377e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 00:10:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b112c8e7fa74f2097955b63d51738e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2374070879ce41faac3b98f803a6ebce : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbb5598e37e04968b27dde5663ef5cc3 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a2465600d934d5ca91f01527a3f5e3a : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f42d6b2b504cd8ac3d5750a46cfc1c : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7bb81db689345218c023fe9d6e9a00b : SUCCESS in 2m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f611865867d04d65b5904fdae38682c3 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2f5ea3da48b42e8ae0f8e1b84902c3e : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e920f8e6ab8641a89986e061e29437dd : SUCCESS in 12m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e12455e5da1b4f648f082c9f9dd21f9e : SUCCESS in 7m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6d580fdab329427a97b67137e1dff8cf : SUCCESS in 15m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9c8ec737dbd44b78ba93867b34b0796 : SUCCESS in 13m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4cc38e5697c3434898aa159cd1b0c098 : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c907b095423e4f6fb5ece32cb07379c2 : SUCCESS in 21m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e07eb2ead2554eb59323963882529483 : SUCCESS in 12m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a6a1400cc61d459bb5a4c28766d8cfaa : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/96b8dc0dff3e4ae1925edf8a76ebcafb : SUCCESS in 58m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/780cd97614024d6f908408bd7aa906da : SUCCESS in 1h 10m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a0e0748466e4892b22066a41e401f4c : SUCCESS in 58m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/93dbf857d2e1493dad96fa9fdcc434e9 : SUCCESS in 1h 02m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bfdfb111eff947609a63b8574189d7db : SUCCESS in 2h 12m 13s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/871c10c2bffb49b8bbf8851a7902c388 : SUCCESS in 2h 14m 19s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3b03e7acf0e4c579d0c041bbd5bf36f : SUCCESS in 2h 06m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/517f2986ac5d492eb7de2ad73612c58c : SUCCESS in 2h 02m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dbb60a70ff7149f4beaeb0e3d742f1ca : SUCCESS in 17m 33s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/842ae6eff8dd44ca8075044f3c70613a : SUCCESS in 16m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/345d4136dff94507ba796c6cfc51fd97 : SUCCESS in 20m 22s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cd38e57962d74ffaa63734e4080dce01 : SUCCESS in 16m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1757742e21784d9881c6ed5134aa32c3 : SUCCESS in 36m 01s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8ec77b1d248e4d39b59ec907c8df896d : SUCCESS in 35m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a2166b7a4564c9e94601f19a67dca54 : SUCCESS in 35m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b64571738754455485501a5252feb6c9 : SUCCESS in 18m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a96a65c9fbc4405cba9e11cf17e3d7bd : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94bac5654654436eaa18b7fb0a3a318a : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":3},{"id":"664436dd4e968fc3e3cc1bfbad39be7a6f276e2e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-12-11 09:02:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"257f30aaca0dc6edbeb66339ea099961bb44165b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-12 09:50:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"23329940b7365e6ec35d8c0cc99de8a6e6e400b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 09:51:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b415ca16e906aabd1bcbb62fff45391eb3ed3620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:05:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/667a1474e45b4ed48e2df4adc51fef26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa48e0109e049fe948323d9ab610125 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4023a202993343c2ac855cd8b466226e : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03fc9a39f1c4b639d0d6399f27087a4 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e4dd1c3dd27481ab02dd2d169406411 : SUCCESS in 2m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ab4bff219ec4a3f8065bcd610ce7f60 : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b33461cdecdd4d5e8b462aacc938bb09 : SUCCESS in 8m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e4569a80cb74361bdd53c43d01b3421 : SUCCESS in 5m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b7fb69ae86d5492a97ebc97a3939cfc4 : SUCCESS in 7m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8be94d7a332144c9aa79278e3e74bff9 : SUCCESS in 14m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6e01403d6f44885908846f91f81a137 : SUCCESS in 11m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a008be4c044745d1a6a372c0c898fbb9 : SUCCESS in 15m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/70f2972814624e70a80fecb4afd25fba : SUCCESS in 20m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d77883c5ca274624b6b6778d4b4cf115 : SUCCESS in 7m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ad9273ff8f834731b2801b68a317fc83 : SUCCESS in 1h 12m 20s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/8f6806b5c78c40748ead8be8a945ec0e : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e27b7756f5814cc6bd9682df5ae550c9 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a60db6f0dba45fbb5cc5955fc12ca61 : SUCCESS in 1h 04m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/81712a48a3f6499d93955141d0b2f7ce : SUCCESS in 56m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8db8588d96f54a92973eb9fcb16fea60 : SUCCESS in 2h 07m 05s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/df6347ae1e5740a68c6c3e748bff8eee : SUCCESS in 1h 52m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a258dc1ad23d4661bf71cd15cb9f12ba : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a71f3f185344bbaa9ae18940f7525d6 : SUCCESS in 2h 09m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e043f365e02242b281f616d2cf0f8a79 : SUCCESS in 18m 54s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/da2441d7fa4c493dacb51bed713756db : SUCCESS in 18m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae6d388feb474c1fb01c71e0905430e1 : SUCCESS in 15m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/94ce9140a6ba47e4ac8aacb4a34f8b70 : SUCCESS in 7m 55s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e816d55218341eaaa44bbe9f8ee6307 : SUCCESS in 33m 18s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/aacb1a226342414981f20372350a4fc2 : SUCCESS in 35m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aefaeec95d3743b697e94afb6dfa666a : SUCCESS in 38m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a514222b7e5b4c35b6a0aa4f5bf59a9d : SUCCESS in 16m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5aed74ff99774dffa8cd31ee098d1d39 : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1f22eaa974646c0b00db51722b630b8 : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"f855cb6abd55023409ba1d3624e8015775c31e0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:05:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b8b63b4352bcd3f1222378bc1a133ba2298dcbbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:06:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/285595319a28481d8aa776a8019f8e3d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f426819dd84307a2803215a2590a40 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc7a37e5b52f4a89aaf8a5ab7d9d7dc4 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"78d21cad1148bde59eec03f81e0564dd441be216","revisions":{"d3fdf8ef96212df52be49526846d61c5c8b1554b":{"kind":"REWORK","_number":1,"created":"2025-12-01 13:02:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/969042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/969042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/969042/1"}}},"commit":{"parents":[{"commit":"1785385012df43131ad48a40e4e5478552ff5a3b","subject":"Handle empty remote_hosts in kayobe connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1785385012df43131ad48a40e4e5478552ff5a3b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 09:13:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 13:02:48.000000000","tz":0},"subject":"Fix Bifrost host variable file generation","message":"Fix Bifrost host variable file generation\n\nIn deployments without an IPv4 gateway defined, for example when behind\na proxy, Bifrost host variable file generation was failing. Fix by\nallowing ipv4_gateway to be set to an empty string, which is supported\nin Bifrost.\n\nCloses-Bug: #2133489\nChange-Id: I988810541b4620ac196d216bf54896a774282fdd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit fa13678912cf57ff275768379a4af4e6d775f10d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3fdf8ef96212df52be49526846d61c5c8b1554b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3fdf8ef96212df52be49526846d61c5c8b1554b"}]},"branch":"refs/heads/stable/2025.1"},"94719afd7d0790f358d7c4fb92497d70b7ccdd05":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-10 09:48:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/969042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/969042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/969042/2"}}},"commit":{"parents":[{"commit":"ec431df188cfc721d54be3871997b724084347b7","subject":"CI: Fix SLURP upgrades due to 2024.1 branch renaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec431df188cfc721d54be3871997b724084347b7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 09:13:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-10 09:48:29.000000000","tz":0},"subject":"Fix Bifrost host variable file generation","message":"Fix Bifrost host variable file generation\n\nIn deployments without an IPv4 gateway defined, for example when behind\na proxy, Bifrost host variable file generation was failing. Fix by\nallowing ipv4_gateway to be set to an empty string, which is supported\nin Bifrost.\n\nCloses-Bug: #2133489\nChange-Id: I988810541b4620ac196d216bf54896a774282fdd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit fa13678912cf57ff275768379a4af4e6d775f10d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94719afd7d0790f358d7c4fb92497d70b7ccdd05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94719afd7d0790f358d7c4fb92497d70b7ccdd05"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"78d21cad1148bde59eec03f81e0564dd441be216":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-10 21:41:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/969042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/969042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/969042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/969042/3"}}},"commit":{"parents":[{"commit":"e4fd6fb22b9428da747a488720cb3c42757902fc","subject":"CI: Ensure firewalld is unmasked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4fd6fb22b9428da747a488720cb3c42757902fc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 09:13:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-10 21:41:49.000000000","tz":0},"subject":"Fix Bifrost host variable file generation","message":"Fix Bifrost host variable file generation\n\nIn deployments without an IPv4 gateway defined, for example when behind\na proxy, Bifrost host variable file generation was failing. Fix by\nallowing ipv4_gateway to be set to an empty string, which is supported\nin Bifrost.\n\nCloses-Bug: #2133489\nChange-Id: I988810541b4620ac196d216bf54896a774282fdd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit fa13678912cf57ff275768379a4af4e6d775f10d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d21cad1148bde59eec03f81e0564dd441be216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d21cad1148bde59eec03f81e0564dd441be216"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
