)]}'
{"id":"openstack%2Fkayobe~969043","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I988810541b4620ac196d216bf54896a774282fdd","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-12-12 09:51:04.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-12 12:21:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2025-12-12 16:49:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I988810541b4620ac196d216bf54896a774282fdd","subject":"Fix Bifrost host variable file generation","status":"MERGED","created":"2025-12-01 13:03:36.000000000","updated":"2025-12-12 16:50:08.000000000","submitted":"2025-12-12 16:49:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969043","cherry_pick_of_change":969022,"cherry_pick_of_patch_set":1,"meta_rev_id":"c85b773cfa3fa7c23de11f6ed1600db8937a7a88","_number":969043,"virtual_id_number":969043,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-12 16:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-12 09:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-12-11 09:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-12-11 16:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-12 09:51:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-01 14:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-11 09:03:16.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-12-11 11:25:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-11 11:25:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-12-11 16:03:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"f974e45bb31b307d691d478ede2f012516a5fc57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-01 13:03:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e1a6bfb8107b79048f034237a48c3bc8b3d51aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-01 14:21:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e4637b24fd74f8a88b5a2f561e51a6e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88769303e3124bbc81423de09a524baf : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7786e1bf5eb14353957df2e9d2ff6c1f : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22d906401ec742298d0ad15e2b4924d3 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e649752e1ff14eff915e3ca310f53e51 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050650730b634b3ea70147d1f6b037a8 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3866b900902a49e1b27a1551b8a17b5c : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b50abc48c1ea4963b829e92dd67da064 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5351c1abef43426c994aed8e90ec0c65 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3a5294895e3439bb664060966fe39fd : SUCCESS in 9m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3712b129352c41fba0a15d80c106d573 : SUCCESS in 1h 00m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/258e35f76312406babd38f92d0e2f374 : FAILURE in 5m 47s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f8d6c4551134f09a2edc8e157cbb065 : SUCCESS in 1h 01m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5d8f6b551f6946e4852698cbc42cb3a9 : SUCCESS in 1h 12m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/da4169f4994147b68d10d202b4d46386 : SUCCESS in 12m 16s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e208afd50bf445a4b039990f3ed674f4 : FAILURE in 6m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad09f19c9796414da1ab995fc8df597c : SUCCESS in 9m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6f4d08e08f6a418bb67deaa5ba30b940 : RETRY_LIMIT in 2m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06604e6a4b1648398145c4b3f021e379 : RETRY_LIMIT in 1m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c6eda6f263a44f88bea1d25d82c8a7dd : SUCCESS in 21m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9b3249367f44ee4843c86844623fb45 : SUCCESS in 20m 27s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b807202be7f243a58c95b3025e258e83 : RETRY_LIMIT in 3m 54s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5f20c5bc20b41e596f460f6836581b7 : RETRY_LIMIT in 1m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e34ab7658c5b47ebbc7404967cde5039 : SUCCESS in 17m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/679be2197acf4e7d8fce2659d57663e4 : SUCCESS in 16m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2b0cde87ddbb430db6de91043434646b : SUCCESS in 17m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/603ab8f196c945be9f9300b8f8949f93 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2f6b6066971c9c14df8f8e530889e9b6d999661f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-10 09:48:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7df1de58bd3c9e4f72009a479ad7589cdb3d1592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 11:49:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afcb5c93b11e4bb4a7ad5336e2f63f6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afd8084bf50547e5adf857b825f1c2ee : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed182f62325402382272d03c7d72cab : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a3740e8483a4128a87b47a3ae8f351d : SUCCESS in 2m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73d32e3861e74289a5fc63bd75e246c5 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197a40dce7344592b7a32542c6db6f4c : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7e16275c083443da625586c3af41143 : SUCCESS in 2m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09a7eee7acdc40f48fe70f5a2992ec74 : SUCCESS in 3m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2e389764560c4d758736b1b845d22e78 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f12f27093ea4404b0fd6b3f47ed7fb5 : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8c84788db84246eeb7194634ddcc1685 : SUCCESS in 49m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/82e2aff6eabb4cb1a650e2713c1490b4 : FAILURE in 7m 45s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7330490f25f4d1c95501fd88ba65d8e : SUCCESS in 54m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7ede7a5b8d274d7f9a2e69f58b1ea095 : SUCCESS in 1h 23m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cc9bcf87a3ea49d4a11af900bb1cfe2c : FAILURE in 15m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5921d248d689433ea95ea4e431a33673 : FAILURE in 5m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ec120f6dd5f428090b401a604d781e7 : FAILURE in 4m 29s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/32730222e2ad40b3a3b0b79a5bfd49a8 : SUCCESS in 1h 56m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a22eb51c67444fdabcd1fa8ecc5269e : SUCCESS in 1h 52m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/46def24e2ec644bbbdef40e46863602e : SUCCESS in 13m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a342dfb147944d3d86c919edd49e67b2 : SUCCESS in 12m 04s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/29b0d4584c104748aac20ec999d4b829 : SUCCESS in 34m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/831331cffa674d02800eef78e395def8 : SUCCESS in 19m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eff9f3f9ccf3473ebfa1444fe79e8a98 : SUCCESS in 18m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5276f38856c41c59048a697205086d9 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/65857b3ed3124df1a52d4c087e062850 : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f843c0ee0ed64f20b8f56235f1e8f8ce : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4d11d281553c0063c93c0fbb520be3dc491cbafc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-10 21:41:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a034831f2ea2228e4f54b126a14985b9f631f67c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 23:50:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/912dc94f424d4306835928c218becd67\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50999b6ac5584e49a129479e991ba859 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f078b55ad7654272bf83802e5648c514 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bb5827a616143cabbcf1e42198c6b16 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d8d9a60b2ae48b191a000b6a7b917dc : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ace5f98e074bffb1e4bd0e5d918079 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4e3b05f80584340adf521bc128c7d49 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8176349e57f64b1f9c4b5831edc3b942 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b4db177da4f4f4a91f3bc26f53a613f : SUCCESS in 13m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5627b8a2b0c4e189f55b8e63655dfb3 : SUCCESS in 10m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/72f5ed11e3564af7b197a4ec279f2652 : SUCCESS in 35m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7401cff56aa843dd93d6f97dd0cf8f8e : FAILURE in 6m 11s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24d8bb6dd21d4663aa16fd1c242a206e : SUCCESS in 1h 01m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0be388fd1dc34fc9be1fe68aedd955d2 : SUCCESS in 38m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/882ced2fddad42139ef94028b4601940 : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6048e2c553a14a2e844a9f682ad258da : FAILURE in 12m 11s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b260c637385474ba8dd7b3be7cfb08d : SUCCESS in 8m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7fef3037128644af8f00bcd8ffac549a : SUCCESS in 1h 47m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/496cadb12c884f9c8bda8cf27f5d2cb4 : SUCCESS in 1h 50m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bde61a21a6384f2a88e4a8ad389ee160 : SUCCESS in 13m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22d82b72fff743129a4e6ddb80313f7d : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d5a96eb505e541d0991aa1f36074d9fa : SUCCESS in 25m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fb55d0a20df44918f0513ad2dbacc58 : SUCCESS in 35m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6fc59fd10eef4b739b02cf8d0fd7198b : SUCCESS in 19m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/964f21e6ce0649639e21f67ff5084f67 : SUCCESS in 15m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/35bcb6215dfb47b4b512a01da9ce4e7d : SUCCESS in 18m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dbe15475271541d0b486d35467251b5a : SUCCESS in 16m 20s","accounts_in_message":[],"_revision_number":3},{"id":"e9d4492e6c376fb313d0a63994a20d475997fc9e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-12-11 09:03:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"57ce6c12ada853f9a7867cd0dbd15ac341842325","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-11 16:03:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5d397513438e07507051b8e8ca97859cf40c56af","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-12 09:51:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dec95cc46b3c96fb0014253a72d8ab3de33ded8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 09:51:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"baa6d94da920dcf58cd1eb1481512791e703da3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 12:05:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5081b28c3f746c0b4af8c5c6d6e24f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a42ceb02024980b1a8277c8a53e430 : SUCCESS in 2m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd331499e5bf44b2aebe695ea0c5d588 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb3aa97146a4aaf8f35009b30a0145d : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa635c5ead0d48dbb1ff2321aec35d0c : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/afde7964c92c43be81ca582fed2371d4 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c64e65cbc2f41f9bfb14dd7b6918362 : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a146424b81ed4f5191e4bdcb38c9aac7 : SUCCESS in 5m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/38f23eca218147d09437d1a9fd8a0cb7 : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39dc22b8db4b4fe5b3e88f6353895ffe : SUCCESS in 1h 00m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/14e5616fd4ed4e37a51e1e963c76ac8c : SUCCESS in 39m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5d37a6332ee84d50939e87c5807fa8c6 : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d36a8fe8d672411db374a558ef9cf879 : SUCCESS in 19m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/afeb6334bc1e4d249f57e1aaa7ec99c3 : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba72ae225b0240b7a2f888b867543a6d : SUCCESS in 1h 32m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cf384d2127be418a8f5b4b505a3abd4e : SUCCESS in 13m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/966a62a3293b4f47a7031988d8b0f86c : SUCCESS in 12m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/73cc68bd471e4d288c718067bcee5786 : SUCCESS in 32m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a85998d52164b2c91f71d0e8db2aed9 : FAILURE in 12m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9cd41516f1e442fdb01d7def5afb7e69 : SUCCESS in 20m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe5e8384bebb4dd68d11d65894a8d4d7 : SUCCESS in 11m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b5e8f0f2d8a42a782db00950ddad7fa : SUCCESS in 19m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7c637eefec7462796d13ba03dc41263 : SUCCESS in 16m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9e2b62a80bbd5a9fe24dba3c2a307112e5dc0323","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-12 12:21:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"617fd4d5ba112364cb385c88f488dd9f6a202728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:25:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a127f9fdb5f24e648e2ff36a910de466\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1342f3e690244323851a668ceb667a52 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302c160ce76242319a0fb3ba91f1d467 : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ee0ffc97966458b841462d14d6ec366 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bac69b516b344d9db5989cb4be6322b3 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d76a9b711e45479e5dbb66ea1c7ab2 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/541c5e8427f1421b8e0e85494a2fb4a1 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74aeaa54daed488786e4eff5a738b0c8 : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37b1a85038a34acba7eb251e5691ea14 : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/13ddf8cc32d94e50835c6c21d97de1ab : SUCCESS in 8m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b0dcb2bd69e84667b07dbc891619f8d9 : SUCCESS in 35m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8edc6e1c651b4197a6659c3c0490267b : FAILURE in 9m 32s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbe300e1bdd14775aa2a7f0148fd93a9 : SUCCESS in 1h 11m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a2ecc3e8d59f4cfea0f4b0b30f688866 : SUCCESS in 1h 14m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7a2188aa7ab145d899f9c804b5a2af41 : SUCCESS in 19m 03s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/54e02e139c504039986a0734122a4997 : FAILURE in 10m 48s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/823a4ff46ac741368e8b943d4e1f473a : SUCCESS in 9m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e2dcb1c22234050912053f982777ae2 : SUCCESS in 1h 58m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1717060dc15348959409e0d1e77f749b : SUCCESS in 54m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fca8edbe270e46da8e5e923fa07e8ddf : SUCCESS in 29m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f619e53caf0f4d54813f99ebd0fd2d14 : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/41ab05b93f9f4f0cb1c8c477c0195c00 : SUCCESS in 39m 50s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5373eaadae764fd294fee4ef343f269f : SUCCESS in 30m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82f44ba9669e41c0adb051a849ef7d30 : SUCCESS in 17m 23s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b77805097d49435c9273eb284c005642 : SUCCESS in 10m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b16b2d3c540f4e77817a6ff241783a4e : SUCCESS in 11m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ee1cefc1ac64664a01e59b6c4a80e54 : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":3},{"id":"aff3a78cb02330a1a07eb5697887c30242d54a12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 14:25:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"10abcbd53bd2a70b5d09a008e5ed2c5732d93bc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:49:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20f315dd02d43f2a8a7855d3ec62cd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200e4159e3df44ccabd6fee537146477 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd65f9138958418abce6df196701d05c : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b1318518a3846e6b376cc972ad87243 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e161b43948646b0aaf60e08b5b4cd47 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92ef4a2a9c2b41c988bac0c1bdd96e93 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/441ecdcbd21d4aff913c22fc03708816 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78a69deb628e47488a4eeb35b4811aed : SUCCESS in 8m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/39a828a584ee4eddaa585c095ffd14a2 : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c264bfa4e30942c5a55b2147a8aa5624 : SUCCESS in 58m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/09fd24d31def4259bce2ae46b3302cb0 : SUCCESS in 1h 13m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3bce0564698c4512ad68d5409ae7a3eb : SUCCESS in 18m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a15e1e858ac34b17b6408fde94b919b4 : SUCCESS in 14m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c2809740145c418298e777b510cb7b37 : SUCCESS in 2h 23m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2504fd95a854173b1c7d456a8895cc3 : SUCCESS in 1h 41m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/82e9a9b43e1e412abf96bcb843222877 : SUCCESS in 20m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a864f7cace3473ebaebd3532090e016 : SUCCESS in 18m 40s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f949e4d4ec2649f98071f22d2d6e74aa : SUCCESS in 32m 57s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17bd18995490418ea31de236ba15d5c8 : SUCCESS in 29m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a6084c17c18c438a95196c1f3efedd17 : SUCCESS in 18m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f191fe70d704851bee48fc1a8b3c7cb : SUCCESS in 16m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f7fab923b3694ec9aeb0dc45177eebc2 : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f23dd28653c74e13a3947729f0baf55f : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":3},{"id":"7d4dbd976a4b2e067ad160b96688e684d7db9be2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:49:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c85b773cfa3fa7c23de11f6ed1600db8937a7a88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 16:50:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eefc0daa4b6f4a7db56d6ee4fd615ab9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb84a211ed214b74b8960ebbafd7c307 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ea8bb56d6a54bb39143a22d17fdcf17 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25b2aa74a088562be885dedf2bcbbfa1bf80d2ad","revisions":{"41826c2ee93d507233cce7193419c6781f28060c":{"kind":"REWORK","_number":1,"created":"2025-12-01 13:03:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/969043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/969043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/969043/1"}}},"commit":{"parents":[{"commit":"a09e6763910267e95d03a79b4887a73087729c94","subject":"Handle empty remote_hosts in kayobe connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a09e6763910267e95d03a79b4887a73087729c94"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 09:13:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 13:03:36.000000000","tz":0},"subject":"Fix Bifrost host variable file generation","message":"Fix Bifrost host variable file generation\n\nIn deployments without an IPv4 gateway defined, for example when behind\na proxy, Bifrost host variable file generation was failing. Fix by\nallowing ipv4_gateway to be set to an empty string, which is supported\nin Bifrost.\n\nCloses-Bug: #2133489\nChange-Id: I988810541b4620ac196d216bf54896a774282fdd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit fa13678912cf57ff275768379a4af4e6d775f10d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41826c2ee93d507233cce7193419c6781f28060c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41826c2ee93d507233cce7193419c6781f28060c"}]},"branch":"refs/heads/stable/2024.2"},"b406e399fd2961c44bba2202c78ed8f35bb6be39":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-10 09:48:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/969043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/969043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/969043/2"}}},"commit":{"parents":[{"commit":"809c4e3dbb22322b1b82ca8f10d59579923eee27","subject":"CI: Fix upgrades due to 2024.1 branch renaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/809c4e3dbb22322b1b82ca8f10d59579923eee27"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 09:13:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-10 09:48:33.000000000","tz":0},"subject":"Fix Bifrost host variable file generation","message":"Fix Bifrost host variable file generation\n\nIn deployments without an IPv4 gateway defined, for example when behind\na proxy, Bifrost host variable file generation was failing. Fix by\nallowing ipv4_gateway to be set to an empty string, which is supported\nin Bifrost.\n\nCloses-Bug: #2133489\nChange-Id: I988810541b4620ac196d216bf54896a774282fdd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit fa13678912cf57ff275768379a4af4e6d775f10d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b406e399fd2961c44bba2202c78ed8f35bb6be39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b406e399fd2961c44bba2202c78ed8f35bb6be39"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"},"25b2aa74a088562be885dedf2bcbbfa1bf80d2ad":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-12-10 21:41:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/969043/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/969043/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/969043/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/969043/3"}}},"commit":{"parents":[{"commit":"b0589b563197854257ccc0e05d5342d47fa8d2bd","subject":"CI: Ensure firewalld is unmasked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0589b563197854257ccc0e05d5342d47fa8d2bd"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-01 09:13:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-10 21:41:52.000000000","tz":0},"subject":"Fix Bifrost host variable file generation","message":"Fix Bifrost host variable file generation\n\nIn deployments without an IPv4 gateway defined, for example when behind\na proxy, Bifrost host variable file generation was failing. Fix by\nallowing ipv4_gateway to be set to an empty string, which is supported\nin Bifrost.\n\nCloses-Bug: #2133489\nChange-Id: I988810541b4620ac196d216bf54896a774282fdd\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit fa13678912cf57ff275768379a4af4e6d775f10d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25b2aa74a088562be885dedf2bcbbfa1bf80d2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25b2aa74a088562be885dedf2bcbbfa1bf80d2ad"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
