)]}'
{"id":"openstack%2Fkayobe~969573","triplet_id":"openstack%2Fkayobe~master~I36e92d7f4eea44b7f0b1606e9f09a2e6fd6210f3","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-08 16:51:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I36e92d7f4eea44b7f0b1606e9f09a2e6fd6210f3","subject":"Revert \"[release] Use OpenStack 2025.2 release\"","status":"MERGED","created":"2025-12-04 10:14:25.000000000","updated":"2025-12-08 20:32:29.000000000","submitted":"2025-12-08 20:30:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969573","meta_rev_id":"0333e3445fc441785ff90196431587aa35228641","_number":969573,"virtual_id_number":969573,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-08 20:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2025-12-08 10:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2025-12-08 11:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-05 08:03:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-08 11:31:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-12-08 11:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-04 10:17:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-04 11:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-05 10:10:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-05 10:10:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-12-05 10:10:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-05 10:10:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-12-05 10:10:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-12-05 10:10:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-12-05 10:10:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-12-08 11:31:28.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"06c28e665c37f08d3e58b778fec31672b49aab6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-04 10:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7d2e9a05ec7dbfbb97e42d9baeb5c0363bdbcf0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-04 10:17:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"47bfdee3ce9ef553482c8901cfff57ce79863d9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 11:44:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0c5f5654cab46d1a57c8e735d1f98a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d177aec6ad49444eb14a56b78740ea61 : SUCCESS in 9m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8fc8f9a0454144cf9da4fe2a6ccd37eb : SUCCESS in 7m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7fe4fab454f41eab4fc80866b76c56e : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7350fbeca1c14c53b041e65b7e34df20 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c5cac6a87124defbc8ecc6631b76cd1 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e8646b8702e4506a55765b308dddd54 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1b2f9b7e3e04ff3aa9d7a1c7e4c7c2d : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aea490e153c1486fbaff29d60e826cb4 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9286c97216f4b6f94dfc2283e9d8bb0 : SUCCESS in 11m 09s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/db253402713e4c479d706bfa9811edac : SUCCESS in 8m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/210bd0a237b347c1bfb93103dffa8798 : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6129124511d040a889a534dc71732b0b : SUCCESS in 25m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae012459198148719e3c752f3cb4d7cd : SUCCESS in 15m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e4025febbe3c424cac9b59e5480e2f6b : SUCCESS in 1h 21m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2315d1f915ab4207b027d0eb67540b71 : SUCCESS in 1h 19m 38s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/14155b3ca4e24ecfaf198d2277307e32 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68a137a025a846dbbaac2a9ed7da1636 : SUCCESS in 1h 12m 54s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/85d3b9dd12ae4baba20ad12777efdd6d : SUCCESS in 1h 12m 03s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96a98ce2cf5d4261926df471ed183e9b : FAILURE in 1h 02m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5fe1ae3abc53444fa644155d1676da2c : SUCCESS in 9m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/91184495c6814a938ca216b1ea5e41ce : SUCCESS in 29m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa0b1234b60e472c80fd985edcbaf2ca : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ea675502cd4487ab68a224fb0e02c59 : SUCCESS in 27m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf38999137b74676b839ab64d9952b5c : FAILURE in 11m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b0c3353ae28d44599805e5a13666e2b4 : SUCCESS in 8m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df6756daae844582b7ed614afb6fe609 : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fdba6a9972adce4cae2bdac74ad8358110e2d9d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-05 08:03:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f86ab4d7d22a2ddb6d17bf872017610239db4cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 09:43:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/169321f1de4446a192c1c27afbe5b481\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360e515a415d42f396b9078706f19d4a : SUCCESS in 9m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9795959ea0b441792966789d118da3b : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36c8a986ad294bb2b99b6ddfe7c30753 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80fcc30d2d494d068feb3b8f3180765a : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f352be49965d457988d8d310cb845a64 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/962d2969048e449ab7cd13b9588b6091 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/157451a66623499790cbbcfe6b543ff4 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d9514651a1b48c58b9740bc1b9f4ebc : SUCCESS in 13m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40be511d28e84209b4124465a0a04d30 : SUCCESS in 8m 57s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/81cac04b12804eb5bb55790d83eeb078 : SUCCESS in 14m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/537521619ba944afb20d99b8640fccec : SUCCESS in 14m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e44d04b82d634b4db467987d7bf06439 : SUCCESS in 10m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b06b724a2ee4b0588bae828e340b729 : SUCCESS in 12m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/52bdbe7d0fe24e5994105a4f37822639 : SUCCESS in 36m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c06795de4b8c4398a2d1163bf6a331c2 : SUCCESS in 1h 21m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2b71033b92a84740b4a73bec5b10ea78 : SUCCESS in 57m 58s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d93d132375e148049c5a5e870e88b373 : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/85f0a86360104a5181424cba71954b2a : SUCCESS in 55m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec8d0b1863ac4b19927dfd1dfbdb5640 : SUCCESS in 1h 37m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/baf9a492c0c74202958f24f916291012 : SUCCESS in 9m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/38d63ba4bbf84f699d8d39334859d45c : SUCCESS in 9m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d65a2397fcc414a97c3a9443a7edc2f : SUCCESS in 7m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b8d1c882cbf94458a39e3a92f998753a : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36cea367485a4dbfa41ce54811690cfb : SUCCESS in 14m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6fd6db8febf649e7937d12d567c9d061 : SUCCESS in 14m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75f76cf3b790488497d6ce13401541b7 : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":2},{"id":"31c48cb0c6134d48889ea54b0b52d0ee69a6d79b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-12-08 10:42:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"af8c0a2284091464e2f021ce5f2bb50179c50b64","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-12-08 11:31:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9117edc90268ba0f0c0766b288e70a9d37cee292","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-08 11:56:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"330ca0059bcdeba32d70e550d15df84b6927e275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 11:57:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"800c786b2141db8fecca6af561c26fdf3e037d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 13:29:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d3f9665331a4a98acadf711afab4d03\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e417e7b75855457482366eb9f470c921 : SUCCESS in 4m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6f8a6b40bd124d6395086e5cd068d67f : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/599fed0f3acb470c9730b8e853361210 : SUCCESS in 3m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c7ab543d54e4b9794729697fc7bb06c : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66284fc95b0f4bcfbae0c243ceb63b58 : SUCCESS in 1m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c3673684f334e66b0c15d51b4f88157 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9bd549c24c80463e9162b9c8741d20fe : SUCCESS in 7m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b2e2225d4474d1b85e91ea3976b9f36 : SUCCESS in 10m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca72cdb91b79473eb8eb80de55db81b1 : SUCCESS in 14m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5566f524b486444289e7bf8f56988f63 : SUCCESS in 13m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/96ac00ab7ef8425c8087896ef1e90088 : SUCCESS in 13m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0948af4ab80346fbac58c9ed1a4dd24d : FAILURE in 13m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/47c8967770c8422d8036377a98625ee5 : SUCCESS in 1h 17m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f55ba7a2c7bb4e18bbd9410c916b96da : SUCCESS in 1h 02m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d2c9d7629e0742799ceba3b8fc529bb1 : SUCCESS in 1h 23m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d0c7fde43e544d7a592bb2b266ca1e4 : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/51544884789543c58e537779c1a61130 : FAILURE in 24m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcc7c0edbdae44fe81b81ca127bcea1b : SUCCESS in 53m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7c80b7c779044a54acc8c94ceae4f6b9 : SUCCESS in 10m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/377da4ab9bcb4d2592c2cca3c8f9ec27 : SUCCESS in 10m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02946a7f971b404fb108eaa4a6ed9163 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c9a136ccc06344c69a57284da64ff799 : SUCCESS in 7m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/284c40546dae44beb92cc765f62b83b7 : SUCCESS in 30m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0865e6a14a3c4d71ae6924576bb4caa0 : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d22905c07d5410abcd0e0a82ba65fb5 : SUCCESS in 11m 04s","accounts_in_message":[],"_revision_number":2},{"id":"fc1d3cab6c415ab2022ef2d3f3d8e6e59f88b5ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-08 13:36:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4561c810a2e50d5dd89f43ce07804b00d03e4ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 15:38:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a16481a425574292b1083ff0cf95d05e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d21617c6cb40539112829c029641bb : SUCCESS in 8m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5361eb3729fb47199871ff82b55c4e0e : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0101fcdbe0a042ab933bd79e301233c8 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c37ddb574c1f4d378b01b41b900c4905 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/146e438b5f104795a5e470387b3102ed : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/263af2a9325e4b61957cb5db0b0f8b86 : SUCCESS in 2m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19115676929b4183b53efb0ce46b086c : SUCCESS in 2m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a97b9bbbe20149709ff2526f8b0d8719 : SUCCESS in 5m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/864a5eccd7fd417f8bb8fc3d71167288 : SUCCESS in 5m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f8293c1a22fd49bfa8e62b4bc49cfc0c : SUCCESS in 8m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce5df54ec8704369b543bbdd16f2dc27 : SUCCESS in 12m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7bf7187849284b88954d83be12d29df1 : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ff704ce370e466ba6fdd9190d2d7335 : SUCCESS in 14m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/02a107eeaefe4170b54317c00816f9f8 : SUCCESS in 59m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1ffc1223f638468e9f9a9de5569bec51 : FAILURE in 26m 28s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bb784f598f3449e091170c0d805d29a1 : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08bc5c79207f4442be681a8b9f99ac01 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bee445f05d5e485ca41b91c49a1362f5 : SUCCESS in 1h 11m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11e1de52aad64a61a4512410a425d466 : SUCCESS in 2h 00m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2ec1d62000814afe84e597afcaaf4cc3 : SUCCESS in 15m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ef50c896b17d48aabf48745392a81fa8 : SUCCESS in 8m 35s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/259802b8782c4abd8c2f8ffd033adb97 : SUCCESS in 7m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7d6967eea40844f2b3a6cd0445b13051 : SUCCESS in 7m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3c388d3b61840138e929f4778e9a107 : SUCCESS in 19m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cd2d343e014c467e867db897aeb511c8 : SUCCESS in 8m 29s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7347ccfbff9c46258deb497fb7ee6122 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d95a64c39fe402e5c8348b6c36aeb2b06323cf30","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-08 16:51:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99cdf6d0432e65bc3902fdb837f20c7bfb2eedb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 18:36:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a3f81248a1e49b1bf02ee9ee847f127\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f17139a5b904fdd841d6a3f00a9cd62 : SUCCESS in 10m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/85c2540b63c74993ad8e120f63226a4b : SUCCESS in 6m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7677eaec93024efe8499c3fb324663e4 : SUCCESS in 2m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9bb31d5377c4285b01f77ebe0471998 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6893a8424436409dbc9daeca5dce4611 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/886721943ab843569d24aa8f78471adf : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/18f0c6bb909c4330a35beb8de9594498 : SUCCESS in 2m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/29a0c7f6d9624feab54ccc1dd1c83a46 : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc8d0257b4d34d8d8e7ac0a57012c063 : SUCCESS in 9m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/41325642698745caaf34350380f86ead : SUCCESS in 12m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74aabdc7cafd4c9db9ac6b9003779756 : SUCCESS in 11m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/54d3692a0bb44ccba4632e2419ba3b4f : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed45cb81eeee47c183f811b072f234ae : SUCCESS in 15m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4308d157581b45d787ce013fc6953f2e : SUCCESS in 1h 18m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cfef721ace8c4db9b8329847dc704f84 : SUCCESS in 1h 19m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d6a3bb26422b44d6a2f570eae8cc53dc : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/346747ab2c85430896a91e896a715b59 : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/07906ff8d0d94d1db89122b254ce1883 : SUCCESS in 1h 08m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75d7a768d421489fbce04fa5752381fb : SUCCESS in 1h 41m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b83b96d784004a02aeb20bfd27e158ed : SUCCESS in 9m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/23cd601942da4bb1918d236e2f4e5265 : SUCCESS in 22m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d003ac371eb4d4fa6bacb07a27ea1b3 : SUCCESS in 14m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ec0915d002543bd9904fb1b7b2add6a : SUCCESS in 12m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ee7fa53e1aa436c8f221144accc1410 : SUCCESS in 20m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/95be489be7ec4a179694a9491e854c11 : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/204d937fbdc94baf8126a9b5c21758e9 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":2},{"id":"75796df4b798178caff459207dd742d21ddc6556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 18:36:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"24514320317d310f53d78f10003a096d0010d6dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:30:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab746a6d0e24d48b6163daed93745d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0db2c9a83d47df992f588da4ec8494 : SUCCESS in 8m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/803bfbc268014ebe93833f9ccbd32d0a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9011aa01bf3c4521a66b58fc86aa70eb : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/987c0e6a0e5347ec879bc1fb247b1753 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a65b2d5327cd4ec8b4665151262678dc : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d99d6ceff554be2b49304a5670f8454 : SUCCESS in 2m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/038fbd76d0f249ec82bfcb536b69a25e : SUCCESS in 13m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d410441ab5f64721a6fccb095f833317 : SUCCESS in 10m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a7942f466eb9489daebe830f2425b2bb : SUCCESS in 17m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/addd7d9aed814c69a7578e9549793cc5 : SUCCESS in 13m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8d7f3ba4f716489dbf323d10122d14af : SUCCESS in 10m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0e7e831856949ba97d5a4d0cf64e3a1 : SUCCESS in 15m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/3a981e7cb3d74562b08f0a3643370e47 : SUCCESS in 34m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a2bf95e97c484c8ba399b49a9127e63e : SUCCESS in 36m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a8986dd94d70417d9028cce377a1e424 : SUCCESS in 37m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/477db8da8e3c4c84b253d18f99948246 : SUCCESS in 1h 20m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4e6608aa06354771aca6523fd09416f8 : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/131368aec5134adf963acd99c38f6697 : SUCCESS in 1h 49m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/767f8b726ee34412ba53ba0483945378 : SUCCESS in 19m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a395fdbf5f2a44dbaa1e736c57dbe14a : SUCCESS in 16m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e7b6bad48ac4043bdd7712b455c417c : SUCCESS in 16m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d9482c280b7e4fea8e9536e73ab9b016 : SUCCESS in 7m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/faf0eee774a44a0d83ada1d8802e44b9 : SUCCESS in 26m 38s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fbf38a39fe4b41de9c57f93f99890ab9 : SUCCESS in 8m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f2ec8a9c4994398b603c301ea2ee162 : SUCCESS in 12m 46s","accounts_in_message":[],"_revision_number":2},{"id":"df30eba05e2962d7469efc62447de15e6300f706","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:30:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0333e3445fc441785ff90196431587aa35228641","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-08 20:32:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf793e0a45c4046aa073ba77c79bcde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6c6fcc044648ea8f433776a8269ade : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfee159d6f84420d94a600ac7124e07d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec066451766f43873081b6f1120791ddfe05ef09","revisions":{"ee73c64d0fcf9297fc76f34aace21018dbcc1784":{"kind":"REWORK","_number":1,"created":"2025-12-04 10:14:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/969573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/969573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/969573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/969573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/969573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/969573/1"}}},"commit":{"parents":[{"commit":"fa13678912cf57ff275768379a4af4e6d775f10d","subject":"Fix Bifrost host variable file generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa13678912cf57ff275768379a4af4e6d775f10d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 10:14:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 10:14:13.000000000","tz":60},"subject":"Revert \"[release] Use OpenStack 2025.2 release\"","message":"Revert \"[release] Use OpenStack 2025.2 release\"\n\nThis reverts commit 237e8b07e046b1736f74f648e4c489644e4aad3c.\n\nReason for revert: 2025.2 has been branched and released\n\nChange-Id: I36e92d7f4eea44b7f0b1606e9f09a2e6fd6210f3\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee73c64d0fcf9297fc76f34aace21018dbcc1784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee73c64d0fcf9297fc76f34aace21018dbcc1784"}]},"branch":"refs/heads/master"},"ec066451766f43873081b6f1120791ddfe05ef09":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-05 08:03:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/969573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/969573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/969573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/969573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/969573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/969573/2"}}},"commit":{"parents":[{"commit":"88f30f24ad40c3dff5e02030931755c5c99e863b","subject":"CI: Stop using infra cache servers as registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88f30f24ad40c3dff5e02030931755c5c99e863b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-05 08:03:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-05 08:03:35.000000000","tz":60},"subject":"Revert \"[release] Use OpenStack 2025.2 release\"","message":"Revert \"[release] Use OpenStack 2025.2 release\"\n\nThis reverts commit 237e8b07e046b1736f74f648e4c489644e4aad3c.\n\nReason for revert: 2025.2 has been branched and released\n\nChange-Id: I36e92d7f4eea44b7f0b1606e9f09a2e6fd6210f3\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec066451766f43873081b6f1120791ddfe05ef09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec066451766f43873081b6f1120791ddfe05ef09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
