)]}'
{"id":"openstack%2Fkayobe~969587","triplet_id":"openstack%2Fkayobe~master~I27ba36000d4f5842f27fc49230314fd54e434dcb","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-04 14:45:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I27ba36000d4f5842f27fc49230314fd54e434dcb","subject":"CI: Stop using infra cache servers as registry","status":"MERGED","created":"2025-12-04 10:44:35.000000000","updated":"2025-12-04 18:13:22.000000000","submitted":"2025-12-04 18:12:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969587","meta_rev_id":"773fec6fad1cade77088de81ebf41ed86e6d90d1","_number":969587,"virtual_id_number":969587,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-04 18:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-04 11:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-12-04 13:02:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2025-12-04 14:00:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":2,"date":"2025-12-04 11:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-04 11:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-04 10:48:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-04 11:01:09.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-12-04 11:37:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-04 12:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-04 13:02:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-04 14:00:54.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"db3c77c3d2dcac4c515d4cce6d154181896ae9be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-04 10:44:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc375cced79e5b9adab6786a8a2672ca99a7d889","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-04 10:48:56.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"07074042208ea88e9fed61c97cafaf048c9d5de1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-04 11:01:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4380279a75285aed6314d58563b3b356eef96ae9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-04 11:37:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d81cb249ab413657dff1d84fc0908979073d5c73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 12:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce7191e4c9894ff0a3769339172f2418\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63510bbb47ec4d3aa1cca215176e3e92 : SUCCESS in 9m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99a0689b4d514bdf9b25c19ba1a8f7a8 : SUCCESS in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef8e36149785496aaf301630fa3c6cb7 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a663c7298ee4924aa543208920d8fd5 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11511e1381314286a9c10ade8113e061 : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4c324c80d2946b7a7ad8607f774a921 : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d56f57c2b82b4f4e8f0bd081b2ed8fcb : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/785074b15a5b436099b453119d7e101c : SUCCESS in 11m 39s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/71ad65574e664fefbeb030bb0d8a9e4a : SUCCESS in 16m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64fbe80a3fd7452ab7fcf24faa382480 : SUCCESS in 6m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e94142fa30534478847aeae9ed669035 : SUCCESS in 25m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e278ab4c67140d280ebae992efabb95 : SUCCESS in 8m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cd9dfcf2faa74116a867cb670a42b14f : SUCCESS in 1h 20m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f46f4439cab543648756978ca6c0c2f0 : SUCCESS in 1h 16m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6528443601f349509efc01bac96abf6d : SUCCESS in 37m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abfd85a1955a44968a122ae6e6571328 : SUCCESS in 33m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f157c31ab0a94c79a7832291352e2778 : SUCCESS in 56m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24ade8f964ad47d89f6e3dd12f5d9a8b : SUCCESS in 1h 39m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/38a2497c8d94490ea56e307d798e0e8e : SUCCESS in 10m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ab8ca2f356fc4011a3bfe2845fcceb02 : SUCCESS in 22m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a48689b4b814747b2c678364285c410 : SUCCESS in 8m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/79a90f64e32c4069b060a1d169039b5f : SUCCESS in 15m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ae6fa624bfc488ba7f28d5a8e9b6555 : SUCCESS in 15m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab67f7c104da4907ad0eebe5dea6bd7b : SUCCESS in 8m 37s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c40b336dfb341208f1c0b37a0d65d5f : SUCCESS in 16m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f418f517c74c3cdb902c4851c6c307553fd83d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 12:33:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ceb451de5dea9dbda2a453c7d29565ef817f6162","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-04 13:02:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c044541aef03e37a1a2fa398504716843c461cb5","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-12-04 14:00:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3fb91866f77a5817860a78090e8c167e133095eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 14:22:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95b60c7b53df4ac4a7114b2c6747e391\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0602c2d7fff4f1aa804b78ba5185c82 : SUCCESS in 5m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2783242a355c42e2b2818e5068642f39 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98fcfd73ba94c66abe30a7b5a4f38bd : SUCCESS in 5m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/54185c2a82624e8bb4aab86902e795d6 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b9bfb34e22e45e6a168669e9de48b01 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c653610e648406d8cd087b09b595002 : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5ccc3755df24846ab0ee987bf1a0aec : SUCCESS in 11m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e65d053b6ba64dc3aee224f13c8f6832 : SUCCESS in 9m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/839bdd1d7f264101bc29e11bba96325f : SUCCESS in 7m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5a89fbae4b754f97a558e3ad0a8c44aa : SUCCESS in 10m 45s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f994ec6979d744afb757cc24615e530e : SUCCESS in 15m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2d57c85f0d09435caaea025d5dbcdea1 : SUCCESS in 55m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/04703be4108745dba59c59b5d02f5ea2 : FAILURE in 7m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6d5326ce373e47cc81a7a3ed52e26ab7 : SUCCESS in 37m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ca58245519c47628a55975bf7230281 : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1c3ee7826644fbd85be9b4c0622bc09 : SUCCESS in 1h 00m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9168e3d0197e4c01a427dfba8f77bac8 : SUCCESS in 1h 48m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/25712305fcf54840ac260ac62f85ca17 : SUCCESS in 13m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3de0db685f8b47399fe3b09565444856 : SUCCESS in 10m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74e554dce5aa419899af37fa801c3331 : SUCCESS in 8m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efb2a6444d7c4d7e989d1b10b6e40768 : SUCCESS in 8m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e739302a76d4106981604755bc169bd : SUCCESS in 25m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f3432782abd4f9d90ca360a5f1f0561 : SUCCESS in 9m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffa3cca362d344e1b32728084d2afd5f : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3b36e216e2e146aa03cf851deeeffb185bfb2a64","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-04 14:45:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7d636f09dab9f5eed70037383f87de4ce2fc002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 16:52:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b1622096e52490f95d73cd0c24a37a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667e9a8f41014077a1ff561ae4fb764a : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea8a3d826ccf4ff5ba07d36124351ab7 : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1790255df29f4623b2ad416ed9cb8149 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff184693028749b79c4393125aa05854 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d32658fa7c09491790039d2c3405a78e : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/074fc52991b04d99ad2c491407ae7358 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/298ff9efbbb74e0981809736f0ef95b5 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6249592c86bd4069b5611bfd981dc355 : SUCCESS in 11m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4736d7c0ece64d1cb0b385abc62ab57c : SUCCESS in 8m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb06b2acc0384747a27f314c453ca270 : SUCCESS in 6m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d0edd479b4bc4095a14c20d332c25749 : SUCCESS in 23m 52s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17fc05d1c59641bf9dd0198710a88228 : SUCCESS in 8m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c17d0d0dfaa648768e38bfb1e3c6599d : SUCCESS in 38m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/152d47ae674e48ab987e9d3a15ffff0c : SUCCESS in 1h 21m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/98a4bc8b2e68425f88426933cb76ca9b : SUCCESS in 36m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/652d1cf5e22e451da259e9e8c49f118b : SUCCESS in 1h 08m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d65808e20455414789abdcd5b207d058 : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80d76ddbe2cb430ab7ffdae64141fce5 : SUCCESS in 1h 43m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a467fa68b3d14d60b3a1787fd3393e44 : SUCCESS in 9m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bd44339de9604176a130d8e87741c22e : SUCCESS in 9m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcdc339cd5e24c989914cea16a9eaf44 : SUCCESS in 16m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1979a5f41d65443eabfec2f7ef53d682 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5132e5598c6443c29297f56e0de6a812 : SUCCESS in 29m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5af5da3c271642a99fba38611da85414 : SUCCESS in 20m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/092c7acbe9f3444b95cf64f742657060 : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f86dab88859b0ddf2b352c4daeb039a0b7eaaa92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 16:52:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a879142524644bbcfc5002f410cb112fadd886bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:12:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e322d905ef497cbba2d658b0b1e493\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfceaa36e6ee45e48f0bac88ac234f3f : SUCCESS in 9m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8e74d0d93ff64b56b45fdbf4375d0bf3 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68436d83429e48e2a8da3c3d7fe69b51 : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20fdd6912d91448c80ab1c644fb8cbee : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a173eb1b39184237a211ba51230089ad : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/54fee1c88c1d42878daed454688cbbc8 : SUCCESS in 6m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/491e9e1415464fb19b78ecdce8368f5a : SUCCESS in 11m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3429426292dd4931a833e42593f47d3d : SUCCESS in 19m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50bf9894f0e1478c822d213f3b440835 : SUCCESS in 6m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8b9c2793690d49a69d41cbaaf8fdc517 : SUCCESS in 17m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f561ea9c9c843a38ed0bb798e843447 : SUCCESS in 8m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5c4b6ce8cca74853b36d369aa07c6321 : SUCCESS in 35m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b77ee869204343f88fe813615ac0a867 : SUCCESS in 37m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/991b5148e3b7492198b1231809db7014 : SUCCESS in 37m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4dc9aa27cb147beb508d08dbc126ebf : SUCCESS in 33m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7bb6ea7e36214e5296d9357b5f03d5ee : SUCCESS in 1h 15m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa28f1450cc94794aad3f5de3e64b70e : SUCCESS in 52m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/16b3c32db70a47c98f18f99d4d8c609d : SUCCESS in 23m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/14ac2168aeb84e0694729ca62844ff6e : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09f82f7c6f4a46d9bbeeb48c826351e4 : SUCCESS in 16m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d71beb9b17dc420e9c73edc151157c15 : SUCCESS in 15m 51s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35df8524b3774bdea62b274e0ec60c93 : SUCCESS in 28m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/766cb604a1774468a8884d535313bb75 : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19c7912892f741649c2af6dc66be5e48 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d35e6ee87b9ae486e617cac776580b8a8c78b827","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:12:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"773fec6fad1cade77088de81ebf41ed86e6d90d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-04 18:13:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe7fb2d4f9494b118033816b41978360\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c988068847449ea9d58ca7f1f7d5349 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88f30f24ad40c3dff5e02030931755c5c99e863b","revisions":{"88f30f24ad40c3dff5e02030931755c5c99e863b":{"kind":"REWORK","_number":1,"created":"2025-12-04 10:44:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/969587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/969587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/969587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/969587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/969587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/969587/1"}}},"commit":{"parents":[{"commit":"fa13678912cf57ff275768379a4af4e6d775f10d","subject":"Fix Bifrost host variable file generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa13678912cf57ff275768379a4af4e6d775f10d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 10:42:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 10:42:16.000000000","tz":60},"subject":"CI: Stop using infra cache servers as registry","message":"CI: Stop using infra cache servers as registry\n\nWe are seeing frequent CI job failures due to timeouts while fetching\ncontainer images. See similar change in kolla-ansible [1].\n\nAlso remove the kolla_docker_namespace variable from CI overrides:\n\"openstack.kolla\" has been the default value since 2021 [2].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/967159\n[2] https://review.opendev.org/c/openstack/kayobe/+/812947\n\nChange-Id: I27ba36000d4f5842f27fc49230314fd54e434dcb\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88f30f24ad40c3dff5e02030931755c5c99e863b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88f30f24ad40c3dff5e02030931755c5c99e863b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
