)]}'
{"id":"openstack%2Fkayobe~969859","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I27ba36000d4f5842f27fc49230314fd54e434dcb","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-05 11:19:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I27ba36000d4f5842f27fc49230314fd54e434dcb","subject":"CI: Stop using infra cache servers as registry","status":"MERGED","created":"2025-12-04 22:16:48.000000000","updated":"2025-12-05 11:21:05.000000000","submitted":"2025-12-05 11:19:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969859","cherry_pick_of_change":969587,"cherry_pick_of_patch_set":1,"meta_rev_id":"60a98e063c52a33a36baeb89e37271459763ed6a","_number":969859,"virtual_id_number":969859,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 11:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-12-05 08:25:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-12-05 08:17:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-12-05 08:17:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-05 08:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-05 00:45:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-05 08:02:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-05 08:02:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-12-05 08:02:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-05 08:02:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-12-05 08:02:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-12-05 08:02:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-12-05 08:02:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-12-05 08:25:47.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"7f679e62b5215c540bd2ad79008546b2a87378e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-04 22:16:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8fce367d5b6de86b2876c83512599020a62a226a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 00:45:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fca321bdf9db4245b1f2ef4e1d6f2a42\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63bd7f211ba4421cb69ff36bc3e3ff4f : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f9c81aa20a40c3809ffb31f117232f : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54d6bba7028d47fe985c03c8d5020ada : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f90ea429a74443a7b191c00eb9743a : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99b05643c75549ba96460d594948f785 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed6d45e084a943be82d545214bb6f344 : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/48776639cbe14217bda748be3d867715 : SUCCESS in 13m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e13d4b9bc1f040228f4c736b516dab4c : SUCCESS in 8m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b09ff11d2dfd4506bfb727fe8156c2d1 : SUCCESS in 14m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b592e3b08a54379bf0a6886ae49a7a7 : SUCCESS in 13m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a2bad1d137604cf293dd406fc01ab742 : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0cfabcaeda0c47aba9c9fede2a339a72 : SUCCESS in 10m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/786e51c1f4e6474ba8fce9dc3cd05afb : SUCCESS in 14m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8cd233ad020c457c9694d38129667427 : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/4f58719a3c034760b7921ee16104ac7f : SUCCESS in 33m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4488c7fdccfd40eca830f622b87caa85 : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69bd8205553041ebbaf6c1395f1cb679 : SUCCESS in 49m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6b87f30c6b6448ea086c59afe201893 : SUCCESS in 46m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9682cfd198a74b9e847e75cf133cad0b : SUCCESS in 2h 12m 02s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/501d7115e8794e7c90b3e869f409ffd1 : RETRY_LIMIT in 2m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47bbdbb2255446b287a3816b74a573e1 : RETRY_LIMIT in 1m 35s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ae18916f43e47a2a81d6cc230499a56 : SUCCESS in 2h 25m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/82a424422a554c36abce5e735e84ec7f : SUCCESS in 21m 27s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/3c86010915244f91a8882a3f76f4100f : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77e8b8f041d0468680e1fa57f8d23519 : SUCCESS in 8m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f6b863cb7e3e42278ebbe1325300a854 : SUCCESS in 7m 33s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/63299f790f7b465fbd4fcfb1a07c5a28 : SUCCESS in 21m 33s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c3de30379a374151b1836fe470a8d9f1 : RETRY_LIMIT in 3m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35e68288ceac46e1970f5bba7e6a523e : SUCCESS in 24m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a4a517432e0d4339a0e61e61f1bd5da4 : SUCCESS in 12m 06s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/26ce463161b0413c9f175b2d07a09d40 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52369dcc5c064a47a79dd51a0e0c2718 : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0d98e9aaf28b5b4a9dcb456703618cc148e47318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-05 07:57:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e623cfc65652da264db291f6f2ea02e6bfd2c18c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-05 08:00:15.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"eab9b5830412f4a2fb3ee3b9867c7d24e1f27a84","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-05 08:02:26.000000000","message":"Patch Set 3: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c9ed4a8f8e6c90253fb8adbdc606445d2f96b946","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-12-05 08:17:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d26dda51d6c95c86869e83eb556422d756423752","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-05 08:25:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"41f723d3a2a7931eae5441d76568423cec2553b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 10:03:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78743a16e5f14efabe4463d7470c58ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d563d27526499e87dcee92711cd96e : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17bde123b6c495695d29af0366bff8d : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5950702364794dff90eb0eec46b6775f : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bc864877514fcf988e18eb124a2947 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2c7f97173d14296ada8dccd9ebc83b9 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/15a70192199f4579927792a97f445905 : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71c379bccd2b4f40b7cb846389f17238 : SUCCESS in 5m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/136d8d79fa7c4a0ab5ea19cf041c29fb : SUCCESS in 8m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cb50b11c5dbc40cb90422d823a191cfb : SUCCESS in 13m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f6758be32e84a1799a97d235e29dd7a : SUCCESS in 10m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ac3999d112bb4c2a8eb42393979649fd : SUCCESS in 10m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4952a10519c945559d3b7cbf31558cde : SUCCESS in 19m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4224f615daf44abc8051b359b5f32a72 : SUCCESS in 13m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9be095560b1c4f228df4afdb971c78da : SUCCESS in 31m 50s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/470c5a0e4ebe4123a0131aabf20604aa : SUCCESS in 1h 00m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e0c0b180055042b0915636135e1ea78f : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0d2e8ca9f9d4d9a9ba30370bf339626 : SUCCESS in 57m 35s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8bad02daefdd45708b6238a65241aea7 : SUCCESS in 29m 59s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d70ca86090064c81a69f6ca051203bf4 : SUCCESS in 2h 00m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/035ab2c7e06848ad8e8f570c88bb38e4 : SUCCESS in 1h 08m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e7eb4eb4284742ae8c7b20af0de70b4e : SUCCESS in 9m 55s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/71fe9b5b218f4613b0f5599410207038 : SUCCESS in 9m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7113268f00f149d284cc1cbd87a1329d : SUCCESS in 15m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/81df859c813049d0843924aa763cbddf : SUCCESS in 7m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/57b73c1f84d04732ab172c2fc111ebaa : SUCCESS in 30m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27f94deae03a43099f39cf3508da0d36 : SUCCESS in 38m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e44320ae2b29404cac7248c6314e3309 : SUCCESS in 14m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d81014e1c774a92aab67d6839816476 : SUCCESS in 7m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dee61cd73c424620aa29cd8631ed6441 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":3},{"id":"312527f2ee941730ee2acb11ac0d50e450fafb68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 10:04:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1aa0408543f60f2265492c56efb3ff761438daf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 11:19:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfcd80ce41d4fee92f01e78145ee05c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75975ab521047ba917ff602b4a2af8a : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12cea0fddb9d4095a8dd66a36c293fb1 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62aae9f0744747aa8eb89d497c423e7e : SUCCESS in 8m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a975cf4ca9a46498bc60ad3befc0a20 : SUCCESS in 3m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49e8628763c6415d8f265fdc757dc987 : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb74c89dbced4da2adfa9e48a78fc36b : SUCCESS in 10m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/406e977af3d048eda1a4d03d14feeafd : SUCCESS in 13m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/aab5d589782c4d07aa197aae1fe3555d : SUCCESS in 13m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/800934bf21a847098879d9a3530de373 : SUCCESS in 10m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/014999b801314aaa97330c76feec70c9 : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a0eac089799d410cbf637b94fe0d6636 : SUCCESS in 20m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14b2df97425f49d58b6d7f7156d12524 : SUCCESS in 13m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/88067b4486354e2ea01d2c0885c56429 : SUCCESS in 1h 11m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/aae63ff68bb6417baffd0f1f59765430 : SUCCESS in 1h 03m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/66c88a217de143639cffdb2d27ac2bb6 : SUCCESS in 1h 10m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6914a6318a0f4b4991f8e87a594f48c5 : SUCCESS in 58m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a08b6111556545c58502397f46df57f6 : SUCCESS in 58m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dc8beddf99d544e186882e067afa3211 : SUCCESS in 1h 13m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/460d85d310224773bf864c46715a6c65 : SUCCESS in 1h 07m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d42469e57c9b496180ccd226f1f37cde : SUCCESS in 18m 49s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c83e7780c5ac4de982928351e23fb278 : SUCCESS in 17m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1ff51e2aca342cbb7300bd04852d90e : SUCCESS in 15m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4563fdd4624a42e1b1b1462b04faa59d : SUCCESS in 14m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/76aabce17772419e8f7d6295754d1913 : SUCCESS in 21m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dbbb40ce07248cca27cd054c6de27b2 : SUCCESS in 22m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/feaa8d68349c4e758838417a6acf5dc2 : SUCCESS in 12m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/42097bf0de5e4c8faa43ca59bef4a645 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f295a5b7a1b431dbed47100a38cc0c8 : SUCCESS in 12m 01s","accounts_in_message":[],"_revision_number":3},{"id":"afd56a422994dd7e85dc5f7d5928e2a123e86acf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 11:19:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"60a98e063c52a33a36baeb89e37271459763ed6a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 11:21:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65ead1aacf214b8b96476e9f3c3b72ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a75832039ab4e92bffbec00e68e9e4c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6f3995e95232d563bb0a8eec80fd48d009b5947e","revisions":{"a494de34dcde0dbfd6804ca7290a9b070b766184":{"kind":"REWORK","_number":1,"created":"2025-12-04 22:16:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/969859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/969859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/969859/1"}}},"commit":{"parents":[{"commit":"1785385012df43131ad48a40e4e5478552ff5a3b","subject":"Handle empty remote_hosts in kayobe connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1785385012df43131ad48a40e4e5478552ff5a3b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 10:42:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 22:16:48.000000000","tz":0},"subject":"CI: Stop using infra cache servers as registry","message":"CI: Stop using infra cache servers as registry\n\nWe are seeing frequent CI job failures due to timeouts while fetching\ncontainer images. See similar change in kolla-ansible [1].\n\nAlso remove the kolla_docker_namespace variable from CI overrides:\n\"openstack.kolla\" has been the default value since 2021 [2].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/967159\n[2] https://review.opendev.org/c/openstack/kayobe/+/812947\n\nChange-Id: I27ba36000d4f5842f27fc49230314fd54e434dcb\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 88f30f24ad40c3dff5e02030931755c5c99e863b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a494de34dcde0dbfd6804ca7290a9b070b766184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a494de34dcde0dbfd6804ca7290a9b070b766184"}]},"branch":"refs/heads/stable/2025.1"},"fc99715ecc2056fad93aa265bd777d8f5f542e27":{"kind":"REWORK","_number":2,"created":"2025-12-05 07:57:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/969859/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/969859/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/969859/2"}}},"commit":{"parents":[{"commit":"1785385012df43131ad48a40e4e5478552ff5a3b","subject":"Handle empty remote_hosts in kayobe connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1785385012df43131ad48a40e4e5478552ff5a3b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 10:42:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-05 07:56:42.000000000","tz":60},"subject":"CI: Stop using infra cache servers as registry","message":"CI: Stop using infra cache servers as registry\n\nWe are seeing frequent CI job failures due to timeouts while fetching\ncontainer images. See similar change in kolla-ansible [1].\n\nAlso remove the kolla_docker_namespace variable from CI overrides:\n\"openstack.kolla\" has been the default value since 2021 [2].\n\nDisable SLURP upgrade jobs temporarily until 2024.1 is fixed following\nits move to unmaintained status.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/967159\n[2] https://review.opendev.org/c/openstack/kayobe/+/812947\n\nChange-Id: I27ba36000d4f5842f27fc49230314fd54e434dcb\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 88f30f24ad40c3dff5e02030931755c5c99e863b)\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc99715ecc2056fad93aa265bd777d8f5f542e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc99715ecc2056fad93aa265bd777d8f5f542e27"}]},"branch":"refs/heads/stable/2025.1"},"6f3995e95232d563bb0a8eec80fd48d009b5947e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-05 08:00:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/969859/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/969859/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/969859/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/969859/3"}}},"commit":{"parents":[{"commit":"1785385012df43131ad48a40e4e5478552ff5a3b","subject":"Handle empty remote_hosts in kayobe connectivity check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1785385012df43131ad48a40e4e5478552ff5a3b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-04 10:42:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-05 07:59:57.000000000","tz":60},"subject":"CI: Stop using infra cache servers as registry","message":"CI: Stop using infra cache servers as registry\n\nWe are seeing frequent CI job failures due to timeouts while fetching\ncontainer images. See similar change in kolla-ansible [1].\n\nAlso remove the kolla_docker_namespace variable from CI overrides:\n\"openstack.kolla\" has been the default value since 2021 [2].\n\nDisable SLURP upgrade jobs temporarily until 2024.1 is fixed following\nits move to unmaintained status.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/967159\n[2] https://review.opendev.org/c/openstack/kayobe/+/812947\n\nChange-Id: I27ba36000d4f5842f27fc49230314fd54e434dcb\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 88f30f24ad40c3dff5e02030931755c5c99e863b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f3995e95232d563bb0a8eec80fd48d009b5947e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f3995e95232d563bb0a8eec80fd48d009b5947e"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
