)]}'
{"id":"openstack%2Fkayobe~970361","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ic72f2e3a262d6491da8bc39de19e23c3acbedf89","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-10 16:23:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic72f2e3a262d6491da8bc39de19e23c3acbedf89","subject":"CI: Ensure firewalld is unmasked","status":"MERGED","created":"2025-12-10 11:25:14.000000000","updated":"2025-12-10 16:24:48.000000000","submitted":"2025-12-10 16:23:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970361","cherry_pick_of_change":970306,"cherry_pick_of_patch_set":1,"meta_rev_id":"2620584fa8332e055c731b40d619438d25d2932c","_number":970361,"virtual_id_number":970361,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-10 16:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-12-10 12:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-10 12:01:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-12-10 12:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-10 12:01:43.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-12-10 12:35:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-12-10 13:56:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9023be3a86590e06bc57357eb0a6dc72e63f68c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-10 11:25:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f19d71dc08d3759d50dccd0ba5cee4be8790563b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-10 12:01:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a1ce5006af47688298089dab31130fbd92ba00d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-10 12:35:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e724e117b0a89c3c0d0864ec63cf470321e9b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 13:56:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/045313fef9714f469a021f7b3fd3f45c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/046ab50e356e40189b7d543b46cf8588 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639f72ad039c4388abb779744c6d3dac : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c9dfd17df704dd8b3cafffc033ac258 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce61983f6154a1bae0e5945dae8be54 : SUCCESS in 10m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/894174cc92d64db8bca74a1a9566a126 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0ce77cbc82c74790945c2dcc3fd1fb47 : SUCCESS in 6m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06e62aebafce460996060c5c45100c8f : SUCCESS in 5m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/655ba125d9764f399c18a891fb16ba58 : SUCCESS in 16m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6e51bf2dcee64d659932e684ec006d19 : SUCCESS in 14m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b9788576324995b0376f3820e7360e : SUCCESS in 6m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ef5a0a35801443eaaac856725c934b19 : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bd4a1f26fef849baa827991313208b54 : SUCCESS in 19m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75675ad9c87a40a9988bae7ac14dc795 : SUCCESS in 14m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/88a6dbc377224d76a7be9c94d538b504 : SUCCESS in 1h 04m 04s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/e1ddb237fe55441faa03256c456145a8 : SUCCESS in 34m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/91fb4ac3b40847b79980e46af7847adf : SUCCESS in 35m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad5fa4f1e33741a88caaa4cee163ba7d : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4882fd5492af484fa8706e7d3e6ea67d : SUCCESS in 58m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ebb9506be7a54a88a30d6be4acbc60a4 : SUCCESS in 2h 20m 43s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/3c13a837dce3409899ca6268a1da52cd : SUCCESS in 2h 21m 31s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8d1b74f42474779b705b8418f4120bb : SUCCESS in 2h 08m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bddd9b059fe4bcca24c59706d5a6972 : SUCCESS in 2h 02m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6410e309e42844a6ac2b6bf0072492e8 : SUCCESS in 19m 57s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/ea838f060fb54eafa33da5746ad5834d : SUCCESS in 16m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/331eb4f5c5fe4e8b9140a59f591e5242 : SUCCESS in 15m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/54f7a2c65f7141c98f29ed3166af1a13 : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2ed9e829cdbe4c40a7d8336b2c852818 : SUCCESS in 21m 18s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/14c68d0eb76b4987b2cb1d7968a0d17a : SUCCESS in 21m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f548a64a8194dfd905c7798c6e062be : SUCCESS in 29m 50s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ae0bcc9871474e1c806aa82a5379c5fd : SUCCESS in 8m 28s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e013279fb63e49c59c8bb534aadf23f9 : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8676787baf3a4ac1bb8f6f3af10fcca2 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"820a302620cc717a788d2b0df14ae6b7f0a04926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 13:57:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4a15068498e21293b61cea373cba9ea80cce59b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 16:23:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2218427d58e044f9ae70d41b040398a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b86f314f9fd8489e9f5a1a2905d044f5 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c36fa273b4e84acaa16dc2a02976b0c7 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e20dec9e2524d8894450dfdc3941298 : SUCCESS in 5m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a09253925090454e9b13823a2d072f31 : SUCCESS in 2m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2f4f4f474e4447090db46abe6ab0404 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35b5a7fbc93b4a5cb70d99211f6bfcb5 : SUCCESS in 6m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/11bf2b5e05db4e9c810c390c84c86318 : SUCCESS in 8m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9474f74e0fed4099ad92e119f852473c : SUCCESS in 8m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72e4ade6d6654412b504b0317a966ed7 : SUCCESS in 9m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c06038874a814a56b4ece8d9b8013fd5 : SUCCESS in 16m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6fdd04bd3b704075b7fbf71be58a71bf : SUCCESS in 11m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43c719c7971d43a4b55b7ffdcd42472f : SUCCESS in 13m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a10e67db9b5d4b219ea36e2790959e79 : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/54a34d950e404f3088a9c8e26a834206 : SUCCESS in 1h 03m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fd9623f16f544cbb8a4d0b34cd42195b : SUCCESS in 36m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ca00450307749dcb66667fade76f771 : SUCCESS in 54m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/276535d6bc834f5dbccc5c10053753a7 : SUCCESS in 1h 08m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1b75a34db2144512b7f298fdba4c5c09 : SUCCESS in 1h 17m 09s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/103d5c37bb88406687670ebbc206a3bb : SUCCESS in 2h 14m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/267a1892232041e38ee5469c49c81fe0 : SUCCESS in 2h 21m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0365f994967e4b78b41167cecdd60e51 : SUCCESS in 1h 59m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/616ecdc02cd94bb2bec12494cf8c6d7b : SUCCESS in 16m 35s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/f45f1bc4d3024c3c90b4a0f68b610941 : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30617f5d9e004b68b95a0f15c6dbcdcf : SUCCESS in 14m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2382c2815aab4bd18db47b56276ff778 : SUCCESS in 13m 59s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e2dd6d9e94e04c28ae34682bd1760431 : SUCCESS in 32m 32s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/16d7df6a542544c38c514e78179a7734 : SUCCESS in 33m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7578012fad94f48bbfebc0546bd6009 : SUCCESS in 33m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dcb43404891a4a3e8ef54bc244fe2a6f : SUCCESS in 9m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/064d9a26606a4550baf4a2d7257ea5a2 : SUCCESS in 15m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b171dddb55047efb97e806e876f6847 : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"26c7576bff0c268f711f3bf8f44b0ed111476b0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 16:23:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2620584fa8332e055c731b40d619438d25d2932c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 16:24:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f2aae8878b4c31852b9dffd0b0517f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b164b728b5494f7fbb327d1da85362e7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4fd6fb22b9428da747a488720cb3c42757902fc","revisions":{"e4fd6fb22b9428da747a488720cb3c42757902fc":{"kind":"REWORK","_number":1,"created":"2025-12-10 11:25:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/970361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/970361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/970361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/970361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/970361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/970361/1"}}},"commit":{"parents":[{"commit":"ec431df188cfc721d54be3871997b724084347b7","subject":"CI: Fix SLURP upgrades due to 2024.1 branch renaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec431df188cfc721d54be3871997b724084347b7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-09 20:52:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-10 11:25:14.000000000","tz":0},"subject":"CI: Ensure firewalld is unmasked","message":"CI: Ensure firewalld is unmasked\n\nRecent nodepool images are masking the firewalld service [1], but we\nneed it for host configure jobs.\n\n[1] https://review.opendev.org/c/opendev/zuul-providers/+/967962\n\nChange-Id: Ic72f2e3a262d6491da8bc39de19e23c3acbedf89\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 3c2cf92507ca0a493d1454b34dcc47f2673f830b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4fd6fb22b9428da747a488720cb3c42757902fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4fd6fb22b9428da747a488720cb3c42757902fc"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
