)]}'
{"id":"openstack%2Fkayobe~971003","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I92bf70e1444b17d310a9d63ad53c5edf523a9bbf","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-16 12:20:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92bf70e1444b17d310a9d63ad53c5edf523a9bbf","subject":"[2025.1 only] Deploy centos/rocky containers matching os_release","status":"MERGED","created":"2025-12-15 06:10:50.000000000","updated":"2025-12-16 12:21:45.000000000","submitted":"2025-12-16 12:20:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971003","meta_rev_id":"0c62a507552730b445ef2f4c0ba65e1e1f54dc1d","_number":971003,"virtual_id_number":971003,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 12:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-15 10:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-12-15 15:31:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-16 10:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-16 10:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-15 08:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-15 10:29:56.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-15 15:31:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-16 10:00:58.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"9c7192a2fb1412b6383b575ecc3d9691158d5628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-15 06:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"042dd85bbb1f8c5dcb928133d1fea1c91525a73e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 08:47:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/499d10b6d9584d0ea973e1a8adf04563\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e3a5a866e3948918668738c7c265dd6 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4118db9d2841e499185c48a5669cfd : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d4896bceefc42c887bb53252f151644 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444dbd85a2614da8828c35fc41da7cc1 : SUCCESS in 8m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c46e72eda8264e6680f6f9432fc90781 : SUCCESS in 2m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8959ea46a274ee08ca21f624652a123 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e11233caea04ec8a8fc8b7f23e6b975 : SUCCESS in 6m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e5f208325ceb42a1a3e8c05298b52a7b : SUCCESS in 11m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0b3e748ec0d7424aaa08784867dafb72 : SUCCESS in 14m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f637d87c20dc43f4aec0cba077d0c5ed : SUCCESS in 11m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f4a1d8defc184ffdbb23893712837a60 : SUCCESS in 9m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4afc2fec0b90411b957b1c63af9462b1 : SUCCESS in 11m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfe2f36ec63242e39959667ba07cc686 : SUCCESS in 10m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e9f6fedf40d749289873e46fffa53b6c : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a663cd6eb1a54db8996537f3e643e39f : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/58aef5b4d9b34fab90c8bb5f6852c657 : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30fcd388ec754e1eb8e1728c983b90d2 : SUCCESS in 49m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/60545b2ebb394c92bc7239ce23e4bb7e : SUCCESS in 58m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a0d671ffcf9d4263a7a8ff2ebc173abc : SUCCESS in 2h 00m 53s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/0b1221c985e245f3802235d475d0bd19 : SUCCESS in 2h 28m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1a97ac16faf44b1a83e762289ca1642 : SUCCESS in 2h 09m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92825b5eacb54744af95782c05796531 : SUCCESS in 1h 55m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4ae28858e62644629fe9366b02e77198 : SUCCESS in 17m 11s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/edfff7c443a24c86956092d15f0923e5 : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66cc7c1c502a4575bc4f924484e2b5fa : SUCCESS in 8m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/43bd8d0904424a74a2a58ba8b0dfcf47 : SUCCESS in 14m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/50301c451aef4de7b291a9c758544f25 : SUCCESS in 34m 46s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a28338dd8e6f41e7bed70239ede888e3 : SUCCESS in 27m 38s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d796027a978a4dee8d16b0e4efdf4dfc : SUCCESS in 32m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fe1abb0d84f142ac9af848f31f21b81d : SUCCESS in 7m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/05e452f2dc634e9c80eb43c073731f7b : SUCCESS in 8m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/031c544f7e4443dabf3ed17a7a9d15e7 : SUCCESS in 9m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3ecb907ac330af061f6159c604b9993aa83a0784","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 10:29:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27a49f4c06e69057987bdf35455457deef9d0bb3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-15 15:31:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a488218d1107cfc7569494398455f103de7c89b0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2025-12-16 10:00:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72a4ef875f90e6cb7dd1fa31cbd03a31b009f5a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:01:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e66734cb023816e8b95189f8120007aaa6f6165a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:20:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be6165fb1784d7c9af6119b195d31ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53751f06c51043129bca2c346a169126 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91af3d78f8264bc2934e81632fe34908 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7156db74dbd464abf6af578f35363c9 : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57a8a72c6f63446eb0d78b05b3a9385e : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e2107c8836943309d64f4a8ba246b12 : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36d1c53140984d4c9838c8b782e77378 : SUCCESS in 8m 24s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ab79d137bc5f47b88e22534ae4137e0e : SUCCESS in 13m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8e131571031b4fb0b9e87eec8c2688a0 : SUCCESS in 14m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a04446760a9a49ada1f0a248537b4bd7 : SUCCESS in 10m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7aff596efd6a48f8843d0c3fbeacf937 : SUCCESS in 10m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5222187efa7346b4b9f039c227e49492 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dacffa45d944428b0503fb842a871d3 : SUCCESS in 13m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ae2de711de744870aa72cd2124bdd4f1 : SUCCESS in 32m 50s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/cc65aab4ce3e4540bb58efdd46c948b5 : SUCCESS in 34m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cf2a40dcbd5644be80fed9e5f525fe8f : SUCCESS in 1h 06m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/51c673cd39324df89851cf61c5bb3fa5 : SUCCESS in 29m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c53ef7e083a444ceba6ba06295625fc9 : SUCCESS in 58m 46s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/dbd50948a9524a4090fa854271fb28be : SUCCESS in 2h 04m 36s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/bc4a066cd0de44e3b04017cb8b69485c : SUCCESS in 2h 13m 10s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a3095dc1e2c47b284fe6680846aa750 : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a6864132c944861bd889fa57b2e1194 : SUCCESS in 2h 07m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ecaa210f1d804ed69890518a1f080f54 : SUCCESS in 20m 05s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/d40195c7c8e649968793cf260826f181 : SUCCESS in 17m 18s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09859c7d716f4831a2586f5ccaa3e881 : SUCCESS in 13m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1eb928958ca44b5a853aab4d78621202 : SUCCESS in 7m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/081218acb0904ea9ba8e32cb5a45c2da : SUCCESS in 34m 02s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ffb9e1bebb92461c91b1c02440c074b0 : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/733dba6009c94849ab42d71483c33878 : SUCCESS in 20m 07s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0614de522c00404bbb6031f8865375c0 : SUCCESS in 12m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f66259906d4847b5ae7cf3a2595aebe7 : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/da3517c34b784680877dcad898c1165e : SUCCESS in 11m 38s","accounts_in_message":[],"_revision_number":1},{"id":"de2ba53490c91659feb3bf91c0105f355369ae91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:20:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0c62a507552730b445ef2f4c0ba65e1e1f54dc1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:21:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d89348b6b4134bcb8650f8236228e82f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9caeed6e86f6440291a748893c0fb718 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8d79c011147b042bd75fa9eec0fcf0e90d341ba","revisions":{"e8d79c011147b042bd75fa9eec0fcf0e90d341ba":{"kind":"REWORK","_number":1,"created":"2025-12-15 06:10:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/03/971003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/971003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/971003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/971003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/971003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/971003/1"}}},"commit":{"parents":[{"commit":"78d21cad1148bde59eec03f81e0564dd441be216","subject":"Fix Bifrost host variable file generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78d21cad1148bde59eec03f81e0564dd441be216"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-15 06:09:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-15 06:09:57.000000000","tz":60},"subject":"[2025.1 only] Deploy centos/rocky containers matching os_release","message":"[2025.1 only] Deploy centos/rocky containers matching os_release\n\nChange-Id: I92bf70e1444b17d310a9d63ad53c5edf523a9bbf\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8d79c011147b042bd75fa9eec0fcf0e90d341ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8d79c011147b042bd75fa9eec0fcf0e90d341ba"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
