)]}'
{"id":"openstack%2Fkayobe~971004","triplet_id":"openstack%2Fkayobe~master~Ie39a9975a6d9bec8e118d1d7e7a67fe2095b66a4","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-15 11:16:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie39a9975a6d9bec8e118d1d7e7a67fe2095b66a4","subject":"CI: Use centos:stream10 image for molecule","status":"MERGED","created":"2025-12-15 06:19:29.000000000","updated":"2025-12-15 11:17:48.000000000","submitted":"2025-12-15 11:16:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971004","meta_rev_id":"107aa37f53e56720e63cc09b391ea55d0fc043b0","_number":971004,"virtual_id_number":971004,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-15 11:16:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-15 09:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-15 09:46:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-15 08:12:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-15 08:31:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-15 08:31:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-15 08:31:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-12-15 08:31:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-12-15 08:31:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-12-15 08:31:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-12-15 08:31:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"65c0bc339b0abd098c65a3f17df21c875dfaf6b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-15 06:19:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8495d34aff12c813b13c39e81c773ce63d48202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 08:12:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aae73dce9123403ba2e06ee7fc013f1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc2c226904147c7be371ff067a386db : SUCCESS in 9m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a9ec3004ace84563bb88dbc753f49b78 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4b59f2bf0ec4daf8dd23da5e20ab94f : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adf196b2f8244de90c7b46ab82aff90 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9cf1c374816c407983af088c178f818c : SUCCESS in 5m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24009ffbab504285a0ff10aeeb7c315e : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3be92ad809c46dfabf310a6cb4a665c : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b873b4213bb44cba73ecf8551b33ae6 : SUCCESS in 5m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7634a574c56d4bf2b5c288ba5372f6e3 : SUCCESS in 8m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2dbee8875284a3fbe359a58af2c4581 : SUCCESS in 9m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b65c0a828e3841a3a843b50fa2216cfd : SUCCESS in 16m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81a9e4279a574265a452a6b6a453e0e5 : SUCCESS in 15m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ac6ccbc42dd84573911f0bd7474b0965 : SUCCESS in 37m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/282fd8129d5d4e37aff5dbebdbf80a81 : SUCCESS in 1h 18m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a7c0b2df5bb34d39a3db5ad32612637a : SUCCESS in 59m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7e2893a67384f3f92e2c7c0338968c4 : SUCCESS in 1h 11m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/def20d1950ba4b3aaa17dfdea021e18f : SUCCESS in 1h 17m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8184c0e86ba64e6fa824350e8062a762 : SUCCESS in 1h 47m 59s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/88dd46c3921c4b76818609250ce0db9f : SUCCESS in 23m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b73feaf33a604566983194036a62680f : SUCCESS in 14m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1532cfb96e74736a41cb76467d9f12a : SUCCESS in 8m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e98a16ce58ca4501aa58632e684e5dfa : SUCCESS in 14m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/356326e65df24398bbc2921019da8937 : SUCCESS in 26m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/834795cfaf5e4792beae9c0072dc4861 : SUCCESS in 8m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a4642f36fc144e2a6bb59a2b3571dc4 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e08f6851e30638fc8168932814c0a0efb41fbdbc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 09:46:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87a09387b8719c9319d73744f6f3f1ad09c181e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 09:46:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56da36c1150f051d654fd26c7045a823d719759d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:16:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9769cdfc69864503baf719ffa01c5122\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56daebdbc161489fa044bde2c1a47a1f : SUCCESS in 5m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/212d9547dd514c9299291099b829b49b : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/666586e5d95448f388f4ff85ec2740f0 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bc9f5fbc54f42e08d357c51b6f6e7a2 : SUCCESS in 6m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2200b2f8a3a48989767985f0b866226 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0934793eebc64f068f7836e632f844a0 : SUCCESS in 12m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab0bea4a27ab4ab7a9e4fe485fa61903 : SUCCESS in 6m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0fcfa382da6f4ab38c52defd37a82ff3 : SUCCESS in 8m 34s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1009353eabdd43b0af135339af38db4c : SUCCESS in 9m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c9af796433d24bb1be0813a7445ea036 : SUCCESS in 18m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ae439b5326943329297baef3ffaed95 : SUCCESS in 8m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/05222e1a2cfd4b3b860e82b4faec15e9 : SUCCESS in 1h 15m 13s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/31a4b625b82849b1a7ed4c4d86b01e53 : SUCCESS in 1h 16m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/69e23acd670643d0826bfe6f25269d6a : SUCCESS in 37m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74676fced74c49219816340fef50a895 : SUCCESS in 1h 03m 58s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a4be18bbef1f4361b4319e94002ce9ef : SUCCESS in 35m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/630c1e9ee2a3443fbaaef80b36a631d4 : SUCCESS in 1h 23m 46s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e51dff7a57a34b778016a80194ac5383 : SUCCESS in 21m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4d6deee4197c440fb5770da2f2dddba5 : SUCCESS in 9m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ff90f87dc7e4e04b3c9cdd981768b3f : SUCCESS in 15m 00s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/40d15a955e934052a833f76081c6ff75 : SUCCESS in 13m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/720617bd634f4d46b9a331bbde6f720a : SUCCESS in 15m 12s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1c76bcfb23f94e2d9fac6dafe80cd24a : SUCCESS in 18m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f1e68e8533442bdbd1b43dc97ee2f0a : SUCCESS in 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3da6535c2e5bcb4552d50367cfa3a10e31f7b98f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:16:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"107aa37f53e56720e63cc09b391ea55d0fc043b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 11:17:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9feca45e30429886c66945d9c22a1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba474e71db74d3ebc3ad9c118c4e0fc : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f773cd04888ebd6b3d728ac418ff320cba7252aa","revisions":{"f773cd04888ebd6b3d728ac418ff320cba7252aa":{"kind":"REWORK","_number":1,"created":"2025-12-15 06:19:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/971004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/971004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/971004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/971004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/971004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/971004/1"}}},"commit":{"parents":[{"commit":"3c2cf92507ca0a493d1454b34dcc47f2673f830b","subject":"CI: Ensure firewalld is unmasked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c2cf92507ca0a493d1454b34dcc47f2673f830b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-15 06:19:05.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-15 06:19:05.000000000","tz":60},"subject":"CI: Use centos:stream10 image for molecule","message":"CI: Use centos:stream10 image for molecule\n\nChange-Id: Ie39a9975a6d9bec8e118d1d7e7a67fe2095b66a4\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f773cd04888ebd6b3d728ac418ff320cba7252aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f773cd04888ebd6b3d728ac418ff320cba7252aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
