)]}'
{"id":"openstack%2Fkayobe~971027","triplet_id":"openstack%2Fkayobe~master~I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-21 17:40:21.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-21 21:02:04.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-04-16 15:18:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-21 21:02:04.000000000","reason":"Change was submitted"},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-04-21 21:02:04.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-02-17 17:48:26.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2","subject":"Support configuring DNS on Ironic Neutron networks","status":"MERGED","created":"2025-12-15 11:49:27.000000000","updated":"2026-04-21 21:02:57.000000000","submitted":"2026-04-21 21:02:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971027","meta_rev_id":"01e8051079153aa1a9e32e0294f73b11df1e600d","_number":971027,"virtual_id_number":971027,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 21:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-16 15:18:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-16 15:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-21 08:57:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-04-16 15:18:54.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-16 15:15:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-15 12:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-15 15:29:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-28 10:36:32.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2026-01-29 14:02:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-02-19 15:38:14.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-04-10 08:52:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-16 15:28:08.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"9a02943a30a275b127adb70822f9e1c641bbb7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 11:49:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00105a90577bbafc13501844ff05a42c1077e8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 12:53:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/662396b435d84876a7011eba4b39787e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3e976e877c435185cd1e9c7cb597d4 : SUCCESS in 5m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a872080b6649449f98fd2981d24ef6f8 : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7ec671254ae4eda922c03d34d5a9232 : FAILURE in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9021ceab3864e9b877d2ea1ccdd3463 : FAILURE in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e72b7fa999614f4f9944f7bceaa9e3c7 : FAILURE in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fbb22fae174142f4a1d6a9ab02a0fee9 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdf74e21a26346aea2a032b6412cdc7d : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b82476cbeca2441591187e06f49a6e8f : SUCCESS in 12m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ce7d7de481f5422d96b8bbbbead6acc8 : FAILURE in 4m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7eb38b96c674bac95077c3fd614ebe5 : FAILURE in 6m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5ebec1eb438b44ef9c3bd604499db70a : FAILURE in 5m 58s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33b7649f6e8e48abb697e795929d4884 : FAILURE in 4m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bb623d0320994f5eb23758b9eafe9270 : FAILURE in 5m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/82fab5c14c9e43c09695c57b9022accd : FAILURE in 8m 26s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/609c68cdabc74b58b0ed2122fadace6c : FAILURE in 3m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f89309b1668d43c2a379e258bcae4c1d : FAILURE in 6m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a1c41a1d8fbb4c059d0d7865ee0eef67 : FAILURE in 7m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdb500f078d2467089694acb62d81728 : FAILURE in 59m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f3b0dc9739ef4c0a8bffaed2b71a30b5 : FAILURE in 4m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f1395e8ad4db4040a907b00f7303b948 : FAILURE in 8m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43eb013a8ce04259a9db1e07cd8ac610 : FAILURE in 4m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/76c761eb7578424fb827890f023ac1dc : FAILURE in 8m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/003e59e425c949f3ba47a64d32218897 : FAILURE in 17m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/58ec0185e09f4736a03cf680c61cb6bd : FAILURE in 4m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7809a9f898ee4ab89c8c928dc070134e : FAILURE in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"12686e8ed522fccbba70a02a632090d896746d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 19:46:21.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba5d7f41a3432b1c12f7cc8b75196f0e63f06a6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 19:47:23.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"a4b18033d143772b76df6cad40ca539e0e288772","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 19:47:35.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"c895b7f927911e6be710c646536e5976f8b3e02c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 19:48:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"14489370e7831b0f091e3217dfd73a2a92539f66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 21:45:15.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94dc7fa520c041d0ab10d949d99fe53c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6efd1d181d7f40dd9c07b7ab2ae7f44f : SUCCESS in 9m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c7ec165c091481ba08689b85efb7917 : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed4b78713bb8401a8b39c62efcdef470 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be4136d7b4c484f960058670cd0c60c : FAILURE in 2m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a11707b7ff94a7ea68ba8debabd6665 : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/27f6fb6340e6424c9100ba77483e8dcb : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc73e8e676e2435bb249f491f2ec6066 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d76e225184104b76bda15bd0718963d6 : FAILURE in 5m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eaa690155cce4921b0d09874d1f5b195 : SUCCESS in 7m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0270d73bbc674ec09749d19474d9bac2 : SUCCESS in 6m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ccf349f44bf84eaa8d0e2fafc4c6b0df : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/754f2b6a5c3d4d2f83df3beff8fac23d : SUCCESS in 14m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/aec429c37ae74a90a34ca0a3cdda043b : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5cfe425693644db6bb549a2053c15619 : SUCCESS in 38m 16s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/eaf7a5a50b4a4e768fa9c3c31517ab72 : SUCCESS in 36m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7667299feda49de93fc6fa89950d30e : SUCCESS in 1h 11m 36s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/913d8d4d24554707a0730f3950c43da3 : SUCCESS in 1h 10m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9b758ba752db4b8dabb91059c672cd36 : SUCCESS in 1h 46m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ac53b2af742f43048555cd48b921cb9e : SUCCESS in 8m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/bccf460c264e4c9296b2e8bead23b829 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22661998d33d47dab70de4f0c301e744 : SUCCESS in 16m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/84a8bc005641460c92667321c584dd62 : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cf51c762f614f6389c8adc71b55ede2 : SUCCESS in 26m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4d2d0eb6aa6e4a6da17c2770ba2c45fb : SUCCESS in 16m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/216f780874d04b88ac03726172bfaa0a : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":3},{"id":"bf3ef15ce5c6244997238f8c06280b3322ff6046","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 17:56:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d61c7e08814e6dfccd842a525258e8373de90e4c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 17:57:43.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"e97cbd16514980dc49a5aedefc09d61a7a0abcf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 19:41:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5732891e6626442d8a489679a949b8ad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/979929c374c2495ba66bf61686bc5649 : SUCCESS in 7m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/582851294b5e44d4a414e66e23e4bbd2 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fd3329f48d845cfbc4a954b5df7f81d : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f986042d9490415fbe7caaa03c2e704a : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1a588ea226f48b39843a0d53e6bfe25 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9ca938a4fc14589a2fd39dcd0073a09 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0df5a1d705034be0a0aee84d983b7207 : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7d3de33e3ad4cddbce5ed6d57790c16 : SUCCESS in 7m 27s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/326888d274f243a289882fce6e5417fc : SUCCESS in 14m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9437d0c486e24259b06bebf329768328 : SUCCESS in 12m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/54a66266e4bc4ca4a9ba53887e24a3a8 : SUCCESS in 19m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc0fc172e2da44c49d8e1c75413c1243 : SUCCESS in 10m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fca4dfe85177475d933bc65e6d459b89 : SUCCESS in 55m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d56e547934ea4e13ae20cc308b3ab25a : SUCCESS in 56m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d3e3c095d2bd4973b8273aa134624216 : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/551f1bc7dc894bab901e7002850a9a9d : SUCCESS in 51m 48s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/62ae68a6dc3948f0b1a87eacc0536cbf : SUCCESS in 56m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1d56cc5be96542c9b7fd19990c3348dd : SUCCESS in 1h 43m 17s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/76f96cd21f4e46be80b32f3407293081 : SUCCESS in 1h 27m 47s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c095ca81522c4f81bdc900a2abd4190a : SUCCESS in 1h 23m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b39d42bf18064a799adf267f71eedf24 : SUCCESS in 1h 29m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/da25ac07910a44b39caa99e2a6898415 : SUCCESS in 15m 58s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/67c34f1d77924115bc80448711170be5 : SUCCESS in 10m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2e706a807f8431dba8002ae428d8c30 : SUCCESS in 15m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29e87847b4ec456e816c235942d2e497 : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/867b92b611874d45b4b4d44e844e8503 : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/380ff9fda1594be09b1e43da00d981e0 : SUCCESS in 26m 19s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b137a3318ac49dc80e41a83bc454b15 : SUCCESS in 25m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0124cef2c4a40818a225d9c435c4567 : SUCCESS in 31m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/299e511dae6e4350a04541127225bba1 : SUCCESS in 13m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72b8b46ac15e477398e5265680248615 : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":4},{"id":"f935cf331c7fa8427ef317211a8b1d30fbcad321","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-27 20:00:50.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3020275f77bfae2fbc79c6aa1719f433daa2b4d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 10:04:37.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2fb5bcc842dbc8b9f70e71513819fbf7cae217bf","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-28 10:36:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a88d695cfcebd5c10090451198d567c6720f60f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 11:40:59.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"2debcf4888c2a2e2c3475b81bf9cf16d46c52d6b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 11:41:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"897f8e161ed40193a1d76015a4ed5aed68c25bb9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-28 12:13:05.000000000","message":"Patch Set 6: Backport-Candidate+1","accounts_in_message":[],"_revision_number":6},{"id":"e4e6a362f74dc67dc4386b691547a9587694a905","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-28 12:13:10.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"8c42ac6c459dcbca4d16830e10850b94a9542681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 13:19:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd4dd461ba140dc8839740ccf031dc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482df271601144f39546dbb3f9caddd1 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4458a8efd48b4a46b6bd4d63b6c065f4 : SUCCESS in 5m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf6db253917d41f69dd723aa2d53bca9 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35610501d5c94b0d98a0b7805bdd0461 : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9a45fefb0904466a5bae7fd5af142be : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6561dbb2f5724686b7b916157353c2b4 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/adf5566d480a455e984d25686a37eb27 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e07c70a354048e5ac58e9b87be11ce7 : SUCCESS in 5m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1ac03d5d1b5f4dd89ae5e86e901cd830 : SUCCESS in 9m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0e6df61349c4c8e865db4926d290301 : SUCCESS in 13m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/81869df3d48247e6be096ecdfe223f54 : SUCCESS in 11m 14s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4505e61e18ac44a7b5fd0e52dcecc57f : SUCCESS in 15m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2dd616319cc748f38e0e921e93efb239 : SUCCESS in 50m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6a8e3368819e4b1193b84ad6477cd61c : SUCCESS in 51m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c2f9be0da0b4451bb45b5b409099785c : SUCCESS in 58m 33s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be1a746298124a53901ec014d13ff240 : SUCCESS in 49m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a43a0ce7169c4b1eab0c45cb4a5914dc : SUCCESS in 54m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cd4cc85cc712421ab648e82045befcc9 : SUCCESS in 1h 37m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8b45d288ab82491a9a836b01f82bc7cc : SUCCESS in 1h 28m 27s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/958cdef98a5a41c382e2094feab184df : SUCCESS in 1h 23m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44494111e34e4662b71aef87d27d8871 : SUCCESS in 1h 30m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f7ffdc979a2c45ffb5df3ba1ce978890 : SUCCESS in 16m 05s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/be34d0a7965a43638fbd10a78549635c : SUCCESS in 16m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b8702c3ca104be7b6ffa30825f8fe90 : SUCCESS in 16m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ed5860d9083e4e43831c79bace0aa3e6 : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/23bafa5e84fe450a903e8777e40f24d1 : SUCCESS in 32m 08s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cd8c00c05a4e49c79181be6ccabbf148 : SUCCESS in 33m 50s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edeb7c54842c44fcaa6a332fcfe1ab9b : SUCCESS in 30m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7092af0cc7d746339857d85dc597394d : SUCCESS in 15m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/409a6fa6a42143d382848180abe37dc2 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa40696e60f640c68d7a1eeb62462040 : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":6},{"id":"21aba25519ed18486a2f0f1bf6b93af92769460f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-29 14:02:11.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ec0f32c284a910e03907c6a66ff51c842178c6e1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-17 17:48:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"754b67e73bc4716e1ddebb5fa37772642b45e2f8","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-19 15:38:14.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"b72bc391d0429194a276f7a14348c1425f4d2258","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 16:14:50.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"b6f50a619a3fcf59a2a6be883be0de015d097b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 15:59:41.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"1cf5a7ea1fd90b8e580863f0f1df342324216a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 17:00:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcae4de993184708b6f785d61ce6f70a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7c438cc60b40c6938fa7ec57f207db : SUCCESS in 9m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f6fa3e2de21a4cafae1dafa6788f740b : SUCCESS in 4m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b8e91e9bb194a0883669c9d98b1fbbe : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c754a006f1493e9945df13b6ff5415 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74a9921e23bd4bd09342c50a17540b8b : FAILURE in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43181e6b05fd45b99669fe63a8a05ab1 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9cc0646aeaaf41c0b8809f12ad49493d : SUCCESS in 7m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c7b7d32dc243b9a26586c686e10859 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c410fb078be14640b5fadf3f65e5f261 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6ae6c4081e64aceaf09ba9d519b751c : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1cc8683462af444bb76a1adde9186898 : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/090049ff37ba48d6a02b6da96c108014 : SUCCESS in 14m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c8a166a3ff94535aa7eb6f188abdaa9 : SUCCESS in 22m 09s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/861d2a0120104e5e81c6dd290310308b : SUCCESS in 13m 12s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8fe915dea92442a94cb413a8653dc26 : SUCCESS in 24m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/413271f21b1e4fa787c595de132fb32a : FAILURE in 25m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3086a56bb78e4e60a809839d6b967792 : FAILURE in 24m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/336977df5da74490919c1475148b3e8a : FAILURE in 30m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c95096453bc4a8eb74f22840d21028a : FAILURE in 24m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52e0f27451cd4a3b80b6adeab7ab7241 : FAILURE in 23m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c7c2cabb1b4d40a2b67044d8dba32e71 : FAILURE in 40m 31s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/30198a32ab214510a96fa016a516ef9a : FAILURE in 40m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f312666794c4d9fad999c3e212484b9 : FAILURE in 59m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7a896bc4a284b88951286f40586954c : FAILURE in 38m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5759eddf3c5b4ce1b48c6503c82d92d3 : SUCCESS in 26m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/be5e26fc27d1434597b26526b0ac4c14 : SUCCESS in 20m 58s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1558b0b950c940d5bab5a653030a2c47 : SUCCESS in 22m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bdbd7500ec564438b16710e7e8da6b61 : SUCCESS in 23m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/19abe60eb71b4d80b4754ed644eb341d : SUCCESS in 27m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0fffe74a153c4fc981cf94f4bef79922 : SUCCESS in 28m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b3fb01f0fbf47fdb21b818baa6f01f5 : SUCCESS in 31m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f31d112edcd24ebcb062a25785450e9e : SUCCESS in 32m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/608e9c3c1e7941558caeaacfe5ce0047 : SUCCESS in 28m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eab986db178e42669725cac919ceafed : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":7},{"id":"3da5a6a37cb39581f52935df77e6e9ff03256135","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 15:15:23.000000000","message":"Patch Set 7: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3e63f007896d89777628b377fe6e500bead143d0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-16 15:18:54.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"a3957dcf1573b580599835a98d2254dec5c04861","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-16 15:28:08.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"96112e126acfa4b98eb68adeb12fb7e5986ac894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:57:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ff4c9c0c1c84e34a6d6bf5c4717d371\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d491a7682e9540c98c5f438260b20605 : SUCCESS in 9m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/404a4f496d3a429f96290e3826eefc7d : SUCCESS in 6m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/862181ef888f4f699a003dfbf6e59bbe : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0229ed6fd59e4cf4a9ceb1a650cfd63c : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca281f8bcc424cde92751ee5e2cd7d4c : SUCCESS in 6m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d45247fc8a8842e3916600bc3a9167ba : SUCCESS in 7m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/870886e52b9a4c029c955a130fb8e339 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ea53c97716c400c8435fec4ec42fef5 : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd34a4f25dae461fbce34668faf82f0b : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/783784266f3b4e77ad1f916e70fd7d5f : SUCCESS in 20m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1506a7b591324b7e9e8ced1e56cb6dbd : SUCCESS in 22m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d7d391ed80143a58db3906a5cee982c : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c5350aa141634c0eba6833d6d50b8dd9 : SUCCESS in 25m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdff10f6ce8f4674a2c6a4fa9545f62e : SUCCESS in 24m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b941d5d9e173419395669e15f6f7ea75 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7f3212613aa449a78547cff0b0816bcb : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5c8b880b1f534b6e87f69da30661fc86 : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8aa1f59e7e04e7486124e66dfbd76ee : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e8e9c821d5134adeaf900665c9337ac5 : SUCCESS in 1h 05m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/33a796e521a14a74a13b6387ceb14d9d : SUCCESS in 1h 37m 13s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2508653feee741f6b6b526d3dd78c686 : SUCCESS in 1h 31m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/23acad3db66d42348702a2fb71fd58ed : SUCCESS in 1h 34m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/128879330f3549ba8d741f251107ea63 : SUCCESS in 1h 40m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5563ad38c3a34fee82cbc8ceb0531b37 : SUCCESS in 24m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1a933d1bc90b42689bc51d3a121c6dcf : SUCCESS in 27m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf9387d71e33459b9b691895824de54e : SUCCESS in 14m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5794aee14fb44669b821e6fdc091950a : SUCCESS in 20m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/316df73bf343414ba8d37d05045233e0 : SUCCESS in 29m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a7ced7e37699402ca480fca61d4ead3a : SUCCESS in 31m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc7e78ace89042ffbfd25445fb358825 : SUCCESS in 36m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17e5f39abe96456cb2d78759d62a4d79 : SUCCESS in 31m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bcd658ba1afe47609546d76f52644a53 : SUCCESS in 14m 50s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b92dbfc2c22438ea32ec050efeceb5f : SUCCESS in 24m 37s","accounts_in_message":[],"_revision_number":7},{"id":"162c3496ca412a915625e07b2127eeb295dbdffb","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-21 08:57:50.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"dc84c47982c06abf6bfc42955cfd7df54e05682c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:58:13.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d52e03ddc0d2e48e033696230c888c8148cb627b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 13:21:35.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db3b6b18e01b45b9a63d783186e6d439\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a86161a9fac4d8fa2aa1089b5ffb1ee : SUCCESS in 10m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36dc8b8beca7464090847da86dd0a40f : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9629a8c5097c4e429c9cb9997011e331 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38cf91c2ed034525b2c6accb6f9d5419 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e0f9e0b5e104bc3ad0ee544f1c8039b : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e79b63e47e134932b1b51ec42c7e33c0 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/072ebfab8dfa4bcd9b88b368ae0a8287 : SUCCESS in 6m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d01dd2e88ee14ee1aaf3a82e27763145 : SUCCESS in 13m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0fbe0336eee44370b2dc5c30d165d3ae : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c59bd0d2438747edbed2ae8996bd1615 : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c36c8238ffca490c84947acda604d672 : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5386e6ae9ac74740b84e197e82398cac : SUCCESS in 18m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a6bc13a106e14badbba481d4587a3f71 : SUCCESS in 44m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ebe09a6b1a52430cbea3cdace63131df : SUCCESS in 1h 08m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/743b02514d94467d9797f866fc71f0f5 : SUCCESS in 1h 11m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/98c50b9deed44977852e7daa52a335c3 : SUCCESS in 1h 24m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5f84553487747758052f225840dbec0 : SUCCESS in 1h 48m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b657ebcde26f43cd95097948056a2257 : SUCCESS in 1h 20m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c17c50d46ce641d1bd876b65dbb8f177 : SUCCESS in 1h 42m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7588a11471444b24abf06eda483778ab : SUCCESS in 1h 32m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4b66dddfc2148f999fa2015311624f6 : TIMED_OUT in 3h 03m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c4f955b349064eadad711bc3dc2943d2 : SUCCESS in 15m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0f9102c4b74e4f84b9f32246e11d33e3 : SUCCESS in 21m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/411b0a4fa29d456db3a726a8c16eae79 : SUCCESS in 23m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9b54446a7b8b4a50be189894fe20b1c9 : SUCCESS in 20m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5fdeff98369e44d288e5c2c30c7a9ffe : SUCCESS in 17m 39s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3ca56c76818849ac93b6cbc678a47cb3 : SUCCESS in 31m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6a446faf28c4a1c8f3f8cd983b3da3b : SUCCESS in 33m 18s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/995ea8a3a21e4334b2516644bf8e8993 : SUCCESS in 28m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8c13e4518bdf46d787f12b08da80e7fc : SUCCESS in 22m 27s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f2bb720460024897b05219fa7b29c387 : SUCCESS in 22m 04s","accounts_in_message":[],"_revision_number":7},{"id":"2d4f302f66e5230d61f3ff6837ec5e7d130b447e","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-21 13:58:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"957e52513d1ad1f0b98980e866a5528ff5a96861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:41:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f496d9efd5ea4db1aeaa3b14b00022a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea9a39d46a9c4907aaef38d58691f4a2 : SUCCESS in 10m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/660e9aecbbff44758998622d377ead4f : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17c9759fb38448c3813fb70f56ba64fb : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6b0fd8c4994e178fb6aa896d49364c : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/437915da187c4410b4b99c46b061e168 : SUCCESS in 4m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7270b32be1fb461b9235d3076317d7a9 : SUCCESS in 6m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6ad930acbad44e6b64374f47b9a4844 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4fc96e84dee442b9983a66783f82c5cd : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f1be75be95248ebb9c08f8197e17608 : SUCCESS in 12m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f49df6856c5495fa32fb698fdc194f8 : SUCCESS in 11m 14s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/014d056e1d8a4a56b0a086506ee528e4 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7021f67a21b24256a3a243e33a079770 : SUCCESS in 18m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a4b14c9aebf844e195737ee2a655e7c9 : SUCCESS in 12m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a3e3469622f046b6990d819d7896bab7 : SUCCESS in 18m 06s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e2b1cf1c2657491dad7942140e833813 : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/31ebb9132cd04ba6a933131031b4f7fc : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a9adcf92d3a04d78acc1f2af580e3d8b : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff4c3caa28504cc4ae5e239f74968622 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d636dfe0e4cc4a15a595079c2eaab669 : SUCCESS in 1h 09m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3a2782450e9f4c1180c180bffb14af1d : SUCCESS in 1h 39m 04s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/384cfd893da64a8c9e4b2f7d884e8482 : SUCCESS in 1h 32m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1afc99d37e34415b19db6ec81f42cfb : SUCCESS in 1h 30m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62c1298401864df9bf3a175500a057fd : SUCCESS in 1h 41m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cefec5034d904df68b6d0e1446d68c6a : SUCCESS in 28m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5da4ebdf0bae483994665d27b34483e2 : SUCCESS in 19m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5056dc8cda34022b8a96c03db472a0f : SUCCESS in 23m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ac78963ac8042e6aad70d4719949c34 : SUCCESS in 21m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/620747025e814ea9b3009eeabeda00a3 : SUCCESS in 37m 31s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9caeacc7e35346b18c4995fa5eb24b3f : SUCCESS in 18m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dc1186c768349baac00b1858477f1e6 : SUCCESS in 31m 23s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/421ee3d82ce746278cfb9cb0e1b9377d : SUCCESS in 33m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ea2e625eb70b478abb0f1969126e05d0 : SUCCESS in 23m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d39c61dd2911490a8952e0034464d492 : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":7},{"id":"791ee8ce57cb7ca8506837f132678aaefd8ca478","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 15:42:00.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"d509e5a379322a687b539c66f306242e6b5544f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 17:24:26.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d78ced4dbc94b3a8119f5b86d8f63b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea43488aa444adf8116e2b7913c40e0 : SUCCESS in 9m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/20f9822f223d41b1b8e08ba63296ec63 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4df2e3f751b244eeba87610e4be1ec40 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4572117a2d8466b8f04849555d58500 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32fdea2f82a14e849fa1d424e9461947 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7b7b56193de4c67b8f45ed7a638bb3b : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44338d283fcd4c669927433b004ac9b3 : SUCCESS in 6m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00a1b0d2f5864a5296770e882eda7d49 : SUCCESS in 13m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d1e685c469b24d88bf86db175c7d6867 : SUCCESS in 14m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7403352e8b24b0db5e9d1de28d42c66 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/84cf4782c5e5490a817961c78481d0db : SUCCESS in 12m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd3b789f2e0f4e43a992d8965fc9cd3d : SUCCESS in 22m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0f5fbf1deae343548ea47af77fd0efe1 : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b5ae11d9af444bec842d878f56d3f49a : SUCCESS in 1h 02m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/022b844444a04fac90058748adf5709c : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3caf91fa466741b1b4eb0bfb12a1f2d0 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dc5ab1e430f84ee4957e08c254e975ff : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/79e91e1b51bb48548dba7d143e2e5be5 : SUCCESS in 1h 40m 08s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f11a7a9f89034754a103b21b4ce091e2 : SUCCESS in 1h 31m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69409ab893f84692808cfbef3d27c012 : SUCCESS in 1h 33m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f847fc2756154c189a0bc40015e6efd4 : SUCCESS in 1h 39m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b71bfed4bdcf40b5a7732350697b67cf : SUCCESS in 31m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2abb7a2ee8c84c85a9f1f5e91aba78c7 : SUCCESS in 14m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d0b6775af3d445e940a407bde1d5562 : SUCCESS in 18m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fb46d41ea1964653aedcd1128277ffdf : SUCCESS in 13m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70b61602b95248ddbcd7c0a4f790bd0b : SUCCESS in 16m 35s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/634f23e8334f4b36bc98892eab3ba3f9 : FAILURE in 12m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed548d423f854144ad216f4b8f5e3c39 : SUCCESS in 35m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/facc95b333bd4d028743baac3b75a44c : SUCCESS in 28m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/06e83091db1545f1abcffef4b8d94800 : SUCCESS in 16m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d330eca1217943928727a6000a1108a3 : SUCCESS in 13m 25s","accounts_in_message":[],"_revision_number":7},{"id":"5005ccea7d7d230c5c4b17659f4767fb1d4ad42b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-21 17:40:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"53d3b6bf59c52247f641fcec44b90fa5bb159ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 19:20:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d85699cc44194bdd808a3b3435bcc07d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2242262754194632849a214a39ceb7fa : SUCCESS in 5m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1b0095ea9c0f488ba0be151d3414e7a6 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60f4494fdb304a02be7ccf8262eb19ea : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7861f32acf614cb4b30ff0c128c0721b : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aec084a422574357ba9338b629980abe : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc33d12f149340f38c54bf13e118cf53 : SUCCESS in 6m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8c994eca0284216ad34a06bce8e3c88 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0c9f33492a6348aa825b8e972f266deb : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/34b4cd8e912044d0b4ac6bb19c60fd12 : SUCCESS in 13m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/239b7f5acb634b8b95a3b681fb18b5d4 : SUCCESS in 9m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/15d072684af64177b26288774736c7ef : SUCCESS in 14m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bbe292be11634d28ae0d796cdfa1da8c : SUCCESS in 12m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a38ecc5871bf47f2b2d4268fb3578521 : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48c5629c802c4ad3938066fb762624ec : SUCCESS in 11m 45s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b278772644624ca4b8bdb87e021bff32 : SUCCESS in 1h 05m 12s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b1c78d81cfd149ed90a9740580cbd317 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/65bf271527c1445486b14f844af3d750 : SUCCESS in 1h 04m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2336298299d6424d9e3cc3d6ba5a39b2 : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1c687b9b8e7247e081e2b2b2e22d31bc : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/702fb8171cf34f3fbd87b1257e5aa112 : SUCCESS in 1h 35m 29s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c8f85266645b4828840d3d7f6c98b2a7 : SUCCESS in 1h 31m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/137ba38aa9464644b74f578c4fc104a2 : SUCCESS in 1h 32m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/48de806e582f4142937e2376720074d4 : SUCCESS in 1h 39m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/33af45207de543849086c2fd63072f64 : SUCCESS in 19m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/93660ac76be14c728115568b572d665b : SUCCESS in 14m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/261892898a1d4e4fbec9bd2c1282cf78 : SUCCESS in 24m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efc908770ca94800939663551b667da0 : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0d95b8959cd3407bab440e5dc78f1f4c : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6f5f861c942446ad9c228cdbafa5b0c7 : SUCCESS in 17m 59s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68ad134f7df149558f1c4b97b9ea00f7 : SUCCESS in 31m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bd125f1d4e64483a7760b03284b8cb5 : SUCCESS in 15m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/73c032d442204b529a938c4cbf855b99 : SUCCESS in 16m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02010c20ac13427cae62402a20479781 : SUCCESS in 19m 55s","accounts_in_message":[],"_revision_number":7},{"id":"136d2cc803686b87f59d179cdfa75c8fe14aaa13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 19:20:44.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"871fc20986ebb128cfabd1d64bdf6b8ebf2af716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 21:02:04.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b961c5386764123bc56924c5d045632\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ef8c56ee0848caa65f303596921198 : SUCCESS in 6m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea284fc3556947f695376ee4ecd711fc : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffcb1b67fb0a4642b75b7648fd488ef3 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afeca42df88d435089a46df2dba340aa : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c8986dd07654f9fb30edb1a78e2c856 : SUCCESS in 2m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a79500bd9c6843639b2b102cf2edd3ea : SUCCESS in 3m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27de57c138f14e95a6eb740819add054 : SUCCESS in 5m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8c0c914c9c77423eb4daf7e2f931a7f9 : SUCCESS in 9m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a9f72613d17e472f9d2b9937f8290c3a : SUCCESS in 15m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b2660823fdb4093b3fb882a48d98b6b : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5673408d8fda4689a4a30e2228e081e4 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7f87031933d431fb1b62d9387203e2e : SUCCESS in 13m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1b12ecb8ff8a41408c7a95787370f5b6 : SUCCESS in 1h 02m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9286e62c2d074ee6a16b411f68c8a14e : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/01ea9afe548f4455b5d4b0d1c0e33084 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59e1fa46e39541fd9380275ce2c40437 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87ed21dfbead451782429da715bc85ba : SUCCESS in 1h 08m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0f6b6e4616244b22a7cafb8bdd90621b : SUCCESS in 1h 34m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/35704ed06c2349f2a647c028b70f18e1 : SUCCESS in 1h 34m 05s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30a799c8c45142f79152ec1f531e38fc : SUCCESS in 1h 35m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b29ce2b4711c4b2a9039419357c8e9c7 : SUCCESS in 1h 40m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7f44ae30587f4d0190cc221fa396b418 : SUCCESS in 15m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4e35e15b086c4886ace023cec2a8e896 : SUCCESS in 15m 16s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93a2a3776d65469d8539aa913cbf0a59 : SUCCESS in 14m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5a2b2040adf84f40b20fa429db72719e : SUCCESS in 14m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/16710c8ee39f454ab6d688b7831ad6b6 : SUCCESS in 27m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/855f45eaf53f401583754f3b37030632 : SUCCESS in 20m 47s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b4c650b7804490aa72c9850acd342ff : SUCCESS in 32m 32s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e65de4acbc943db932ef4fc6a29635b : SUCCESS in 15m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b3bb8e732aa44fb7b1eb823f38ecd8b5 : SUCCESS in 14m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb3e5ca7aea24e269e0a373f09918549 : SUCCESS in 14m 36s","accounts_in_message":[],"_revision_number":7},{"id":"4b6698daee8c956f960f39808ef61b76eaa65e51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 21:02:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"01e8051079153aa1a9e32e0294f73b11df1e600d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 21:02:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77299e81d5cc422dac29cddc424b5d0e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/648b5d74499449e4af012e1f1ba6a726 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d6cc2d941484ad187f570be50bf941b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0fdc9c2edc3f865f2aa9cd331f8d19bc1d0fd060","revisions":{"33d01d5166edd440e160e1be22bdde2b57e1be18":{"kind":"REWORK","_number":1,"created":"2025-12-15 11:49:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/27/971027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/971027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/971027/1"}}},"commit":{"parents":[{"commit":"8cef26d5b5558b6820c2c551e085370f3877dc55","subject":"Remove undocumented playbook for creating external network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cef26d5b5558b6820c2c551e085370f3877dc55"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:20:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:47:09.000000000","tz":0},"subject":"Support configuring DNS on Ironic Neutron networks","message":"Support configuring DNS on Ironic Neutron networks\n\nThis can be useful if you are customisting the fully qualified domain\nfor the internal API so that Ironic Python Agent can resolve the domain\nto post back data to the conductors.\n\nChange-Id: I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33d01d5166edd440e160e1be22bdde2b57e1be18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33d01d5166edd440e160e1be22bdde2b57e1be18"}]},"branch":"refs/heads/master"},"797b9b33ef96e753e8151442dd8c0d43cebcfd9c":{"kind":"NO_CHANGE","_number":2,"created":"2025-12-15 19:46:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/27/971027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/971027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/971027/2"}}},"commit":{"parents":[{"commit":"caeac1eed009c7561e7b0ea567483d500e668f93","subject":"Remove undocumented playbook for creating external network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caeac1eed009c7561e7b0ea567483d500e668f93"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:20:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:45:58.000000000","tz":0},"subject":"Support configuring DNS on Ironic Neutron networks","message":"Support configuring DNS on Ironic Neutron networks\n\nThis can be useful if you are customisting the fully qualified domain\nfor the internal API so that Ironic Python Agent can resolve the domain\nto post back data to the conductors.\n\nChange-Id: I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/797b9b33ef96e753e8151442dd8c0d43cebcfd9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/797b9b33ef96e753e8151442dd8c0d43cebcfd9c"}]},"branch":"refs/heads/master"},"2273cf8f5925b6a1060ede1e4f1bb962eafd635d":{"kind":"REWORK","_number":3,"created":"2025-12-15 19:47:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/27/971027/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/971027/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/971027/3"}}},"commit":{"parents":[{"commit":"caeac1eed009c7561e7b0ea567483d500e668f93","subject":"Remove undocumented playbook for creating external network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caeac1eed009c7561e7b0ea567483d500e668f93"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:20:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:47:23.000000000","tz":0},"subject":"Support configuring DNS on Ironic Neutron networks","message":"Support configuring DNS on Ironic Neutron networks\n\nThis can be useful if you are customisting the fully qualified domain\nfor the internal API so that Ironic Python Agent can resolve the domain\nto post back data to the conductors.\n\nChange-Id: I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2273cf8f5925b6a1060ede1e4f1bb962eafd635d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2273cf8f5925b6a1060ede1e4f1bb962eafd635d"}]},"branch":"refs/heads/master"},"d56d893cb2f00498d3670ba366ce617f993ebedd":{"kind":"REWORK","_number":4,"created":"2026-01-27 17:56:35.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/27/971027/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/971027/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/971027/4"}}},"commit":{"parents":[{"commit":"caeac1eed009c7561e7b0ea567483d500e668f93","subject":"Remove undocumented playbook for creating external network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caeac1eed009c7561e7b0ea567483d500e668f93"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:20:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-27 17:56:23.000000000","tz":0},"subject":"Support configuring DNS on Ironic Neutron networks","message":"Support configuring DNS on Ironic Neutron networks\n\nThis can be useful if you are customisting the fully qualified domain\nfor the internal API so that Ironic Python Agent can resolve the domain\nto post back data to the conductors.\n\nChange-Id: I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d56d893cb2f00498d3670ba366ce617f993ebedd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d56d893cb2f00498d3670ba366ce617f993ebedd"}]},"branch":"refs/heads/master"},"fa939fbb4773c6d6d90320a34a4034f129b139b1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-28 10:04:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/27/971027/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/971027/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/971027/5"}}},"commit":{"parents":[{"commit":"f1cea2ba29ded2ef389c3614fe5257dd02bb6e34","subject":"Remove undocumented playbook for creating external network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1cea2ba29ded2ef389c3614fe5257dd02bb6e34"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:20:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-28 10:04:37.000000000","tz":0},"subject":"Support configuring DNS on Ironic Neutron networks","message":"Support configuring DNS on Ironic Neutron networks\n\nThis can be useful if you are customisting the fully qualified domain\nfor the internal API so that Ironic Python Agent can resolve the domain\nto post back data to the conductors.\n\nChange-Id: I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa939fbb4773c6d6d90320a34a4034f129b139b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa939fbb4773c6d6d90320a34a4034f129b139b1"}]},"branch":"refs/heads/master","description":"Rebase"},"744f3c88a9cc3860e2d599c3e9262a26f6b72a7d":{"kind":"REWORK","_number":6,"created":"2026-01-28 11:40:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/27/971027/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/971027/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/971027/6"}}},"commit":{"parents":[{"commit":"f1cea2ba29ded2ef389c3614fe5257dd02bb6e34","subject":"Remove undocumented playbook for creating external network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1cea2ba29ded2ef389c3614fe5257dd02bb6e34"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:20:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-28 11:40:55.000000000","tz":0},"subject":"Support configuring DNS on Ironic Neutron networks","message":"Support configuring DNS on Ironic Neutron networks\n\nThis can be useful if you are customisting the fully qualified domain\nfor the internal API so that Ironic Python Agent can resolve the domain\nto post back data to the conductors.\n\nChange-Id: I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/744f3c88a9cc3860e2d599c3e9262a26f6b72a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/744f3c88a9cc3860e2d599c3e9262a26f6b72a7d"}]},"branch":"refs/heads/master"},"0fdc9c2edc3f865f2aa9cd331f8d19bc1d0fd060":{"kind":"REWORK","_number":7,"created":"2026-04-15 15:59:41.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/27/971027/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/971027/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/971027/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/971027/7"}}},"commit":{"parents":[{"commit":"f1cea2ba29ded2ef389c3614fe5257dd02bb6e34","subject":"Remove undocumented playbook for creating external network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1cea2ba29ded2ef389c3614fe5257dd02bb6e34"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 11:20:22.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 15:52:10.000000000","tz":0},"subject":"Support configuring DNS on Ironic Neutron networks","message":"Support configuring DNS on Ironic Neutron networks\n\nThis can be useful if you are customisting the fully qualified domain\nfor the internal API so that Ironic Python Agent can resolve the domain\nto post back data to the conductors.\n\nChange-Id: I45a578804a8b2c71ad4a81b3eae6b6a4f874a1e2\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fdc9c2edc3f865f2aa9cd331f8d19bc1d0fd060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fdc9c2edc3f865f2aa9cd331f8d19bc1d0fd060"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
