)]}'
{"id":"openstack%2Fkayobe~971121","triplet_id":"openstack%2Fkayobe~master~Ib18cf699d2b477fa2d4272e3fe690d17dbe17173","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-04-27 11:21:05.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-04-27 11:21:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib18cf699d2b477fa2d4272e3fe690d17dbe17173","subject":"Adds option to disable registration of networks","status":"MERGED","created":"2025-12-15 19:46:21.000000000","updated":"2026-04-27 11:22:25.000000000","submitted":"2026-04-27 11:21:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971121","meta_rev_id":"36bec6e7c5742ea32c08b21119a50ca53f32c3a8","_number":971121,"virtual_id_number":971121,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-27 11:21:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-04-17 17:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-04-15 16:00:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-22 07:56:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-17 17:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-04-16 15:15:35.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-15 21:01:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-09 17:25:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-04-10 08:52:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-04-15 15:32:20.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"e57f65f311875d28696f7486dc13889cfdd465a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 19:46:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f19a9ef6bf5dc56dd0541e236ca0d47be8745be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-15 19:47:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"460503ace6059d48e902b7bacd72df5b43294679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 21:01:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8371ed5226674bbe98c4e47a1ce3b00d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f965b51b6e40b18d848241c51e405f : SUCCESS in 8m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f8269472c8b042e9b9bdf226ea0ac8de : FAILURE in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88b32a91410c42748baee7a0693ba92b : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47d8c9fd05c4d5f9781b669f1b9c1a1 : FAILURE in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b48199847ce8412a98b2bdf367581256 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0038d770a7c4990859eee16c7397dd8 : SUCCESS in 2m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c383b49255b43f0accab88f1beef572 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3bda707fc71747d795d60f93eb0180d6 : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/057852c63edd4724afbb906b2c3093e1 : SUCCESS in 6m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6781899c7e424d26a3d443f4c0b0be11 : SUCCESS in 18m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56b233fafebb4c33945925e2aa69305f : SUCCESS in 12m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1341e12303ac417fb5d3a82c5a4f687a : SUCCESS in 10m 41s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/333fa89cef5a4a5eaa70fac813a14844 : SUCCESS in 20m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/acace8bfd6d949809ac4ae6497172a5d : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/43adcc2c948940389da52d44f2d1b8bf : SUCCESS in 40m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/43094294aa764204aa7fd04224a5363e : SUCCESS in 36m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a9db4dcec774a2fb5c670472ca65ee2 : SUCCESS in 54m 38s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f2946e3d9c064256add2cc52e5814a95 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57a9714cf9c742a6a1d63a4be2059410 : SUCCESS in 52m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/66ff065d616d42d2bb6d5053d9237abd : SUCCESS in 21m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/847b7dd979694c4c8457657f52f3af88 : SUCCESS in 9m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fd1a73592254311b9f4a74323bf3baf : SUCCESS in 17m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b97763b4c6d480fa77b952d7a76018c : SUCCESS in 15m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a201ad0798464403a24729833d8d23ca : SUCCESS in 20m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2584584056144eb1a3919992fc3e67a3 : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/641a5379098d451da0f7b21f122a06c0 : SUCCESS in 11m 46s\n\nWarning:\n  Comments left for invalid file kayobe/plugins/filter/networks.py","accounts_in_message":[],"_revision_number":2},{"id":"9d694abf34ca799a9440ee2bbd8db4df5f294f07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 18:02:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b65a5371808ad33dd2229b8b6eb15f9f6103415b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 19:36:38.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb6adcc9dc584e56be13f57338ec323a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95db52119cf94ac9a619fb01cd0db8d0 : SUCCESS in 9m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/05ca37bab84e4d56905b0ec1bbf2fe4f : FAILURE in 5m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ffe4e9e6ab04132b46c5ab52fe5867d : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efdd81a054e4ccbb2f5aef3f1dc33b4 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/39ea9d3b525942f7be2a9a5e94a04c37 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9410a2481f1a4ecba64915a2a9d34a62 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a834f602c4374c3695d3976a05ad2943 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a8eddd043d44537baaca2ff216b7685 : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea82e61f09b84f35a0542f05d8510815 : SUCCESS in 7m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8af7bbb168674dc4a791ccde6ba15978 : SUCCESS in 19m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f19bbc95e6649d8a5c34c5b89402f3b : SUCCESS in 12m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/795d64dda09f4694b52b735186e9bbda : SUCCESS in 16m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3679272df0b4d279b13891e7f67bd55 : SUCCESS in 15m 27s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7afe5cc86dc64477849ab4cb461d17ae : SUCCESS in 51m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/133935007d4c4078b00bc43d6e2b8152 : SUCCESS in 49m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3ea87d733d934600a17b9b6e167e1d5f : SUCCESS in 59m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44ae7d3fffe940659c81c1930ce5fb25 : SUCCESS in 51m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f60f299a49314cbdace5c7448708af9a : SUCCESS in 51m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bdb4d98c10524f82a307918b496f3336 : SUCCESS in 1h 33m 02s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/046dbfd6e0eb4582a311ef90c2ddf317 : SUCCESS in 1h 31m 22s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/028975cfb8c0467aac309eecb8b38978 : SUCCESS in 1h 29m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80fc2d2ec22848eb8f4a3abb8ae513cf : SUCCESS in 1h 32m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4bd6ac865bf742e391a0246b9cf91051 : SUCCESS in 22m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cf04005a71ad4cbd988ccff9e0ee75f1 : SUCCESS in 14m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e13402f1b0c4c538a816ad72a552379 : SUCCESS in 9m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ba93439b9d34ad987619e9cb3191374 : SUCCESS in 8m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/570114c72d234e1697979e8ec003cf6c : SUCCESS in 33m 45s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bd97ad07a8d54a978e1fb8e3865a6944 : SUCCESS in 17m 20s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4615e9c74d9479d82eda778e3b236d1 : SUCCESS in 29m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c96cc928a50843b7b08f5f2a631cdf96 : SUCCESS in 15m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1d6c807c19cf45caaa9309110b3a4002 : SUCCESS in 10m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abdd5e9fd40e4bf989c8a1a2b906b152 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":3},{"id":"27a5f0e74b2719dbcb33f8b9322663cd7d5a6677","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 10:04:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b828d0e34c8ffcc1279026bff5246e8b54f1166f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 11:47:33.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec911cea8e264a2cbcbd61ce9c0f5ade\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65dbfaae91c434a93f264aa9dcf3dba : SUCCESS in 7m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7a746b1185243d4b373b115c2902965 : FAILURE in 7m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed733e8cc8ae49e2b0486369bc769c7d : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39854d2c85a64746b3ca311c219b7d47 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb6eaf42f16a4c948741dd30f284f428 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01bc0a7dd39e4ac1a3c34c35293b1568 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4bdba010f3bf45aab4a7a699272bbd7b : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db86dd4e52004ab8acdcfc8ff36814be : SUCCESS in 14m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a06f5ebd66748049a6f07fd69092158 : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/492b8cfb5b074602856b46e65dd1e539 : SUCCESS in 15m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75446e352f9843e69f00448d702eb482 : SUCCESS in 13m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/934a22e680c54208ba98cb6f5ce30918 : SUCCESS in 12m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3d22ce9c07e4187848be558ba856598 : SUCCESS in 15m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/77586d6cfb384b98be018fac97243de2 : SUCCESS in 50m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6b6ca859e52a4371b9ff4a420e72b15e : SUCCESS in 49m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bb6071ee96e14b9297ee44897db5227b : SUCCESS in 56m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b24ad0e56953454e936753d74ac77db1 : SUCCESS in 56m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/54ccaebabd604bc7865188ec43daaa2b : SUCCESS in 53m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3ff0013188bf412a82ea02623aee63df : SUCCESS in 1h 39m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/578b517dd760423aa832f6a9ddc5b3a3 : SUCCESS in 1h 35m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfdf7e3de7ad4d04928480b3f31c9d2d : SUCCESS in 1h 33m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba9d0c5ff38e4e66ac84c38bf0d7344e : SUCCESS in 1h 38m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c7663cffaf0d48af8d8095f08210c73d : SUCCESS in 21m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0295164f4e4b4851882a93bff8a6683f : SUCCESS in 16m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b68699890f64960bf28caf337b4dd9d : SUCCESS in 18m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4c6a35c0ae80436e802213ebfd0e22f6 : SUCCESS in 14m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de03d549415142fd80aaf3273c4ce7cb : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3d78896cea3e4fea8ba5d35f9b484d34 : SUCCESS in 33m 57s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bdb8f809545842b1805c38f7bc8f8ff0 : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/565fe4b0d34a4e9eb2aaaa8e8da103b1 : SUCCESS in 27m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/65b776ef3c50400198f92b725bcc1b31 : SUCCESS in 18m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2595d6f0391a4540a1be0fb96a05d90d : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":4},{"id":"01fde46e4d88e6935d6c53fdbe074162965956d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 18:00:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1c6b5afc94b94e4d61054fb08eeffab469857558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 19:35:31.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/99d717deb2e14fcba8759c8b9291838b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b53929278e1d471a904ee0b75ba86fdc : SUCCESS in 6m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/92959b2659fd42e2b6bcaaa779923752 : FAILURE in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1c63c5258f44c0a9878749ad5b85284 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e56a996bbc64193af27c5d63db492e7 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b488e9187e94cf3866d9fa267b25591 : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c45ff8a658d6428d8939a7690d8700ee : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2aa66815bfe046829b235d78c55ac951 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6e833d6f5ee42498296cffe07e66b6a : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d3c2eafbe114ce19b572559da0b72e2 : SUCCESS in 12m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5fa3f2d8e8004293b0d54a440d700021 : SUCCESS in 15m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0e9534c02b448ef89c2004f7dec567d : SUCCESS in 13m 29s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0f8c17515aeb4a22b4aee3dbf13ae3d2 : SUCCESS in 24m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a70e3ded6ca84791ba0442fa6446577a : SUCCESS in 16m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5325e0019b1d4433bc43bd2d49bab043 : SUCCESS in 59m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/53c3a0ba87034fbb93fb0fecd0753626 : SUCCESS in 59m 46s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1f57d93ce71d4f9aa2084938ee831feb : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df3fe04b97bf4f4a9a1814a1a9829def : SUCCESS in 55m 21s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/58659243f67949ff92a6708344f3e810 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/816d0c25ae3047a5ad7608eee1692d9a : SUCCESS in 1h 28m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1959a38550824f7db59cfe63edee7747 : SUCCESS in 1h 30m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f23793fc186a4b62bc9a31b8ca7c2f10 : SUCCESS in 1h 26m 20s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de3b09d4c23848b7b59147845d5dbb8d : SUCCESS in 1h 25m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/58ef2f9e1bf8412fbb5a29ee4b5a3661 : SUCCESS in 11m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ceefde297e9d45648dadc10baf53b1db : SUCCESS in 17m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/caff08581dd64db983f8c1c521cabacc : SUCCESS in 12m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7df8e0095fab44f88969e8151dc2c462 : SUCCESS in 16m 36s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bf2844ba1e0f4e63ab63bf6d5a75a048 : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/eb1e4ae6ac904c5ab3b4941a2d3bb9f1 : SUCCESS in 17m 36s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3534488c214048f8815cd8f0f6ee7c16 : SUCCESS in 16m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a77fbc5cf0c748cb94d3a157ff14b7ea : SUCCESS in 14m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b1e7788d63394b64ad31c50e1cce037d : SUCCESS in 9m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a94e33defe984bc583df1e0e7cc74399 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":5},{"id":"b80d53b5cc939903481d43f7eaa2fbd67ba34166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-29 12:42:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"82e76fe415a51658dc972eed948e1f0a3dfdb2ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 14:14:10.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7113da2cdf2c40b8b567b586619e2a0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2769f81d8845c7abc9df807e144a45 : SUCCESS in 10m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/752c71602703428ba1bf0bb825802949 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74201f52d3fe4a58898c1423ecae0cf5 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0336e3e0b14f6791c72259e4ff9aff : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3593e6cadd04c178092519d410fbd8a : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5decfca00aca4dc68bd26d494770e54a : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6eff00b6d7a48c4bf38a950c864503e : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5178ae194b4409293b10cfffd15f4f6 : SUCCESS in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0571b38a9e834cb8b2c9edc515621dcf : SUCCESS in 6m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dba0cfeae5c14fd0af9afe76380109c0 : SUCCESS in 8m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3298d8cc868c4fadb8147e08516ec411 : SUCCESS in 14m 15s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c11c87335e0e405898f7f9ec8b9b8813 : SUCCESS in 23m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e55e63e36a64aaab95d2946feb1dc33 : SUCCESS in 15m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/149a9271a78744e88045dcf3e28a05d9 : RETRY_LIMIT in 3m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/af6c559d92874987b9a860510c662fa7 : SUCCESS in 56m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/fe1504e8e39242828384303c007feb5c : SUCCESS in 59m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/388570e53612429292de5da29d63f04e : SUCCESS in 57m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/255e78f747f94e7391e653ff37748638 : SUCCESS in 1h 00m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/88573c9f7a114370b77be21d93d69ef5 : FAILURE in 5m 19s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b26c04b9bed949d89162007355ef2897 : FAILURE in 5m 16s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a90867324280413b9f3e926461e4b4a4 : SUCCESS in 1h 26m 40s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b633bb6a81554355976ad00f2887b3a7 : SUCCESS in 1h 30m 13s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/606e2aca161c4d4e9a3a00ef42410c09 : SUCCESS in 22m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/60002cbaad9d461e9a42ab5d07f52d8e : SUCCESS in 11m 47s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a84f07f809734d2386a5cea1e15dc669 : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8c1bb7e884e42a2b9f68690bd4307a6 : SUCCESS in 15m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3bbf761d530a4d58aaf5ead03539c58a : SUCCESS in 16m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/40eb7dca62e247c38a4d3e8fa10fd886 : SUCCESS in 40m 30s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a16155d67639470a9d8dd732ed9f6554 : SUCCESS in 36m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e365c2aeeec43eda9e73d0d9513929b : SUCCESS in 35m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/41b0b55538af46a589ea9e668c4a45a3 : SUCCESS in 15m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8eb7ea309d742109976e0711fc43ed4 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":6},{"id":"e265ce953bc5f150fae72c92439d089045b9bdea","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-08 17:25:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1a5250e6ca920ea7a3d2dec02db616e213d6d656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 19:07:47.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3abdaa6b40e14302ab5367884511ba37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06b1b7881ef546e988fb1cb9f6f4dbf3 : SUCCESS in 9m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/497b7f1c6a0e4934885ed537cf32bb82 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a754deb203f9457590a4003798cea1ff : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53694e4bfd6449fac88d16e0ae1b958 : SUCCESS in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/271c2ebb50df408f937565f8a0005928 : SUCCESS in 3m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a64b360ad6474caea80cb57e14110821 : SUCCESS in 4m 52s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73d06ab57f334a2b9ed20358db4e31b7 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3cdaa1eb39b43e3adb47c7c1109823e : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/138e43ccf68b4a6f832465f82eabdcba : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa2b614bdd824e2ca2c353709a095c55 : SUCCESS in 11m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ea590d56f47a4af498daf5e53915096f : SUCCESS in 15m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed17e44bfe514f07b77626f93964a880 : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/13339fcae0b3492f815b32f649c92329 : SUCCESS in 11m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddf45bbd67f040a7b4e83521fc56c228 : SUCCESS in 17m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/cca564a9107b4562b8dd197d73e22b0c : SUCCESS in 59m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/adc71c03f14c47e2bdacae9456659cd8 : SUCCESS in 57m 50s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8e1decbb9bfc4ad088ff7247eb2f8f0a : SUCCESS in 1h 00m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/652e7093aec741e2af08a45ae16d470c : SUCCESS in 58m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a07e443f79af4ff5b4b3516713779cbb : SUCCESS in 59m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4d7613e55c3c4ccc94811760cce357e9 : SUCCESS in 1h 36m 09s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4e2f095f551c43f6b699d2f6be7bfd43 : SUCCESS in 1h 41m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3ace726a0e74b11922bc592519ef2aa : SUCCESS in 1h 28m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c495903d1564f4a8867534f0e17afba : SUCCESS in 1h 34m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/01cec23e4fab457d8061f2ef12058da2 : FAILURE in 13m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1d805e3030af42599d4ef0e4e5fab8e9 : FAILURE in 10m 11s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab33af5c0cad47d8ba14ee01ca50230c : SUCCESS in 10m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/143fd94fbb18429597f780be32ef2ff7 : FAILURE in 11m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3af5856ff2ff432891b178bd3f5ee060 : FAILURE in 18m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/31b53f4efa28461f9a55520bbe547f58 : FAILURE in 15m 31s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28874ba4af0a479f83eb912df93b9144 : FAILURE in 16m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/613d0cb0ee3c47dcbec0ff30450e4b86 : FAILURE in 11m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/74ab64a703fa4f358943423d8f99e749 : SUCCESS in 17m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d85b5925cd94654b9919da81cc38d20 : SUCCESS in 14m 21s","accounts_in_message":[],"_revision_number":6},{"id":"7cf72e485fd2cff4e9e998fa01af35c21f2901f6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 08:34:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e8d7193f4b44f3c9d2e71cd92981872651c4c4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:16:07.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45278017bff04e9cbd6e6b48a411a80d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b112fae11b141fba65b75796d74baa3 : SUCCESS in 6m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/84bafc9605ae4384a1174c98db5c6006 : SUCCESS in 7m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48294c088bb5455095bd8875825c4757 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0835a5e599644ce29dd82c4a210afac2 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0e2d1e920184573ba9b29f063953af4 : SUCCESS in 4m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0523ea8c1b1b444596c3f95291192b20 : SUCCESS in 7m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3860e93092b04766aa8dd515d619c074 : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0b6c93f8c6344e9b4acbccaaa2db102 : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fc3408a10ea4deb818f786653f41636 : SUCCESS in 15m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5025cfc4ca1849e6be3afceb316d555f : SUCCESS in 13m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8f604d7dd77d4e3d8a9ce744b36e8ad0 : SUCCESS in 13m 41s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3534e8cfabc041108c9ab75a7868ae99 : SUCCESS in 13m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/08cb211aeafc47f68aea0706ea1478ad : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc737b56a3a4436786d15740e0e2d735 : SUCCESS in 17m 15s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/41ab88f8ff404203b7d8a2bf97c20e0d : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/774c4972ad7344d7b176a146372dad2f : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/39c7fd9a52f541ec9a8908151c2079da : SUCCESS in 1h 02m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/017ec4d4372f4c0cbbe228627f317a2b : SUCCESS in 59m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8756040221d8457d959c16177cc786d9 : SUCCESS in 59m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/031a75f03a2a458daa5b81a645bdc20e : SUCCESS in 1h 38m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c5f4b9987cbd4784b835c1df39387683 : SUCCESS in 1h 31m 24s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41abb7c298174a6dac2dd21b67d15772 : SUCCESS in 1h 30m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a71c304435eb42568d2fd6a8115d4ef8 : SUCCESS in 1h 36m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/80fdc7f53b6f4e8586805d0d1a131d5d : FAILURE in 13m 46s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/600ee01f14b44c75872091abeda782e3 : SUCCESS in 17m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70bdc768746a4062ab9e7a58339d5cdb : SUCCESS in 20m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9f5597b1c65446f096d058ae700f2468 : SUCCESS in 19m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/44e2f043b9be4ae3926f1f643a41ff05 : SUCCESS in 31m 04s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ee7c0d9eb2464ef5b4ccad99028fcccd : SUCCESS in 17m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb34be78e98b4368901dad9d615521c8 : SUCCESS in 30m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db5df31ca3bb43f8ae95c8d93f143faa : SUCCESS in 34m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c5824f1288ff4a56a2f537751b586b80 : SUCCESS in 10m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a58e867e1144798b1f11b05326e2bd9 : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":6},{"id":"4d75e2a4de74acecffb6ae53d75a96dcfcfd6e47","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-09 10:49:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fc1946c14d20c21aaa52e5f9162485b63f99c6c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:31:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e428825ac7554017ab4d48c0c2729f2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9fe6d4243e146cab5f9be1150b50deb : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/17664e41be324835b06b5209c5d5fbfb : SUCCESS in 4m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06e16ce17f67460e9bbcb860d9491eff : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db53df3652a54ba38420f76a0d4fc23f : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/581571477d624efa823a6ba9e2974790 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/42b179b2c0e44f609fee29b28d9d2847 : SUCCESS in 6m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c1d75a0252b40b08afef05bfd5a6106 : SUCCESS in 2m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3e47e66d1b743faa56b3c5dac94fe09 : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c18148d5eaae4e77b133512685c5c79b : SUCCESS in 12m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7db828ecaa854dc19d28066636dd80df : SUCCESS in 15m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6c0751b6bfb749568314905d5d1a74d5 : SUCCESS in 8m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2209afd1b3e04d7c97f938528b7af1ba : SUCCESS in 14m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/38ffd2adb70248afa17e7dca5fee6c3e : SUCCESS in 9m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6de8dc7639ff4b4d8149475963208532 : SUCCESS in 17m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/75e27339988c4d79b2e8d97af361c992 : SUCCESS in 58m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/68a6fd18cd1247049a5fdadc5649fe8f : SUCCESS in 59m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/723034451f8d4729a8b216069285f885 : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/905c51edb1a1487da2148810b7a048c5 : SUCCESS in 59m 12s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d3e81fbf68554ce7a159fa1df2ee6ec6 : SUCCESS in 59m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/18c41b29d06f49de8a8aeec230747042 : SUCCESS in 1h 35m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/33fc2b45450d4a6a8e1bcc3364b7bc13 : SUCCESS in 1h 29m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae979ce697a343229f0270eec82eaa1e : SUCCESS in 1h 25m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54a3e451e76c445b9dae4372357789a9 : SUCCESS in 1h 41m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2e96cab7d0694e8a863f087f9d5b459c : SUCCESS in 11m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1d9efe6a461041569ec837a387d401fe : SUCCESS in 12m 01s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc933d18e4f94631b5cd71b077c5bf76 : SUCCESS in 10m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/bc85702e0c5f4360b75f2d700cd7b350 : SUCCESS in 10m 34s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c5d2f30719d14fceae078907ce522d27 : SUCCESS in 20m 14s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/398b9145847a476687169fc3bf62b3c8 : SUCCESS in 21m 40s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0be3c5fc13e24b309a38672c8a25dc70 : SUCCESS in 30m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9852301d975d4916b2702d41e7a3524c : SUCCESS in 28m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e69931a9a9044439a267abbf858d0d15 : SUCCESS in 21m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dd55f9907e64caf8ee0adcd99741a1b : SUCCESS in 15m 35s","accounts_in_message":[],"_revision_number":6},{"id":"986a93ec1145f2e6656d9d8f854260af8d480fad","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-04-09 17:25:05.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"366b9318616b3bf66df6ae1a59da35d6d47ba980","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 10:41:09.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"5e970c2648dd84d7fe7ce40058bda45387a3c4ab","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-15 15:32:20.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"544b64c92144e6212819dc456e2ef6adb97fcf7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 16:00:45.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"718a7510c4ac9736d33f45dac27b6f884813c9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 16:43:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c96b0fdf39df45159e69ffd59eb92643\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc00163746ab44cd8ade20081f4e389e : SUCCESS in 9m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/60d7dc35ab5a4bd7917b13ae21f017bb : SUCCESS in 5m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f367018c053042d1b100affa41a6d326 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1867e0861076409d923f01c7e520853c : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e24def9e263746e5b53be4cd982cdf69 : FAILURE in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3fdd3c4d83c64dbc82aa3470b1c23f41 : SUCCESS in 3m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f58dd64fbf349e5a633439ed921c533 : SUCCESS in 5m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72506d2d19c64ec2be6069496fc0a0c8 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/72e9cb0306754cd1a115567d30da8885 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/935446bcca644c03a54543d227977d5b : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66b06639dd4440178d6b4f57b87431f3 : SUCCESS in 9m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/85d91bc6fd304b96a64cdb9cdd8b0849 : SUCCESS in 14m 10s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/391eea6fd6f04a6393b39886aecbc628 : SUCCESS in 18m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7e17f42ad09b49d3bd52421e6da63848 : SUCCESS in 23m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d463d0e01b3e4e448477f63dc80a80ed : SUCCESS in 18m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/add4a5e48054460db46382c23a50a811 : FAILURE in 27m 32s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2a3c4b5a72b94c7eb890f38b14ba2a9a : FAILURE in 26m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/24e36d5c910a438cbc14b1193a55f4e4 : FAILURE in 33m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/711c1cfa81374fa2b577965bb81a1e56 : FAILURE in 24m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3ebad40642e24eed9a92593de5276181 : FAILURE in 25m 58s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9082afeb26014b4386eb5f4aa3806f77 : FAILURE in 39m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0d09821d521b46f68911bd8f722f37b4 : FAILURE in 37m 14s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7741c6ab76f42eb9f9af95f64dcc814 : FAILURE in 33m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cea5c8a96aaf48289512bccba9126bb8 : FAILURE in 37m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/13a92a4cfc40458cbfe74cb552e9a72d : SUCCESS in 14m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5967737935244e5ba86f60ff983d5945 : SUCCESS in 24m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80e291f9646e4c9e9c144dbd31b87349 : SUCCESS in 22m 59s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d46141465cff470a86b992979092c782 : SUCCESS in 21m 15s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b3e887a1ce944fc9af6c5ace4661840 : SUCCESS in 25m 09s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b041db5ca1f34569b98b090c39d03e9f : SUCCESS in 28m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e64e24f7903b4ff2a85501b43c0b167d : SUCCESS in 30m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/512ea84583a54cb5b96864399fcd6a51 : SUCCESS in 28m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4ac5402e06c644a6b4f15bcee3298cae : SUCCESS in 22m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8176e51c8521417a8f0b5af449f1008d : SUCCESS in 19m 30s","accounts_in_message":[],"_revision_number":7},{"id":"b2467d8061a5109c6cfc948c1066e8de3a72cccf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-16 15:15:35.000000000","message":"Patch Set 7: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1b09d8bed29cb429796fa0d9e665f5072cce68f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 16:58:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5537af3b15d74ce7b2faaadec5174ba9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3437da68e6c4ec394d11534194fe447 : SUCCESS in 10m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9d0d46b1a4c14f1087632a3f9831b787 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec59bbda4c314c9d959df6a5e8049eaa : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2145b65a7f4fdda961162badeee33c : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5884da9a1fb04b769b0a4d4bbcba19ce : SUCCESS in 3m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0cf58a0996c439e8d82103f3b733c1a : SUCCESS in 6m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020b903172da4acbb396bab474812aa2 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b34e470f20841a989073aa04c135b0b : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7ccf987ce0c54e2eb95534a26d52fdd3 : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4e4b9e411ed441dd8d0299577ae740ea : SUCCESS in 20m 38s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/41084a87e7f54ff1ba29269839f4bc45 : SUCCESS in 15m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67b75a4557a74f9589b637138a712763 : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fe3cc024a2b14133a44b8017f8878a35 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c95cb5b9afc4175ad9c9e6c77ab8405 : SUCCESS in 16m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/02c5d090e9e848949e358a02aad6fd33 : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e1954992760541248addd726a969f3bd : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a11263dc9a5b4d1d8c75d220f9f2799a : SUCCESS in 1h 10m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28e9312f612e448089fb6cceed185daa : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/81e5c05a4a8f4d28bfc622da3af5bacf : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/daa504d058c546d18753de8081f462c1 : SUCCESS in 1h 39m 38s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/68fbd5a0bc7947bd94c841e9c2d7ae9b : SUCCESS in 1h 30m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a5843013e364f43b3cbd5ad26e8eeb7 : SUCCESS in 1h 32m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/065937f58a594f5d857469d94d6739b5 : SUCCESS in 1h 40m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ec0d0201c48f44378ad057fd4cef5770 : SUCCESS in 14m 54s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ebeebd67b0e0442290209329b0d68a54 : SUCCESS in 16m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32a2fcc129d8437bab1390b3d5e299f9 : SUCCESS in 21m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2ee3d05b6c7a41b68f1a33d908d3b12c : SUCCESS in 13m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d5149e13192c4c7fa82efcee6c957c41 : SUCCESS in 27m 25s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d2e2d816b5554d4eb4da3b247384d700 : SUCCESS in 21m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3664134d4140401eae7024b7dd1d0ff2 : SUCCESS in 20m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b7e87ad4c6324ed3aca459e9b69922be : SUCCESS in 26m 23s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/724174215d5f452294df344bab50c835 : SUCCESS in 25m 10s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3980bf80707e49608ed5e49a212333c5 : SUCCESS in 20m 31s","accounts_in_message":[],"_revision_number":7},{"id":"01f69bddfd4531321da5d591d774d3ad0aab6524","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-17 17:22:33.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"91522b7d708b5dffa60da66c3d5df037f33dacda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 08:58:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e0ba8be24d8041098bf9b5c70e0454dce99508cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 13:21:47.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/971027 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"c8feda41c35ad302b95751598d86987611e5e92e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-21 17:40:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7f5745d7569831ff501f97a76d9b55bb4a15f2d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 19:20:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d753fc9d8bc44e42aa1df7b76bab3a99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ffdf08465d64c9bac078832eab18d06 : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cb2cc02747344c5e8adc5f2bbe6b0156 : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcb0474d49724fea99586db54129d3fe : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af8f021daf5d41b085df52cc1f650cce : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d094f6334bd0401ca8366261169177da : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13d7ca5799674ad990e68b0aa967bc67 : SUCCESS in 6m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d067607f43514786a2a6db75dee5f7a2 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc1d2aa5fd134bf7867b541e8283e181 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98b0c4462ad94894a0ab6dccc86eb083 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b981af6d9154f3db9c1cb12b413f029 : SUCCESS in 6m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6b388dda4ebf4203840b8f52a6b547f8 : SUCCESS in 26m 37s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/906b6f4763ce4cdbb10c35b51b145a0a : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ff3766d412c9403081e557aaec1c7e19 : SUCCESS in 17m 30s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9bd202136b640e88971a13bfbb5e787 : SUCCESS in 18m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d01b40c2a2634f4c8fb1c782eb9de924 : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/522359ad1a6745b5ad28418f101c19fe : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/929cbb461168450b84afb27ae6786626 : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6efad20c163b49f69915a535bf1aee00 : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/51a45bec162e4797a90e5a6a0599ba1b : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d507b01522d3454c904f72493ac88671 : SUCCESS in 1h 39m 15s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/18e9de7d78604704a9496d7e9b541a96 : SUCCESS in 1h 28m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae101e574f504f55b235a5b21c5d4d10 : FAILURE in 14m 15s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1828452016214163aef8f72a1d568cec : SUCCESS in 1h 37m 16s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c0a19350c774432a8c2f19fdb863b8f8 : SUCCESS in 28m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5826353129ea46ac915c65c6bf511d88 : SUCCESS in 29m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0109777ff3f14340bf1164e9425d6487 : SUCCESS in 13m 41s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d92ed6d407874ff58a1b9b6c1e201f3e : SUCCESS in 25m 46s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/44f868f2356f438b8b85d189ab2becbc : SUCCESS in 23m 12s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0dd17126e70a44f2995c5a809cd6bb9f : SUCCESS in 18m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03c1fb3872f34f1d8fc31f6306528918 : SUCCESS in 31m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/025cbb86653f48858206de88a69cc289 : SUCCESS in 35m 03s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ee43b4706c82485e9829a1d37f270dca : SUCCESS in 20m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c57f9cc03f374a2990a7834dd8879147 : SUCCESS in 22m 09s","accounts_in_message":[],"_revision_number":7},{"id":"92a0a373716b447e2b75a6bf7e667ff41088f961","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-22 07:56:09.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5c99ddd1029b1706cd237ac6a3071e97577da1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 09:36:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a177b1bab3e24703bbf03fcfcceb1ac3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c573db38083c45ffa4850bed56b0db64 : SUCCESS in 10m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31dfeb8d8af94832aced2dc75762f1fa : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19ca22cdc3f449c2b132aab22b74c1aa : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d704c4ada9194c6c9f3c531bf62b1d85 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5006a51928ea4ec1a3314718d2cf41c9 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b12dd303362450698dfd862ce50dbc7 : SUCCESS in 5m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0ff7e5037b745a2ab89eaa10056cdb9 : SUCCESS in 2m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/11f7dbf2243c4e688efb90ad54a8dd24 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/131c47c457274ba4a1937a53c864e876 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d6ffa9e104d94cfea6c18e47c4b0c5b8 : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9a65fb3dad6b4e0d8ec761a1347ad1b4 : SUCCESS in 16m 06s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/817f8e6435ad492ca6c12731bda1e33d : SUCCESS in 19m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9e4ae58805a7435c8aebdab2750c8bc8 : SUCCESS in 12m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10c34993e67f4d90872bfc200022078a : SUCCESS in 18m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/07ce3648e5814c5abfd1cb727d461ef4 : SUCCESS in 1h 04m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3c5a8fa4229f4c2186092e5728e28fbe : SUCCESS in 1h 04m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b39ee4e51533405fa4f4a60c67c6b4e9 : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fcd3502675d411f93bd35a5c749dbce : SUCCESS in 1h 04m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8a8b6398538c438a8e1ef8c45dbc4cff : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/788752e90cf243daa12f59521d2e1320 : SUCCESS in 1h 36m 54s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/396bff519ecc487b833f144a66cda306 : SUCCESS in 1h 35m 13s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ab9ca974479498b9d77643d6708b5ed : SUCCESS in 1h 29m 58s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64da14ac68af4d1cac9ca1bc3756c826 : SUCCESS in 1h 38m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/175b8a15321745f998cf8ae8802849e8 : SUCCESS in 16m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d777e5d5dd04431cbea474e5a8014c6e : SUCCESS in 21m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fd31845d47c4be79c9f430df103e4f2 : SUCCESS in 27m 25s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cac88d5e8dcc46ff8fbac2720f47d94c : SUCCESS in 22m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70f1d4695be9432885096218bbaf8f7b : SUCCESS in 17m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ceae6db9fdea48e0a3fb496c5a513825 : SUCCESS in 38m 08s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/72faea5b09e347079e4f380a2ff5802d : SUCCESS in 30m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e286b7005fc74060af1e2d713f293df4 : SUCCESS in 34m 30s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5fc5456b493248d38612b0e1d276a070 : SUCCESS in 16m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e36e9b2e0aa4bfe84554cc986a2f711 : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":7},{"id":"b71f94abfe0f03c580b6724c0150bf6e48a5e42e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 09:36:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b7a3650674ad5cb8887b318f44136e47cb23ed01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:24:49.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9df18d270714b628f2a9beceb8875ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785d7dc98ab94085a9f005e04d8e082a : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cbe92a0d1d5e43e9b23ffdaa50fc7493 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ecd81aac864ea297c9e0b489035ec3 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48b625508bc04e95b27b46c780d4dba0 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9afb9e85af174127b35ba037b8d97ce8 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/449310ee86ae4f2683b9fec59fa13393 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98ee9aae0bf849659afe8921293032cc : SUCCESS in 14m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ea0f3a43c534373bc3f2dac383ca4a1 : SUCCESS in 12m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2aabca39a2c14327addcb1cd1b5033c7 : SUCCESS in 24m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b26325c85fe4032a738bd71c5b90f32 : POST_FAILURE in 19m 22s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1672b59b80f448c5b6e1ede845289c44 : SUCCESS in 28m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47e41ce47db0428186be08a265616111 : SUCCESS in 19m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f0c87459f87b4b97bd202df8aca9fc92 : SUCCESS in 1h 10m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/77c78963ac4f46a7a63aca1119f11e6d : SUCCESS in 1h 09m 20s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/88553d54856f4186957d32bdc1d7e2f8 : SUCCESS in 1h 11m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50ec14f1e6af499b92b25a8cb976efd0 : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fff0ddb3e912429baee3263ae3300401 : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5b2b92ca46834eef82b7cf625cfaaae8 : SUCCESS in 1h 42m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2770eedc0bb243c4b91d146fee5eeac5 : SUCCESS in 1h 32m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5d927e7e300442e97f97e1246a8f7cd : SUCCESS in 1h 41m 55s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c020eafde17244339156bc193a147c5d : SUCCESS in 1h 47m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/339c7ec9f318449389909c49a3ca04cf : SUCCESS in 25m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a3c6fedb686f493b82c3ca34442b6928 : SUCCESS in 23m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24d3ba4fe7d14ddcacb11b261ddadac8 : SUCCESS in 17m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/baff10a2d4754b9da5a95115fcee407f : SUCCESS in 26m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/385b9b2ed4c14d049852b4cc04836a08 : SUCCESS in 32m 53s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f70c2626487c401eb38f2c4f504d9332 : SUCCESS in 29m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b873d4455cf64e7f8cc616ff1a772013 : SUCCESS in 40m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41bc05b57c0f4b1bae600b4e45d44a8c : SUCCESS in 29m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16d8f7faafae46cb97b32a22de6f4574 : SUCCESS in 30m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ab899a3f3833469285289f44505a7714 : SUCCESS in 28m 02s","accounts_in_message":[],"_revision_number":7},{"id":"132cf48abf3d6d250cc0567197883b496c4c2738","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-22 12:32:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4ab8b34ba994b8b2cfca4bc44d308901ec931318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 14:11:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3c6ca7e553042e8b182b340855fc9de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d953f9af9540e996937908a2091c15 : SUCCESS in 5m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd4101e50b77458eae343535f6733cc6 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60a70bd2da6f466eaa3df04efcf8fed3 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8275684b69964d51a9e5a39885871cdc : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a53c7961b4764a84bdb0f3578f80b679 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8bd93394637a4f2aa1e25f364b9ae9a0 : SUCCESS in 7m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ce514851be4156b6f597ae2ece33bb : SUCCESS in 2m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4f10f9e6a814ced9e0fd7c3ec8bf7fc : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0edd72b997514660a6ff087366754ff0 : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d31dd9aa0ed4aefa6971c419a6e422d : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/18d1e7fd2bb8474b8da4f96964d8c033 : SUCCESS in 18m 27s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c78ca29f2a741819fb2a0af7b64f870 : SUCCESS in 12m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e1653204174a40b5b8d451f3774893dd : SUCCESS in 11m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/785dca142e9f4543b97fec074a9dfb5b : SUCCESS in 11m 25s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f6b7b91c6a95430e897a89a850861d36 : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f4e08f16464a4c5b9f365f48aa78f63b : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1554932af7c84c51b93f43fbad59573e : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c53c87f841c946cdbb2e171389c8ee18 : SUCCESS in 1h 11m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4021d69e570642afbcae84a2f696ad9e : SUCCESS in 59m 37s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4875cadb0f0743cf9aa5271db5a391c3 : SUCCESS in 1h 37m 52s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1a6fdf6114644abd8e6d041823d08c85 : SUCCESS in 1h 32m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/78d18448d9eb4d9d83f9e58f0af2ce80 : SUCCESS in 1h 29m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/467d70b32df24c7fa563c508df4e1539 : SUCCESS in 1h 35m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2408c94e0eb14705987dfd071e73c298 : SUCCESS in 19m 01s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ba76121b4102439a93394d21115a3741 : SUCCESS in 19m 31s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20fc77b760a1483ea408c27b831bfffc : SUCCESS in 13m 16s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1eb5f76988ba4b9aabf9228089a838e3 : SUCCESS in 13m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0a4fc685ccff403a9475f8ff78df50db : SUCCESS in 16m 54s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a66b1f63e1b448ed8288526492afa7da : SUCCESS in 20m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cac3cbb61dd94a9ebbef461d786af837 : SUCCESS in 17m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05f819e3b1784b1182d3ccee7def23c2 : SUCCESS in 14m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3173f82d65794b709d591bfccdb6d3fc : SUCCESS in 22m 01s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9b4644d82a84cd5a3c94cb433223042 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":7},{"id":"c7d1092f5b1aaf063a2015d3fadd6b38887b6712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 14:11:53.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6f32c5c41c5a6ebca8cbbb6ff9faf43358cb73e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 15:45:35.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4188251560c046d59e55ac37e2497fb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3cd1af2f9b74bc6a0fde498cbbc20e0 : SUCCESS in 9m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bc0b4335f053439b9035c534b0b53b6f : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7655375d358c4410bf121aebb0846461 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bfcf403cdd44724bf85f7ac190c24fd : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3799fbd61e43cea7a2309cee69d05e : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9273fb0755bb4b749b80f6281150dbb0 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a77ca82a1314e5994d07e942f72cfc6 : SUCCESS in 14m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ecfa7777d8d147ecb634b9da5df45a32 : SUCCESS in 7m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6684120bee704b5787abb1fc664bf9e6 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cfce61a6802469ca5508cbf81b39b32 : SUCCESS in 16m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ca36eb39aead427b91b527f0c6274f2c : SUCCESS in 26m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe8193083ded4b6eb7f9f8dbe1e66d83 : SUCCESS in 17m 10s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b1857e5dbe7849439b9665cf6039f064 : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/917989d07a774032bf5b281662f91b71 : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/1f4c2182c92b47a5ab738c474cd00df2 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b5c93d747184e5fa23c5885260e6fc2 : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a87c5360fdff4d2db03ec173c3a55245 : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/de066f1014f24302adab2cd82a9815a8 : SUCCESS in 1h 33m 03s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ec5475b1f5844926b9ca1e8ec1122726 : SUCCESS in 1h 30m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/420f13a0857b4f95b78cad37f43053df : SUCCESS in 1h 26m 41s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9ed7f3352f04fc385b522ad7db3fa51 : FAILURE in 15m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e03b535adc764124945672554be7f30a : SUCCESS in 23m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/82f342bd0cdc4f88a93a27fbce576867 : SUCCESS in 14m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6591a877ba29453ba66e4a0ca9c2ac04 : SUCCESS in 19m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06f0236d9237450c81e1f41d67edcf81 : SUCCESS in 22m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5619f0363c9f4fe59655b242ae3bf237 : SUCCESS in 26m 02s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a7f06440efb24a69807404452f8d7fc2 : SUCCESS in 37m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56aab4e124324b9ba6dff6c216a11848 : SUCCESS in 30m 14s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12f3cc610d5e447a9418d296316537bd : SUCCESS in 29m 40s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e2040eec5af942ca83332a1c2285e144 : SUCCESS in 26m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e0a3df32a2e4d17b6113f3a9d17d2de : SUCCESS in 16m 37s","accounts_in_message":[],"_revision_number":7},{"id":"7fe237bb3d75f1f4bafb95b94cbb68b5115786db","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-22 17:20:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a5ff1caa4bd21732e1652cec3f3c87aee630e600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 19:15:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f156d5153f064797ac28265d16b8d6e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba336ed9d332406d8b71512aed6f6e79 : SUCCESS in 5m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65ba4ec06784459489b595c1f245f22f : SUCCESS in 6m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/836a9395bfc94a7cb4bc0b6b3e9ed867 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/410b671b805447fe912ca51cff5f2dd8 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db71f526d1074ee08d89b6dc6db3de97 : SUCCESS in 2m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5597aa01a941450fb047c07bb97a54c9 : SUCCESS in 7m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a751ae5e9286422b80d42410745f5935 : SUCCESS in 2m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c22c321777eb490b9a7075e180bf0b4d : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/388a8542c3de47c7be2b5cd6428ec125 : SUCCESS in 13m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c176d83beea14007ab35741024e6d69a : SUCCESS in 12m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fe88bfdd933c4ef1a81bbc1edf4301e6 : SUCCESS in 14m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5bb6c12e37446d5a3c44437d8787ceb : SUCCESS in 15m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b6d8d842c40045af903bc8782063dba7 : SUCCESS in 12m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24566a2b19c64d7fa8c77917550696b2 : SUCCESS in 20m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/de5f78ea9d624f90889ad8a4fb4fcdfa : SUCCESS in 1h 22m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/464c6267c36d4915b28268294abf60ae : FAILURE in 31m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ae3104f84fb34751ba4304cead800152 : SUCCESS in 1h 07m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cc4054544c740a2a59d229cf226034e : SUCCESS in 59m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ef23439505864b4590f6d558efd1d90c : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/68a4706afde342e6814704fe3420efa6 : SUCCESS in 1h 39m 20s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/c30486e1f4814bd99efdd243e6b73e6f : SUCCESS in 1h 52m 24s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd1ac07dcc8f4bf4aa81cdb76b0d7f8a : SUCCESS in 1h 29m 18s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64f49b0e6ee84039ace46e8722150cdb : SUCCESS in 1h 36m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/57e20705fd204a65a5c59e59064b5e17 : SUCCESS in 21m 12s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d9746e5032f149bc9634a17e235dc3ce : SUCCESS in 14m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ea7ebd3f5594cdfbf4ab025c840402b : SUCCESS in 14m 39s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ab749bf960f470c995f42d9e8a8267f : SUCCESS in 21m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/41da1b51a0d24a62912ff144437e2241 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f874994592954585a72a29fc026f9942 : SUCCESS in 31m 26s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/afd5f91db311497b8db948ca1deae7ca : SUCCESS in 31m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a9584e7312d4434bb80135c53c6fd56 : SUCCESS in 32m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9cda340903ef4333be76e978b9b38b47 : SUCCESS in 19m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/79d4dc7889bf412d9816ed93869cd191 : SUCCESS in 20m 41s","accounts_in_message":[],"_revision_number":7},{"id":"043a37bfa3d5ca8f6fc00aa4fc56d2a652a85f0b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-23 09:33:23.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2097b82c67ffb9e577448b600d2fbde10839d46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:29:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c89579971740e78c0558408befd3e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c06388e2dc64ec7baa6582d9d3311bb : SUCCESS in 9m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6d228980d044747bd36ffa8c5d839f0 : SUCCESS in 7m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0966b64f2ced4b21901d58bc6b1ec22a : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1da9a561b32415e8b34e78de5c05c93 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/07251ab44d1b4bad88e6cff4c37fcb1a : SUCCESS in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3bb6def066b4afc92388117b18c8c01 : SUCCESS in 8m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/326b4e7f481349f4a0de5a74af6fb2a1 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/16db01040a24455b9c7cf7db1e3cf96d : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5b7ca84706f4e378a6e1e97acd1a724 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed69c6a9fe7444179579ac5e1e7f0530 : SUCCESS in 6m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6b72ae704fc4449d9591bd9840042f4b : SUCCESS in 13m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3671403ffce94aa7b876326d46c7c72b : SUCCESS in 15m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d103c16e3bc649e3971007fb8a952000 : SUCCESS in 15m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d021e024c25a472f96b603831473b8a1 : SUCCESS in 16m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f7a77ff09e1f4ab88da4034de3bbbbc6 : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c2b8cc7aebd04e0197b0e397eceeb3ab : SUCCESS in 1h 15m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b03c2e98bfe0461dacfb64295bfc25c0 : SUCCESS in 1h 13m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2cc50780c2b48cb9c2db9440c64dbd5 : SUCCESS in 1h 05m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa8b6943d9544bc393f99103fb707e05 : SUCCESS in 1h 12m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6a2d74ddfdbe48dd8ecc06ceccc0cbe0 : SUCCESS in 1h 53m 57s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0a3ea7aa3a894c8d8a21420035db4f47 : SUCCESS in 1h 35m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a532567327cc4b8f9864122c37241285 : SUCCESS in 1h 34m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ddd8cf9ef7c4b4ab7e8bde4bfa9cece : SUCCESS in 1h 38m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/540788e9264142c6a131254fa28cdd38 : SUCCESS in 14m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/89d0990d1d034e31b54b8fad01060d8a : SUCCESS in 19m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7b74d514a994ea0b0d422ed4ab248ca : SUCCESS in 21m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/68bcdb569e9a4c59948019cb54ae115b : SUCCESS in 23m 04s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9b95aea6af4a46f5a669f61a02d3b644 : SUCCESS in 27m 06s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/7177bc74821b45ca8acd6994820b075b : SUCCESS in 31m 03s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe71336b397c4d32868fe980a2ceceb4 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a66ef6405ef49419c219e5e24782556 : SUCCESS in 28m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/518fa5ef0ea242159f180983c76b2117 : SUCCESS in 19m 54s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b968349867de416aaf60807b2a847822 : SUCCESS in 18m 46s","accounts_in_message":[],"_revision_number":7},{"id":"6cc22b800519a51a2d38c3929296c3f5f07e2c7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 11:29:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"786101e950524ba13115993ac26b7a125d783175","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 13:15:26.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7b2f8cf0a214a50a03767020ba6bb9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e69f4dcdb04279b9a4b4214fa207d6 : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/35b70d0a483f4eb2b2bcc475dd72f0be : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f104d8a7bd8c449689306cad6236d657 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9091d39d0630446e82652b18b9d4ba9a : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52a0a133fd67454da31af3f6d0af8433 : SUCCESS in 2m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c893bad98cf46eb87f838cf5f453e44 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/06b1ba022e3a4e389dae438b61ee9287 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e30e057e579f462ab76d8a040ae64586 : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a5e5040f3d9942769b3337d9147bc2ae : SUCCESS in 22m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e8b2e7246054fbe851f4743efc639a8 : SUCCESS in 22m 55s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a06cd3f49ed64796a3c69a465868a2aa : SUCCESS in 19m 39s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25e153cfcfe841668c62ab0fe02e7562 : SUCCESS in 18m 00s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d9594b8164d14c2e90acf7287c02d4d0 : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1cdc917ea9a343e0bf8edec2b399f4a6 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/14e336a3599744eab740b94944119487 : SUCCESS in 1h 14m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/257c9cfde6f04be3b86237a7ad75f51b : SUCCESS in 1h 03m 14s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31b2985c77dd4f87b3a065bd96e94f84 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3763ea0f2f794786b3673e546358e7a9 : SUCCESS in 1h 34m 53s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/576ab87101694cb49affd854f24012f9 : FAILURE in 56m 54s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a44d8f2c75f414d9f820169c4a1bae0 : SUCCESS in 1h 36m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed2875c238144241a4242fb47c96c862 : SUCCESS in 1h 45m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/478d1bf194214e528a2593441763311f : SUCCESS in 22m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/17f842f8ed424081807c322a49517ce7 : SUCCESS in 20m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b98111efc106464986347f7563023555 : SUCCESS in 21m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc56b9724cd84d66a7d2cc463cdcfcbf : SUCCESS in 20m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0f93726b29e24885938acad4e8fea35d : SUCCESS in 28m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4c6cd7036063455eaf1ae0096fe8d2f5 : SUCCESS in 36m 10s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09d82e76c639427190becfa06bf090c1 : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d16106b2c5104a8e92c8926a4fa534c1 : SUCCESS in 27m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/cd362cd491274cc091f44ebf1e496d32 : SUCCESS in 20m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04630d1dd15541749baa2f15bac6acc2 : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":7},{"id":"9ce644250dc80c5e50abc5dfac051de2b228962d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-24 09:46:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c7badbbce4a01c2e3454912bd1d4732a564c4847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 12:57:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b3c5d5bdcb514a7cbb9f1bd0d266cb55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11a8bf05580475883da5431cfcd77c6 : SUCCESS in 10m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dfa8f53cd8a24886b436e2252ee307ab : SUCCESS in 8m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e45184bb6ddc47519577b98b9a31128b : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a320e4083d40d199a721256fbdf080 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01538e08256d49d4a283e694317b2dc6 : SUCCESS in 5m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c6881306a464ba89610de9a4abe950e : SUCCESS in 7m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b022fe07f4d4755b1a4a267885e5931 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a095047d1aa94b7c992dc365771295a3 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a0343cfb74d47b28d1d1a2568d04386 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6258bcf432544748a2742b1a267ef75e : SUCCESS in 13m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/55a7a0101cd94476b019c1cceb1aabe7 : SUCCESS in 25m 01s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5ce9767480346669ebe1faf5476066b : SUCCESS in 18m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/00b73c85c2d44f52a93ebc2bca112c91 : SUCCESS in 21m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f11770a155041698abf69910a4a93cb : SUCCESS in 19m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/7b97427f53f240c4a5b0c4c825351d4e : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f514cf90ac264cb49546cc65ce58d97c : FAILURE in 14m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/40d4d56560444d1c828d16f2a2420488 : SUCCESS in 1h 26m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb786c198baa42739ddc04b801e182f1 : SUCCESS in 1h 22m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/65b7f9180ef9444dbd751472941d1d61 : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/501673f2fd344218be69b5f11a436624 : SUCCESS in 1h 16m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/58ff8c1b6d03485980fdc8c2fa4f4ba4 : SUCCESS in 2h 02m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e0338d22b98449038a184fb3db0f38ac : SUCCESS in 1h 36m 17s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a1927a778224ab09e3f3ac09a75f358 : TIMED_OUT in 3h 03m 29s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f430efa52cef4877914dbc41ce0cb445 : SUCCESS in 27m 07s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/6db88c9633da405e8afa4dfcae63d166 : SUCCESS in 25m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/43934ec420f045838626fbdb7ea4f365 : SUCCESS in 17m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7cb921b4f9764abba6b7fbb891f6c7c2 : SUCCESS in 19m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f068352f862d456783b95c6060823282 : SUCCESS in 17m 20s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6ecf13982a53419f9268932768b97f9a : SUCCESS in 31m 34s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cb9506d8594411fa3e7b6366f641bfd : SUCCESS in 36m 33s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a2ad0c7428147198895fc761fe2f898 : SUCCESS in 27m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e6850850fd20434eb7439cc5858b0e08 : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/131ad0a3cc694f4681a093d938ae7804 : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":7},{"id":"eb8aca64066aa2ba3849e05d951365bdf3f30f76","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-27 08:06:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3c3c02d167b8b38b529ba8a7304c82259e98e3d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 09:43:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a05402208e7c4e5eacd78a13519b7454\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b723cea2eaac4863b14dc7369b24942d : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74557b6164ca4e539594ceb19bcf0f8a : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7e768470d2f44b19bc60fa5005487ca : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58861c4db5544cb693bfe76fd4e14950 : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/272650c115974ed8a733aa4f5aaacdd9 : SUCCESS in 5m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24eae8f559024df3a36cdae8c9a8a94b : SUCCESS in 6m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e4418b0a314496e8ad550fdd2c2b8c5 : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee6d47b96aaa48ff8246340430265cd9 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/573ce12214f74816a7b78e3ac4d6dcd3 : SUCCESS in 6m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7b0899988a54854abbd78b9e4573bae : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ffe0cbe71c1433a914490b5d39de020 : SUCCESS in 19m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3be9cebe7e74e53ae3862222218f59c : SUCCESS in 15m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/49f3771893494291b5314a4b7a985bbf : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fb43569d9594ccfaa0b442b64ab126c : SUCCESS in 20m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2cbb4c315a114989a6306db44b986d11 : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/38b7a9d79f5244fc96ac79a9b68acfa4 : SUCCESS in 1h 01m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6c730d65a24f46499ef7e2ebcc497b8d : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bd4e04e7e9c4ccab0a5129798c9ab49 : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7cb568e99c0942bf85024aca36353b4b : SUCCESS in 1h 04m 23s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3c366b55d7d743509f511b5a6fe48b85 : SUCCESS in 1h 36m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f095f31ff6874cffb383d8c6843f2de0 : SUCCESS in 1h 32m 26s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a609046b26b4770a3619cb9fa41d5e7 : SUCCESS in 1h 28m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54eae628b8a34513a224cc0980ea58eb : SUCCESS in 1h 34m 21s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0c7c419a8ccd429797a42f325c577144 : SUCCESS in 19m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/730853f3de014bde8a30889306140225 : SUCCESS in 27m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f88eb7ea71c9439683204d79855019e4 : SUCCESS in 23m 10s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c19ba4c050b4c65bb6f9e381cf0834c : SUCCESS in 21m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9c08b0fffcbb4ed88efea292da1c3eec : SUCCESS in 35m 59s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/42124b5f18d74568be1d3368b73ea936 : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/687286707ff940698a66acad7e3d2988 : SUCCESS in 17m 17s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e731991ccdfa412fb3fea580e802c5cf : SUCCESS in 28m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8d3aa4990e0e43899ddf60e467d00f11 : SUCCESS in 14m 34s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a89ea0c55494f0e9f2fcf8240fd5d2f : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":7},{"id":"6ab7fa3b496b49c7ee9f597ba03c52465e94fbb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 09:43:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9245b3048ef820d1f718e165f98c276ec6a0cb2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 11:21:05.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704e2314a92142a489ec48b4ff3ae2ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85586c684dcd4763956c86f3524c85fd : SUCCESS in 9m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b50afd7ca5a548718b874006dde42000 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/805c49e6d1bc4465b73af9aa638f2f52 : SUCCESS in 4m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4c2d5936e51498f8650e20324ca1c05 : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a97d460d1365459084659d59a3c80c80 : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/abe69f0cd7cd4f3fbf0ebfa530b76379 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23686d7bbe694bdc881050a765c5f6f0 : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea98477e31e9437381f0f5580dc0fcec : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2527c11011864b15a2f715ca33407d73 : SUCCESS in 14m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bc85a18bff74bbba2aa9dd15f415f93 : SUCCESS in 14m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/599aaf59c3dc47b68ab3a9927a640640 : SUCCESS in 13m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/663f1a35cf1c4f579c208ee199191e0c : SUCCESS in 18m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/41b705494c9e4a6ba483874d87e2e363 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d1e5dd05537e4c58817b61fad362ee37 : SUCCESS in 1h 03m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/13ec8fed1ed14655a66a3ac7d8b14cba : SUCCESS in 1h 03m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3864642aa28545f3a4ca85223795e305 : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c63f719bfc3c430d83a02cdfc544a42a : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3e05f08fae014ba9baa8512e876c9c22 : SUCCESS in 1h 36m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a49ded668ee549f1bb40b484cfa0446a : SUCCESS in 1h 32m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6283062746b3484db55aecc1e641fb1d : SUCCESS in 1h 30m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87ed6fbb1a044d97ab3c3d1ffc1fe11a : SUCCESS in 1h 36m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b46d2e8491404993a4163b8b79b9e680 : SUCCESS in 25m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a3caf424c8624f5a921612497ce14bb7 : SUCCESS in 16m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bd9e33c18354ca89d0624b7bdfceb00 : SUCCESS in 24m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/29a8a54df3de401da78dac4542a11f48 : SUCCESS in 24m 23s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/126c1924383f45f9a6278bc19147e4d9 : SUCCESS in 15m 58s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a523f98189e84e0b95cdabf9f53ddf50 : SUCCESS in 38m 27s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d60ce2fc599d4f889307d8f71137e0a0 : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a5dcf5e28c94e4fad5b075313c63fa2 : SUCCESS in 34m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bf4a8f776f4a4a54b183df743fd24ffb : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a919b968577b4670b21ce99814c8651a : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":7},{"id":"f788c8c5eef33c8669a038e72ea2bbf38d9c949b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 11:21:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"36bec6e7c5742ea32c08b21119a50ca53f32c3a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 11:22:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc721c68f3754b989b647a8d9fc4d5fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170f4d27c2fa4619af4f5f1bc031f01e : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90a6d2189e6b4a4d9297462fa6878eb9 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"1bdb5ed4ee7513b26c12ddac3b84f827830180dd","revisions":{"c6cf46a215c0c1f2425bb1efe031357e24a758ba":{"kind":"REWORK","_number":1,"created":"2025-12-15 19:46:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/971121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/971121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/971121/1"}}},"commit":{"parents":[{"commit":"797b9b33ef96e753e8151442dd8c0d43cebcfd9c","subject":"Support configuring DNS on Ironic Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/797b9b33ef96e753e8151442dd8c0d43cebcfd9c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:44:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:45:58.000000000","tz":0},"subject":"Adds option to disable registration of networks","message":"Adds option to disable registration of networks\n\nUseful if you are using OpenTofu to manage these.\n\nChange-Id: Ib18cf699d2b477fa2d4272e3fe690d17dbe17173\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6cf46a215c0c1f2425bb1efe031357e24a758ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6cf46a215c0c1f2425bb1efe031357e24a758ba"}]},"branch":"refs/heads/master"},"3c2e1efb40d47b78e6a1520224ac8b459bf7d819":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-12-15 19:47:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/971121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/971121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/971121/2"}}},"commit":{"parents":[{"commit":"2273cf8f5925b6a1060ede1e4f1bb962eafd635d","subject":"Support configuring DNS on Ironic Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2273cf8f5925b6a1060ede1e4f1bb962eafd635d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:44:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:47:29.000000000","tz":0},"subject":"Adds option to disable registration of networks","message":"Adds option to disable registration of networks\n\nUseful if you are using OpenTofu to manage these.\n\nChange-Id: Ib18cf699d2b477fa2d4272e3fe690d17dbe17173\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c2e1efb40d47b78e6a1520224ac8b459bf7d819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c2e1efb40d47b78e6a1520224ac8b459bf7d819"}]},"branch":"refs/heads/master","description":"Rebase"},"b7236b8ac3d8578f9320ab646076fb436fbfdc44":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-27 18:02:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/971121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/971121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/971121/3"}}},"commit":{"parents":[{"commit":"d56d893cb2f00498d3670ba366ce617f993ebedd","subject":"Support configuring DNS on Ironic Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d56d893cb2f00498d3670ba366ce617f993ebedd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:44:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-27 18:02:23.000000000","tz":0},"subject":"Adds option to disable registration of networks","message":"Adds option to disable registration of networks\n\nUseful if you are using OpenTofu to manage these.\n\nChange-Id: Ib18cf699d2b477fa2d4272e3fe690d17dbe17173\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7236b8ac3d8578f9320ab646076fb436fbfdc44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7236b8ac3d8578f9320ab646076fb436fbfdc44"}]},"branch":"refs/heads/master","description":"Rebase"},"9b03310327630fbf6f9e64d65364502fbd212d1c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-28 10:04:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/971121/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/971121/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/971121/4"}}},"commit":{"parents":[{"commit":"fa939fbb4773c6d6d90320a34a4034f129b139b1","subject":"Support configuring DNS on Ironic Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa939fbb4773c6d6d90320a34a4034f129b139b1"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:44:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-28 10:04:37.000000000","tz":0},"subject":"Adds option to disable registration of networks","message":"Adds option to disable registration of networks\n\nUseful if you are using OpenTofu to manage these.\n\nChange-Id: Ib18cf699d2b477fa2d4272e3fe690d17dbe17173\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b03310327630fbf6f9e64d65364502fbd212d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b03310327630fbf6f9e64d65364502fbd212d1c"}]},"branch":"refs/heads/master","description":"Rebase"},"46deb81045cfcaa40876e265db00990ace577db6":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-28 18:00:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/971121/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/971121/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/971121/5"}}},"commit":{"parents":[{"commit":"744f3c88a9cc3860e2d599c3e9262a26f6b72a7d","subject":"Support configuring DNS on Ironic Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/744f3c88a9cc3860e2d599c3e9262a26f6b72a7d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:44:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-28 18:00:23.000000000","tz":0},"subject":"Adds option to disable registration of networks","message":"Adds option to disable registration of networks\n\nUseful if you are using OpenTofu to manage these.\n\nChange-Id: Ib18cf699d2b477fa2d4272e3fe690d17dbe17173\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46deb81045cfcaa40876e265db00990ace577db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46deb81045cfcaa40876e265db00990ace577db6"}]},"branch":"refs/heads/master","description":"Rebase"},"3a6f80bbdc459a67f122987817d36190773b2eb3":{"kind":"REWORK","_number":6,"created":"2026-01-29 12:42:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/971121/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/971121/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/971121/6"}}},"commit":{"parents":[{"commit":"744f3c88a9cc3860e2d599c3e9262a26f6b72a7d","subject":"Support configuring DNS on Ironic Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/744f3c88a9cc3860e2d599c3e9262a26f6b72a7d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:44:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 12:42:16.000000000","tz":0},"subject":"Adds option to disable registration of networks","message":"Adds option to disable registration of networks\n\nUseful if you are using OpenTofu to manage these.\n\nChange-Id: Ib18cf699d2b477fa2d4272e3fe690d17dbe17173\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a6f80bbdc459a67f122987817d36190773b2eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a6f80bbdc459a67f122987817d36190773b2eb3"}]},"branch":"refs/heads/master"},"1bdb5ed4ee7513b26c12ddac3b84f827830180dd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-04-15 16:00:45.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/971121/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/971121/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/971121/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/971121/7"}}},"commit":{"parents":[{"commit":"0fdc9c2edc3f865f2aa9cd331f8d19bc1d0fd060","subject":"Support configuring DNS on Ironic Neutron networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fdc9c2edc3f865f2aa9cd331f8d19bc1d0fd060"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-15 19:44:16.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 16:00:45.000000000","tz":0},"subject":"Adds option to disable registration of networks","message":"Adds option to disable registration of networks\n\nUseful if you are using OpenTofu to manage these.\n\nChange-Id: Ib18cf699d2b477fa2d4272e3fe690d17dbe17173\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1bdb5ed4ee7513b26c12ddac3b84f827830180dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1bdb5ed4ee7513b26c12ddac3b84f827830180dd"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"3a6f80bbdc459a67f122987817d36190773b2eb3","theirs":"0fdc9c2edc3f865f2aa9cd331f8d19bc1d0fd060","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
