)]}'
{"id":"openstack%2Fkayobe~971318","triplet_id":"openstack%2Fkayobe~master~I5a2b001803bfa2f83cf823e0db8f27af06a99171","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-22 11:46:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5a2b001803bfa2f83cf823e0db8f27af06a99171","subject":"[DNM] Test removal of module load","status":"ABANDONED","created":"2025-12-17 12:44:42.000000000","updated":"2025-12-22 11:46:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"718915b9642a05de261da91478789d815174649f","_number":971318,"virtual_id_number":971318,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-17 14:00:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-17 14:00:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c159065aa54d9643b68f8fcddc7f79e79853ae8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-17 12:44:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"947444e5a0002db1f1c2bc40feb10b1748cab9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 14:00:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9723702db7f84931a7f142455328c82e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f322642a7d6c41f09eb1ae019eeb89d3 : SUCCESS in 4m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a4e0995a0f3a48dd9b5ba185b9f31727 : SUCCESS in 5m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f45aa62340d4ffc85f152e2b3d94335 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccee8509dcf94b93890843f0d58bd9b7 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b6c21592b924565aaaa8e596f106697 : SUCCESS in 6m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca07cbda4dcd4563976cc4fb575d5c53 : SUCCESS in 2m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e03aabf54934fae8fcbd949680c86b0 : SUCCESS in 5m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34f75bd7b4084f1a9218e827aaa336ff : FAILURE in 5m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e6a72d90e99d492dbbfeb6e6de4e8161 : SUCCESS in 18m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/431d491ae4814d65ab61c534c60bcceb : SUCCESS in 6m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b0cc0a28d87b4c9f9f10c5a1f93388e8 : SUCCESS in 10m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/823d2d286bfa4b1eb0ad260c220a2417 : SUCCESS in 8m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6dd314995b744a069686a2941eaa8347 : SUCCESS in 35m 08s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6f4fc61261c34718ab9b60f9b5823d0f : SUCCESS in 37m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/bd7f55dc15d8428ea545994a92e42a58 : SUCCESS in 57m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7f0cb965c9b41d0a5dd3a5746c57311 : SUCCESS in 1h 06m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ccdb893eb1b49b4895d7c6bf37e4c1f : SUCCESS in 1h 10m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc0901fe8b924ae9a7865df97f06a92f : SUCCESS in 52m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/6c85df5722c1447687be00830c5edee9 : SUCCESS in 9m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5d4c6448fa004b1bb32c8f5f4b6ccb92 : SUCCESS in 16m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7995ea34228948a2ae64524561b58daa : SUCCESS in 19m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e07ae886f37b4184854be1ddcc68466b : SUCCESS in 16m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ca5e341c9bf423fbf83dc24bb393ebb : SUCCESS in 32m 05s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/89215983de364c108ea6b97e541fd205 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1361973d16774e988fccdc1c62a71a43 : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":1},{"id":"718915b9642a05de261da91478789d815174649f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-22 11:46:01.000000000","message":"Abandoned\n\nWorks fine","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c21e0835c8cd85e631f00b655a2d256add8d66af","revisions":{"c21e0835c8cd85e631f00b655a2d256add8d66af":{"kind":"REWORK","_number":1,"created":"2025-12-17 12:44:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/971318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/971318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/971318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/971318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/971318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/971318/1"}}},"commit":{"parents":[{"commit":"cee36ca383be4eb144058829adc127e37db8a950","subject":"Merge \"Add configuration for Ironic introspection DNS servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cee36ca383be4eb144058829adc127e37db8a950"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-17 12:44:33.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-17 12:44:33.000000000","tz":60},"subject":"[DNM] Test removal of module load","message":"[DNM] Test removal of module load\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/971317\nChange-Id: I5a2b001803bfa2f83cf823e0db8f27af06a99171\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c21e0835c8cd85e631f00b655a2d256add8d66af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c21e0835c8cd85e631f00b655a2d256add8d66af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
