)]}'
{"id":"openstack%2Fkayobe~971329","triplet_id":"openstack%2Fkayobe~master~I953ab956250d6dcc6e96212057db01f438c69d78","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-12-19 11:58:30.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-19 05:27:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-12-18 20:41:21.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-19 11:58:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I953ab956250d6dcc6e96212057db01f438c69d78","subject":"Use opendevmirror/ubuntu:24.04 image for molecule","status":"MERGED","created":"2025-12-17 15:00:07.000000000","updated":"2025-12-19 12:00:30.000000000","submitted":"2025-12-19 11:58:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971329","meta_rev_id":"5587504875e31c445dc466b3a97c7adf7e46afc4","_number":971329,"virtual_id_number":971329,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-19 11:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-19 10:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-12-19 05:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-12-18 20:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-19 10:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-18 21:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-17 15:02:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-12-17 15:02:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-12-17 15:02:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-12-17 15:02:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2025-12-17 15:02:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-12-17 15:02:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2025-12-17 15:02:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2025-12-17 15:50:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-17 16:23:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6b4eb931ce9e6de843d312ab033dad15fa643d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-17 15:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d19f6db28298f0c2ece5b3d10fbdd1e1f04db96","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-17 15:02:27.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"4a1e50f9bef1d205f4dd8a6658fb22f50fb2d7ce","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-17 15:50:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"57e724f4d95022e5d5c65038b2a1dd5b1bca9a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 16:23:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a08688ffb3e4cc5a9e5aba938a03693\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be86beea020441f82273f5dbbe57d26 : SUCCESS in 8m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b6435cf9418f4713a00129332c3e3f6e : SUCCESS in 6m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db92c1f128d94f6db85959d63bbc9968 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060c65916bc4486aa1a12277bc76e812 : SUCCESS in 6m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e54b315a736143978f08d7ffe4c691bb : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/097dc025be014d2daca2585cacb80659 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf5806bc96ae4f748290483e54912ccd : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1aa8338c164b45e8a79bd94ac45843a6 : SUCCESS in 5m 43s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3750793052fb460f810ed651c42bf7be : SUCCESS in 18m 02s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f106dcfb325416da841751ef36134ef : SUCCESS in 12m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fc6ab2729e8141d083ecdf61e91da44e : SUCCESS in 18m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35ae6199f4fb4d479e00f965d2f75abb : SUCCESS in 12m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c4c34fdd7d984808ba485739d84ca0a4 : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fabb51c10b064567adf10d553a02c036 : SUCCESS in 1h 09m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7dcfcea43ec44354a12830511cadf70a : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fdcf6fcdfe54bcabf34d6b92469a52b : FAILURE in 23m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fcbc88a2ee994d4293d202ccd0142a0e : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3138474b503340c7a6b5f2ed144a7d4e : SUCCESS in 1h 22m 40s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2b0d29400c474445b574ecd2061559ab : SUCCESS in 13m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2698257f89944ab08109d85aaaa92cd1 : SUCCESS in 9m 56s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc235fc818d0413b81bb7261559875d1 : SUCCESS in 15m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5ea2efa918f448f980e3a1d3f784de1b : SUCCESS in 18m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6ac0a27a2cb44898ed8293f7f0c57a2 : SUCCESS in 15m 46s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/16927c21cb284e7ca35898a5de99b2c1 : SUCCESS in 13m 40s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf5e19b8dd694ae4acd4f15790b49c84 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1167e4a0371b97153fc6e8e5b3eaf885706f1ff1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-18 08:15:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"28a573fd22860822f9af8e99ee5299f2b48aa5dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 09:59:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab7a68b700045d79c4743f4183451fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b5e6c73ebf4d65a808646192c75be2 : SUCCESS in 6m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ced338be47a4f13870f52df27c2f9eb : SUCCESS in 3m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3903929693494d7bafcf04d5fe24aa89 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e18e23510ba247c8b52c6f7078679c0f : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b5f2141930e43db83e4499d36cec986 : SUCCESS in 8m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b5318ddac7e465584e43fce7d539a9a : SUCCESS in 2m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a43df72479da4627b07c5eb0f48399c4 : SUCCESS in 5m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3a6f6c9e0e243bba4656c12de313433 : SUCCESS in 10m 22s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5a0d345bc041417e8b2b2a980e63c342 : SUCCESS in 18m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae4fc985c8044d6b817f3c996bfccb7b : SUCCESS in 14m 39s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c7cfcd07660a4535860f8f94c53f04b9 : SUCCESS in 10m 51s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c644aa02163241fdb2a63d01a6395d9e : SUCCESS in 8m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f51b0def70df47a2bb175cfb76be045c : SUCCESS in 38m 43s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa0028040fae4e0b89612cb96e4f86b1 : SUCCESS in 37m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d2c8924280454d5aa248dc26ef82a4fb : SUCCESS in 37m 47s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44ae79ab43ae4f6a9dab15d17f113023 : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/512403a51b8e4acead7b4bc93bd6e192 : SUCCESS in 33m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d38bb446c9f48cabaa3e459f6bc0626 : SUCCESS in 1h 39m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3153047158774a63ab0cc0e587c21a2b : SUCCESS in 9m 18s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/af8d463720e14662a66e206668f58f02 : SUCCESS in 9m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de37dad9730047c09cc8cf5e8fb63893 : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e993d012d2940b69c603c00b1f0b9e5 : SUCCESS in 14m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95e01f086e6d4f5d82f2295edd3e55ec : SUCCESS in 28m 02s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/548f3fafb32d411bb06dac1ecfb28bb0 : SUCCESS in 18m 35s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2d3903de90a4e8db976e56df59b51b3 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f7f1545a5fe73e60ff9cb7e9618a791defb242a6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-12-18 11:51:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97fec57571844b5fbae6976fc543eda3be9ec1b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 11:51:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb696c006942c1828fa4d7729ccf8ff26118ceff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 13:06:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27afd13fcd9647509634f13d8ca38867\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead4172474a54e00bd56ff93e5e62321 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/275ee13a4b2e42f5b96acc5a06bcdbcc : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be34b930409642869c33c7f59c4ed42e : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a09c8c2ab3be4e2bb75282438f4d8659 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e95fa26d00b34942a220ecbf26c7558e : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3eb6a235078b417f9c6feb9ae97b94eb : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d281011b2b5146b7800277c3535415c5 : SUCCESS in 5m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/62a7b536468c4761ba4888d30611202b : SUCCESS in 17m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cdd3edaa5dab46a8b8e977a2db68dc72 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3d19814389664d9cab25469f9ac0fbf5 : SUCCESS in 24m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84f04c78fa9e4fd8b4fe160a6f9003b8 : SUCCESS in 8m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/bb0af0cd100549c79980a69fdb904cf4 : SUCCESS in 1h 01m 44s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/947d8c4e52824c888d310e343cabf9f3 : SUCCESS in 37m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/78840cb6315249f2979a39c116151021 : FAILURE in 8m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f87af07c223f4d139f86ab4d2fd42b5e : SUCCESS in 55m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7030052df09a4e5ea8ab6e6611cc703c : SUCCESS in 1h 10m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/def94514ebee416f9dcf8702e8e80173 : SUCCESS in 53m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3f2afb4b0faa4e2a9f927516763a5139 : SUCCESS in 10m 26s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1fd55b79ebd5400c946acf0bfebf3a3f : SUCCESS in 9m 50s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/502f4d8128c949958a6feb5a4bf3c52d : SUCCESS in 14m 07s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ab24dc30424409fb660f49e7cb47abf : SUCCESS in 7m 22s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8f44f26dc558486da359cfe221552ddd : SUCCESS in 25m 56s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/99f9b831ffae462699b60ecc1ee95b00 : SUCCESS in 9m 47s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd34383362bc470a8c75feb6dd46e2bf : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5f57cdd020ec0b5c56f6dc85273d9cd1072d1950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-18 15:16:15.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9b663e14b58eab462bed42c9e5b486f9c6907fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-18 15:17:02.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a0419d15997da056492cb90748e2bb9b9503450d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-18 15:36:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b081564495d94bd48d309541e43ef0e8e3d73c08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-18 16:05:07.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1d7405bf0030cac3a27d582505f22e73ddffaaa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-18 17:58:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c18ac2e604e441df867255a8beee1c72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644c5fce02e847d9855ad1a3d907a4aa : SUCCESS in 8m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2addeddc5c914b6f927d0d3084ddec28 : SUCCESS in 5m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b18f429623da46bca3b6e5a3b1448591 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e315fa9e4c4fcbb2d4d1f677c2f272 : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d783cc2280bb429cb4d2575beb06cc71 : SUCCESS in 5m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbecb15ce81249b7bb4bd497fdd518ae : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a929cbeabd44840b430eed1f04a399c : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72a6e48ce67d4fad85d934c0f99ef272 : SUCCESS in 11m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a1c424d10464c87a41ecddeafa6c7d5 : SUCCESS in 12m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b568c5249d9d4213bad1499c26e46486 : SUCCESS in 6m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a8e420c916ad458da7f7dd148b30f9d9 : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a2b54271b474592b5e63f2b1d7bf4c0 : SUCCESS in 16m 05s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d5183837eac44050b68beeeedb62b7a7 : SUCCESS in 1h 01m 03s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/53b5540a1e4641279d66d9f5d3387fe2 : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/903dbbfb3e0f4b599bfd53f63af9ef69 : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2927aa2a8ff2400c89b053bffe68c157 : SUCCESS in 56m 28s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6456d34b284f422eaadd735c1887ada4 : SUCCESS in 1h 16m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cd001234c7f41d88884e3958940ae6d : SUCCESS in 1h 48m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/703807b6bfc24daaa7b5e334b1e52db2 : SUCCESS in 13m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/05cbc357c23e412abc29d27f3c4d1d51 : SUCCESS in 9m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff127d9411da4c9c8a9c6061f5ac5410 : SUCCESS in 15m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aefd05148d0743a092aedb12186e2700 : SUCCESS in 17m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67b01ffcb0f44333b053f7ddd204000b : SUCCESS in 28m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c94feb744223492d86afeb20c2130ce5 : SUCCESS in 17m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfc8aeb10d59421190d5b026e2f5ab35 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":4},{"id":"f91512a19ad7568db32af1f1898efcfd639ac0dc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-18 20:41:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e4e9b45e292c3a0f19ad2ac3015e86a59d97acd3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-18 21:26:35.000000000","message":"Patch Set 4: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f15607227168ccff75b045aa8cf246b510749a6e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-19 05:27:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b3d92ce857511d7b05c66ec522e713f60fa8b388","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-19 10:07:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"82535f2f2f8fc619ab6845f63274f0f0f2efa44c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 10:07:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2e31a3bb530e80675ffec50bb290118f83199ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 11:58:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d22d98427334b999ccea6d7a0f91507\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd67fa4f87d4f3da559c1a76885b9e0 : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54c8770f7a5a41088817e419e6487b60 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7694d94d6d244d9cbea615dc04dd09d9 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/789654461ec44392a4cdcc67a7e4a2f9 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f7cbfd753d764bb598826fd4fef86b34 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56c76f83669a4c8fa8423cedcdf19b5e : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/175fba0a07474f4faaba628194786f35 : SUCCESS in 10m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f16e9888cd91431e8199a0a2766513f2 : SUCCESS in 8m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4597ab5c4d548289a579d3c05ed40d7 : SUCCESS in 14m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8a1f1721d1584bfa86dea1b7866cebd5 : SUCCESS in 10m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95d01649fe5345bf9874835ee70ed4dd : SUCCESS in 12m 26s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/15d4dadce1ce4a7b9825d96bb57d8908 : SUCCESS in 55m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4b1aa988dfc54aec851aa61ca47e32ac : SUCCESS in 1h 20m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/74d950204fc247daa1a2ed3c930f24b1 : SUCCESS in 36m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f3716f47e82453c81f04b42f05b1d13 : SUCCESS in 1h 16m 25s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3b07b4ab11df4d5b8e96924349e3d43a : SUCCESS in 36m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27dab1b5cde441e8bc53d4dc11dd0e5d : SUCCESS in 1h 46m 01s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1480d5ac0ed545b480326d969fa26cb3 : SUCCESS in 9m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/dbc7764dbfdd47a4a30ac0b3dffddd44 : SUCCESS in 17m 13s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2c7e32229e6e490183f317032a2645c9 : SUCCESS in 17m 06s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/056a546d593446c185548fc943223b54 : SUCCESS in 15m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1cbfed4dfe384106b86f84275972ff7c : SUCCESS in 27m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4fad60fff38d4767895226681ab6f283 : SUCCESS in 18m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c087107ddeb14ebeb58a19182a1589a3 : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":4},{"id":"c74c519efe234525134a1cffd92699a53e4d8e43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 11:58:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5587504875e31c445dc466b3a97c7adf7e46afc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 12:00:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d95ab9a3eec4f9ba6364e3de5f44637\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6dec23c3b44efea89f7d9ee1d24a30 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1444d7947b154a251167cdb7aa54ef0b253aa079","revisions":{"7e2a4abe6d9a0750d283c2ac9ae27d5be985b773":{"kind":"REWORK","_number":1,"created":"2025-12-17 15:00:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/971329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/971329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/971329/1"}}},"commit":{"parents":[{"commit":"cee36ca383be4eb144058829adc127e37db8a950","subject":"Merge \"Add configuration for Ironic introspection DNS servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cee36ca383be4eb144058829adc127e37db8a950"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-17 14:59:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-17 14:59:41.000000000","tz":60},"subject":"Revert \"CI: Use centos:stream10 image for molecule\"","message":"Revert \"CI: Use centos:stream10 image for molecule\"\n\nThis reverts commit f773cd04888ebd6b3d728ac418ff320cba7252aa.\n\nReason for revert: Jobs failing when CPU does not support x86-64-v3\n\nChange-Id: I953ab956250d6dcc6e96212057db01f438c69d78\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e2a4abe6d9a0750d283c2ac9ae27d5be985b773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e2a4abe6d9a0750d283c2ac9ae27d5be985b773"}]},"branch":"refs/heads/master"},"a10da0f44dd37c591831bd0f5395050317fafdd4":{"kind":"REWORK","_number":2,"created":"2025-12-18 15:16:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/971329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/971329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/971329/2"}}},"commit":{"parents":[{"commit":"cee36ca383be4eb144058829adc127e37db8a950","subject":"Merge \"Add configuration for Ironic introspection DNS servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cee36ca383be4eb144058829adc127e37db8a950"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-17 14:59:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-18 15:15:08.000000000","tz":60},"subject":"Use lib/ubuntu:noble image for molecule","message":"Use lib/ubuntu:noble image for molecule\n\nMolecule jobs were failing on centos:stream10 when CPU did not support\nx86-64-v3.\n\nChange-Id: I953ab956250d6dcc6e96212057db01f438c69d78\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a10da0f44dd37c591831bd0f5395050317fafdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a10da0f44dd37c591831bd0f5395050317fafdd4"}]},"branch":"refs/heads/master"},"3306353d5976d79e4ae94c7feba74633cca9f93a":{"kind":"REWORK","_number":3,"created":"2025-12-18 15:17:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/971329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/971329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/971329/3"}}},"commit":{"parents":[{"commit":"cee36ca383be4eb144058829adc127e37db8a950","subject":"Merge \"Add configuration for Ironic introspection DNS servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cee36ca383be4eb144058829adc127e37db8a950"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-18 15:16:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-18 15:16:54.000000000","tz":60},"subject":"Use lib/ubuntu:noble image for molecule","message":"Use lib/ubuntu:noble image for molecule\n\nMolecule jobs were failing on centos:stream10 when CPU did not support\nx86-64-v3.\n\nChange-Id: I953ab956250d6dcc6e96212057db01f438c69d78\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3306353d5976d79e4ae94c7feba74633cca9f93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3306353d5976d79e4ae94c7feba74633cca9f93a"}]},"branch":"refs/heads/master"},"1444d7947b154a251167cdb7aa54ef0b253aa079":{"kind":"REWORK","_number":4,"created":"2025-12-18 16:05:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/29/971329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/971329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/971329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/971329/4"}}},"commit":{"parents":[{"commit":"cee36ca383be4eb144058829adc127e37db8a950","subject":"Merge \"Add configuration for Ironic introspection DNS servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cee36ca383be4eb144058829adc127e37db8a950"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-18 16:04:33.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-18 16:04:33.000000000","tz":60},"subject":"Use opendevmirror/ubuntu:24.04 image for molecule","message":"Use opendevmirror/ubuntu:24.04 image for molecule\n\nMolecule jobs were failing on centos:stream10 when CPU did not support\nx86-64-v3.\n\nChange-Id: I953ab956250d6dcc6e96212057db01f438c69d78\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1444d7947b154a251167cdb7aa54ef0b253aa079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1444d7947b154a251167cdb7aa54ef0b253aa079"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
