)]}'
{"id":"openstack%2Fkayobe~971528","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I953ab956250d6dcc6e96212057db01f438c69d78","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-12-23 12:34:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-12-23 13:12:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-12-23 19:25:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I953ab956250d6dcc6e96212057db01f438c69d78","subject":"Use opendevmirror/ubuntu:24.04 image for molecule","status":"MERGED","created":"2025-12-19 12:08:29.000000000","updated":"2025-12-23 19:26:01.000000000","submitted":"2025-12-23 19:25:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971528","meta_rev_id":"b56dbed94e1597f21b278a88f82d1be32e3e57b8","_number":971528,"virtual_id_number":971528,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-23 19:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2025-12-23 12:34:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2025-12-23 13:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-12-19 13:54:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2025-12-23 12:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2025-12-23 13:12:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-19 13:54:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-12-19 14:10:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-22 11:21:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-12-22 11:21:16.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"b90b5a45ab8a5a31eafcb998b0e9009d9e398037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-19 12:08:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c1fcdb1e69cafd70a67d63a1dbb3ce173a98ff4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-12-19 13:54:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bde8b8a0aaf4b97ad6d44303c2e1800c376d9da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-19 14:10:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6908f9fb3e8e4397b55b5c2ed87bf585\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df60f84f0154aae80f6f83691224741 : SUCCESS in 9m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0e3d42b10d5745828244a70d878c8b6c : SUCCESS in 7m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38239f38eac04db4b7e3ca09d4e020f6 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56ad420be46740f68dae97ed8b566d05 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95c092f3f38143d3a449a2c7afb88e87 : SUCCESS in 5m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcfac95c5e5644ec9b17bcc5fce88ad8 : SUCCESS in 4m 10s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cba869df363c42eea42c3c5b29d1456a : SUCCESS in 3m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dacf1e620d0c472f8f1a70e4f4258727 : SUCCESS in 5m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bbf3497a02d94b439cd39ce35ea090bc : SUCCESS in 12m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/65b4d25c803241e2a0746bf54b20cf18 : SUCCESS in 8m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d27f648026db464295b65cec51bb2187 : SUCCESS in 6m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/57e0a7edf2dc4e0da873020d435b0ed5 : SUCCESS in 11m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/738ac78b8a6f43e49b2208a49c8f498c : SUCCESS in 18m 39s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fd4011f4562f4e2b81d0809852f3df24 : SUCCESS in 1h 15m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f8fd51eac0db431882cf3ccaf742606f : SUCCESS in 1h 20m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/da0f562fbc114780836fec34cbd6aff6 : SUCCESS in 1h 21m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cad871661dc4b70a96b875145266bc5 : SUCCESS in 1h 16m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f0e43dc75c36443fa2a59d8fdb04bf7b : SUCCESS in 1h 20m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/106265b82fb34369b6a2d127e1b8d4ec : SUCCESS in 1h 57m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f77de4838070427eaa47f31551b28c80 : SUCCESS in 21m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/98ac6cb62f71428da5d2b16ed96c2078 : SUCCESS in 9m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4d23298eae04e419b1177c77b94939d : SUCCESS in 19m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/14dff60873d3416fab0db5143bc96f0d : SUCCESS in 7m 56s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65d22efe5eee4348bc9bd4db878cccae : SUCCESS in 32m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fe2599002f1b4846aeb2ed7ad474854f : SUCCESS in 8m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eebc2acf2012489ebb108addfc95cb9b : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"eeead03536d261ccb2fc50af2a3f51a6296df6b0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-23 12:34:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1875400d67492f5c905731bdaa89b3cd1b5cb194","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 12:34:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc8ab02a47383d1fa5510d4ea2daf38cd8025e4e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-23 13:12:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a851a0be14822ddfc2a6eda057bed1f1c90a56ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 14:19:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee95b948084744b9957d3177a731a153\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3668712820e94efc9c4988ad863a9091 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9bedf2de3684d7a8dc7114d93e15c93 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1dc916ede724a2a9cec36f8c1bd5f16 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4560e4db1db4412896a1859b2f490088 : SUCCESS in 2m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/02e63570829340a1a88ae24a437e77e9 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f98823aff5b945bcac97f0007cbf4bb9 : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08c48b21d3a441c9b3f41804ac903231 : SUCCESS in 8m 54s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9ec418a3c6cb4b6fbfcb46131f545785 : SUCCESS in 17m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b7f455f43404588a9fec6acaefbd3a3 : SUCCESS in 7m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c322f7aa0d5f4087a63231b37c6e2802 : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75c84ab43bb247019825092e3e4c2738 : SUCCESS in 15m 53s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/31012d9ed7c44c2b90e848180485b716 : SUCCESS in 1h 16m 01s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/2706e287b2f64171894125b39cc3fbe7 : FAILURE in 5m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f7f55ea7e56a4379bb5554fc1b6ead0a : SUCCESS in 38m 30s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e546fb44f7648468c629acb9fcfdbc9 : SUCCESS in 58m 20s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8edfeb301521411a85a3eff4be2ca572 : SUCCESS in 1h 12m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b77820c53ed242239e2209b913d5a294 : SUCCESS in 1h 40m 58s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/eca2ca8dbabb434eb874ce3eb8ecda19 : SUCCESS in 14m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7fb6d027f4c946ae8e9c33c77cfde1db : SUCCESS in 15m 09s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/65fbda5149d94d568b19d47328c3a463 : SUCCESS in 8m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/96cb6e4d09004ff1937711548d748f7e : SUCCESS in 14m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fc03b2e6f4e4203963ffd839629c51a : SUCCESS in 26m 28s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d8b2f535286745b096cb2018ddbdc6d4 : SUCCESS in 14m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/642710b3331840d7b8eed794d7e17112 : SUCCESS in 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"99b3f7837cad898139c2f73dbfedd33198d5141c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-23 15:49:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8eacf351627d619d7379a0db26fa753f7489b2d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 17:36:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/727a15e8039a4042a982d6c19f3000ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f6f3d8395c64c9bb2f62cc268a22ee6 : SUCCESS in 4m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/862cb123396442ddaf6dabfa598342e2 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/154acab4b7a14ea78789ae9a00be713a : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a39065910d4846f0a8686fefe841a62c : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/145af23789d24176979338309b489d00 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd3f29a60cee4ad28cbde690374b0ed2 : SUCCESS in 5m 20s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/143027127cc843c9ad992af6f3c6881d : SUCCESS in 2m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/208f04fa0779435887a01120aaa7065a : SUCCESS in 5m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9fa334198370459cb35f32068e83bb8a : SUCCESS in 5m 42s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1b840b9792e64a168844da76e1744384 : SUCCESS in 12m 40s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97ded7ba028a4b63803569c069b5d1dc : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e7f549b235e147398ab0128fe69be344 : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/982fc66dc4c049c5952189602b41d3d4 : SUCCESS in 9m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d4f111d9bbe64f7c9a24eac4fe5abc34 : SUCCESS in 1h 15m 56s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/211dc46d1cdd4d0493ada644cad36c8a : SUCCESS in 1h 15m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/89757073618140a9a48ff657d19d96d1 : SUCCESS in 41m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5da8a70918a9459ca9ace34dba136611 : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/32b7a7c4519c49ce90b9e17217a41ff6 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6a017920a4e405b84299ca5ef48e74f : SUCCESS in 1h 42m 50s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1d27d617830449cd8a63482543ccf478 : SUCCESS in 9m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0e66e482ad6e42b6a5f9fb46e37493d6 : SUCCESS in 15m 59s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60852f9603ba49ffaba2717a6d492faa : SUCCESS in 11m 51s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b0433e72e3a14d37ac3d0f068ca51730 : SUCCESS in 9m 21s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e97fd8f8176a4dab94b9a9464ca0dc50 : SUCCESS in 27m 54s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/76a86a36a59940c0a698600aa6f64ed4 : SUCCESS in 9m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c21f48a758a64043b03bfb66bb0eeff2 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0ccbd61a8725aa31dc06d2c9c8a1656a2d2aa8ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 17:37:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b3a319057ba4bb2c4ca19501633bea2815325d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 19:25:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20de61bb461d4aadab345828230226ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c142baf87d764ebda8444f9ef79ea9ef : SUCCESS in 8m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c8abc3089ab46a6aef2d940d23c321d : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c722f03260c49e1bcd8dfc398f886ad : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/937de286a12e4c6f94b35eefaa33d036 : SUCCESS in 2m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0f5f17cb66e4e4da056b0ff89404c79 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b617b9e4bb04ac1b170e34110658a39 : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bacc4bcefce74918b5b4c70b4ee1652f : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ea9d5ad19fa542b386426b577e768103 : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a181c84632944276b0ae872e406c2c7c : SUCCESS in 12m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/126672ae272448d68abf29ef9a77a25d : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6c155a88ac44ff2bbc7fb713cc88950 : SUCCESS in 15m 44s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/431c82056728444797cf37fde167303e : SUCCESS in 1h 06m 33s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fb264fdac6ad42ea95b9670ba368b747 : SUCCESS in 1h 09m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b33169f0483e43549cffe3e2008badcc : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b43d4ceb1b2431caa2b5ef6cd605c3f : SUCCESS in 1h 16m 40s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/03da8b530f47420a93fa85295bba1771 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4db7a46186fb41f0af9a212a57b8c1ff : SUCCESS in 1h 43m 32s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1bd76c7cf1704813bbc7eef2e87de2cb : SUCCESS in 15m 35s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa73ebccf95c4c208ef438a840309f61 : SUCCESS in 14m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dabd4b912d6b46b28d911689e48127f4 : SUCCESS in 15m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0d1bf0b6dd8e4b3b8500cd4a6fa15099 : SUCCESS in 15m 59s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cd92aa4148e4c49878b69811a60ad29 : SUCCESS in 24m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1f2a56260c474f9197f5e15b5dbcfdd5 : SUCCESS in 14m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa77de62d855467388a7c63bd0c53951 : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"08635d5e6cde648efa2c316fad03310aff43f231","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 19:25:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b56dbed94e1597f21b278a88f82d1be32e3e57b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-23 19:26:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19aac45fbd8e49c1923ea1e169fb4d40\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86931ae1fb184565913d822c9eda037a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca8c14f25d0c21e5928622afeed961757392bb29","revisions":{"ca8c14f25d0c21e5928622afeed961757392bb29":{"kind":"REWORK","_number":1,"created":"2025-12-19 12:08:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/28/971528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/971528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/971528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/971528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/971528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/971528/1"}}},"commit":{"parents":[{"commit":"ebd1638d3be9d76e1ec791f9cc7b82ec665f0ba4","subject":"CI: Ensure firewalld is unmasked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd1638d3be9d76e1ec791f9cc7b82ec665f0ba4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-18 16:04:33.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-19 12:07:50.000000000","tz":60},"subject":"Use opendevmirror/ubuntu:24.04 image for molecule","message":"Use opendevmirror/ubuntu:24.04 image for molecule\n\nMolecule jobs were failing on centos:stream10 when CPU did not support\nx86-64-v3.\n\nChange-Id: I953ab956250d6dcc6e96212057db01f438c69d78\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 1444d7947b154a251167cdb7aa54ef0b253aa079)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca8c14f25d0c21e5928622afeed961757392bb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca8c14f25d0c21e5928622afeed961757392bb29"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
