)]}'
{"id":"openstack%2Fkayobe~971883","triplet_id":"openstack%2Fkayobe~master~I6dd6808f8ad5325314435650db47dc71e126d6d5","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-22 13:51:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6dd6808f8ad5325314435650db47dc71e126d6d5","subject":"CI: Always use openstack-ci-mirrors element","status":"MERGED","created":"2025-12-29 09:27:29.000000000","updated":"2026-01-22 13:52:51.000000000","submitted":"2026-01-22 13:51:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971883","meta_rev_id":"3622514a3857e023fcfca88f32cb60c44262d126","_number":971883,"virtual_id_number":971883,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 13:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-01-22 12:02:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-22 11:26:39.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2026-01-22 11:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2026-01-22 11:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-29 10:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-29 11:24:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 11:26:39.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-01-22 11:35:13.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-22 12:02:37.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"7b3bbd85769e8e7d08154bdfacb6addac258ca0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-29 09:27:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a00b718bdf4e605e68babd7bf985337e6759bf75","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-12-29 09:28:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9175c46a0dc4b589cda107b70d1ad74e520a2546","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 10:30:15.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/250a204f92e14cb3a66dcebf2d5b37df\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/66e47200f95449fd80e4b339f1e13c61 : SUCCESS in 11m 32s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/de36d6ca5268406f8943fca8f266c52f : SUCCESS in 14m 36s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/2886ca30a81a406298fa5cc257f73a36 : SUCCESS in 18m 43s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/10bc2d7081a04e8bbfc94e323bc2b70a : SUCCESS in 17m 44s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/45fc3103ec1842a2bb394fda8c86c6df : SUCCESS in 36m 10s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/b081154fd6134e1bbcfa6f2b461e812e : FAILURE in 5m 45s (non-voting)\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4344cb8b61a843c09b3d396879f92de3 : SUCCESS in 1h 00m 54s\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/d7c187baecd846bb95b1a7df97cf5e4a : SUCCESS in 15m 08s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/e4c3c93f3bde4a00a4d61dff7c4d8f93 : FAILURE in 9m 24s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/fb6e6b82653540029d50e14d219c6a03 : FAILURE in 9m 19s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/661bf3dbb73547708dcdac6eae2b9b82 : FAILURE in 14m 40s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50d83082c1174a1f888420ab35efde96 : FAILURE in 44m 21s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1457885e476f47d598d98670cf458e1c : FAILURE in 13m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/38e309d743664fd398165ce69b9265f4 : SUCCESS in 15m 30s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/65272ac9864b4a2aabe39ebb485ec9a7 : SUCCESS in 11m 53s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/93d4562651da4029ba58b508835a3508 : SUCCESS in 17m 56s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/00bc96887096486c9e6883b8c8eae35c : SUCCESS in 12m 06s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/e460debb0d3a426e8928569436e45bdd : SUCCESS in 16m 54s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/e374020a7185469e966d405386635d8d : SUCCESS in 8m 35s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/36dd43909ca141a2965b30b7a7eda285 : SUCCESS in 19m 14s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/a2e6b48f3c264a9994d969692d97ec15 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8f265ced0080b0155b328724964b7f116e2cb796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-29 11:24:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ca412a879c34236b3fb3c4a198626d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7308352dcb3f464b8bb71d76591dca74 : SUCCESS in 9m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d84fcf5876224c50b5e769a20833779a : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebc73b7b597d410594a3601320659b26 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/720a8ae216c848caaa14db0b57d93a0b : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b190c87f341a4a3d91200186d098079b : SUCCESS in 4m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e34f220ef2f40c5952d024b0ec970d3 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c997116c99e4ab28a3bdd4df5d70118 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f09480b00ac842e783db43c35c3baac2 : SUCCESS in 6m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/52e50db0e8e649a89888d7e6461f2589 : SUCCESS in 12m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4ef7a99324a457b9900b2c6745a0839 : SUCCESS in 9m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5bc4bc0d79e74777ad7ca9168e8ae9a6 : SUCCESS in 11m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fcf51212ae24e1e905f3635852c0d64 : SUCCESS in 19m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9270ce7a9cb640ed946c9a3a9647e803 : SUCCESS in 51m 54s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/547cbd23a9ce4f13afd4152622f55664 : SUCCESS in 40m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/abb25aa89a5941dcbe73a4d7f17833c0 : FAILURE in 6m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/912d1ee0dd4b4d18b213c33ae51e1ea1 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0630bfec811e4cc1b01c1c6724af75bf : SUCCESS in 1h 12m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b44393700d4e476aa7c3c1da3480f35e : SUCCESS in 1h 51m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/37dd7c7fabe44fcbb574d7ce742ab316 : SUCCESS in 10m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/904a56680c974c9c9b8ccde1b4c3c58e : SUCCESS in 20m 44s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/831e8ccc7cad47b78a04650bb085c42f : SUCCESS in 17m 24s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/867027f9c3af4c3d835fc40c196c0e22 : SUCCESS in 7m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cac1ec830b70426baf8166ce9afff15d : SUCCESS in 14m 34s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aa34f76453024255b2fedf5f912cddd1 : SUCCESS in 17m 59s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/342b6f1ad9124b199945e131f1020dc2 : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1c873ae28dc6c14efdffe3e6ae7aadaff3670c5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-07 18:28:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba1de6c8d22e6f096962128a389f4b40d42cd9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 20:44:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a73352537f94b2888ca6b66e5cb6537\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab64d5316e1d41f4869e9bf139729ffd : SUCCESS in 7m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/34d55c56a5344f5885eacb5a2eabec1a : SUCCESS in 6m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b3f094f51748a284cf44ddc6d11dee : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c516f595eb4630bda5e9de50d165d3 : SUCCESS in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cf33ffcb9704ca19f50e0b31cc87f3e : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb76b346f71f4e2ab73ebf94d1aee9bf : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44a6bf5c0a5f42db96dc0a7e2d975100 : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b57390dc033e4717993e3b65d026bcfe : SUCCESS in 9m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5660ac0edf324fbc87e6d876604c73d8 : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b40b2be821ea4226900e5e2406455e46 : SUCCESS in 14m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/79f386bab99f4144a3064df7332366ec : SUCCESS in 19m 57s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eb8fe45ae0840e7ae3027f6622a50bc : SUCCESS in 14m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/625b06715c414b35a78e465964ecebe2 : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/116663699569447f8f0700b26b1a6bab : SUCCESS in 59m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/edcdc384f93946d89d8e2d792801d118 : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa4a6534a07f40d187e5a3e91ed2fdd9 : SUCCESS in 56m 50s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c104726f83a04821a2c36926c32b3848 : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b3e9ee8ef3d7443ea51fa13481422b15 : SUCCESS in 1h 36m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20ae0704224049cfb7949552e0b235ed : SUCCESS in 2h 00m 51s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1b4c245e27ff4af69faac4dfc281fc21 : SUCCESS in 21m 15s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/552be4e154074eaa9594b9a6a7618250 : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5e81afa068ea42eab1048e183ccc084b : SUCCESS in 9m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/afaae3436aa84557aff1e9e549d706b9 : SUCCESS in 12m 51s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/10a462e0ddeb47e785300c2454f32148 : SUCCESS in 28m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ddcf598670644afa5b72f5d8745d88f : SUCCESS in 25m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e161dbd68c4846ca9e1197344f6acbef : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84ad0f55e207451f81313ab5f453dbd4 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e474071d1798dce3171b96bd3cd743899bf67d69","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2026-01-22 11:26:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2c1da62092b9e685c4d379d88e81edb58c26853","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-22 11:35:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a8e96bcb51661e24815dc29616a718bb5de23185","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:35:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec5310ca0fde16b9e22c85057740fb36097bfb12","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-22 12:02:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fbc0ca4ff31c6160fa92a5fc0ed06f9843d0b2d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:51:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07cdd6413694ba3a9a71225c4a6c7bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/833bd6b13c33459f8991fae0d63225f9 : SUCCESS in 9m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/082324827d5a4a0e9a5e1dc4485f9cb7 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75ccf5200464da5887d4435fb6f7fab : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6af2ab127f2b41e5bdf659adc22b5475 : SUCCESS in 4m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46833c6ae77a4c12a791cdbc1fee65ee : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1f67ab21d194fdbabff0b9f7d6b79b1 : SUCCESS in 12m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14a77d0c62f146258e02bc958ab28641 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c83cc21d6bf244dfbd20ad2100968636 : SUCCESS in 10m 12s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c5b10c35d7d4c62a0d439e0939d6270 : SUCCESS in 13m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/418992fcca284c56bfccd072f0b0d340 : SUCCESS in 24m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c806f84a94124f548bbdbc3e60981524 : SUCCESS in 15m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b198a4df114248c4a4c71ae9e8d37598 : SUCCESS in 33m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/57ae5206a53a474695cefee1128c4781 : SUCCESS in 1h 00m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b6ec2f7558d34fa9bf4a8be6f7324a4b : SUCCESS in 37m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7a933e958cb4b33b7c350ad8348ef59 : SUCCESS in 37m 31s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b2a348bda3a341d88afda8e60dbd2e52 : SUCCESS in 58m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/65adf8a3f3c94aa1adb95becedcc802e : SUCCESS in 1h 46m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73b4af13dcfa4c0ca18c09026a63e99f : SUCCESS in 2h 02m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3727483d5a12437e8690558a2c5a48cc : SUCCESS in 18m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/18c7e2b399aa4b628bbf18041f6c511e : SUCCESS in 17m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/978327f892294827844889d8bb14d9e6 : SUCCESS in 12m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0e4a004e543b4c06af5f4e237d5c1d89 : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8617bba2be3745fb8f5825aeece97635 : SUCCESS in 31m 30s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebb674617e1c4aa1b2e582238ccef3d1 : SUCCESS in 33m 22s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0aab4f36c4b24e82b0616073d901ca8e : SUCCESS in 19m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4299bb4846540aaa6ba00e1d20f0bba : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":2},{"id":"565c01e06aa953993f3ebea432ceb5abe32c7277","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:51:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3622514a3857e023fcfca88f32cb60c44262d126","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 13:52:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd2f70f71f74e7185b49b7ae4524c72\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a5a38662bd439fa462bd7cac896436 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d62a9c9dd9813f22792b07cc9e162e5adc618da","revisions":{"815ab851e04699d8c74c2a90a739fcf0b29950b1":{"kind":"REWORK","_number":1,"created":"2025-12-29 09:27:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/971883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/971883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/971883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/971883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/971883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/971883/1"}}},"commit":{"parents":[{"commit":"1444d7947b154a251167cdb7aa54ef0b253aa079","subject":"Use opendevmirror/ubuntu:24.04 image for molecule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1444d7947b154a251167cdb7aa54ef0b253aa079"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-29 09:27:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-12-29 09:27:24.000000000","tz":60},"subject":"CI: Always use openstack-ci-mirrors element","message":"CI: Always use openstack-ci-mirrors element\n\nVersion 3.40.1 of diskimage-builder includes the fix to the\nopenstack-ci-mirrors element.\n\nChange-Id: I6dd6808f8ad5325314435650db47dc71e126d6d5\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/815ab851e04699d8c74c2a90a739fcf0b29950b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/815ab851e04699d8c74c2a90a739fcf0b29950b1"}]},"branch":"refs/heads/master"},"6d62a9c9dd9813f22792b07cc9e162e5adc618da":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-07 18:28:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/971883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/971883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/971883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/971883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/971883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/971883/2"}}},"commit":{"parents":[{"commit":"afc25e4eae3877e66b6ca6c7a40b281c44d4d46b","subject":"Merge \"CI: Fix overcloud deploy with TLS enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afc25e4eae3877e66b6ca6c7a40b281c44d4d46b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-07 18:28:12.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-07 18:28:12.000000000","tz":60},"subject":"CI: Always use openstack-ci-mirrors element","message":"CI: Always use openstack-ci-mirrors element\n\nVersion 3.40.1 of diskimage-builder includes the fix to the\nopenstack-ci-mirrors element.\n\nChange-Id: I6dd6808f8ad5325314435650db47dc71e126d6d5\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d62a9c9dd9813f22792b07cc9e162e5adc618da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d62a9c9dd9813f22792b07cc9e162e5adc618da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
