)]}'
{"id":"openstack%2Fkayobe~972040","triplet_id":"openstack%2Fkayobe~master~I9c27e7ca3b258cec45f27095ba5006b0dbd1cf92","project":"openstack/kayobe","branch":"master","attention_set":{"36781":{"account":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"last_update":"2026-01-02 16:22:26.000000000","reason":"\u003cGERRIT_ACCOUNT_35264\u003e replied on the change","reason_account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I9c27e7ca3b258cec45f27095ba5006b0dbd1cf92","subject":"Add ability to enable swap memory single source of truth","status":"NEW","created":"2026-01-02 11:41:47.000000000","updated":"2026-01-27 14:59:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"dad7aebad9fef4552d5a220b371ea53dc9bdf767","_number":972040,"virtual_id_number":972040,"owner":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-01-02 12:52:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"all":[{"value":-1,"date":"2026-01-02 16:22:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-02 12:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-02 16:22:26.000000000","updated_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2026-01-27 14:59:04.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"}],"messages":[{"id":"d8f40064bf04d708f9254f859b875d9cf39d8578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"date":"2026-01-02 11:41:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3191a270db8d2ac8169509fa9914e8d90c6aa803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-02 12:52:50.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/234bd7b43a5d492cb7c456a9b841bd5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1aee922ee644beb7f28baf5ec6de64 : SUCCESS in 10m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65a4daab8b594531b0731f5e0b575dfa : FAILURE in 5m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0224ec6ec123494bb4cebb5fba31ffec : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226406b4c4924b08b0ab2668ded7b193 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0dffac6e3ae4768a3f4d1065805fa52 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28d23c0d37484da493fe51a85c1e8d88 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c054a846f414cfdaf6419284b3724bb : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61a2faf7956e4834b9050ffbb8b9e8ea : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee84a8068cef47718a02c658501aeec4 : RETRY_LIMIT in 2m 37s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1af2266d12594edebe37311eb5a4dafe : SUCCESS in 13m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2754c4e64d984903b8dec1da2157e609 : SUCCESS in 12m 20s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8b113de7fc194b7e85e8c4240db143ed : FAILURE in 8m 32s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dba9e72381814ea29a7d093bc3ecdec5 : FAILURE in 7m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c2249310f47b45fb96fd624b7983e949 : FAILURE in 6m 29s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a49937c9de2245849b1d2a3f78dde7c2 : SUCCESS in 37m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/75063db2a73844e295baca674d66c8be : FAILURE in 13m 01s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/188faca9eea34686aa3602846332bd19 : FAILURE in 13m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7370fc5f3fea49f1a6345e71f82f436e : SUCCESS in 1h 06m 34s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dffc7d85481e40f094a9ff4ed31dbda6 : FAILURE in 11m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c2d3fad73ba4f949d1ecbeabc927a69 : FAILURE in 12m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9237e8ba19224b9d81cec8bc10790e78 : FAILURE in 13m 34s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9a18425d31224fea9bc86c49ef7831ce : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efd3ba73b19d4b3ba19f00f72439e1eb : FAILURE in 10m 52s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/517d7ca28c7b41b19ea10fc98d2d4e9e : SUCCESS in 15m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4c0011d45f3b436093693aa75a38fc1c : FAILURE in 13m 43s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b28bf4fe2cd4491e850c1cb51e3d4908 : FAILURE in 5m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a0409488cc9b43bdb5661fb0268ec14c : SUCCESS in 12m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f3e83a9a2364f9b931a10c0366bde02 : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dd34aa903bb15d9a048a701b770a1484d9131543","author":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"date":"2026-01-02 13:31:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13b305b59f48296e1ddb7daaaa8a794cf4de4540","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2026-01-02 16:22:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"dad7aebad9fef4552d5a220b371ea53dc9bdf767","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 14:59:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"569216be992acf75d0ca3db4e174ac386454573a","revisions":{"569216be992acf75d0ca3db4e174ac386454573a":{"kind":"REWORK","_number":1,"created":"2026-01-02 11:41:47.000000000","uploader":{"_account_id":36781,"name":"Massimiliano Favaro-Bedford","display_name":"Max","email":"max@stackhpc.com","username":"MaxBed4d"},"ref":"refs/changes/40/972040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/972040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/972040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/972040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/972040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/972040/1"}}},"commit":{"parents":[{"commit":"16e21b0ad81622ab4e575c4c826fb7dc99fa0908","subject":"Register additional new swap items","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16e21b0ad81622ab4e575c4c826fb7dc99fa0908"}]}],"author":{"name":"Massimiliano Favaro-Bedford","email":"max@stackhpc.com","date":"2026-01-02 10:54:40.000000000","tz":0},"committer":{"name":"Massimiliano Favaro-Bedford","email":"max@stackhpc.com","date":"2026-01-02 11:41:26.000000000","tz":0},"subject":"Add ability to enable swap memory single source of truth","message":"Add ability to enable swap memory single source of truth\n\nWhen applying modified swap config it isn\u0027t necessarily what will be\nreflected on the host. This is because, by default, the swap task can\nonly create and add swap memory files/devices. Meaning that in order\nto delete/remove swap memory, it would have to be removed from the\nconfig as well as manually be removed from each host.\n\nThe addition of a boolean variable ``kayobe_swap_conf_only`` (``false``\nby default), which if set to ``true`` in kayobe\u0027s swap configuration,\nwill treat the config defined as the single source of truth and remove\nany swap memory found on hosts not defined in the config.\n\nChange-Id: I9c27e7ca3b258cec45f27095ba5006b0dbd1cf92\nSigned-off-by: Massimiliano Favaro-Bedford \u003cmax@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/569216be992acf75d0ca3db4e174ac386454573a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/569216be992acf75d0ca3db4e174ac386454573a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
