)]}'
{"id":"openstack%2Fkayobe~972074","triplet_id":"openstack%2Fkayobe~master~I9463cc1ad0855e59f858c355e6e636466d6a490b","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-08 12:58:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9463cc1ad0855e59f858c355e6e636466d6a490b","subject":"Bump up Ansible supported versions to 12.x/13.x","status":"MERGED","created":"2026-01-05 08:54:39.000000000","updated":"2026-01-08 12:59:32.000000000","submitted":"2026-01-08 12:58:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972074","meta_rev_id":"4c99b89fc4024070edd0f8c66dab8eab10b8d32a","_number":972074,"virtual_id_number":972074,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-08 12:58:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-01-06 22:49:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-01-08 09:04:21.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2026-01-08 11:06:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2026-01-08 11:06:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-07 18:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-05 11:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-06 22:49:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-07 13:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-01-07 13:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-01-07 13:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-01-07 13:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-01-07 13:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-01-07 13:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-07 13:42:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"33ef46586235b6681237f367a740d2f0c26ab054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-05 08:54:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb81f4d4f2b01728c378c18d350bf019d1dd18ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 11:00:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4b44e642204b53a31f14425c4e629b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d078e87dbbb4686a494757443576407 : SUCCESS in 8m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6cda3d2f26a549f08306567f3d87343b : SUCCESS in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d807927083a417e80fe57b8c77c3f1b : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf973d370c3140f49877e614860f7172 : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47049d797a4a422486c52ac84631917d : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f1103d728114de298b4d4e55c9ca079 : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b58a889e7c047dcbb4feef7851f21c3 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/29ab608a881d4fe3b4c8e6472755722a : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b045c9437e8f4b4eb3724634464d39ac : SUCCESS in 6m 56s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9356215581a44892bb44b6b88076929f : SUCCESS in 20m 29s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77b4aaf2125846f9819c125ffd3a597e : SUCCESS in 12m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab3ae000ac8742e59495cc638cc97f68 : SUCCESS in 21m 03s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aa81f7c0bb14ad59e1c7bdfc6545822 : SUCCESS in 15m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e7f62788bfa142e0b3846af6b0a4424a : SUCCESS in 1h 13m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c305e3a8b67341be92d006fbe048f761 : SUCCESS in 47m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0d24f7a555194317ab29f095900c1b7e : SUCCESS in 1h 12m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fb81b4fd1c264233a1d861cb9c87da03 : SUCCESS in 1h 08m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9082f07fb33e4e6fb3768de7ce535847 : SUCCESS in 1h 04m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/db978b77dcf74c449cca00816d6e0283 : SUCCESS in 1h 59m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e17bae51e2da4c5db3fa2148a132a110 : SUCCESS in 1h 42m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/83063492f10a4d48a3e17204e70298d5 : SUCCESS in 21m 48s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9bcb1f5dfffa4ee596c176be3c338076 : SUCCESS in 10m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28f726caeb6544eba7786d1d6d74af72 : SUCCESS in 15m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/338cefde641840d9ac9b4710909d3907 : SUCCESS in 15m 17s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/44654f95a9d94d62b1ae76c82326be1d : SUCCESS in 18m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4402767f95c843eba4f797c9c7f93b67 : SUCCESS in 15m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5d56a63e1e2d4db2ade972d70de19166 : SUCCESS in 9m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a9be3c2355149989fc4e65f10296bfd : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"323daec3d54805df02662dea82b1e03bf4881e5d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-06 22:49:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ee77370300d92ded224dae1eefe000558266c13d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-07 18:27:10.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"b185f37b8a5649f5862e16ef092235575bc4ca05","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2026-01-08 09:04:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"201f0d8d02409b108800a51e6d23d91eff14be23","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-08 11:06:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a062317ffadd014a351dcebaa921711a3bbd2271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 11:06:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b1c7a5f0cf6ec5443fcb5d46755a89c96d8241e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 12:58:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4919eeab8fc42b4a5a037e7397b400d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd582b0f9f564918887f66a7e40ff104 : SUCCESS in 8m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ea1385530f94f2e908491a627f4f8f1 : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed8793f7b1c4170b41352c6b4fc53f8 : SUCCESS in 5m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e6ca398405445c9bdd1f28d03ed58f5 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a1e6a72a99543c082a2c0fab83d1f9d : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c601756cc4414bb1a439fd2944ee5e2b : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb2ed6b2cd934d339c23f0bd2be815cc : SUCCESS in 7m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93e75f734b2d493c8f6f0c2ed963ea4a : SUCCESS in 9m 35s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d3727c00c91441ed94239e71e0813346 : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b39f2c231cba4b93a8452908fddc10ac : SUCCESS in 14m 37s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/10a0be041ed346fcb22c23963e377ac1 : SUCCESS in 16m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efd708eff42c46e3bf1782a47af91b59 : SUCCESS in 15m 07s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2f2db412900b40a5ae80d91cea4fd527 : SUCCESS in 35m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/1439f39505b94301a0f33bc6b3c46a58 : SUCCESS in 46m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/830f86a6528440d9a9dd6cf586d04eff : SUCCESS in 1h 20m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d48007dae8d4c15b3ef0fc884e7f97b : SUCCESS in 1h 16m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6fc1f7092b3a43679610aa6e46c5ce11 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3cc8b9b9794d4ddebc3a3f596363957a : SUCCESS in 1h 41m 47s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3b6974c703b43f99f389a16e15f0981 : SUCCESS in 55m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/847087a780164f4085fcfb9436565f11 : SUCCESS in 9m 19s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/418f661033ff4db5ad228993b342f463 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53ef4ea590c34533b832d3836927456e : SUCCESS in 15m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2868b1d07b8f4f71bd98f869300f231c : SUCCESS in 7m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/98061f753bd940188cf0d703258575d1 : SUCCESS in 28m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d4c29dded47488498d7d7ed610477d5 : SUCCESS in 29m 17s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/db72c8eb586a4201a95b97a0c28aa095 : SUCCESS in 19m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85bf369837da4ef28478808e2b9b3c0e : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0231eceb95f5e537b0326d2f7335ff98bd840c43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 12:58:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c99b89fc4024070edd0f8c66dab8eab10b8d32a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 12:59:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d97f8e50cc4f7ba9fbd68a26285507\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78a978785a4497980f89c39148e534c : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6faad69225114427ab782f0161134814 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21213e77b21c7dd126c1434737bd79e5714b6d57","revisions":{"21213e77b21c7dd126c1434737bd79e5714b6d57":{"kind":"REWORK","_number":1,"created":"2026-01-05 08:54:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/74/972074/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/972074/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/972074/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/972074/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/972074/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/972074/1"}}},"commit":{"parents":[{"commit":"afc25e4eae3877e66b6ca6c7a40b281c44d4d46b","subject":"Merge \"CI: Fix overcloud deploy with TLS enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afc25e4eae3877e66b6ca6c7a40b281c44d4d46b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-05 08:53:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-05 08:53:16.000000000","tz":60},"subject":"Bump up Ansible supported versions to 12.x/13.x","message":"Bump up Ansible supported versions to 12.x/13.x\n\nThis change bumps up the maximum supported version of Ansible to 13.x\n(ansible-core 2.20.x) and minimum to 12.x (ansible-core 2.19.x).\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/971606\nChange-Id: I9463cc1ad0855e59f858c355e6e636466d6a490b\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21213e77b21c7dd126c1434737bd79e5714b6d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21213e77b21c7dd126c1434737bd79e5714b6d57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
