)]}'
{"id":"openstack%2Fkayobe~972685","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I3936250ca7b9f851db8e4d73e1e8c28ce823fa4a","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-09 15:44:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3936250ca7b9f851db8e4d73e1e8c28ce823fa4a","subject":"Add infra-vms to ntp group","status":"MERGED","created":"2026-01-08 13:40:20.000000000","updated":"2026-01-09 15:45:54.000000000","submitted":"2026-01-09 15:44:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972685","cherry_pick_of_change":972470,"cherry_pick_of_patch_set":1,"meta_rev_id":"1b925949e4da676933daf52c37579e38d23b2fff","_number":972685,"virtual_id_number":972685,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-09 15:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-01-09 13:11:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-01-08 13:40:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-08 16:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2026-01-09 13:11:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-08 13:40:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-08 15:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-08 16:25:53.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2026-01-08 16:26:01.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-09 13:11:36.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"2e1a7ffaec617d4f3449cf29c2e2b858684e4a9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-08 13:40:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9ee734d959536e7a6f419874bd90e3afb8ac9a75","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-08 13:40:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fcabbf8f5a69695e1935bb1da4a8f3abe88e2a0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 15:40:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07536aedebe24e459ce7df7dfa41fd68\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1a354a2e2344a6e825091883b556154 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806db3f852724190be7da0e6986b3482 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ed2ef04f205464099e7bb5fa88d79bf : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/263751a2886f49bdae24efdabbad814b : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950b319a297f4971b3a602bf6802e57e : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f62c2c61c894f49ae9c0091da5bf3a4 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5ddfb603edaa4210b4fa904520987070 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec195313475649b3a2812af52d96efeb : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db4b6efe3b69459fa483eece782df6e0 : SUCCESS in 8m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/47cf7275ea414f43a6c39883954dcfc1 : SUCCESS in 1h 02m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/523bbc88090440fa88522a6e166db589 : FAILURE in 8m 51s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8056cab100514e4098e604de828b1014 : SUCCESS in 58m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fce8596927fb4e0c99ed17c9186efd10 : SUCCESS in 1h 09m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5426cff125b545f0945f57f59adea305 : SUCCESS in 12m 24s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/94fb10aeb8c647c5aed08fbb9b80a5a6 : FAILURE in 9m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e32b154964d74d0bab4c10aad1760591 : SUCCESS in 10m 30s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab46354c6fe34eac89893fba26c58503 : SUCCESS in 1h 51m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d88f131ea85481886d1d0e0c3c9e0c1 : SUCCESS in 1h 39m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b60e41d2803c41679669689a48a96de5 : SUCCESS in 20m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ead223433ea747fe8e53c11997505e9b : FAILURE in 15m 17s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/39f8a2f8d13a43989c39642020c95696 : SUCCESS in 20m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0fd930fb5dc3481fa9fd8b1b09867047 : SUCCESS in 30m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5bd110859fc94c1f8d530f737798e5b2 : SUCCESS in 18m 36s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93d2db970a154a568d1af903d4fcee1f : SUCCESS in 16m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/08b1df745b7a432f85298b3177721249 : SUCCESS in 19m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3553ea203a441cdb8e10895f7455f3f : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d27f784cce0d02a20f45bcba04a10b241c5dc122","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-08 16:25:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce72612ca4fd73355bb1c7160b6917c8609e5aa6","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-08 16:26:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6115f244b4d1dcd1073af02186d6107cdc0af43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 18:42:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b99d705ce76497fa3ff11b7dd102a15\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/717b424a69344a658bf52ecb39f8e582 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25fe8bcedb454e7091c2e5745a3f66da : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/beec1467db8b445a89f94abda741d7f7 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5372233f3c6245cebe6e8b7d0ad32312 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8acb0f3ae3f44e1bbc3451f97ef4b303 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e61b595f0f3471798fe0efc00eb3b6c : SUCCESS in 2m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f53c0a9a27734d7c8d250ece5df2a2d9 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92c509e5cd784a3fb708a238e58d25df : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9b1181497d34e44aedd7c163b2d8cfa : SUCCESS in 5m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2c57801ad1e049579ed7e2c146101cd4 : SUCCESS in 1h 11m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/63745216a3d742a59e768194ed8e6340 : FAILURE in 9m 06s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f30107db1c843589cd5dde5c39ff287 : SUCCESS in 1h 03m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dbe916e62d9c4b0f8568a42a242d997e : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0c0fe9f870084aababf3f4ec82ea61c0 : SUCCESS in 19m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0aecdaec8def40cca47a9f19455aaf42 : FAILURE in 9m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd601006393d42fc9582467e7d1a0219 : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b642095119694ce386819049d35922cb : SUCCESS in 1h 54m 15s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd5d2b643be04f3d8d0791a5702b7655 : SUCCESS in 1h 39m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7e16691dc8264f9ba5f49683d1a503bd : SUCCESS in 25m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ea7cbb25df04ec5b4db0a724f88deb3 : SUCCESS in 18m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/12b2770f29ca403b8d808cabc6015b28 : SUCCESS in 35m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c4233b1fa2f43e1ac0e36de8595783b : SUCCESS in 28m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a6aa5be41cf4cb790746975de13d3d3 : SUCCESS in 12m 30s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e59dcf30bcf3455da120881705e0ed9a : SUCCESS in 15m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f86f0f466cb54b1f8edbcdef686bd068 : SUCCESS in 14m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/501ef218267e4d61969b696db8a6709d : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1f7aa00981ca2e1ef0b69adc7d4beccc78230a92","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-01-09 13:11:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1208a5b36137d88ca133a1d9e19b52f3b030dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:11:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed0f14db0d1031f2c5899162cbeea08fa6ae6bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:44:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f3057df83b74b6f92dc0256cb9dbf0a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25cab361585a4b0eb854c2ebe2a27ceb : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3044899482b4af5bc2c91a52355463c : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8ac623f70f2453cbf81697daf600e16 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964d6053a63a43faa6e4e9825e3dc7aa : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b0aaef940c94d75a82a3f976a29f10e : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9943bb2098b4480bbd844398b5d52a90 : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a64203526e5045cc8a6625ee555d8630 : SUCCESS in 9m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/946a10656dd94e5f9efabab25cb7451a : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b988df4ac3b4604a32ffc0d32d8c5bc : SUCCESS in 1h 11m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a7bd4443c35b42c6a8b356887c0ab849 : SUCCESS in 1h 13m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3500ed21ea2d4fe8b4e0c75504669f23 : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/414e1c1fb8ad462cb3819edabd4d17bc : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/000b668b1d0c4cf1ad5f74b27b4e6f9e : SUCCESS in 2h 01m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b528bb4292864cb09c3b3a4c2ed9713f : SUCCESS in 1h 37m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/229d7508ad8a4e67aaab85d2129561ac : SUCCESS in 21m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cebca70a5ea846b89d8b5d6cad277cf3 : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/935f004cf5644c7fb2b8c9e0f8433316 : SUCCESS in 37m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be4b0b55b2324bd7a4eb0c2a65e69192 : SUCCESS in 18m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/37957c69c0e7486b92206112c06847b1 : SUCCESS in 18m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16b646d52af3443083d68bef7fc984f1 : SUCCESS in 16m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/699e8ba75c4e402583260cc8dbab4aad : SUCCESS in 12m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f8243ea11684ea283f3aaf2e97f14ec : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"47dc95405916dcabcea4bce675877eb4abbd6880","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:44:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1b925949e4da676933daf52c37579e38d23b2fff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:45:54.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c86af70b7a7f49d9bba966c30291c84b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b993ede994747c894c45aab9f54546c : FAILURE in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/522199108b22493da94d3b3783b2b28d : FAILURE in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"755beae3b1702d5ecdbfdceca468495ea20cfb54","revisions":{"755beae3b1702d5ecdbfdceca468495ea20cfb54":{"kind":"REWORK","_number":1,"created":"2026-01-08 13:40:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/85/972685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/85/972685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/972685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/972685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/85/972685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/85/972685/1"}}},"commit":{"parents":[{"commit":"25b2aa74a088562be885dedf2bcbbfa1bf80d2ad","subject":"Fix Bifrost host variable file generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25b2aa74a088562be885dedf2bcbbfa1bf80d2ad"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-07 12:32:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-08 13:40:20.000000000","tz":0},"subject":"Add infra-vms to ntp group","message":"Add infra-vms to ntp group\n\nCloses-Bug: #2137637\nChange-Id: I3936250ca7b9f851db8e4d73e1e8c28ce823fa4a\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit c69eaf93427412209aab6df97ee2dd872fc26f43)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/755beae3b1702d5ecdbfdceca468495ea20cfb54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/755beae3b1702d5ecdbfdceca468495ea20cfb54"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
