)]}'
{"id":"openstack%2Fkayobe~972843","triplet_id":"openstack%2Fkayobe~master~I3f33b6471313a9a9ebfa679f4a9028547b0dec30","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-05-14 15:21:52.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-05-14 17:10:47.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-05-14 17:10:47.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-05-13 07:56:20.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I3f33b6471313a9a9ebfa679f4a9028547b0dec30","subject":"CI: Add kayobe-ansible-control-host-configure jobs","status":"MERGED","created":"2026-01-09 11:04:01.000000000","updated":"2026-05-14 17:13:52.000000000","submitted":"2026-05-14 17:10:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972843","meta_rev_id":"41ab30abe82f575a60e44a9892cfba9134219dfe","_number":972843,"virtual_id_number":972843,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-14 17:10:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-05-14 15:21:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2026-05-12 16:39:39.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-05-13 07:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-14 15:21:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-05-14 15:21:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"CC":[{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-09 11:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-09 18:22:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-30 13:16:02.000000000","updated_by":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"reviewer":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"state":"CC"},{"updated":"2026-04-13 09:49:01.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-04-28 15:50:51.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"32efbb2b7c028f1177980098aa3c093af2892114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 11:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a982e08fa48938dcf0e9eff9f269f990ad58949a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 11:18:35.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"1744a45e8eb6d057f04360c618275a91a44e0079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 11:20:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c618c2c0b10f663ad0d68330eb620045db758512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 11:47:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4a525ff2941413fb0855a238b15fc50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b84469edec4f7e95f7dd04150b2bdf : SUCCESS in 8m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6272533bfd4a4687851471d63f5dc7b1 : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3109d72c88c4958b88af09046943930 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66a53c2d58824c5eb1190fb19f060267 : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02d4e92e5eaf479987051df885d81336 : SUCCESS in 4m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/667db516972148548d868937b84282a6 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9947055fe8ae4dbf95732a623a4fb6fe : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/caf7f10beda840e7bdf22bc9d42ddb10 : SUCCESS in 11m 20s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/5a51411e97cc45de918b791c08f2ed9b : FAILURE in 5m 45s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ab40e481d9904473829db29926d09da5 : FAILURE in 8m 40s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3901f78774ce4a4fad335413e496af62 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":2},{"id":"2e41f4340f9e215d38bd7d7cd972c8ca0d448432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 12:01:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c1c9a7e0d5ce046001193de06742c3965969bd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 12:20:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2d30200fe79ff99b9bc1dc2dc6f1d3dc41b151e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 12:40:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/614a202d94ef490f9b845db60f381822\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd22568f3fb5440f9e1e906b75a8bb8c : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/90eb2d004a3247b39f9fa053b5ec49e2 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd2076e700ff4530871d7e9d7ef3cac6 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c934269df0e4d21b230475474caeae3 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a506a0f402a40e0a3cffd974f6ad058 : SUCCESS in 7m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e18aaf0103d2498f82cb336add115645 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7cab47927c9942a7a911b588b8cc45cf : SUCCESS in 6m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f08215012334ee4a5ea1d89b9c42ffb : SUCCESS in 5m 56s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/fcf8edaf48c442f28285f0788a37d252 : FAILURE in 8m 28s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1f99a78bf85843c0a7c259895ac8e8ab : FAILURE in 5m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3354d31456714b43858fc8d47aff9e1c : FAILURE in 8m 58s","accounts_in_message":[],"_revision_number":4},{"id":"da1d2f7160c57cd9f20ee2af7c2a8b251fc5a562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 13:17:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0c6903a1a2d50256cd66489145ce869a7f972dc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 13:42:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/413b435ee98545dfa25cd13953d4ac2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4b84d4e392428ab50deae7c556c0b6 : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e98f5eb8a44747c586090d88e65e52b8 : SUCCESS in 7m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/787d91694c204edeac8cd550b3695094 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98816af47064d828eaafd8904d9d85d : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1825861105ce49efb47995bb0bcaa4f2 : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ce237418189448a877187b57187e997 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04460692d6214166a5ce096321700b10 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f8c9ddbfc25462bb878256c885589f1 : SUCCESS in 10m 40s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/0c67bad558b14cb1889e75881c01a42e : FAILURE in 11m 50s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2cf26f6ddae64c0f8c468e1c13fba4b0 : FAILURE in 6m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/380a157193cf475fbcea54789ca7386f : FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":5},{"id":"bc3c398bbc1f1d0e58de1ddbedd4fd7a7d0f1e43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 13:44:40.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"455aba0927be79b2080ca58350003ad17ca05457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 14:29:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c96247cfc975422fab0766b9906e0956\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824969b047e54cfe9e4b36d789c3f20e : SUCCESS in 9m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e6a863d234d8454abc5a3db62cea72a1 : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce104cd278a84cb88c484c130ab53954 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a375cd5a79fc44b29bfebc21f0b0422f : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/950aef4db2c147f5895797a6387d05f4 : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a26ebc32f464e5c9464d824dcfbde74 : SUCCESS in 7m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b06c4359ec341a4930f8095420bc700 : SUCCESS in 5m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd27827527b141ca9a2d14b33d4e5e97 : SUCCESS in 11m 02s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/082db2b282aa46be854b2e531227fb54 : FAILURE in 7m 31s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/65559b8d58644cb4b7a043a31727609f : FAILURE in 7m 14s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87a705c2e5b4421586695d0069c5d2b1 : FAILURE in 11m 11s","accounts_in_message":[],"_revision_number":6},{"id":"86d13a58047393bf58a61a994484870f26b4f1ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 14:35:08.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b7251470796f59013352e97dcfaacdff073bde87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 15:41:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a49996371e74c90a08270b861f89756\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e5b56f090042d38467c627d2ebc153 : SUCCESS in 10m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1256fd85be0c46c2a0a6adccf635c944 : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f62a187297984143891485dba9862aba : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36737ed3ef894abaaf9308e39720f2fa : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9795b38d73554fe79979d0fa0089649d : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/faa78425ecb8402d89a6bd2d173ce298 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ed942fd258a45e1b45ca5335796f1b6 : SUCCESS in 7m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/701afe0dc05f467f82dfd3d9b7f58dac : SUCCESS in 13m 17s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/cb9eb3b8a8484e66ba78c29ea0379131 : FAILURE in 17m 01s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/64cdaaf4725949f484aaf42303cef89c : FAILURE in 9m 08s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ccd387c849349d2ba41136fa87be76f : FAILURE in 15m 44s","accounts_in_message":[],"_revision_number":7},{"id":"827124a2ccf3e43f7371211148c66f353ee43294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-09 16:10:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"5e484217cf95bb26ffb5fe44e68fb975e5adc5db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:44:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4e2e6e23d154dcfa138775381860bb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5835f5a1b31e42d991488a3ff9c0dcbf : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/960dab1bec0e4573ab335c61acd40eec : SUCCESS in 9m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34bd107d9c03464d91c0ba9c3ace038b : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d577f95008648f4a55307e02230d6a9 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ecfb2c608314fc9928a5578a7e79fae : SUCCESS in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad98b1bdebf242099ea84eb44b853e72 : SUCCESS in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81bc1efe5c084f129ed58be91258031c : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/adc30259130f4be8921254d7cd0e7db1 : SUCCESS in 7m 39s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/48dbecd828504e9ba2a89b543e189cf8 : FAILURE in 8m 20s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/693c257512744aa6a5eb40e502ba99cc : FAILURE in 16m 55s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47def4b0b5ce41a3a00ecaa7b055496c : FAILURE in 8m 25s","accounts_in_message":[],"_revision_number":8},{"id":"2042b83e48f2909a336639a19669317e9f5be3df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 10:34:10.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"54fe24f2ce948faa379331ba7fc48125f4a35ba4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 10:52:11.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffd7e8bc1202411d9f1a2127aac5d753\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e736d0b2302e4546a7ade43bac6b3792 : SUCCESS in 4m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4c197b01c9684d0cb6215f493732188f : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60ddd12cb6a84425ad6e70e2167dec55 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79840242555d494b90e2104393b76d91 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23a5597375ae43b68c8b2577a2f2d7c9 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/858f53bf80aa4dc5885438b7d6f88dad : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8b56630b73334289a5fe8985c33dc521 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f2ecf5dc6d04033a134b3998b873b61 : SUCCESS in 6m 18s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/cfc0108c8fba47efa44ea253baa406d4 : FAILURE in 7m 11s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/97fbf830cf1a482ba997950277a62329 : FAILURE in 12m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ac51d20a95c443c48167b52193f83c5c : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":9},{"id":"d984d0bc524cf00eed74a49fc723166d122866d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 17:31:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"244c802f9f9a98999c8ce760b72119abaacd418a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 17:35:12.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"939fbccca000af54aa876f5e8879c606025482e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 18:21:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e978a6324f44422abf3287ec736d8dc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1304e732804d9f954687e661406368 : SUCCESS in 5m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/69c4177c174c457e8b93cc1d7d40d7cf : FAILURE in 7m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b31d29cc3a114b9b96bbdf5f08f56473 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ab84e6892134897abcdb29561c9c76a : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b335bd9c48034d75b37aa08afe9e3f6e : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0d39bc7a2054133b2b8d0f916ef965a : FAILURE in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bebfedac9b214fb5b087dc57b890171b : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/264a754a23874eb98516878f03c46a30 : SUCCESS in 11m 25s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/ec075c48a80545858933745785e6df44 : FAILURE in 10m 10s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3265254ef4ae42fcbbce83b7d99ce15c : FAILURE in 8m 04s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3c215862a9fe465ea5a7e0493c3de025 : FAILURE in 5m 28s","accounts_in_message":[],"_revision_number":11},{"id":"dd9a288cfa441b9f6382f64284a5230525e5c070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 20:33:06.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"4424f1d1306aee8b5b6774512098b79cf7b86c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 20:55:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd504a5002104637893be6300194e345\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c72cf5e5cb4e2699aee0b35f77f5c5 : SUCCESS in 9m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8630c404a53e49aaba4877d29b6e905c : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2afc1fc3c824f3b96142bd6b9f29b8f : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1858854b1540a5b005d27130cd1337 : SUCCESS in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5322b87d2ff3481ba1eb0f8454843e41 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0b7f7ca9d39c47aa9372b2d4c7f874e2 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/85f3e8bdc086443dbd684bd083802da8 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/762cdc8b7e584ad1ae9c31cc5e80717b : SUCCESS in 11m 26s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/eb66f0fc8f13493192a7a8a6b3c0204b : FAILURE in 8m 03s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fcf4854c73c84629b2d1379461856506 : FAILURE in 16m 36s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6f189ebea9d4dc1a342c4275e293758 : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":12},{"id":"b8c397da114a9bd449fcbb62ee1ae8cea0b5f7e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 21:04:46.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"3ca7f534e7e395e7842f72513ad13852b2dff18d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 21:31:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee656945136648d28ea791f2ec0b2282\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2e78c697a144e0f97e0a2d89cd788c1 : SUCCESS in 9m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/de10a36681b447e5b6fd467c23d25337 : SUCCESS in 7m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a6d929e3a5b41fea9f67c6681d02621 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b8cee4704e54fdfaf25f38583bbf3f0 : SUCCESS in 7m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6edb2a43f25c449ebb8ccc57804cbe80 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dfb6a7f4aef3449f89e15c830bea4ab8 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e46d047fd71431fbba71298cb94eeeb : SUCCESS in 15m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/96cd1f2334ab4432ab7810c6f2a42d31 : SUCCESS in 11m 03s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/00a59ef93f7140d3b2321ff6c9a665f7 : FAILURE in 15m 54s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e0645088eccc4c6c827387386c90623e : FAILURE in 16m 05s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00116d73a71a441cb1bd58871724b1be : FAILURE in 7m 13s","accounts_in_message":[],"_revision_number":13},{"id":"bc39f29feba485acfb1a02663e25cfcdff2eab7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-12 22:16:32.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"e765b4c48a6e6f2a23f1c9f388c1f9b7f74b013b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 22:37:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1588da057a2248ec855015d5b4090740\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101b7602c95544faac35b4da018ec737 : SUCCESS in 10m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d703bdf6836044cab1bc29cf7882b67a : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90bbd5a7aa4840878e2f937878ecad1f : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ef999ace5342c9962cab8cada09c92 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c637667f02a04e298a8ca0ebff9dbe2a : SUCCESS in 5m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/757879076a7b4b62b0d4b6643b41d53f : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/538d250a1f4449a599ab599830ccff1b : SUCCESS in 12m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/706a6167f4134caebcbf9dc929cb7d2e : SUCCESS in 10m 55s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/26a36181fefc4e8e986ad9a3f821789b : FAILURE in 14m 57s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3252c78ef9684f00a7ac0340b1e620f3 : FAILURE in 7m 50s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73a2af9e6f744b04919a79ef6372a6e6 : FAILURE in 6m 38s","accounts_in_message":[],"_revision_number":14},{"id":"14f02b26cfce95ef6d77ef78ca7edd2ca79d6b6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-13 08:51:32.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"675796e84c3db378ccd478b5818062637182420e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 09:18:23.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89b6f9c6be594b639899ca39c66de9d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4e26aa43527465aa9cbfd4711fd1e3a : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9d2254b4df1047baaaf06dcd384631dc : SUCCESS in 7m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af1d185fe2344363aa81c4f466aec44e : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2351c0bde74148fe8df45d3449b1d84c : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbf4e1a6c3b04a33ba4be933ec6093a6 : SUCCESS in 8m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a58d24e6bfa4ee5849a269a7210b7c0 : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35543b95ea234e3287ed7e6c2631e59f : SUCCESS in 11m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a10233fbf804466da27b18084f09446f : SUCCESS in 11m 28s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/cf009421a6f54a0a8f3292ddd0acf9bb : FAILURE in 16m 01s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/32e13bf2aa7543a1b18428ea57de2738 : FAILURE in 16m 01s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e343a4d452d042ee89a570960241fe82 : FAILURE in 13m 46s","accounts_in_message":[],"_revision_number":15},{"id":"6a979b850c76b28151c6c012efaac31ca0473ebb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-14 10:45:49.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"d0188b07c1d051f869983ea6ebb02e461119dc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 11:20:10.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da22f5bf6b8545f3904a6eae00268c1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a10c7d845b40b1ad583ab009adafc0 : SUCCESS in 8m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ff14e7d28d3474687d50e4395cfedb9 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75dff6d68f0f4d8a874771cd8038201f : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887d24ce5a4a4eb387db06a1db9d3b8b : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab860afdce084cc79ed6ef423bce09ee : SUCCESS in 5m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f01c74e39f54a6582fc19574043b31a : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3ebf628b6b774f00a33b34a97ffa68c8 : SUCCESS in 12m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3fc4bf5deba438d87a78004c42a1b60 : SUCCESS in 11m 15s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/7d1af1487b704d7c8e9534955bbd5c4f : FAILURE in 12m 01s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dfdbf4f6022c4b9599077b2209ee29e0 : FAILURE in 13m 26s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a1ef333ee2642fb89adba8f2e5d5edb : FAILURE in 9m 24s","accounts_in_message":[],"_revision_number":16},{"id":"1d76d8faed21d183b6568b8ffd34d1aabf41a2af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-14 11:22:44.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"29a72a4c06432ea1991fdd96a55551cbf358c5d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 11:42:35.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efcbd07dd56a462a8e00eb6ccc2472d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920c7e301e4d4800a8d34e96505883f0 : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb1747c57c814e84a55d8adfbeeb46b7 : SUCCESS in 6m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39936e527e214c9aa8827fda2d0a467d : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6c0389e00245d1b57814357708fa1e : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5dd7e172701f462c80acf91002ce69b6 : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5100e65e67224724aeb3018fb3a662e8 : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/925bfdf4e0744fba9447ca04e163f84c : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dac8caa6f3034ad6bbb2ae68cff98667 : SUCCESS in 6m 24s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/05678eb68d894498a52ba028a6572029 : SUCCESS in 9m 28s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/71cccb31e63b4a95a08ea43a8e05174b : SUCCESS in 9m 36s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e62e2d7dee94803a685c8859ba36457 : FAILURE in 5m 30s","accounts_in_message":[],"_revision_number":17},{"id":"72f57372e0b07bb7e7b442e20c88472a69f7702f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-27 22:07:01.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"990a0190fcf648360fa877471884106b583dd92a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 22:29:59.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67cc4b5401864011b56d746cd5681e1e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34ff7ceb7c64bc69f707903631bbb8a : SUCCESS in 8m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30f3084b0cf94d0f8164bf00c69a61a7 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/929dd86ecf1b483c9470a14deed7e337 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f4babcaf892436ea24de4f831f9e94e : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e5e5fd9ba2d45cca349fb021e34ba40 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99b83a4daee1419bb6fa1d778a82c3b9 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/26a9440ef36e4481b0637857792837b7 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8332f5387ed44dc9d7027e428578940 : SUCCESS in 10m 41s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/d899bbfd5c7e435cbfe5dabe1cf94904 : SUCCESS in 10m 31s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6442e06a2a1e49a7921f2c763983d0be : SUCCESS in 14m 48s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7db7fa36428244a0b1381c0910f25d31 : FAILURE in 9m 21s","accounts_in_message":[],"_revision_number":18},{"id":"4c5cd474c25c26854229f319ab5e53d65df2b967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-18 09:53:37.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"bc15d9dfc86b305a67bbe31aeb728910908ac96b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 10:16:29.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f8d45156da4016b96fe88509211fed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33d5db2f0da4292a2350b06749e124b : SUCCESS in 8m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/76edeed1e99845898744e65b140ea4a7 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce6388be99454cfca7f24bf5f7e2e659 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb063fd811804b9d9f5bcf238330e2d0 : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02d2fcd6b7204c0c9d9b0f32b02b138e : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/213f6cca0e684954a99b786c6866fd9a : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5803922faed4430288ea9dddbf2aab35 : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/733476790e324f31b09343d66b8ccac0 : SUCCESS in 11m 11s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/4730166e150a493eaff54a99fadb26a0 : SUCCESS in 16m 24s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d3517b98ef7e4b4db17084ce01e3078c : SUCCESS in 8m 38s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2ad7d8e64554bf2a13ca86b46ef5edd : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":19},{"id":"9138a2ba4dffdf506f8839cc71b133844dcf7b7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-19 11:51:32.000000000","message":"Patch Set 20: Patch Set 19 was rebased on behalf of \u003cGERRIT_ACCOUNT_15197\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"_revision_number":20},{"id":"fc51fd255f994c8c085262932bf0b9c59a7b7ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 12:17:43.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd244645935c4ff9912192359ebb2a23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366a999620e34ed6b628229a893942d1 : SUCCESS in 9m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f6ad5d1b705b45578fee0dfb137fb794 : SUCCESS in 7m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/757cf8c5728f49f88cda2979433a7bf2 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee2664ab0d3b42baa8260f58dc368cf1 : SUCCESS in 2m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe784158a2ca4dc5b0be044a9b293bff : SUCCESS in 4m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7536a0979f2f46c0ab8dbc99f4d4b32f : SUCCESS in 2m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f930d75b1f1145148d943b2ee0cdcdda : SUCCESS in 5m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7a31e7e5feb4243b21488d257442fa4 : SUCCESS in 10m 15s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/0f54aa05de3841c291b20f84403357af : SUCCESS in 19m 32s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/45d18b00555a4e2fbd0b1d8c9b1d5a6c : SUCCESS in 9m 00s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d92f21a810664c248c21e691564db091 : SUCCESS in 17m 21s","accounts_in_message":[],"_revision_number":20},{"id":"f69b44cdc79959c9dab4deca4ff71ab4b4790784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_author":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"date":"2026-02-23 08:14:00.000000000","message":"Patch Set 21: Patch Set 20 was rebased on behalf of \u003cGERRIT_ACCOUNT_15197\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"_revision_number":21},{"id":"c4fa4293aa0191be05c1032e6b3221bc613266a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-23 08:39:26.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac14c9102a334573b340a1e74f22191d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eef7752076b40639c30030bb4189a5d : SUCCESS in 5m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6aa88f9e04d641648f489f756e9e2b03 : SUCCESS in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdd1ea749de44d738dab48b7593f8e08 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b51a65e452f4f99a204878756fdc61c : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb02b0a45b18471a8a9229d0497ff10d : SUCCESS in 2m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dadadbdc673945788321af831d4c3e2c : SUCCESS in 5m 37s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a0aae77823c4cf6b07f8ee69516b935 : SUCCESS in 3m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/adf567eb85324eaaac31cb7a69986cf6 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d126a9e5e563450ab9a49c1846e367dc : SUCCESS in 9m 37s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/2816b6c8cfcd4ec1b96e8675f2d88a85 : SUCCESS in 8m 19s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f4a4829ae87f44c4b3911dab036914f3 : SUCCESS in 8m 47s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b491b48e10434c70a8e346b15be8cda6 : SUCCESS in 16m 30s","accounts_in_message":[],"_revision_number":21},{"id":"a9a9d935272c44364c43fea590f82a374ce344ee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-24 18:39:53.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"8476516c4d83e333ab8714a6b47b7e2174f36cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 19:28:13.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b9f67aeb0cd408f84125806cfd2fb45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee95814e81e247af8003beda04594299 : SUCCESS in 4m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cec0a5848b284aa7863c94b15a3de4e8 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/180baba9ce754f4184a48058a219eefb : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f7d8107f3946d480a8fcea7b1e7367 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87b58df5d378430698530b6a2073d4e3 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10455c5ddef44cbc8f4b4377fd349e5a : SUCCESS in 3m 47s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c5039b486ad4453b2994fd07738b5fb : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5e3d440051e4b5dbce6c1917a332896 : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b01b2112e06b436ca8855fb42df350a4 : SUCCESS in 10m 24s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/4b3908e8c457455f933969af64b0a034 : RETRY_LIMIT in 4m 56s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c2188f4381cb4bc489718f9a1f836ded : RETRY_LIMIT in 4m 06s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17cc66fbbe7e4b9d88e0e7c95ba947b6 : RETRY_LIMIT in 4m 28s","accounts_in_message":[],"_revision_number":21},{"id":"8c42fc26033158ac61d5afefb9ae08b009e7059b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-13 09:49:01.000000000","message":"Patch Set 21: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"885b3e36e0f77a5a64fe898803261a997c949020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:21:06.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2df7b06fae44e12aa55b72f69821bbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb304950f0945c0b88dc6ca51c6f524 : SUCCESS in 9m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/496c88d7289b449a866e15466f683a63 : SUCCESS in 5m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5622e75c03d4bac963b174cc4ba9b35 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0101d6a1137d4e3e81ec788fb7040ed2 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/638cc82d49494a7f91ff6e5853059cfc : SUCCESS in 5m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a2e042fa1964bdc8a7cb4add0d1ed1b : SUCCESS in 7m 06s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fffd99e860a443249bdf4d0e9a0f6cd7 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0424e92cc83406f847cc7585e8b30fb : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12c6bb2f659349bab026484cb93bcbec : SUCCESS in 8m 23s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/9526d771fa9647219081a365e751ef68 : FAILURE in 15m 40s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/23498a7fced84ed2951d74a9d83db7c6 : FAILURE in 10m 40s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82c7da49b5524d91867ff15cc1278e09 : FAILURE in 14m 32s","accounts_in_message":[],"_revision_number":21},{"id":"2bcb0149e609430d5087b3c56c892eca98157902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 12:00:06.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"989721b996afda666892482b97a6db0942171f10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 12:06:31.000000000","message":"Uploaded patch set 23: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":23},{"id":"8ea832702b8b9fe7f065d57b7b3725f88f979ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 12:42:11.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3466c17c542848a29e941c3085004127\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372d43e39ee54a859480b59a425d0863 : SUCCESS in 10m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7dfc3d5dbd6b4dd1809b860764339269 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ad38937107a4a29bdc317dd74df0dda : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/530fb8377a5c492e8ad929bd4d235a4a : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63ee236af80443e79c530de1673ef9bd : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2b3dbadda665485ebc1a0350fa922bb3 : SUCCESS in 6m 57s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7807feab5a35490aa08efd8eaa71ef7a : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/933a554aa1154abcbfa72799c8653a3b : SUCCESS in 6m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9bf23723a432412aaaed261a413e9c77 : SUCCESS in 11m 28s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/0f5d1593feb543108c6f4e68ae84da01 : FAILURE in 29m 44s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8037b978a01141a9857686b6940d4d72 : FAILURE in 28m 10s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f1256bdc5d1e417986a6ae10f2ffe8ba : FAILURE in 15m 52s","accounts_in_message":[],"_revision_number":23},{"id":"2b34a5462ac93d5cd927bbd7b0a199044a42bb34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 14:28:21.000000000","message":"Uploaded patch set 24.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"38accc2a6830b3959fdad5589fbef4905b10a713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-14 14:30:27.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":25},{"id":"424d038662edddd4cd81a6cb9e0e57af6a7414ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 14:55:47.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/44092df032e444c4896266a3e19933ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc37d9d32614cb9ac372adf76341ced : SUCCESS in 5m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/27e60a991a4b4bad93fd4fc01a1d072e : SUCCESS in 7m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/341cfdc095c141868bc01b56c54ffcb9 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7699300ace1432089235437cb4935bf : SUCCESS in 5m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/066794b0814f47a694ba7fb8aab76a84 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85334c71e65f4daa9c77a54047140a50 : SUCCESS in 4m 51s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa2810576200472a91ee6d332b758cfc : SUCCESS in 3m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28c6ec56ebf44606aa1f9b9d447dbafc : SUCCESS in 6m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb9c0908e99e4453ac458a7016e25c6c : SUCCESS in 7m 20s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/9da291aee5fa4c8796aca3309e4abb5a : FAILURE in 11m 12s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/61724f38e83b4c2ca98ae8a7a7b8f188 : FAILURE in 11m 40s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c0782a2c9b42d893ba67e3afda1e5d : FAILURE in 16m 55s","accounts_in_message":[],"_revision_number":25},{"id":"e17d2b7aeb829c3713c2189fe0b6f244356cd673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 09:02:16.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"143da4a939ebdb14cf504fc614a52d94f50276ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 09:25:12.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7d119cedcd54bfc9c790881136f8ae7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69fe9cfd14bb4424b14cd7e7bf8cf1ba : SUCCESS in 8m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2a2b8eccf8bd4340bde004edf78b8538 : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef934be536814e45818897371e298f76 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0fca0576e4407c9b4bd349257e4e98 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c513d8a8d1b4d09b7b758046ed97dd2 : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a0f109c30eb401485208c4f2f8df425 : SUCCESS in 4m 45s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f320214ad83f4985acbd1f20515b0bd6 : SUCCESS in 3m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/486662ba07ba43f19ef3f4b293e5c0c2 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cded305f60d4b88a69994ef80db1384 : SUCCESS in 6m 58s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/6369cc24cd3a4017ba3753e75529795d : FAILURE in 11m 20s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0e72a8351fab448f9c3a627604cb0388 : FAILURE in 12m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8224fdf456e491d83e4b3cbbde28de0 : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":26},{"id":"6394a7b7adc964042cd29e7fcb1ef57d7403f29a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-15 09:59:54.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"8177fda531aee05b9e6c1c209734fe17cea32cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 10:21:05.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5af60e3afa04405da07817ee5f7c4713\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73683c97befe4cacb6044fc27873ded3 : SUCCESS in 9m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dcf06da23a6449258b76eb62f09d86b7 : SUCCESS in 6m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/457e9e727d414e2ebd65303782a164a9 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8212d1890bd545c2b9d1dbfd9eacae17 : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/977d5ef253ec4d69af99cdd3eb2fbd36 : SUCCESS in 3m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/44e5725b13e64c7d9dafd6e7b82014a7 : SUCCESS in 4m 25s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ad222ef2e7924ecc91ae1da5fa86e362 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c3ace8ec8f74420bbbaf69f63a13e16 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c68ddedbe534db2ad3db22bee8e7802 : SUCCESS in 11m 56s\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/78b6d168c12a4358875d63ac9d20b3ec : SUCCESS in 10m 01s (non-voting)\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/36c9ba065cf54776b1c62998a7675b42 : SUCCESS in 11m 52s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5492cab78d24391b3428c239ff4fd17 : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":27},{"id":"5335f76494b18c20c17d5241498cdad6c5ae1ad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-27 17:40:48.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"14503abecf67c06c0f72a0fba2f7f648e8be4b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-27 19:24:10.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbdd54aedf0746e9b2c50ff0177dedab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c92d132d5714809b6e7691efed720fc : SUCCESS in 6m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a7834ed58ad94a5286d2713efec1a97b : FAILURE in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54f8b831e7f340c083cdaa72b5a92084 : FAILURE in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4904a79411fd49c1916144a3fd3377c2 : FAILURE in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3ca9e3e52cc4a8d9d70da522dbe2348 : FAILURE in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd5197c10bd64a3f80b27a9fc06c7ade : FAILURE in 3m 30s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a90d83dba544932a07ad714c6a2a2ca : FAILURE in 2m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3151535c8b9348c7979a3357e4bad752 : FAILURE in 2m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff865b6e0e6d44809bbbf955a80b34b2 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a039f879e76f42b7b1c5ca107ef1f904 : SUCCESS in 14m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02b9684ea94a46bb8bd9bdfc95ec0e56 : SUCCESS in 19m 45s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8c2b6984a2ff4e18aac89ba45821b49a : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1edca48892a349a5b8ea1fcdba42e253 : SUCCESS in 18m 13s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e4ecfa03dcfc4fec9714dcf23ac63e3b : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4c94a8e8a7044928911693530fd8a2f1 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/76a76f5eeebc4d429ca82dc3968ac32a : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad54bf5dc45847d4b1c29cc7ad01b60a : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d2ea2924eb8a49ccbf398f77ab89708c : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b594724f7071472a8cd08498f3de2363 : SUCCESS in 1h 37m 45s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/569bff9a7adc4d938ed6dd56ce5ec33c : SUCCESS in 1h 31m 52s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7278997800d4cce8d130293a5427208 : SUCCESS in 1h 26m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6545e4a163544bc7b05c9ecb9ac53db8 : SUCCESS in 1h 39m 07s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ece42a0a47c54451b66b264b69ac6cd7 : SUCCESS in 20m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/189b3bc6534548f69626b8fb2aaadbc3 : SUCCESS in 19m 38s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d2bcd9818544b8eb47842ac79a320ad : SUCCESS in 15m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7f016d6dce2a4c45a48ae9921cfa734e : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/678938311a154f6fb39623299a5ab6c4 : SUCCESS in 25m 53s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/cedd99deb96c44ca937159527d892e3b : SUCCESS in 18m 35s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/132d41117fe8444b982586998dee9295 : SUCCESS in 30m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4c5d982a132a4cf3bc95b926a8543d18 : SUCCESS in 31m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bc964c2c80ec4b0e87bd55deede17004 : SUCCESS in 14m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8477a1a371574eb1bb3995b7ccc863cc : SUCCESS in 21m 03s","accounts_in_message":[],"_revision_number":28},{"id":"b4282bdae0a9b2788b182f84ebcb09a02ef04e32","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-28 15:01:48.000000000","message":"Patch Set 28: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"c41f05f7e50e4ea12a95ceb842ae424923255a12","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-04-28 15:50:51.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"d87197393541a6627095dacf61cc1751887de0d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-28 16:42:23.000000000","message":"Patch Set 28:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9552845ad05d41838d824e636da73838\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d331e0f7846a4e12b86ffbf5dc2e0d0f : SUCCESS in 9m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c983bd4b510a45be8f099c3e37059f8a : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f087d670aa234b43ac7d627d39781643 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e65c2a41664284bcbc2a3675774722 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f89c081ca3d54ebcb94e563d5a49ba4a : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96dc461c76a843469a5ff59d1b5ea47c : SUCCESS in 8m 11s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/155ddf19f08344f38ce9a5f4bf2102fd : SUCCESS in 3m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a275ea4a0dd448f5b67563c1425df607 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f68905e572d94f64849c08a8b53dc475 : SUCCESS in 13m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/639713d087d84622bdbf495dfd1341f9 : SUCCESS in 14m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d05535e9227d4e6ebb58943e655cf570 : FAILURE in 15m 19s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e689a16096ee4f8cbede0efb174394bf : SUCCESS in 16m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ec82a1395ac74bd58e496875102f146d : SUCCESS in 19m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/efb9f7ec27ab4f94b45cfeb35dc6ae0a : SUCCESS in 1h 04m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f27da6dd20f04a79a5fe256bfa10594b : SUCCESS in 1h 03m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/524296ccb2d347f6a217ec1ac3561ab0 : SUCCESS in 1h 06m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f86199cb062486e90b3a7e0582ec4a1 : SUCCESS in 1h 04m 04s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/387beed63a9e4c569a7f394619434bd5 : SUCCESS in 1h 11m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/68fa4875167f425a8ab7f1745b558353 : SUCCESS in 1h 36m 23s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dd7b2eda529345a69cd58a60da9cc749 : SUCCESS in 1h 38m 17s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b11fefa753f74d1da2ab4bdb24c1390d : SUCCESS in 1h 32m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6bdd2cee33fd45edbf1205fa2e3b3639 : SUCCESS in 1h 35m 57s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d43dc9bea242423496367293f87e60ab : SUCCESS in 15m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ca96d8e1b7b94062ac2954b2f1f2172d : SUCCESS in 16m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8112f08827144f759af9a51f8b40ac8d : SUCCESS in 27m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/985414c8041e466aa638c1796a884dec : SUCCESS in 23m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b6ed71e227a44436b14c591dcdfe999a : SUCCESS in 18m 13s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/83847ebd3e5647789474da506a113b60 : SUCCESS in 18m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4df3a85256224c77bcf8beb4dd816dc8 : SUCCESS in 23m 00s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d3d5efcccd240cebaab3126abfc4173 : SUCCESS in 15m 24s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/aa0c5398d4fc4e6dbc41123321232e98 : SUCCESS in 15m 16s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a44582ac440f4448ac530c0daaa266d9 : FAILURE in 13m 20s","accounts_in_message":[],"_revision_number":28},{"id":"06fe3251607eb74adbf11bd9c36589901026aee1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 08:26:55.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"16bf6e0a376e95ec086e50e073cdee0dcd4bf937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 10:04:49.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/122f30f23b96486f8a6e8cddbe371120\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff857c0396624557bd1d116c40f9ba7f : SUCCESS in 8m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a1de6df1fceb4dfa94941649c3dd0aeb : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa6186df0ea44aaab5b800818d92f22c : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2baf8c55e7ce48a5b5e8a1bd089aac71 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ade52bc92f184537b91368acd1bb7c41 : SUCCESS in 6m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9f4f08a2e3124c0486b8f107258f614f : SUCCESS in 7m 39s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1713567da01041e4bae216bcb4944002 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc160a50c1ac400c946259493fc8d2f0 : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0944de1b8dbe4ae8a9edec6f25c76988 : SUCCESS in 13m 40s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9f1db104d11940a28281cd3585101ad5 : SUCCESS in 14m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d5747b5e696944b5b1b1d802c298c84f : SUCCESS in 20m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/085676cae6a94d0fbf85265491f5d966 : SUCCESS in 20m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b377eb5ea28407ab16396f90927c021 : SUCCESS in 17m 16s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0a57d4d37ea746d090f6ecdb3a8bc270 : SUCCESS in 1h 03m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a8cc1a457a4a4ba0bce6bfb352b9f96a : SUCCESS in 1h 07m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7076cd8806c347fca9d7c03690156556 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/45c781771ab84eef9b7bc676595f442b : SUCCESS in 1h 07m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3aa3bb03a6c94e1d80600554a315a342 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c0eb2febd59a48bc98ca18d52fd7562a : SUCCESS in 1h 35m 42s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/da92ba63286f4ff7bed4e16936830fdc : SUCCESS in 1h 29m 45s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd7ed14352e84702b01e69e53b3274d9 : SUCCESS in 1h 33m 57s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c704687c0aa04c06890861b63242ceb0 : SUCCESS in 1h 36m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/da4141b4908a40c6ad4f7fd8282223f4 : SUCCESS in 23m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/fe6fa474ca404fc981f3809d980cadd2 : SUCCESS in 23m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be52662264c04c8bbadbf47fd3375608 : SUCCESS in 21m 42s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/972895b6a56f4dcaa56184524b44db62 : SUCCESS in 20m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a1e4b8ec92914c5c8ce2cf3f4d97ee08 : SUCCESS in 24m 35s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/861addc1a88f44d8be41402422c24ea0 : SUCCESS in 39m 29s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9ddaf79859547f99c8fc81d440c47d8 : SUCCESS in 32m 20s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d8308fc18734028b73daed56a85fdb4 : SUCCESS in 33m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0f3951714dbc463f98a916f24ff8f0a0 : SUCCESS in 22m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe1044e42c4a453db050b66f8cb0d831 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":28},{"id":"edb5556d06c33dad44625d7813acf5deb8616972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 10:15:15.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"a9884bc551c4724092700c920fefdd1e1cbe1d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 10:18:48.000000000","message":"Uploaded patch set 30.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":30},{"id":"d0f3219fd2d65744b220e66b39dc18538917602f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 11:59:53.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8876530e9fc040b3934709039fac3ab4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0761e3341d524386a9fa91a258b4568a : SUCCESS in 9m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7b253a369ae049b68df06529253a6b6d : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13d1c2b1f01a4cda99fef29b23e4558a : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/819ed6df67e641359156fcd4fc559cfe : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c60e2726e80d4c5ea767ec6ea9c8ebc4 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73315ce27dec4ca396f083603839e950 : SUCCESS in 6m 53s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c35b9e092ed8460892a8880f84847691 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/272f1cabdf06442ba5164f72e32c96b9 : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ae751f016e1245c9a61630341a7f95c1 : SUCCESS in 12m 16s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d236a75e05d5482caafe1166f8277495 : SUCCESS in 14m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/508f2f181d2d42878de0efb960581b53 : FAILURE in 16m 56s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/733c0b525da343a6ad517f257d57a707 : FAILURE in 5m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1599ab88faf74fa7867cf637135581a3 : FAILURE in 5m 51s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/86e01d724a4142c4915695b14cf23502 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7deedd6b50624a9f9b12534847109bb4 : SUCCESS in 1h 06m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d3b108f305e04b55a099e2bcfe221f4c : SUCCESS in 1h 14m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9680ddaa1599482b9c975dc92d09eb7c : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efd789f9fb45474aa831de3a977275db : SUCCESS in 1h 05m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/77b06aa96c5d42978161b310d4397670 : SUCCESS in 1h 40m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2cb40b91e7e34d87a7c98013781e1b2d : SUCCESS in 1h 34m 48s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b87faf29a6864910b2c8980f8771da0a : SUCCESS in 1h 34m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/347734dc2dac46b2ae57b476e45ff0e0 : SUCCESS in 1h 36m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e37d4b385d9f4f84936d1013fd09094d : SUCCESS in 15m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2ac9aae2f0624f58af7f09881c7c13a7 : SUCCESS in 26m 53s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fb5e73a2ba844bc87909618ea3f87f1 : SUCCESS in 23m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0a464c0c5e394933856179457d5ad987 : SUCCESS in 20m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/62f1da4dcc964884a282c490a53182ad : SUCCESS in 15m 31s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/eb341b302a064db5a580485e66904f39 : SUCCESS in 18m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66c22db1b9b74b619e7391298925a9f6 : SUCCESS in 17m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5375456086a14d0ebfd80e16fa97be03 : SUCCESS in 34m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9226f273eb604d0d9ac3b8f001f51f70 : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b3d6945066054d8b8895cf0cc5d7d829 : FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":30},{"id":"538ca901fdb68fd8e88c79682c3a0dcb7da906e7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 13:06:23.000000000","message":"Patch Set 30: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"d6a0ad1d8fb5220082cfed4824b079fa0a0efa95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 14:45:43.000000000","message":"Patch Set 30:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b815e959f3f2407cb895fac512caf86b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726f8f625474442db1f505a75f62604a : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9c4d788700fb4b51be000726757ba5f6 : SUCCESS in 7m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5783a9855b3f4d9486f31f6211fb7474 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57c3ea4abf14a92aa25da622209ffe9 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1571bdac4b054962b7598eebb2f0ac28 : SUCCESS in 2m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dcc792fb30c341eea75ab1e1ee9a7508 : SUCCESS in 7m 02s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/258ea0fa45fc451181f48a2cc0ede157 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61e33cca64e54030bf86d9fd21e4b389 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5e9256332ad41c39dd05edbfa4239b8 : SUCCESS in 11m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0a862c96bd6f4fcc94b448a43a6cc769 : SUCCESS in 20m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c0303b6f351241029046fb5239b0225d : SUCCESS in 16m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/15c55071dcdc42f5b09802fcc3b648e7 : FAILURE in 5m 21s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66400187e579493da4a7f3072811ca73 : FAILURE in 3m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f7074122df4b45e59a23cf5be8e55ef3 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/73b7e224b8ad42b4a4a0da613be015de : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ced99224ddd64bad935b6ca475d2efe6 : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb372aa96b0b4ee082eb64ab2988469f : SUCCESS in 1h 05m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b5e8dce5262948a3a5d0317f6eae1ee1 : SUCCESS in 1h 05m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b7c1912c07094290a99f8ae1a6624ff8 : SUCCESS in 1h 37m 30s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a46c8018bbcc40c887045f9e9700a74a : SUCCESS in 1h 33m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/947aabf369704d849de6099c071997b2 : SUCCESS in 1h 32m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/874039833ed54cf8ad9e9ff000df0fec : SUCCESS in 1h 37m 44s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0fbad41a320b4122916d0c6d55a62907 : SUCCESS in 22m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/10129d450eb04396bfcb22e148dbf34b : SUCCESS in 15m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85c69b4becad4fe0a5d8ec2951393263 : SUCCESS in 23m 26s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/447befd0458c4e99869e635f7b34b64c : SUCCESS in 16m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5c77d2741c0d4cb295ddefaaeb4a569e : FAILURE in 6m 06s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/46cb9f8894604fb0a5fba83a89a46652 : SUCCESS in 22m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b642bbd55a544371a229a1944a102f9e : SUCCESS in 20m 54s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8b5e2144585489aac4defb6e4b97d88 : SUCCESS in 28m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/071b92fec6a849dcb19ef54edc4a7685 : SUCCESS in 23m 39s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a56371447f0e448c92e2ca5952cc04a3 : SUCCESS in 25m 29s","accounts_in_message":[],"_revision_number":30},{"id":"f59a4ba0bbebd7e02fcbb5c827c73db1ba79934e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-29 17:51:52.000000000","message":"Uploaded patch set 31.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"4bed90386f3a4999b224c1f35ee74b1381c87b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-29 19:51:20.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0115fb2b89d84dbf8289cb42a2649284\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3433faa136534c45ad0546b90cd7095e : SUCCESS in 9m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f6a13bf024004faf8a9a1e5f86c21819 : SUCCESS in 6m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d292a6853804bf19f0b651428d645bc : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03a019f831204db0b18c2d2b063497d9 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d299a3028c26416ab1814f7db5cbddbb : SUCCESS in 6m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8612045e904b43e7b72678fa1075f8e5 : SUCCESS in 7m 05s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f41ed1333a0f4ec6bd5fb069e61ef1aa : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7839355d4962486b91bd39d9dcc3900e : SUCCESS in 6m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7d7259527ea49c4b7b5fc23338c5893 : SUCCESS in 7m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/15b3aae9f55c49e6a140a1bbf96eb7ff : SUCCESS in 14m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd065a78b4524fd6a5a1d214b71e8046 : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2e53e3c99c424502b07c260735ff289d : FAILURE in 4m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/560cf88a7b1344ff87c2b59715182847 : FAILURE in 4m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/50f92c8b382f4af18575893043a82df9 : SUCCESS in 1h 04m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3bf668af08c24014879e903fe0bf5929 : SUCCESS in 1h 03m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6b34cd539a0e43d3be7e25f2e989f44b : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f216439d4134315938501ffa89b488f : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/008dc704a6244166b34bfa5fef9f4515 : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9f78834f30dd46eb80cc23c776e57cc8 : SUCCESS in 1h 35m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/bd0624c08b14445caf250d2cae7d901e : SUCCESS in 1h 33m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fad4acd81814aff910dd7039d221a57 : SUCCESS in 1h 31m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cce3ef072b4a414d93988e6cfefb3c86 : SUCCESS in 1h 33m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3207b09511544477a5817fa70e17e03a : SUCCESS in 23m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b96a041f564a4e9a8d27dbfbf45727fb : SUCCESS in 16m 02s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/169e8c29c94d420eaeccd957b1522e54 : SUCCESS in 22m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/130a84265d134b529d366b91625ed36b : SUCCESS in 12m 59s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6d3af6bf95664e258ea45c07f7f9d788 : SUCCESS in 18m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8601b45aa9fe467c898967aa57a77668 : SUCCESS in 28m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40999825e98e4c49b8366ef5c5d9c5c1 : SUCCESS in 34m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4a1e364564444a14b798bce4f01674be : SUCCESS in 29m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/df89e089566d44f2b44ab3c177b75d4a : SUCCESS in 21m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7de3e07bd72d4ff4aaab77aa250c8dfa : FAILURE in 20m 33s","accounts_in_message":[],"_revision_number":31},{"id":"4924ca7a397d626ae75ea8971957110b00eedc51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-04-30 09:23:15.000000000","message":"Uploaded patch set 32.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"8fbe5fbf5318875525b69bb4caf8bbb4d196b371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-30 11:03:22.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2780e2f1148f455695fc0274f8d1bcc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57eff69ae03b41fda632b35c281f8f65 : SUCCESS in 9m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/410e7ce4a0434074a3aedae9378bdd60 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8af30c999b584643953639b6fe87b778 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a5871dd6ec40daae960f50045d2fbb : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dca5afa50c804e349b2cef435268ea34 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1f4855212c54a4e9042f27263c34b75 : SUCCESS in 6m 36s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b941948b8084193b1f3cfca85b2d37e : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c58a528ed7014c26b3e6355331d2ac99 : SUCCESS in 14m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00825823bc8e4c9785b9352b92d85a39 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c44330e21f5a43cd9686d113b661e0ba : SUCCESS in 13m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ea884c6502d43668f418bc8245cedd3 : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/41a2e207e7254856a7f6b882839f912e : SUCCESS in 12m 05s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e090192c0c14dc6a12196e3b19fee3a : SUCCESS in 21m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/88dfac908cc04aaf935d0ddf87b3bda0 : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ab0c3dfcbe254371b1387f1265d90b85 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8abcfecd4de5487a8efa3641227bee98 : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/052a6a733da4485a862317cf72896e43 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6f1a274eaeab49498cc5b49af2948aea : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4ddffae12eef4c92b963575833fb7573 : SUCCESS in 1h 35m 58s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1d890a2350274ceab14d0012f76c0950 : SUCCESS in 1h 30m 00s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cc770fdbe044e94b0d3c8f94270b6f1 : SUCCESS in 1h 28m 27s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/577975b3ee224160b9be5217fcfd1b9c : SUCCESS in 1h 33m 56s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ddc883e5bbb945de9de8917732b805c0 : SUCCESS in 16m 24s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/de1395aeae094b53bda39df49f179b85 : SUCCESS in 15m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fd23de1626424b96a9eea473b5a4e99f : SUCCESS in 22m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0b37e0f7bfbf4698923d70858a3ccbc1 : SUCCESS in 13m 12s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4386b36f02c742909ddcb3719b97e1ad : SUCCESS in 34m 13s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ae378ab3f9854c8ebacf14b7cd849c3f : SUCCESS in 17m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6220989aa7384159a85ffc7ea5efde36 : SUCCESS in 28m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a627a16c51947aa8283cdb72df9a2fc : SUCCESS in 32m 52s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3336e527c7d74549863c5b794c59a56c : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03a4f01990c04fdea8abdf148b863ca0 : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":32},{"id":"94d26e24b742f46df5cecc3614b80f593f437015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 12:33:49.000000000","message":"Uploaded patch set 33.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":33},{"id":"6ed9c18186a473b72ed3eaf88d3c34613bbc2159","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 12:52:15.000000000","message":"Patch Set 33: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"e2aa07e2bcac0360ed7bd3c9d9d0f47782364b67","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 12:56:18.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"07428d51fe49a259397372d9c45724c78f9b1d64","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 13:50:31.000000000","message":"Patch Set 33: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"06804811ca0b0dce9472a08a53b161576a79bb4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 14:11:56.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d06422bbd9e0415e986afffe2b79ed3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8be764248a34eab9628e464a72fd93b : SUCCESS in 9m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a899770d6f245b2a535edfca878ad03 : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/209625cf61ac4fdbbf91097887fa35f6 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ba24fea4b34769ace2586af328c6ef : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6481b29db6d94cdaa2f6165342665457 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/27f223a6dcf146ae919d7099e033bc55 : SUCCESS in 9m 52s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5749cbb2f0804069ba237b9b74f0a37c : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ca94012402549a8a24250d44363f29c : SUCCESS in 6m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c87e18f063d0459bbf5612e06cefe74f : SUCCESS in 6m 13s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d129a0e6fd304255a0f6fb103c904913 : SUCCESS in 10m 54s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46cb66f0af224289a3f3fd5c42cc6cb2 : SUCCESS in 15m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/07cf051b2beb48b1b42ee0440ec230ed : SUCCESS in 16m 14s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4297c604bdc2489798bcb837a4baddf1 : SUCCESS in 13m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/03ba534db4994401a84643f5491891ee : SUCCESS in 25m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4623b5fcf4244c93a16dc609e854f4ef : SUCCESS in 22m 01s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4ccfa9f39d7040f180f94132ae52ef77 : SUCCESS in 1h 00m 25s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0220f0c63c8f4e418a097a07725ce1cf : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/7cf48396bec44a49be1a89f94a1a8ca6 : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9abdd9561c124661a792d95fffd0e471 : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f10f3c4c770347ce8670c917478e5abc : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/dc8da18395f24e93ba6fb8eae151a077 : SUCCESS in 1h 35m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/04e9b0ef60554f999aad01b7acf85db8 : SUCCESS in 1h 28m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c43250318bce4223ba152c2038c4f81e : SUCCESS in 1h 26m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c874d369314e4b6098a95652cae1d3b3 : SUCCESS in 1h 36m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a561025108f7430b9f65ad0fc47bad69 : SUCCESS in 29m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/85d78c46f2104ef4a60d443063156d5a : SUCCESS in 18m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/797613b2b8964609b5270133a845f6b4 : SUCCESS in 29m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d904672d36b141ee8173255fac88dd2e : SUCCESS in 21m 20s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c68a3cdd8ab244f1b1eb9acc80efa6d1 : SUCCESS in 27m 39s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/56d87b74157a48eaa3911e473bbf0714 : SUCCESS in 31m 42s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83383994e3344ef184661fca05a94d4a : SUCCESS in 29m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db294a44915f4127a34e4c4d2ed8c805 : SUCCESS in 31m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/09536c1e962d48ff87d3ee41f100b47e : SUCCESS in 13m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/86b5c673516f4f2a801315c8abc07060 : SUCCESS in 20m 41s","accounts_in_message":[],"_revision_number":33},{"id":"2edfcf2cdf9137375572f75aba4a393d534b0a9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 14:12:31.000000000","message":"Patch Set 33: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":33},{"id":"bd6b2ed3b0e0b5377d0340aeaab37eb30ea68701","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 14:43:49.000000000","message":"Patch Set 33:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a079ac5cc5c4610805f176653b4c06f\n\n- kayobe-ansible-control-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/837bff28f19c4857910931758490ea8c : SUCCESS in 14m 24s (non-voting)\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/323aef4f19364a66ae3e0a05ece97d52 : SUCCESS in 20m 04s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/6d6e44c2652144d293966f37f5207f72 : SUCCESS in 20m 01s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/bca3b867a9a6442daf4d08cc3608abb1 : SUCCESS in 22m 40s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/1a6dff3f52a3450f9831cd857e5d5f25 : POST_FAILURE in 29m 24s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/ff6bac2065fd421cac4d642645f4f514 : SUCCESS in 1h 06m 04s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/9afb0c30fdae4b9284413e6020afb984 : FAILURE in 13m 52s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/705bd2301a354ce99ad5edb32ff1d6c8 : SUCCESS in 1h 05m 02s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/26078dff29cf40caac1e09b00bd4d430 : SUCCESS in 28m 25s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/6d26c7e43b894f729f19e2c8434bd2bf : SUCCESS in 40m 00s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/934330b79a4a4f38bf4f6294b19de9fd : SUCCESS in 52m 30s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/f42f2a8f2d96410ca959482043403929 : SUCCESS in 1h 20m 27s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f77e59d6461484eaff556fa0dd12c96 : SUCCESS in 1h 18m 59s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/98dc41320e664bbeaa6920cc916dc428 : SUCCESS in 1h 14m 34s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/dfd34d8c68c2495392ac3b142a76cfcb : SUCCESS in 15m 32s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/2ac30b8ed55f403387ba2f125dd39483 : SUCCESS in 18m 59s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/0b432567888c42d3913e7bc6d544156b : SUCCESS in 22m 06s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/d91e74dc94a04d41a0fb68e329b108fb : SUCCESS in 25m 02s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/1bc022b399d645fcb823937b60c341b6 : SUCCESS in 24m 35s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/594a22f217ba48a58629f2a0f61f2cbe : SUCCESS in 29m 19s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/e07c1be8e3124d388ff4680632e4be36 : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":33},{"id":"8b953d72f34f279a184ecfe027d448ad4fc16535","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 15:48:59.000000000","message":"Patch Set 33: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"0dfe245db106821d996098ba95ac56226c48f0bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 15:49:03.000000000","message":"Patch Set 33: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":33},{"id":"cb0c8a4a5d14e83df3b02817b27155ee69fa7083","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 15:49:17.000000000","message":"Patch Set 33: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"7a74de1f6a90998487d8d902cd5ab5e191b3cdfa","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-12 15:55:50.000000000","message":"Patch Set 33: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":33},{"id":"09ae6ae48eaddcb050c30a72de337a941d004a10","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 15:58:35.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"711476894762a48299dc87e9ad1be7afef97e9b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 16:39:18.000000000","message":"Uploaded patch set 34.\n\nCopied Votes:\n* Code-Review-2 (copy condition: \"changekind:TRIVIAL_REBASE OR **is:MIN**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-2 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":34},{"id":"7f0797aee94d4273bce2521b47d8af363db6a402","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 16:39:34.000000000","message":"Patch Set 34: -Code-Review","accounts_in_message":[],"_revision_number":34},{"id":"27383450e55e1b11f9e785b2fc53deb054742091","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-05-12 16:39:39.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"2988c1a1a8872d14aeb9f1d1b5a40d8e1da54c17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 18:29:15.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af64167248d2416fa3376e432ae66ea3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ffba490ed64368a878dc62b744aedd : SUCCESS in 4m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a4223b8bd3d47878c93429586708b7c : SUCCESS in 6m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/108b3eebc6954e289f4da1d937850a5e : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a78e3eb27a5472f8ede7cf579ef03a6 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2aaf9b028744416b0f303c800cd9992 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c7218d9ec08466e8e700b744c465a3e : SUCCESS in 6m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3c5cc41ea874e348616d13ab12ef751 : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c53ed26bc9fe4f15bb61f9b6a679b44f : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f93cd97802b45589a5edbab64c20260 : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cad8628c4de74de892527a70f48b9ba5 : SUCCESS in 11m 28s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/aa6b613ec790430c9452769183979885 : SUCCESS in 11m 46s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f6714c3fb49941a2bec9d0a512dbc622 : SUCCESS in 14m 47s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/de43b054bd0a4ce48ccf83e2c0494d5f : SUCCESS in 13m 48s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09fd051df61742e2a78c9d1ee481285a : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a07e6e0c09d340328f5061aec530171f : SUCCESS in 12m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/518685e3dacc45409dd4629b80827170 : SUCCESS in 15m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f5aa73cf6bfc4b8da9af5b8011573178 : SUCCESS in 1h 07m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d225a129f219414a9dbb9b531e51db07 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/51f0ca95e2a24f208796e30ec812a881 : SUCCESS in 1h 10m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fbf07c41b00c4bb1983497b2c574bb54 : SUCCESS in 1h 07m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f9f992e8f2146b39822f0d48d5f84ad : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b94760be14ca4f7f8b31e0bd9bfe1b86 : SUCCESS in 1h 44m 46s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9573d7b785dd4dcf9e536035a44b15e9 : SUCCESS in 1h 31m 32s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55d4211629844b5a923ec9f1e6d798b4 : SUCCESS in 1h 35m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2861a1415e66423b8e1ee5c5bb76e155 : SUCCESS in 1h 47m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/daa5d31278844ed98f787293196c07b7 : SUCCESS in 22m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0309c7bf92eb4441b04a9805d6db2cc7 : SUCCESS in 22m 29s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d3f63351d224463bc34082dba0c1f05 : SUCCESS in 22m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0f8a59bdb704452c8fce06c3088dc453 : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/01d3feca8ea8474ba02a5e3f7b424788 : SUCCESS in 36m 07s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5d6981dd81e649b2a9a16ec4b0ab45fc : SUCCESS in 32m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4ebdf1a9d5342ad85bab057005c94cc : SUCCESS in 38m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ed79b51d58f4e09b337e7f6769b9e62 : SUCCESS in 28m 09s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b1cfb413d4c342c38f176ee6be67a479 : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/03778c116ba549138766fc65308f74cf : SUCCESS in 20m 43s","accounts_in_message":[],"_revision_number":34},{"id":"055adca86172da9d51f41741f132cdb4bd5a8b75","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-05-13 07:56:20.000000000","message":"Patch Set 34: Code-Review+2","accounts_in_message":[],"_revision_number":34},{"id":"15ff8fe0d052c76f3122636d5ec20e6aa4dfbc47","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-05-14 15:21:52.000000000","message":"Patch Set 34: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":34},{"id":"b7659665bb6021407de53b7dd769009b32dddc07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 15:22:30.000000000","message":"Patch Set 34: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":34},{"id":"d8919e15e3b560533fbe8cc73e2a4640f6e75166","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:10:47.000000000","message":"Patch Set 34: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c58a236413343cc8e711dccd0a25fb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b82b1b328092433cb0245a6b4689dedf : SUCCESS in 10m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f876f582955e488f926b5a58acdb8961 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d641bf12b5e4d288655cc9c1ac06502 : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/364ceced736f4d66ba0ceb1e0e38db9b : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99c30a0cb46040eca2e61352e64665cc : SUCCESS in 6m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28e4a8ac7fe24cba980fe48c629b578a : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9a214d103b84797b4f5110dccc15706 : SUCCESS in 6m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/956309a187b7428bb7cbd7182d754e22 : SUCCESS in 11m 48s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/065e0acbf0764778946ee5c086448a41 : SUCCESS in 10m 41s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d06bdce8b7bc4f8bbdd5fc4e0f75bf88 : SUCCESS in 16m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1b076d8192d1486f9946b50fdbaa8a7b : SUCCESS in 13m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5ae212e85dd4b03985720507f3cc281 : SUCCESS in 19m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/18d93f18588a409d8d46bf0fbc81f71c : SUCCESS in 11m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/50d9058d438544a38d200509475e6c39 : SUCCESS in 24m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/0d01f28d71ce4e50806d449afdfa24bb : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ff13662a24774ffea4cda33d40139d79 : SUCCESS in 1h 08m 34s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/af8ea46c58a045a5afd57ac8f8c9fafc : SUCCESS in 1h 13m 27s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8de229b525074c0a898d1c705ee60024 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/06d202af1a314de6bbd8dc375dd6c0bb : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/045e0c8faf614a91ac0a9963b492c9d8 : SUCCESS in 1h 40m 44s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/867b6689da314d2589fe8e8876fe582b : SUCCESS in 1h 40m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a493006885b042e68722d8b2744b72d7 : SUCCESS in 1h 34m 13s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6ad63b197efb43aaa9fce64b57ac4abf : SUCCESS in 1h 47m 26s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/9f0095ef57f34df3ab98dedb8c9c0ee8 : SUCCESS in 27m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/37686489c8ec41f2b9a3fcbcac6e52b3 : SUCCESS in 22m 43s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cbd5d300127c403ca18f157e1f0dccee : SUCCESS in 23m 27s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/66ce53a66b9a4950bd7623125c88afa1 : SUCCESS in 21m 52s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cc530e6af6b94a2191523df689bdfd17 : SUCCESS in 36m 52s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ac3cfa3e96ce423f911c0788c99eb823 : SUCCESS in 30m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1f9840cad3b44a3e82554ccbe56b0e20 : SUCCESS in 33m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30f79ef0b91847159e2d8c5f578852b7 : SUCCESS in 33m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/29821cb3db8d411985572f497361641f : SUCCESS in 16m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c006672eba9f4290bbd51991894b0b55 : SUCCESS in 19m 55s","accounts_in_message":[],"_revision_number":34},{"id":"bbd324c24fdbe51d96f5e965c1c6729fb889f708","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:10:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":34},{"id":"41ab30abe82f575a60e44a9892cfba9134219dfe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:13:52.000000000","message":"Patch Set 34:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6d6ca4859a24f91a8fc2482b7bf806f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1ef20c9d8041c1b158f5901461e874 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":34}],"current_revision_number":34,"current_revision":"ca2adbf8fa258527f54d46ab1ba9277ec89d7787","revisions":{"b204e9306d2d6a0f2244d602d8330d05a663d489":{"kind":"REWORK","_number":1,"created":"2026-01-09 11:04:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/1"}}},"commit":{"parents":[{"commit":"e8134e8e36c93b7964aa1788614b6c43651e2dcc","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8134e8e36c93b7964aa1788614b6c43651e2dcc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:03:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:03:07.000000000","tz":60},"subject":"CI: Add ansible-control-host-configure job","message":"CI: Add ansible-control-host-configure job\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b204e9306d2d6a0f2244d602d8330d05a663d489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b204e9306d2d6a0f2244d602d8330d05a663d489"}]},"branch":"refs/heads/master"},"8a318721b9784d6a52764d0721a8ede263e44936":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-09 11:20:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/2"}}},"commit":{"parents":[{"commit":"8927da563aa4746f45f04b4401a56fcc1323a721","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8927da563aa4746f45f04b4401a56fcc1323a721"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:03:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:20:11.000000000","tz":60},"subject":"CI: Add ansible-control-host-configure job","message":"CI: Add ansible-control-host-configure job\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a318721b9784d6a52764d0721a8ede263e44936"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a318721b9784d6a52764d0721a8ede263e44936"}]},"branch":"refs/heads/master"},"d63782f5bb0b914b27c81d85c295797d1bc8d13a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-09 12:01:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/3"}}},"commit":{"parents":[{"commit":"b0ea93dad436eb17572e26f3564a30f7e8a02679","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:03:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 11:58:25.000000000","tz":60},"subject":"CI: Add ansible-control-host-configure job","message":"CI: Add ansible-control-host-configure job\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d63782f5bb0b914b27c81d85c295797d1bc8d13a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d63782f5bb0b914b27c81d85c295797d1bc8d13a"}]},"branch":"refs/heads/master"},"f9590c51ca045f248cd7ca09c42070ade733fa01":{"kind":"REWORK","_number":4,"created":"2026-01-09 12:20:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/4"}}},"commit":{"parents":[{"commit":"b0ea93dad436eb17572e26f3564a30f7e8a02679","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 12:20:25.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 12:20:25.000000000","tz":60},"subject":"CI: Add ansible-control-host-configure job","message":"CI: Add ansible-control-host-configure job\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9590c51ca045f248cd7ca09c42070ade733fa01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9590c51ca045f248cd7ca09c42070ade733fa01"}]},"branch":"refs/heads/master"},"0a1547f062ce55ac169daf492843c155d320fbb9":{"kind":"REWORK","_number":5,"created":"2026-01-09 13:17:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/5"}}},"commit":{"parents":[{"commit":"b0ea93dad436eb17572e26f3564a30f7e8a02679","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 13:17:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 13:17:30.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a1547f062ce55ac169daf492843c155d320fbb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a1547f062ce55ac169daf492843c155d320fbb9"}]},"branch":"refs/heads/master"},"f6b4a590e3024c811c8c2df9e2cc3a4a9fe7b3e7":{"kind":"REWORK","_number":6,"created":"2026-01-09 13:44:40.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/6"}}},"commit":{"parents":[{"commit":"b0ea93dad436eb17572e26f3564a30f7e8a02679","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 13:44:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 13:44:23.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6b4a590e3024c811c8c2df9e2cc3a4a9fe7b3e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6b4a590e3024c811c8c2df9e2cc3a4a9fe7b3e7"}]},"branch":"refs/heads/master"},"65461f4cc8a6edda014b1f76ada91fc624c2d60a":{"kind":"REWORK","_number":7,"created":"2026-01-09 14:35:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/7"}}},"commit":{"parents":[{"commit":"b0ea93dad436eb17572e26f3564a30f7e8a02679","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 14:34:40.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 14:34:40.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65461f4cc8a6edda014b1f76ada91fc624c2d60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/65461f4cc8a6edda014b1f76ada91fc624c2d60a"}]},"branch":"refs/heads/master"},"da7b8c1006dd95a1597ea809a71f603a2eba5fb4":{"kind":"REWORK","_number":8,"created":"2026-01-09 16:10:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/8"}}},"commit":{"parents":[{"commit":"b0ea93dad436eb17572e26f3564a30f7e8a02679","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 16:10:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-09 16:10:35.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da7b8c1006dd95a1597ea809a71f603a2eba5fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/da7b8c1006dd95a1597ea809a71f603a2eba5fb4"}]},"branch":"refs/heads/master"},"9529813f6f442b11ded459b173aef30d9d99c9fb":{"kind":"REWORK","_number":9,"created":"2026-01-12 10:34:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/9"}}},"commit":{"parents":[{"commit":"b0ea93dad436eb17572e26f3564a30f7e8a02679","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b0ea93dad436eb17572e26f3564a30f7e8a02679"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 10:34:03.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 10:34:03.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9529813f6f442b11ded459b173aef30d9d99c9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9529813f6f442b11ded459b173aef30d9d99c9fb"}]},"branch":"refs/heads/master"},"d4abf724f0e45f6f46c8b7f7c396064fa5f11a02":{"kind":"REWORK","_number":10,"created":"2026-01-12 17:31:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/10"}}},"commit":{"parents":[{"commit":"4883834b444f3056d0613c926d7521b58f7bb91c","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4883834b444f3056d0613c926d7521b58f7bb91c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 17:16:26.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 17:30:40.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4abf724f0e45f6f46c8b7f7c396064fa5f11a02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4abf724f0e45f6f46c8b7f7c396064fa5f11a02"}]},"branch":"refs/heads/master"},"764c76c01e24e5be6d763e38d086863b792a3872":{"kind":"REWORK","_number":11,"created":"2026-01-12 17:35:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/11"}}},"commit":{"parents":[{"commit":"4883834b444f3056d0613c926d7521b58f7bb91c","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4883834b444f3056d0613c926d7521b58f7bb91c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 17:35:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 17:35:06.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/764c76c01e24e5be6d763e38d086863b792a3872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/764c76c01e24e5be6d763e38d086863b792a3872"}]},"branch":"refs/heads/master"},"aad6ccaa2d74974e64a83b6770c5801eeb034128":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-01-12 20:33:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/12"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 17:35:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 20:32:57.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aad6ccaa2d74974e64a83b6770c5801eeb034128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aad6ccaa2d74974e64a83b6770c5801eeb034128"}]},"branch":"refs/heads/master"},"0a792ba354d155c61b65466f8b2b3bf58844acba":{"kind":"REWORK","_number":13,"created":"2026-01-12 21:04:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/13"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 21:04:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 21:04:35.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a792ba354d155c61b65466f8b2b3bf58844acba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a792ba354d155c61b65466f8b2b3bf58844acba"}]},"branch":"refs/heads/master"},"0ff5c20788feacb576fc84a9b6238158e7379ef8":{"kind":"REWORK","_number":14,"created":"2026-01-12 22:16:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/14"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 22:16:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-12 22:16:23.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ff5c20788feacb576fc84a9b6238158e7379ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ff5c20788feacb576fc84a9b6238158e7379ef8"}]},"branch":"refs/heads/master"},"5305dbc7c683281d6ae2b29ccbef38d38ebb4f24":{"kind":"REWORK","_number":15,"created":"2026-01-13 08:51:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/15"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-13 08:51:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-13 08:51:18.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5305dbc7c683281d6ae2b29ccbef38d38ebb4f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5305dbc7c683281d6ae2b29ccbef38d38ebb4f24"}]},"branch":"refs/heads/master"},"427e4ee0981f9b149d8ca77308333ef2ded15af1":{"kind":"REWORK","_number":16,"created":"2026-01-14 10:45:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/16"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-14 10:45:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-14 10:45:35.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/427e4ee0981f9b149d8ca77308333ef2ded15af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/427e4ee0981f9b149d8ca77308333ef2ded15af1"}]},"branch":"refs/heads/master"},"a226ab4551466113fb637fc5a91c24dde0eb57be":{"kind":"REWORK","_number":17,"created":"2026-01-14 11:22:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/17"}}},"commit":{"parents":[{"commit":"ffd5d0279b3a446a96f13f8593f3c11b2e771421","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffd5d0279b3a446a96f13f8593f3c11b2e771421"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-14 11:22:33.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-14 11:22:33.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a226ab4551466113fb637fc5a91c24dde0eb57be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a226ab4551466113fb637fc5a91c24dde0eb57be"}]},"branch":"refs/heads/master"},"b90ab63c449baf71d10f5443f2cb95657d85be94":{"kind":"REWORK","_number":18,"created":"2026-01-27 22:07:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/18"}}},"commit":{"parents":[{"commit":"a2da6d69171ae2b5cbe2f7d2bbae1ed5943342aa","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2da6d69171ae2b5cbe2f7d2bbae1ed5943342aa"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b90ab63c449baf71d10f5443f2cb95657d85be94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b90ab63c449baf71d10f5443f2cb95657d85be94"}]},"branch":"refs/heads/master"},"0aecbe7b9d277905dc62faf511423fdd66cd75c0":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-02-18 09:53:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/19"}}},"commit":{"parents":[{"commit":"adaeb2c5b3acb72d82043139be3c264378d06b7b","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adaeb2c5b3acb72d82043139be3c264378d06b7b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-18 09:53:29.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0aecbe7b9d277905dc62faf511423fdd66cd75c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0aecbe7b9d277905dc62faf511423fdd66cd75c0"}]},"branch":"refs/heads/master"},"b3a0a3523de1a1cc079c93e8e214577ab7733468":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-02-19 11:51:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/20"}}},"commit":{"parents":[{"commit":"11ff1e829ae5665b63c01bd80dfe8fa83b2f6a94","subject":"Support Ansible control host configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11ff1e829ae5665b63c01bd80dfe8fa83b2f6a94"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-19 11:51:32.000000000","tz":0},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3a0a3523de1a1cc079c93e8e214577ab7733468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3a0a3523de1a1cc079c93e8e214577ab7733468"}]},"branch":"refs/heads/master","description":"Rebase"},"f900644866cad6ff387512f87a9539aa32fa40b2":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-02-23 08:14:00.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"real_uploader":{"_account_id":38288,"name":"Leonie Chamberlin-Medd","display_name":"Leonie Chamberlin-Medd","email":"leonie@stackhpc.com","username":"LChams"},"ref":"refs/changes/43/972843/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/21"}}},"commit":{"parents":[{"commit":"e2279353f9e6ff0b18bade8694d1d8b10d91183e","subject":"Merge \"fix (docs): correct passwords view command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2279353f9e6ff0b18bade8694d1d8b10d91183e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-23 08:14:00.000000000","tz":0},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f900644866cad6ff387512f87a9539aa32fa40b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f900644866cad6ff387512f87a9539aa32fa40b2"}]},"branch":"refs/heads/master","description":"Rebase"},"bb21b5bcc137fada703a3fe54df9bbb3048ba893":{"kind":"REWORK","_number":22,"created":"2026-04-14 12:00:06.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/22"}}},"commit":{"parents":[{"commit":"57893c876d42bbbaad62b48abcd423fa0499bedc","subject":"Merge \"Tweaks to nmstate release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57893c876d42bbbaad62b48abcd423fa0499bedc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-14 11:59:57.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb21b5bcc137fada703a3fe54df9bbb3048ba893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb21b5bcc137fada703a3fe54df9bbb3048ba893"}]},"branch":"refs/heads/master"},"9db2284a3ad3c55ecf722543d9e6cb8409cc43b6":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2026-04-14 12:06:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/23"}}},"commit":{"parents":[{"commit":"57893c876d42bbbaad62b48abcd423fa0499bedc","subject":"Merge \"Tweaks to nmstate release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57893c876d42bbbaad62b48abcd423fa0499bedc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-14 12:06:25.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9db2284a3ad3c55ecf722543d9e6cb8409cc43b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9db2284a3ad3c55ecf722543d9e6cb8409cc43b6"}]},"branch":"refs/heads/master"},"b786ba03536209bd7bdd502793169c8cd9f4de5e":{"kind":"REWORK","_number":24,"created":"2026-04-14 14:28:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/24"}}},"commit":{"parents":[{"commit":"57893c876d42bbbaad62b48abcd423fa0499bedc","subject":"Merge \"Tweaks to nmstate release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57893c876d42bbbaad62b48abcd423fa0499bedc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-14 14:26:02.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b786ba03536209bd7bdd502793169c8cd9f4de5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b786ba03536209bd7bdd502793169c8cd9f4de5e"}]},"branch":"refs/heads/master"},"edc6117ce718eda6a9bbeb67cce9469e8948769f":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2026-04-14 14:30:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/25"}}},"commit":{"parents":[{"commit":"a9c4504488578550dc3bcbb0d6d61d81d7e94137","subject":"etc-hosts: add separate task for custom_etc_hosts_entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9c4504488578550dc3bcbb0d6d61d81d7e94137"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-14 14:30:27.000000000","tz":0},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edc6117ce718eda6a9bbeb67cce9469e8948769f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edc6117ce718eda6a9bbeb67cce9469e8948769f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"b786ba03536209bd7bdd502793169c8cd9f4de5e","theirs":"a9c4504488578550dc3bcbb0d6d61d81d7e94137","contains_conflicts":false}},"70822fdb04208b58060f85dfef251a0ea937c257":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2026-04-15 09:02:16.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/26"}}},"commit":{"parents":[{"commit":"a4bec723378983278dfd355464c764004e396969","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4bec723378983278dfd355464c764004e396969"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 09:02:06.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70822fdb04208b58060f85dfef251a0ea937c257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70822fdb04208b58060f85dfef251a0ea937c257"}]},"branch":"refs/heads/master"},"119b3af12d805ded7d83a5ed3526da3be1ab8990":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2026-04-15 09:59:54.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/27"}}},"commit":{"parents":[{"commit":"199610f528197c2551534ab5e5fae50ce58f7758","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/199610f528197c2551534ab5e5fae50ce58f7758"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-15 09:59:54.000000000","tz":0},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/119b3af12d805ded7d83a5ed3526da3be1ab8990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/119b3af12d805ded7d83a5ed3526da3be1ab8990"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"70822fdb04208b58060f85dfef251a0ea937c257","theirs":"199610f528197c2551534ab5e5fae50ce58f7758","contains_conflicts":false}},"5f97aa49d27c7ecb16c36f7876de2a03ce9500ae":{"kind":"REWORK","_number":28,"created":"2026-04-27 17:40:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/28"}}},"commit":{"parents":[{"commit":"c0a5022e22ec50fb7a8a4307fd1ad7b7e6e54505","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0a5022e22ec50fb7a8a4307fd1ad7b7e6e54505"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-27 17:40:36.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f97aa49d27c7ecb16c36f7876de2a03ce9500ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f97aa49d27c7ecb16c36f7876de2a03ce9500ae"}]},"branch":"refs/heads/master"},"f1009d5a4a7777be019abd566762eed540fed12a":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2026-04-29 10:15:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/29"}}},"commit":{"parents":[{"commit":"adde03e6f617b94af12d485c91e9e5e909cfe12f","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adde03e6f617b94af12d485c91e9e5e909cfe12f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:10:51.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1009d5a4a7777be019abd566762eed540fed12a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f1009d5a4a7777be019abd566762eed540fed12a"}]},"branch":"refs/heads/master"},"0ca9fed624b61582f862059bb30dbf2953b6419d":{"kind":"REWORK","_number":30,"created":"2026-04-29 10:18:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/30"}}},"commit":{"parents":[{"commit":"adde03e6f617b94af12d485c91e9e5e909cfe12f","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adde03e6f617b94af12d485c91e9e5e909cfe12f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 10:18:42.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ca9fed624b61582f862059bb30dbf2953b6419d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0ca9fed624b61582f862059bb30dbf2953b6419d"}]},"branch":"refs/heads/master"},"fc27c107959557152ea3010a4f4d8117540bcd50":{"kind":"REWORK","_number":31,"created":"2026-04-29 17:51:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/31"}}},"commit":{"parents":[{"commit":"adde03e6f617b94af12d485c91e9e5e909cfe12f","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adde03e6f617b94af12d485c91e9e5e909cfe12f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-29 17:38:17.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc27c107959557152ea3010a4f4d8117540bcd50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc27c107959557152ea3010a4f4d8117540bcd50"}]},"branch":"refs/heads/master"},"fcc4a860d632e2d580cee86939f7f2a26afa7416":{"kind":"REWORK","_number":32,"created":"2026-04-30 09:23:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/43/972843/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/32"}}},"commit":{"parents":[{"commit":"adde03e6f617b94af12d485c91e9e5e909cfe12f","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adde03e6f617b94af12d485c91e9e5e909cfe12f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-04-30 09:23:06.000000000","tz":60},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcc4a860d632e2d580cee86939f7f2a26afa7416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcc4a860d632e2d580cee86939f7f2a26afa7416"}]},"branch":"refs/heads/master"},"4538884d19d9d7fa7f538e1318e948191e0fa166":{"kind":"REWORK","_number":33,"created":"2026-05-12 12:33:49.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/43/972843/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/33"}}},"commit":{"parents":[{"commit":"adde03e6f617b94af12d485c91e9e5e909cfe12f","subject":"Extend etc-hosts support to all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adde03e6f617b94af12d485c91e9e5e909cfe12f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 22:06:51.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-05-12 12:33:39.000000000","tz":0},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4538884d19d9d7fa7f538e1318e948191e0fa166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4538884d19d9d7fa7f538e1318e948191e0fa166"}]},"branch":"refs/heads/master"},"ca2adbf8fa258527f54d46ab1ba9277ec89d7787":{"kind":"REWORK","_number":34,"created":"2026-05-12 16:39:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/43/972843/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/972843/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/972843/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/972843/34"}}},"commit":{"parents":[{"commit":"a54ccfc3b7bb2b12e747deb84c001a6dcbd585ca","subject":"Merge \"Register overcloud hosts in Bifrost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a54ccfc3b7bb2b12e747deb84c001a6dcbd585ca"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-12 16:37:27.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-05-12 16:37:27.000000000","tz":120},"subject":"CI: Add kayobe-ansible-control-host-configure jobs","message":"CI: Add kayobe-ansible-control-host-configure jobs\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/972842\n\nChange-Id: I3f33b6471313a9a9ebfa679f4a9028547b0dec30\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca2adbf8fa258527f54d46ab1ba9277ec89d7787"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
