)]}'
{"id":"openstack%2Fkayobe~973422","triplet_id":"openstack%2Fkayobe~master~I9ef77d6119be910710eaffa8bc70bb207323f268","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-24 10:48:57.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-27 06:05:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I9ef77d6119be910710eaffa8bc70bb207323f268","subject":"Sync host configuration with Ansible defaults","status":"MERGED","created":"2026-01-14 18:17:49.000000000","updated":"2026-01-27 09:11:42.000000000","submitted":"2026-01-27 09:10:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973422","meta_rev_id":"6e7848ef57fb48c3a2bf4d35de9a9b938b4d2ad5","_number":973422,"virtual_id_number":973422,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 09:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-26 16:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-24 10:48:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-26 09:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-01-26 16:01:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 19:51:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-14 20:12:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-26 09:18:17.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-26 16:01:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"7790ee0c64a720518dfad08b9c713803b4e4f2f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-14 18:17:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6de81635256987476deb476308b29e480f5bbffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 20:12:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b73273f538c420e897aaefbfb5a6d33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5158d8fb16c841bca0cf5ec38192f63c : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/90d9aed1d0614126be955230125c398a : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f67cf1c07f4415c8029d9c852bc9473 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/340231d9b8644c9bbf73046c2c1fde22 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50b0380182c6459796c9ba2e34a33330 : SUCCESS in 7m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/889878dd13b34ad599e730ce5dbd74f4 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9aa871122417411092002dff8feb044e : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fd6ba3a394d45fda87b37345a53ed73 : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/61d4030015404a2880cf02fb2b5bc64b : SUCCESS in 19m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30cf6cf9670e4abf80e1a49db0c79c7d : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/67a2fce637b4428590c5737d2e102550 : SUCCESS in 19m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c21b46f737d34de4b2cb9f6fb9b06b17 : SUCCESS in 16m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/370be23e625a40c7aaf1cb958f047ebb : FAILURE in 1h 18m 27s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ec571c6e96894bb1896347108891c3c3 : FAILURE in 38m 45s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/76d3a1360d504d2c9a1a0fe47d1ae77d : SUCCESS in 1h 21m 35s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b6cba409bacb43a6a188061b2f77e4ba : FAILURE in 40m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/67bcd72d380444709b66eefd78ba9a32 : FAILURE in 1h 04m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b17763f14a444cca97ff8131a3498279 : SUCCESS in 1h 35m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10fd71824b5f47ffbba224cf63969ebd : SUCCESS in 1h 49m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d53dccf8a07542288a6913e09bb869b3 : SUCCESS in 22m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/17434176706544ee861888f886c4c07c : SUCCESS in 9m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0f8cea519e74ad398abc8c86a916225 : SUCCESS in 17m 56s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dc70d2b8c49c42f984e3a30760d9109e : SUCCESS in 15m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f818313b7c4f4a5d8c8d5314960cf187 : SUCCESS in 31m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b9b86e7569c4dc8a18c1c594ee9d766 : SUCCESS in 19m 25s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/716da044940a47ceab92502a86e71f9a : SUCCESS in 8m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a2d2cf3a57b4f1ab7b2d76f87a76c98 : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"776d1224628f2663c53cd36674f9de09af483547","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-15 07:56:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31197f3c2fa45659848ec0589bd99ca6aef6cdaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 09:42:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f96c90945de47e2982d1ca4c5711327\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deba5cb67d3d43619bffd94ec7182c33 : SUCCESS in 6m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f065e6ed34074e86854e7803c5b8156e : SUCCESS in 6m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3a03f58ce8b49399eee96a50c09750e : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/733da0827cb7461ba821e186b969e922 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d1bc1565731432291edc7b558f0eaeb : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2456f3d2dd845bf98b9ca08a766f09a : SUCCESS in 3m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/821933b6863c43f7996c2e9b1d058c96 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/983368f8b69541adade346c71aa33230 : SUCCESS in 6m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0c323c1119554dc29ed234bad9c0a19b : SUCCESS in 10m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0603560664ab4b3cb315e162ca6bdec9 : SUCCESS in 8m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b43a2a7ac9fe4259898975cd1ce67295 : SUCCESS in 24m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ea5786eb8084b6e90f2db74b8875fed : SUCCESS in 15m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e235c8b1f2f1406c9474809aa6e467ee : FAILURE in 45m 22s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8a5da1619b19410483cdf13b2d98e08d : FAILURE in 58m 10s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/9f3d1180d4ef48d985abb2737513f177 : SUCCESS in 39m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/11e1e3d7969b430e97b93115c9e17cac : FAILURE in 43m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/961822a7916d45339025ae9731f52da3 : FAILURE in 1h 06m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/878184d448924ec59ad6084b10ba3d3f : SUCCESS in 1h 40m 34s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8326916606b24297ae01b547dcd0f152 : SUCCESS in 1h 22m 09s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a5cb00e9888246a8a950c84452547cd5 : FAILURE in 17m 52s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/962065dcd70442babcd19134bba1d501 : FAILURE in 15m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2caa09717292437cb2c748aeb00f0fa0 : FAILURE in 19m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/948361c27abb4b3584b906a7cf16d2d9 : FAILURE in 14m 48s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70985ad847c141fd82e5459e77e4c2c4 : FAILURE in 34m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28afddf218914141bce66d06f21bbce2 : FAILURE in 29m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/676fa47345a549ffb1e38d3d9c16953d : SUCCESS in 15m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e864339f20543bcae21cd2d3ddce4f9 : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a28728850c74222617fcd0c177377e29a3d29178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 17:02:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8d84615823f6c73e31ee9c4d344effce1d5dac1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:09:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0dae7da8f3b4e8e9938db590961330b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2aab994ba1f43db969e0314ba09f828 : SUCCESS in 8m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/612c7d7ac9ab47ac9730be68ef5c9dc5 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dae40b98bf8d4f1b92387a74a48cc272 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd12618ce0b74c0d8e82fa600094f4b9 : SUCCESS in 4m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ec3a7185d69417c84a741601079efca : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0b05f0f2c0d54b4e9ef3dfd843890c5c : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9874ba91ca84d25aa1dc0454fc24e9e : SUCCESS in 12m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d35a12c287c4dd6ac9c69646e613103 : SUCCESS in 10m 08s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5b03384016864c0a9a562d96c93ebd74 : SUCCESS in 14m 33s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4466f57ab63a48e0b8be5fbfa5c4c49f : SUCCESS in 7m 31s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/017c9618eb7e4611b506c783175f68a7 : SUCCESS in 21m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c48996dcca0496ab88075ca06a76ebe : SUCCESS in 15m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dddac096c8b042e18439d486bacc60c5 : SUCCESS in 1h 12m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9da62a7680334789a303fa465278b0f7 : SUCCESS in 33m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d5b6bb1865954639b136f8f5290e0c60 : SUCCESS in 37m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5dfb3c2759de4494afb8cb53d8eae9d1 : SUCCESS in 30m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a50ded2922e54fbd875c85f210ca95f0 : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c255e669ac824cf180e662c4f890af73 : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0dfa61d4397847449a1e3bb07b095721 : SUCCESS in 2h 04m 49s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/be5718a40cab469c807935b69a849956 : SUCCESS in 10m 49s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c3b9c473ff834ecfb874153a74321c0e : SUCCESS in 22m 12s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25d4614b55034ca7bf19b4e5ba4beeb8 : SUCCESS in 17m 28s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c0f24bd524704da9ae148081ac8c7bb9 : SUCCESS in 18m 47s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e9a9c3ea59ca4b92a2fa12994b63e7dc : SUCCESS in 24m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0de7b117f8646548093be2910485752 : SUCCESS in 30m 37s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7a2111074c9f4a96b6ac1b14e8cf42ba : SUCCESS in 9m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c305a48edb094dec99a8c3e54cd46fee : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":2},{"id":"0e541a845e1dc13e14480512cc87ac3da3351693","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-24 10:48:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2ad4f45263757c4fbe96d4a6b8cdd744c4c5f5ab","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-26 09:18:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cfc5fbc9f4d41e1f8224436f93887265f4041987","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-26 16:01:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7b28ad0c8db8bec70d96c5594405cc92f4593846","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:02:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d5d5c051044c40143c9a41c15a5ff88e4720c753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 18:00:13.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kayobe/+/974044 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"1ef8b6c1f28793ccee94e603140018872fcb5cb0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 21:57:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"619dd27474ebfac7f1b83303f7121a71cd5796c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 23:28:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d30c84919434742aa4289e5591e7951\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f202f91725b4f55839f74b337253fd6 : SUCCESS in 9m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a8f2badcf0b4aedacb941bc783263c8 : SUCCESS in 6m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbf8d40e505c44688ac81dd1acd7b6ff : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d23c3da5836b41fcac1a5c2db5f8a988 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad29166405764d9695d0d9b206fb9fa8 : SUCCESS in 3m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a5fc445e46f484fb557a63873b1f48b : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/753d935076e64ce78575f3604b4f3a17 : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d0fc420a9efe4becb45c6203e773df91 : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2c3f04877a344a118eb05e245412d0c6 : SUCCESS in 9m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09c6aba424f64bc4953c9093e7034547 : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/05b5deeda20c481ab84e0c4412ebb3f8 : SUCCESS in 16m 49s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d0f2f5329364dc38cda341b04bf0f8c : SUCCESS in 15m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9a1a9891f8cb40ce8ce77f028087eb41 : SUCCESS in 51m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d704ac13db594949acd177c3585683cb : SUCCESS in 47m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/119ae38e3a6243caa9379ed87d911664 : SUCCESS in 57m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc329182f987448a8a6d55905c4b6c2c : SUCCESS in 46m 33s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/98ba92e23e2d4bcbb8757d7b8fa1e3d4 : SUCCESS in 49m 05s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/927bde60dcb74afbaa33da5a615e6478 : SUCCESS in 1h 28m 48s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/11e9bd10c7854771a168fb1feb69e9bd : FAILURE in 1h 20m 17s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fb03fa3648245c1bd3a13bc0ac366dd : SUCCESS in 1h 21m 21s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40ca66a8e0d1453e8a455e414c23eeec : SUCCESS in 1h 24m 22s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e9103cee29164a7899af7b829d2a3230 : SUCCESS in 10m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9ed3395f12ac4fac94846152e2c65007 : SUCCESS in 10m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6d815ce32984568950c3f2530c507d0 : SUCCESS in 16m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/56be8726c9124fbeb2605d4cc4ae541e : SUCCESS in 16m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ddb27871fe594e16b5e08f23c17c1ce6 : SUCCESS in 32m 26s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/f32c4de4b5f1440eb96ac34c8d4e3685 : SUCCESS in 23m 37s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07479b85d47440b59c63ad361d447f21 : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e640fedb83254a16933bc747f803152c : SUCCESS in 28m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/26fc4e106f334dc8a915022a51229b0a : SUCCESS in 9m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4dec9d7654aa44a4a01188c15cb6b805 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":2},{"id":"89ddaf535572f22206b9716ec41b827efdce6c6e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-27 06:05:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7386aad8eafed147ae9ecd24d27e11befb2ec7b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 07:35:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/080fecb16a5947a083eb38115f90ccb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94529861378c454fb93334c7e160e965 : SUCCESS in 7m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a611d046cc604175b8e12a7aed69ef7f : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73f7b9c7cad44bf5931275132348fe6f : SUCCESS in 2m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38411935b5864dba8b22d6097531e4e1 : SUCCESS in 3m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58d09ca77b1748169e85a9ba3a119277 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a247604db6ca4171b00da496f5ea61ad : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f7efdc48b6d48948023877cb9948051 : SUCCESS in 6m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5039c001a73646ed9cb36fd716fceac1 : SUCCESS in 5m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/265412657b8948888972b81bcdd2cb1d : SUCCESS in 13m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1016ca0b01624c5cb34ac217f49d6301 : SUCCESS in 9m 54s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7b8357ac2c99438d9e137f2580b57d2b : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c993652f6664220b545150d114e4acb : SUCCESS in 18m 58s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f06d7cc142884cc49c37267821619209 : SUCCESS in 49m 23s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e6c2e0271aa74db3ba8f5ca93f426af0 : SUCCESS in 49m 25s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2f4d1355d0b1402cb7d392efd408bd86 : SUCCESS in 57m 52s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcec771850db4e488280aec702d70b17 : SUCCESS in 44m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4f7ecb7bac9a44789b716a5ae2bf9ef1 : SUCCESS in 45m 56s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1b0caf9b3c6e4204b6c96ee484f74807 : SUCCESS in 1h 26m 23s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/639096c64a0b45a7a8ecd673aa22a686 : SUCCESS in 1h 26m 42s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2dc26b46da244c03bdc2c854db8887ce : SUCCESS in 1h 20m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96ff45c9a5454479a3b4859d1a494fe2 : SUCCESS in 1h 28m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f2b02b896a274d8eb933c3deab0480b4 : SUCCESS in 9m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/64227e810ebc41ac966c4bd2d46f7bb8 : SUCCESS in 9m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eae737dfae4e498dabd7ac46e052a3ec : SUCCESS in 9m 09s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/495ff5141ab54a0999b7b5bc902242ad : SUCCESS in 16m 58s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/37fe5972ed8f48aab7b8d3908e6e4801 : SUCCESS in 21m 51s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/89c3fa10260246e190e954402e08659e : SUCCESS in 23m 32s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d9ba29058ee84ee2840dd9de17a88a37 : SUCCESS in 27m 45s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca87e8e27d604601b92641f879f6a4f8 : SUCCESS in 27m 44s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2f1c5a019ec44f2c94cd7793503ca48d : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76104f73841e4bc49a926945b9ee399c : SUCCESS in 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"154659bab278b9b57ad6f790172c0021c8b31ced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 07:35:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ae157750290fabcd048fb41f49bab56d17b9fd2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:10:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b41c1cb56669448689c80a87aa096851\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21693f57beed42c89aa72ddfcbc6cbbf : SUCCESS in 8m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9c64132242643f3a3181a7243603875 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42adb48e09404667aa223287fbaf69d3 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf019e1e5dd3453987e3ace60bb7fbef : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61767e60b44b4ca98d25667b4d4c5081 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce4652023f564dce8ea32de6af158995 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/21305cc0042249658f52c951bfe90d48 : SUCCESS in 10m 49s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a9374f0e42da4f499102edd2a9970ebf : SUCCESS in 8m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff718a776d2e404c9ce50f2f1830dec7 : SUCCESS in 12m 34s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/84067cf55fad4f71a9e157a72df0082d : SUCCESS in 11m 38s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bf7045025205406186902643db4fb9e8 : SUCCESS in 15m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/20fb737f7bb24f53a70549eecaea10b5 : SUCCESS in 48m 14s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/51ef5645f6024affbe3917768b5a073d : SUCCESS in 50m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/22ba5bdae0c64f4e8019807f5f0b1b97 : SUCCESS in 57m 20s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83095f201187437ea932d78c1656c5ae : SUCCESS in 52m 41s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52418172bddf4fbead663d03941a75a7 : SUCCESS in 48m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/67e8f11932d1428d82db00760ce62200 : SUCCESS in 1h 33m 49s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9ec8628657e04f55bcdb2fd1152e121c : SUCCESS in 1h 23m 06s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1434a42432c04d918d22f69f4f1a83e7 : SUCCESS in 1h 20m 56s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9233027a3d7340b4bb803c40e29ac7d2 : SUCCESS in 1h 24m 25s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ece5028a084b456d8d5df5e0f4b9012b : SUCCESS in 15m 04s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/00ef921bcebc4cf995a0068e36af5b9b : SUCCESS in 13m 46s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2c875c9e8de42578cfe4548ca62a0c2 : SUCCESS in 16m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/490d8227d66c4be683d25dc9fc7643bb : SUCCESS in 12m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8b31977575b8438f9a91d2263fe68fee : SUCCESS in 24m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a89a7056ee2042e494b86cc6fbf84c1c : SUCCESS in 28m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e6797233f784a09a6a1983736174351 : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41e25372750e4aef9030e84d51805a2b : SUCCESS in 27m 14s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a7bffcf84c2f4202a985e73c8eefaf66 : SUCCESS in 8m 18s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f75171a7869f428396302a28916bc8e2 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e64331fa4420e5b5ece150f6ad1a07e8a8dacb6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:10:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6e7848ef57fb48c3a2bf4d35de9a9b938b4d2ad5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:11:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2a431980e24774aaa80305330ab7eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc24953131e45d99d029fdf55c7846f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e93afb84a45ce5ac29d0f3e22cfd13bc1663bd3","revisions":{"42c7a6e62108be5d295bcb415a78789c28c04656":{"kind":"REWORK","_number":1,"created":"2026-01-14 18:17:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/973422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/973422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/973422/1"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-14 18:17:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-14 18:17:36.000000000","tz":60},"subject":"Sync host configuration with Ansible defaults","message":"Sync host configuration with Ansible defaults\n\nThis synchronises the following files with the Ansible inventory\ndefaults in ansible/inventory/group_vars/all:\n\n- compute.yml\n- controllers.yml\n- infra-vms.yml\n- monitoring.yml\n- seed-hypervisor.yml\n- seed.yml\n\nChange-Id: I9ef77d6119be910710eaffa8bc70bb207323f268\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42c7a6e62108be5d295bcb415a78789c28c04656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42c7a6e62108be5d295bcb415a78789c28c04656"}]},"branch":"refs/heads/master"},"2e93afb84a45ce5ac29d0f3e22cfd13bc1663bd3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-23 17:02:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/973422/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/973422/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973422/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973422/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973422/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/973422/2"}}},"commit":{"parents":[{"commit":"6b4028754321b3e20b45f93b6fba159985e850af","subject":"Support swap configuration for all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b4028754321b3e20b45f93b6fba159985e850af"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-14 18:17:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:00:38.000000000","tz":60},"subject":"Sync host configuration with Ansible defaults","message":"Sync host configuration with Ansible defaults\n\nThis synchronises the following files with the Ansible inventory\ndefaults in ansible/inventory/group_vars/all:\n\n- compute.yml\n- controllers.yml\n- infra-vms.yml\n- monitoring.yml\n- seed-hypervisor.yml\n- seed.yml\n\nChange-Id: I9ef77d6119be910710eaffa8bc70bb207323f268\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e93afb84a45ce5ac29d0f3e22cfd13bc1663bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e93afb84a45ce5ac29d0f3e22cfd13bc1663bd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
