)]}'
{"id":"openstack%2Fkayobe~973616","triplet_id":"openstack%2Fkayobe~master~I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-01-22 08:27:40.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-21 12:04:05.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-22 08:54:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153","subject":"CI: Re-enable seed jobs","status":"MERGED","created":"2026-01-16 13:27:56.000000000","updated":"2026-01-22 08:56:40.000000000","submitted":"2026-01-22 08:54:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973616","meta_rev_id":"1f58594ce9c69bc67d9c5ef509c7d67fbe35f6cb","_number":973616,"virtual_id_number":973616,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 08:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-01-22 08:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-01-21 12:04:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-01-21 13:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-01-22 08:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-21 09:08:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 16:13:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 08:34:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-21 08:20:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-01-21 08:20:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-21 08:20:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-01-21 08:20:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-01-21 08:20:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-01-21 08:20:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-01-21 08:20:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"e6b8a8b7eed8d6d0cab117d019f0a039bb7c53df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-16 13:27:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46d658e5657d704bd781bf880dcec857b3573cea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-16 14:28:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1675f703c8c82132298e3c01c58b97bf79a138dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 16:13:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aec41fc05d84b9e848c5feb405fac9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cfed44d86624d068162bbbe6842fd3c : SUCCESS in 9m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c9fbe941b202417e9463e46f2049e252 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a11388a9ed6c4197822c1a7fdb3f6dbe : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae6c3039cd54657a07b5bf523e4629a : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1cf975e389642bba74381b76a1aab35 : SUCCESS in 6m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5994a8b37a2b40b498cb16d5f6a5b541 : SUCCESS in 3m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a0b0c4cb58e646e68836816b681156fc : SUCCESS in 6m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27d6004cb9e64e9d940a6bcddf73d101 : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b73b9f623e0f40c8ad0742b690b70627 : SUCCESS in 8m 25s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/be36a20c97294b2bbef022f9d24df989 : SUCCESS in 12m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b0f4020fa4bb4329bcfd778eeb780a37 : SUCCESS in 23m 59s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f502b562f70548e29c7b260fb28bf8ba : SUCCESS in 8m 19s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dccec5592a4744e9991e7cb8b71bcd56 : SUCCESS in 1h 20m 00s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7dcbcab24b604eff90bc3a6c5a1a03a9 : SUCCESS in 33m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5ee54854eaf64c98b40fea43042c13de : SUCCESS in 1h 25m 39s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7fd97bc1d8a4ba891e49b9d6ecc7f31 : SUCCESS in 1h 01m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/288d56397d834a748b2f4f31ac97eef1 : SUCCESS in 1h 05m 50s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/55887016b25c4e9f988a7764c051cc6e : SUCCESS in 1h 37m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db3b81b0ab5c4e03a19c62821270068e : SUCCESS in 55m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/cddc5ecb4ed54d73a1c893a237967f41 : SUCCESS in 10m 50s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d3349bc4cda849ecbbb3e3cc2e6beb17 : SUCCESS in 17m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5b1ae9c455f46be93312738b48d3725 : SUCCESS in 16m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d8cd7d097a5e42f1b11248b3cb544f0c : SUCCESS in 16m 31s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/78493390cb874dc7be616a189e26dded : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89d092ae5c0d4fd88f869ec28d3af38a : SUCCESS in 29m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/654a5f7bfe84496ca6ae01a918c5602f : SUCCESS in 8m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52f3fab2bfbe452caecbfd93611e52ea : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":2},{"id":"d8710b755de604618687a88701487b4c29ad011b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-16 16:52:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"847a0038a1555c3469cc27f621753c7aceb0a5e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-16 17:38:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"16f0dc946dc81fdb71a17e8cf3f7eaab0d087b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-16 18:10:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"66fde47d886b91581de3b545b35e99508ab95520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 20:05:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a230ea44b0cd4824bdf925c7219355ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e466b3cddb4e51ba4a596f6e74417f : SUCCESS in 9m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eb4c5f4910c840b392afdf8696dcaf21 : SUCCESS in 6m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0465d4156ce4f3398af3684b4e9ee2b : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5be0d6fbcb84675b16ba847b74fe866 : SUCCESS in 2m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a71c32c4ced54215854e106c61a723ea : RETRY_LIMIT in 1m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4fa49db5d8c24840a8c8c1c9514d0325 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a00fddc9e4044a089a95332f0ad4bd07 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a64d289e3b674aae8cb34be9a6bffacf : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/9ab67a0f79074039bcff0bf8670861d2 : SUCCESS in 9m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/832a49f7f0644eb0b4deb6002aae7a69 : SUCCESS in 12m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/686a0cd081434247ba1fc485dfe2ab04 : FAILURE in 8m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9529d74bc3ff4e999b6e3c0c014dee50 : SUCCESS in 8m 57s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4898909d5fae47e5b7258a46948a23c2 : SUCCESS in 33m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9887cb6e3d3f4fa2ab1a4566bc06d0e9 : SUCCESS in 34m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/201e8ad504334d0e8092fc7839931b20 : SUCCESS in 41m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c38a4ac73c444728e883d774eea4645 : SUCCESS in 33m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d228eb0a62e240a0b4af3f14a1ba3a14 : SUCCESS in 57m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2943dc5f549f4dd8942b1993c35914a5 : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/87251c5caf554b7dbff7fd361bdecdce : SUCCESS in 1h 44m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8cf940405e234faba0c79945b16c7c1f : SUCCESS in 10m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aa20e00ac4fe4a5e9ea5679adbbb3d8b : SUCCESS in 10m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/edfbfca9adb84a4da9122e9218db39a2 : SUCCESS in 15m 38s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4eb9421191944d95a6e1b2458eb48481 : SUCCESS in 10m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bc70ee06af344de58f8cd3a93c8fdfb7 : SUCCESS in 9m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f89908dab019448ba7c27d72391ee26c : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":5},{"id":"25f41fb55785b67d2eff7683da4a99f954f1ed3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-20 08:24:33.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e5c77f298b833e894f28df31d390b5305b262c55","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-20 08:34:11.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e1fca80b086daecfa8ff18707069f3c466df2337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-20 08:42:30.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"04350465182b3394e7d4a54e42fd3502022073f5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-20 09:20:12.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fbd75203d3ce66cdd3a59bac7d29186f5f85439e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 10:08:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7afb7e35d650482085d13fba39875373\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/842a8e67b94f4e258c16a8c0a01298f0 : FAILURE in 4m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a0cf3c1ce24743b6bf4f1efb47ef73cc : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e764e93f51d34855b6e9193e0583d0ba : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d757e49ea1a4cbca4cf4cd4a2cc4088 : FAILURE in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7badf37053a94d3d8032485616479a57 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b37f6553887f4ffbac414da8eb311c5d : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4a228a97c1643a9bf75d494a829287f : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d92db7a80d094453843a9ff784557a83 : SUCCESS in 12m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/114b0bd64ad3489e9dfb37edd5105bb2 : SUCCESS in 8m 46s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4c9268835a3e4b998ef55092b0e7bb03 : SUCCESS in 13m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b60b6b4866854609834a8741faf55499 : SUCCESS in 14m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/4763ff8ddd7c4dacb32e7ad1fe391102 : SUCCESS in 10m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f9740b4aa4b4890afe6b3d97f6f32b9 : SUCCESS in 14m 41s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/006a046a0308482e835ac2b139f49248 : FAILURE in 40m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/e49b4a6f71e6487785ba4cc121def4db : FAILURE in 1h 05m 02s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/83eb93662e9c4a3fbc1859e7bbedb955 : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82eb6026a89f42958e6781e725394782 : FAILURE in 36m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ebccac6666914a55a29468b001f24277 : FAILURE in 39m 38s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a0d2a715267646deaec48525049d0a88 : SUCCESS in 59m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1849b5a5556e4329b043d1ee2fc2df18 : SUCCESS in 1h 25m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/86b80b03e1034d8b82060bcdcc031ce3 : FAILURE in 9m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/217559e244b64722bad6def0d92dd60c : FAILURE in 13m 27s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/346e9c780a5b451495a64a6008efdbf4 : FAILURE in 12m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f315699fb1ba4e7891f3b758f8e21b8d : FAILURE in 7m 53s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/841234b646434d69bd51ae6b15e984b3 : FAILURE in 29m 58s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf2013dfa0fa4149b05a1711df036d2c : FAILURE in 14m 55s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d4dd93422db045fb8019fed5b2a6f7a1 : SUCCESS in 8m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fca179377a0e4cc394faf552ca14ac11 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":7},{"id":"820701928d7f92166fa8732f7dfa5ea79ba55883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-21 06:44:51.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"be3219a38ddb4635a008c27f8b2eaac3e73dd8f8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-21 06:45:37.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"a7f67c4e5075448c0b90f72e9b396ff18f28d182","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-21 06:46:09.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":9},{"id":"4cf7ec891ac2bb0e128c735ed1287b6d1a758de3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-21 06:56:59.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"aa695fe7cdb161ebc26b410ffc3adfbeffe7e0d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-21 08:20:13.000000000","message":"Patch Set 9: Review-Priority+2","accounts_in_message":[],"_revision_number":9},{"id":"cd2ac5310346b96631223cc8ba06b6554c40a5a4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-21 08:28:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"affa26d8e5c6929d8f2d08dad1506578cd031efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:33:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdd25a8474e849bba5854e697663afad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121e755d9a0f41fb86ec6d8668fcadb8 : SUCCESS in 9m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/de32ffc961ba46e8828a59464be28453 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a667d0169e941dc924dce534149772b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e8c67ccd074a5aaa81bb6916e19057 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f034a79e06844c3aa259d68afd83f816 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95c96f5ed199427d9939840822542a53 : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8652e89b7ead45afb0c9519c410a9ac5 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/14e396ccb64148d093b6cd70338079dc : SUCCESS in 14m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ae31ecf53c54a12b35ba66dd2ba060e : SUCCESS in 10m 55s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/124f0e6daf684352a90acc517d260ac7 : SUCCESS in 12m 55s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bd5aae469e2442ae8b3e3d068c1016bc : SUCCESS in 12m 10s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/50b1605bcd674c958be86475661fb1fa : SUCCESS in 16m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54c2f38eade24d409340e40158dbe4b9 : SUCCESS in 16m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b373cc3499d74830bca73e73e841aec1 : SUCCESS in 32m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/4a0058be28eb4197a2c89d06b4270b6d : SUCCESS in 1h 13m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ede8ac7a260f417fa8cc1ce380767800 : SUCCESS in 36m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a9be92ec704a4089bf6eef2aa0d5b52d : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3686c8f0e7a04b63a112fc75783b7bec : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8a302c982e364ce2a18dd83537c1608e : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/299f8711affd4fc6bd7af5fc9f3c7140 : SUCCESS in 1h 43m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/584e2ff0f34d4861bc8ef204b9f037ed : SUCCESS in 21m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ef2f067accce4741bec23a77573e6f11 : SUCCESS in 9m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a325126aead5435fb322656a17e552a0 : SUCCESS in 15m 13s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/52e43e77fb434baea901a056f7932691 : SUCCESS in 15m 30s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f29609d77ff14a4db30f7261cc100eeb : SUCCESS in 16m 25s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0233d837d8143059671d9ce18d2c07e : SUCCESS in 32m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e758adc21329431c95404da20a920197 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13b12da8d6bf40b396bf009919804ddf : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":9},{"id":"06b369bb13188d409da660aa57090c84f4eb97de","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-21 08:46:47.000000000","message":"Patch Set 9: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c71dd6d4c5837b7b1a22586ae4dab6ffd7eace6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-21 09:08:55.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":10},{"id":"083743949b2eb6175b47ff4acaae81855988989f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 09:37:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c148b1226dc473a8a505a3738c213f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18b4d22104b44069bfc75d087cbc58a4 : SUCCESS in 4m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/054e09fb16224f3282bcee1cb5fa577c : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af2189de363941c78e3a8f536dca7be5 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882b13433a6c45fa99dadd944057a3bb : SUCCESS in 2m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6154b51d8cd94e85b6dfc0198ca6b154 : SUCCESS in 2m 38s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/2ea152a2df3b4dad83f047e763219fd9 : SUCCESS in 11m 27s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1a6aa48d067649259bff4d2e3183aeac : SUCCESS in 10m 33s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02c5fdb7f4a44c1d8c978509385d106f : SUCCESS in 9m 40s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d792891a457c42b882c6331839a6372c : SUCCESS in 18m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0360884316e1484bb8a335aafa644173 : SUCCESS in 17m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/24fe4ada2b774885bab79381fc9f6101 : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":10},{"id":"0028d7891d552396819a968e478791386f5fb2a1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-21 12:04:05.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"0a96e6c8abd43b589e77933139ac55b29ae96609","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-21 13:05:58.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"e703a7cc6cd235e5d1413539f310bb739fa4b1ca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-22 08:27:40.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"c9dd4912cc824f428031e5f22f02c8dce876fcbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:28:06.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"12c818d80dc25c2343e9c396081c2f68b1b0c6d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:54:53.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db49840e1dd24576b183290763cd5974\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c84d5a84c994d98a633e76ece5d7a4d : SUCCESS in 10m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/26787c50858c4772a199050d4b363a70 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41771b9403be496fbeed6ea0a589579c : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d20905130e45456ea116f582fb266994 : SUCCESS in 4m 19s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7f9d59a0f04749259e0aaf76ae5776c3 : SUCCESS in 20m 03s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/80b16c4aedaa43b8b9da6236e376d3f8 : SUCCESS in 11m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/006fdc5f3ff441239ea8ee75f153495b : SUCCESS in 16m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e5b39ce223fe4635b5cefc89d71cf5a6 : SUCCESS in 8m 43s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0ecb260679bc444a8c256468f218e768 : SUCCESS in 17m 42s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4afc93b6e7d4486a90f0c65266a916a5 : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":10},{"id":"7373afac345900a816d72178e8edf284d2b9b728","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:54:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"1f58594ce9c69bc67d9c5ef509c7d67fbe35f6cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 08:56:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b332ffad7e924d04bc2efeca1c0110a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730da6eb45e14ac8861ec6eb06e600e6 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"d7686f5f241244b199749428b3d0fb895238b64a","revisions":{"e78f15e66d9f963b5c47fb284196f4d49d50f2b0":{"kind":"REWORK","_number":1,"created":"2026-01-16 13:27:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/1"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 13:27:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 13:27:43.000000000","tz":60},"subject":"[DNM] Check CI health","message":"[DNM] Check CI health\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/973460\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e78f15e66d9f963b5c47fb284196f4d49d50f2b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e78f15e66d9f963b5c47fb284196f4d49d50f2b0"}]},"branch":"refs/heads/master"},"a4688562047acb1c347d58a5ce6ead1b2ba9f93b":{"kind":"REWORK","_number":2,"created":"2026-01-16 14:28:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/2"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 14:28:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 14:28:29.000000000","tz":60},"subject":"[DNM] Check CI health","message":"[DNM] Check CI health\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/973460\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4688562047acb1c347d58a5ce6ead1b2ba9f93b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4688562047acb1c347d58a5ce6ead1b2ba9f93b"}]},"branch":"refs/heads/master"},"0b689a52c466d62afbac49ebd44a29a0c8c56a76":{"kind":"REWORK","_number":3,"created":"2026-01-16 16:52:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/3"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 16:52:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 16:52:29.000000000","tz":60},"subject":"CI: Fix multiple issues","message":"CI: Fix multiple issues\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container:\ndisable by running the 2025.2 bifrost-deploy container image instead.\nThis registry is unused for now.\n\nDisable test of baremetal machine inspection: it is currently failing\nwith power off timeout.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b689a52c466d62afbac49ebd44a29a0c8c56a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b689a52c466d62afbac49ebd44a29a0c8c56a76"}]},"branch":"refs/heads/master"},"e76dac8fddb204e243aeffb491356aabc21bc1f4":{"kind":"REWORK","_number":4,"created":"2026-01-16 17:38:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/4"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 17:38:35.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 17:38:35.000000000","tz":60},"subject":"CI: Fix multiple issues","message":"CI: Fix multiple issues\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container:\ndisable by running the 2025.2 bifrost-deploy container image instead.\nThis registry is unused for now.\n\nDisable test of baremetal machine inspection: it is currently failing\nwith power off timeout.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e76dac8fddb204e243aeffb491356aabc21bc1f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e76dac8fddb204e243aeffb491356aabc21bc1f4"}]},"branch":"refs/heads/master"},"854c2922037c5ba343fcecbe856c8a34b1211e6b":{"kind":"REWORK","_number":5,"created":"2026-01-16 18:10:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/5"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 18:10:10.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 18:10:10.000000000","tz":60},"subject":"CI: Fix multiple issues","message":"CI: Fix multiple issues\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container:\ndisable by running the 2025.2 bifrost-deploy container image instead.\nThis registry is unused for now.\n\nDisable seed upgrade jobs which are also broken because of this.\n\nDisable test of baremetal machine inspection: it is currently failing\nwith power off timeout.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/854c2922037c5ba343fcecbe856c8a34b1211e6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/854c2922037c5ba343fcecbe856c8a34b1211e6b"}]},"branch":"refs/heads/master"},"668c8f9206161fa3042f8314aff067e91afd6e55":{"kind":"REWORK","_number":6,"created":"2026-01-20 08:24:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/6"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-20 08:23:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-20 08:23:21.000000000","tz":60},"subject":"Disable Bifrost OCI artifact registry","message":"Disable Bifrost OCI artifact registry\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container, so\ndisable it to fix Bifrost bootstrap. This registry is unused for now.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/668c8f9206161fa3042f8314aff067e91afd6e55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/668c8f9206161fa3042f8314aff067e91afd6e55"}]},"branch":"refs/heads/master"},"238e6cbc6d715f8bf3a0ce74a0b19059fdb26e16":{"kind":"REWORK","_number":7,"created":"2026-01-20 08:42:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/7"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-20 08:41:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-20 08:41:52.000000000","tz":60},"subject":"Disable Bifrost OCI artifact registry","message":"Disable Bifrost OCI artifact registry\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container, so\ndisable it to fix Bifrost bootstrap. This registry is unused for now.\n\nAlso fix CI issues related to wrong packages listed in bindep.txt.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/238e6cbc6d715f8bf3a0ce74a0b19059fdb26e16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/238e6cbc6d715f8bf3a0ce74a0b19059fdb26e16"}]},"branch":"refs/heads/master"},"0da95e78064ac8fcd34a244702a695f4ed602bc2":{"kind":"REWORK","_number":8,"created":"2026-01-21 06:44:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/8"}}},"commit":{"parents":[{"commit":"60a71c7adae37139fae4556fcac41963a5334fbd","subject":"CI: Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60a71c7adae37139fae4556fcac41963a5334fbd"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 06:44:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 06:44:38.000000000","tz":60},"subject":"Disable Bifrost OCI artifact registry","message":"Disable Bifrost OCI artifact registry\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container, so\ndisable it to fix Bifrost bootstrap. This registry is unused for now.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0da95e78064ac8fcd34a244702a695f4ed602bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0da95e78064ac8fcd34a244702a695f4ed602bc2"}]},"branch":"refs/heads/master"},"dbee3cf694e37ad74a506208ce9990673c0047ca":{"kind":"REWORK","_number":9,"created":"2026-01-21 06:46:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/9"}}},"commit":{"parents":[{"commit":"60a71c7adae37139fae4556fcac41963a5334fbd","subject":"CI: Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60a71c7adae37139fae4556fcac41963a5334fbd"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 06:46:02.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 06:46:02.000000000","tz":60},"subject":"Disable Bifrost OCI artifact registry","message":"Disable Bifrost OCI artifact registry\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container, so\ndisable it to fix Bifrost bootstrap. This registry is unused for now.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbee3cf694e37ad74a506208ce9990673c0047ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dbee3cf694e37ad74a506208ce9990673c0047ca"}]},"branch":"refs/heads/master"},"d7686f5f241244b199749428b3d0fb895238b64a":{"kind":"REWORK","_number":10,"created":"2026-01-21 09:08:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/973616/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/973616/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/973616/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/973616/10"}}},"commit":{"parents":[{"commit":"60a71c7adae37139fae4556fcac41963a5334fbd","subject":"CI: Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60a71c7adae37139fae4556fcac41963a5334fbd"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 09:08:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 09:08:32.000000000","tz":60},"subject":"CI: Re-enable seed jobs","message":"CI: Re-enable seed jobs\n\nBifrost added an OCI artifact registry [1] which runs as a Podman\ncontainer. This currently does not work inside a Docker container, so it\nis disabled by the kolla-ansible bifrost role.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/961388\n\nCloses-Bug: #2138705\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/974029\nChange-Id: I3aa9f84bedbc5bc7b472e8aefa1c16499f40a153\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7686f5f241244b199749428b3d0fb895238b64a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7686f5f241244b199749428b3d0fb895238b64a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
