)]}'
{"id":"openstack%2Fkayobe~973657","triplet_id":"openstack%2Fkayobe~master~I64c9960564ad1cf2dbf8ca5d1bc44bc95ce57cf2","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-22 14:00:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I64c9960564ad1cf2dbf8ca5d1bc44bc95ce57cf2","subject":"CI: Set previous_release for 2026.1 Gazpacho","status":"MERGED","created":"2026-01-16 17:40:23.000000000","updated":"2026-01-22 14:03:29.000000000","submitted":"2026-01-22 14:00:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973657","meta_rev_id":"8315560b0b0c2efae641a74233aa4d10d73b5dea","_number":973657,"virtual_id_number":973657,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 14:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-01-22 12:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-22 11:30:30.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":2,"date":"2026-01-22 11:35:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":1,"date":"2026-01-22 11:35:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-16 19:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 11:30:30.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2026-01-22 11:35:56.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-22 12:03:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"0246c8c3c4e29fc3a7232980faee40467975cf25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-16 17:40:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2091aa7ea1c27035d360543f673e2f273786e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 19:45:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbe64f8b1d57417681666c8a172785b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0146defe6ed41269a0928149d0e771a : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e4701d07b97c49a9a70f6e55e895b01b : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2758f75e8f5445cda5cedffe2d2ef388 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a64748884121477893ef62cc293ac03e : SUCCESS in 5m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9983272a25004e57b0d40dac134fded6 : RETRY_LIMIT in 1m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/fc4c9ca2af78458890059b4b297602b1 : SUCCESS in 13m 26s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bb8908002b04411a258021079c1eafe : SUCCESS in 12m 41s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5c3bdbeed1264e15a6dee7db253c691a : SUCCESS in 11m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92adf734df7c4e4b8622db7aae216f44 : SUCCESS in 17m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e9a9cca2b90b4a47b2f2508178cd14b1 : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7bc277a24c1944ecab0584801418e935 : SUCCESS in 1h 09m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/84328af374c4404f8a9221bb9f3f867b : SUCCESS in 1h 00m 17s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97aa8123d72948518221a48eb717eb87 : SUCCESS in 32m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d847c22644e64aa0a6e374040ab3a9cb : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c2350b90126147dd95eb75aa64ca4292 : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c4752497e20a457a951ae2090381d2ea : SUCCESS in 1h 59m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b5366d2f3a044d40a07be704cf5e1d00 : SUCCESS in 23m 41s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/aaae1515c24d4ee088550e84b16e2a86 : SUCCESS in 9m 34s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5dfbc9eb2edb41af88aa9d027eac8eaa : SUCCESS in 9m 18s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f7dc4617d9804a2fab1c9c2a9a17edf4 : SUCCESS in 15m 44s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/4bf33382df684aefb52a67fef7d0c95a : FAILURE in 15m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e01fe12aebf54c918c099097c664bc18 : FAILURE in 13m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/88cd8870090a46d5a06bf70e357ed52b : SUCCESS in 9m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/858a5d86a5b24078b973d315efab94d0 : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f5477d6582b7cd52f3f6be4513a2d8e206f513ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-22 08:56:47.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"acfbaf96aa6db4e7b69ff3906fbfad34d81b83fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 10:58:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bccb2c4d85d34ba0959e066da7689f9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3908763573489899576959e17be130 : SUCCESS in 9m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8cc4afcff95c468898466a6228c4de88 : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8396682a1095421585288038a93ffb3c : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8d09bbaf3f4f4bbc0da3cc6fbf2372 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5cdb3dafc7e49eca11c43815551d7cd : SUCCESS in 4m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e5095f6e3ca34c7db2f4d0b61aec9f26 : SUCCESS in 10m 03s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c61031ce629d463e8e0aeebcfade3bb3 : SUCCESS in 14m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c70b7ab183fb4fa7839a8adb72787f17 : SUCCESS in 18m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d69b48a0030455aa5681ca5674b66c7 : SUCCESS in 15m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/4c8ff0b907cc427198b991c6034f90a1 : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6b7be36c02814e1d94a59d750b60feca : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/71384bd8919d49fd9aa5da99c1ccfe40 : SUCCESS in 1h 12m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2cfc45c16ba42d98b4c955d910ac90a : SUCCESS in 58m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7fab1475aff44ca5ab13cf4e7fb9b151 : SUCCESS in 59m 09s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0390cc36047049568d79284c3aabcd3e : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5fb75a6c435543ae95ef05146314df6d : SUCCESS in 1h 54m 30s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f62a02f93a3b4fdfb2f4ad9a457d3099 : SUCCESS in 12m 53s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/51449fd70af44536adede16a0fe5a450 : SUCCESS in 12m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e241abcf4af045efb92e0087ac885342 : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f88ebfbb290644aa9435f880ca88f2be : SUCCESS in 8m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/db49dac0fe0d48bea013ebfff6dcb66f : SUCCESS in 16m 52s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ffdca9961ae44c389cff6b79952d760a : SUCCESS in 33m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/20785c44efa74fc3a48900805c876358 : SUCCESS in 10m 33s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69486275a2e744be956438b1120a49f2 : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":2},{"id":"d839ec7845bae516baf328fe9720189f3887d373","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2026-01-22 11:30:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ffe02da861e72a080bff85f870f33d506f1c7d78","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-22 11:35:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9aa9bf6c8695f5508d9b4620ea81e18c7d21adf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 11:36:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7d5ac80c538ec5ed6f716d6df63c567ee57ea75","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-22 12:03:17.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ab357e2401fbac0cd944664c2fa928bea576efa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:00:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed0609162f0f4048a126820a6cabb422\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8457d52bf57d4bc29458c085187bd3f1 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/847002cd004d4a498306946e993a278a : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff4385fc443d487e9ef243ffdbfb6cd1 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8875082a117f4f90915762c4047fe3d1 : SUCCESS in 3m 50s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4aa5b91f63f74ae6a8f9567475cec527 : SUCCESS in 13m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/982f205643ec47c7affebbdb0c72dda1 : SUCCESS in 11m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b73ce6c16231478188f50a4fe7f38cf0 : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1763570c9fbf46abbf2d471401da618f : SUCCESS in 15m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/724b0e07bda54b1aa2d00a8cee8aa6f8 : SUCCESS in 40m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/58d8479cb5d84881ac2fbe7192d3e93a : SUCCESS in 33m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8871c3d5e75b4d77973113360f37fb87 : SUCCESS in 46m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63ab3a20a3ac48f79a2f72bdbf5066a0 : SUCCESS in 1h 09m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/be8287fc8f93422ebe8d809506b49288 : SUCCESS in 58m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/693991dbb86e43378be8d2b8dd944625 : SUCCESS in 1h 11m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85e03a61bd24474b9bc8d9769998ea17 : SUCCESS in 2h 10m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/4182cd04f7e643e7b67255c36f5518a1 : SUCCESS in 19m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/7e2b0821c9aa4df6ade2aae9a9809c12 : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/377535ea825145f8aa938c97ce226e41 : SUCCESS in 8m 11s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ec65880f888640a1b3c5b998ce4308a4 : SUCCESS in 8m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b56c09aceaeb4ff6b4e10e78cdd10d45 : SUCCESS in 26m 36s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ba104979637747e69e12e29ec6fe3a7f : SUCCESS in 26m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2c57649ebef94631a86d3c19bf86df1f : SUCCESS in 9m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1674dd1e855427daca2dab647919ea1 : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d96f624566889d9af96be2e50dfabce4165110ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:00:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8315560b0b0c2efae641a74233aa4d10d73b5dea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 14:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a295e6b6e8164e04a19315967501899a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21856331de742898dc6bd2e0c043e95 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aeaf7cb5f9be1116e23e19b52f5b50dda2795a7f","revisions":{"9f0f89f28f3c476753ec05343903e75f4f7538c1":{"kind":"REWORK","_number":1,"created":"2026-01-16 17:40:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/973657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/973657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/973657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/973657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/973657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/973657/1"}}},"commit":{"parents":[{"commit":"e76dac8fddb204e243aeffb491356aabc21bc1f4","subject":"CI: Fix multiple issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e76dac8fddb204e243aeffb491356aabc21bc1f4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 17:40:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-16 17:40:09.000000000","tz":60},"subject":"CI: Set previous_release for 2026.1 Gazpacho","message":"CI: Set previous_release for 2026.1 Gazpacho\n\nChange-Id: I64c9960564ad1cf2dbf8ca5d1bc44bc95ce57cf2\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f0f89f28f3c476753ec05343903e75f4f7538c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f0f89f28f3c476753ec05343903e75f4f7538c1"}]},"branch":"refs/heads/master"},"aeaf7cb5f9be1116e23e19b52f5b50dda2795a7f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-22 08:56:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/973657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/973657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/973657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/973657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/973657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/973657/2"}}},"commit":{"parents":[{"commit":"d7686f5f241244b199749428b3d0fb895238b64a","subject":"CI: Re-enable seed jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7686f5f241244b199749428b3d0fb895238b64a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-22 08:56:39.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-22 08:56:39.000000000","tz":60},"subject":"CI: Set previous_release for 2026.1 Gazpacho","message":"CI: Set previous_release for 2026.1 Gazpacho\n\nChange-Id: I64c9960564ad1cf2dbf8ca5d1bc44bc95ce57cf2\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeaf7cb5f9be1116e23e19b52f5b50dda2795a7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aeaf7cb5f9be1116e23e19b52f5b50dda2795a7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
