)]}'
{"id":"openstack%2Fkayobe~973922","triplet_id":"openstack%2Fkayobe~master~I14519b2c511c06b409fe95ec57d42309081575f6","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-20 13:26:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-20 15:41:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14519b2c511c06b409fe95ec57d42309081575f6","subject":"CI: Fix multiple issues","status":"MERGED","created":"2026-01-20 10:15:34.000000000","updated":"2026-01-20 15:43:17.000000000","submitted":"2026-01-20 15:41:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973922","meta_rev_id":"dd4833e5e93b5fc4e83f5f2ba9dc1b2561f80159","_number":973922,"virtual_id_number":973922,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 15:41:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-01-20 13:26:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-01-20 13:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-01-20 13:47:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-20 12:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 10:22:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-20 12:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 12:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-01-20 12:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-20 12:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-01-20 12:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-01-20 12:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2026-01-20 12:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-01-20 12:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"d9f4626f50e21869264ae0fac78dde30c8242483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-20 10:15:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"708d75b47d7937886708e76b60fd669c0c8ed0f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 12:12:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c182dc82d22644de9f73d7d42812ac0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b399d087d0e4a2381b1642e83d65a57 : SUCCESS in 9m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd50bf1eb58842c8b0c101415599bb96 : SUCCESS in 7m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a424fa6df20a49d4968fb3b3b4483f9d : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae1cc2bbee444a2893f4a6a928abd8f : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/118db9afe634414e9444e29e44bd4385 : SUCCESS in 2m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eaba1d71de2c4ca9a8ef4fe3e5007ded : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a12eda3985674cf785c04cc35bc7ab8e : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4055f707e0ed41f18183195afe18e780 : SUCCESS in 5m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/39ba6cad762b4cb6b34a28d1b9b69949 : SUCCESS in 11m 33s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/483633e7f8054e459baca5c9a3cc1d6f : SUCCESS in 8m 31s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/22a120b88395481cb5a0f770a58adb24 : SUCCESS in 10m 42s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5d206c02615a453488fd3e18a23a90e8 : SUCCESS in 10m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56ee3a80005446a0ba9bc3e283bc95fe : SUCCESS in 13m 31s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e2f46deb22044419b3e2971ddf2cd515 : SUCCESS in 33m 49s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f42b3ca5238c42a8bc8c481a4749807b : SUCCESS in 32m 55s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6901e32c79a4435786d72d2417d86199 : SUCCESS in 1h 24m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4b2db966c8f24827a7c8e875e993ec9e : SUCCESS in 30m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/97cb781f564243319a514de141d9cad4 : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/26cc9646ee704b78a783f2bfa1b05201 : SUCCESS in 58m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bc49898cfd8a4f47a7cc54eac33a318c : SUCCESS in 1h 51m 29s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/202fc46fe491476ebcb67cc4de0b1529 : SUCCESS in 13m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/857858328ab946b4b2b9eaff5e415114 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"163afad0d8d68458cc1e1e51d32c826d97d7391d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-20 12:13:38.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"4e5010737b31a5049dfb2a33adfe25eae718e13e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-20 13:26:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6095023c038dffdd2130d297e50c98b30f0f6cb0","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-20 13:47:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d94a2a8b7f631e416688e0068c8da5b5a939036b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 13:48:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edb9d07e0ef81f93b63a881c41ff127c0bc04c04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:41:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a845bcb7ba4f4996bd1ff0d1d332e2a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13b22d78f6142d78f642e432cc33bf5 : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e8bab0cb7f2a4ed18e5464f0e76e81b1 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc55e3b3acfd4ddb9f7a2144e87b8963 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61a00967bfd04cdb98939fdb24349e7a : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f55a8e0d65ec4358a7e65094dbb2496d : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e249c894eb2e4bafb3ae04369cba7dfa : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da17b3796df942a8b0d3bdd06b75b553 : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aee3757ca4a74719afd7fcce44c6a672 : SUCCESS in 11m 19s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4390fb8cd6ef4917844cb5a479c12573 : SUCCESS in 17m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c978d9be86454865b56c97f13a2545c0 : SUCCESS in 14m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/dc8cc075420e461ca53f6f4b307bbea3 : SUCCESS in 21m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4f715caee424b79887adefc704596b5 : SUCCESS in 19m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/d722187ebd034e9eae64e45f0a7f5b4a : SUCCESS in 1h 17m 50s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/0304a738d4e448c097e8e7ffa2e0c873 : SUCCESS in 33m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f58ddeb3de7644708c2305ccb037d446 : SUCCESS in 39m 21s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e549a7ec07a4d789a98dd21201c6322 : SUCCESS in 1h 14m 22s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04c29ece9caf4106bf2aed521dc67bdb : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/baaa13556a224e5c94f69add8c3ed0cb : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3123dc27359e4a2880a57533c928deca : SUCCESS in 1h 48m 20s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1e688f039c344b1d938b6264cd00a977 : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ab29a0abfc541aea31fc9f7d5af3627 : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b9b344fa1d95e961cd833fb44d81ef7581314356","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:41:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd4833e5e93b5fc4e83f5f2ba9dc1b2561f80159","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 15:43:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/606414172c864f3fa0cf12ef56e61cff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10f13c1d8d942b18b49e1438f2f32aa : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4a730fd672c42e9b46cf184ec0747b5 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"60a71c7adae37139fae4556fcac41963a5334fbd","revisions":{"60a71c7adae37139fae4556fcac41963a5334fbd":{"kind":"REWORK","_number":1,"created":"2026-01-20 10:15:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/22/973922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/973922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/973922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/973922/1"}}},"commit":{"parents":[{"commit":"8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf","subject":"Merge \"Bump up Ansible supported versions to 12.x/13.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b64b6e53b9e7afb00735c6e061ffa2fa0cdb9bf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-20 10:14:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-20 10:14:54.000000000","tz":60},"subject":"CI: Fix multiple issues","message":"CI: Fix multiple issues\n\nThe openstack-tox-py313 job runs on Debian Trixie, which does not have\nthe libpcre3-dev package. Update bindep.txt to install it only for\nUbuntu. This package is required for running pep8 and building docs.\n\nDisable seed jobs which are broken because of the addition of an OCI\nartifact registry in Bifrost. This is being resolved separately.\n\nDisable test of baremetal machine inspection: it is currently failing\nwith power off timeout.\n\nChange-Id: I14519b2c511c06b409fe95ec57d42309081575f6\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60a71c7adae37139fae4556fcac41963a5334fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60a71c7adae37139fae4556fcac41963a5334fbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
