)]}'
{"id":"openstack%2Fkayobe~974490","triplet_id":"openstack%2Fkayobe~master~If0352355db140b5c96776a9380af9afa933b9af3","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-26 17:42:12.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-26 18:37:12.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"If0352355db140b5c96776a9380af9afa933b9af3","subject":"CI: Make Tenks nodes compatible with c9s IPA","status":"MERGED","created":"2026-01-23 17:09:51.000000000","updated":"2026-01-26 21:35:09.000000000","submitted":"2026-01-26 21:33:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974490","meta_rev_id":"cf9f874949c94b2bec0404f2d0404cbadbb71380","_number":974490,"virtual_id_number":974490,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 21:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-26 16:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-26 17:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-01-26 16:02:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 17:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-24 10:54:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-26 16:00:27.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"a0ecbe84c3362c3355b1a2ba0eee66526d1f8106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 17:09:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a954d0cba96a5b0295a88e01bbc0b5e5c3c09d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 17:10:39.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kayobe on branch stable/2025.1.  The\nproblem was:\n\n  Nodeset kayobe-centos10s does not match existing definition in branch master\n\nThe problem appears in the the \"kayobe-centos10s\" nodeset stanza:\n\n  nodeset:\n      name: kayobe-centos10s\n      nodes:\n        - name: primary\n          label: centos-10-stream-8GB\n\n  in \"openstack/kayobe/zuul.d/nodesets.yaml@stable/2025.1\", line 8","accounts_in_message":[],"_revision_number":1},{"id":"d9a1d4c6e8fcbc33a65d196c53a7e8fa5f939b6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 17:31:44.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1207998f5391548315d0b1ec5b7a5e5527ca89e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 19:12:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452de55c8402407ea4a8ff5a18b60bf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a1ba3abca3a4eb6a1e0c9062ae2a172 : SUCCESS in 8m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2eea073d486d41f89786b3709382d45c : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9a2f24c7cc04a15a7ccdf09bd2e3bf0 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7528a9b66f444011b492b8f91ab942a1 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25136dc4d3dd4340b613b4230338a026 : SUCCESS in 2m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d2d0fcf38b7490d98e2c9af4442ef5f : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9af730c6ab0a40bb9794088eae39a8be : SUCCESS in 6m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b83ad1491c4f4915b152d8779bd97ad8 : SUCCESS in 6m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/4630c4336f9e4e5e9a9368aa12a57212 : SUCCESS in 13m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/738ee09047f84807bd90e3ce10933fd0 : SUCCESS in 12m 06s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/92ded8abbb5e47c8940251d59064e989 : SUCCESS in 24m 23s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d46baf7e79474c178b937655b562b3bf : SUCCESS in 8m 23s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/df572c9777be454badc6c4942f41b7bf : SUCCESS in 48m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/458fe408d3cd4f31b67e2919870998b1 : SUCCESS in 49m 23s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cb53ca936acf4018ab7c057cbe93feba : SUCCESS in 56m 29s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ac84eea2c2d45f8a495cee33c291d3c : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/388cba5ef6a840beb5183309754fe2af : SUCCESS in 50m 04s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1f982698143c4003b0aca358de5a3550 : SUCCESS in 1h 38m 05s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb08581172e54d84b9e53f5388823d7f : SUCCESS in 57m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1056fa2722c54a64b2ad40531ca75422 : SUCCESS in 10m 31s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0b546ad8c7fd4c95afd1462a245f1477 : SUCCESS in 18m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f923d1eb8b1948fe92a3465f61c0aa65 : SUCCESS in 16m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9cd1f09abdb64312a91003e289f05faa : SUCCESS in 9m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ddde521cf53244d1b8ba347873e1dfa3 : SUCCESS in 35m 01s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d8abc12702a49c5bc601a9da7954980 : SUCCESS in 14m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c72850d2d2ab49b1bdc6f3556df5beec : SUCCESS in 9m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae11e089353345dd97c06493eee366a4 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":2},{"id":"a462111b88303ff5668162a5366b406bb346e384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-23 20:47:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3279ed54bd8969aa6e9caf7eec66d7bddb773aa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 22:16:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27420f72896a452598f5dd09db965260\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b3196a6bdf46e1a7d6de243e31d0af : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d01dacdfa19b43bb8aad0d997b92e3be : SUCCESS in 7m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4d02c36287b4adab51c7f6598775d06 : SUCCESS in 2m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8918523e219a424e8bb0bfcbc1454e5b : SUCCESS in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d70af6407b54f83bd1e74d2db6bbcf2 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/093827d628a048758509cc900297ad1d : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28b3158a75604987b582df4da7b63c51 : SUCCESS in 6m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/921050c4e8cd416081269ed4d5c5dd48 : SUCCESS in 12m 48s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1d1a1cf197684f15b747e877aa8b92a1 : SUCCESS in 8m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/84bc788c431a4939809a7c22a18f36e5 : SUCCESS in 12m 46s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/ee5cacfe8554466fa8da5d0470598fe2 : SUCCESS in 11m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fec845ef3a840ff9d249a56f481cf39 : SUCCESS in 8m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/897c2004f4ed4f75af2560c53b47a877 : SUCCESS in 46m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/f6b1984e779d4b8b96403319fcde8286 : SUCCESS in 46m 04s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/4ceed243e3044100b014d13f9150f685 : SUCCESS in 56m 57s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6700253166249a485f9f902cc5a1de8 : SUCCESS in 45m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5f63b0fa8bf7413db95079730ea44f86 : SUCCESS in 44m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/61d28cb5fc244e73aa02bb85033b4fc9 : SUCCESS in 1h 25m 11s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bec75c86a9314d4fb6ea5732c344f3bf : SUCCESS in 1h 20m 43s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0a250e38e57642da8d0ca74d6031d425 : SUCCESS in 17m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/3b44ccd9fa1644ba9c50076ff4a18329 : SUCCESS in 9m 51s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0025e9aeba9848c18a8aaba216c9b720 : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a7745d1873d14ad59423367efd01beb8 : SUCCESS in 19m 11s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/61d6a066ba79441c81019c28da12b913 : SUCCESS in 15m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a472c4a3b6543be81260e66e0d5b3eb : SUCCESS in 26m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5616e6fb3a50471ea56e56eb327ba634 : SUCCESS in 19m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a55f7808980e447b830e7697e55d83f7 : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c2f77e6cfb3ef69aa67da3e765999aa5dd79595e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-24 10:54:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5cf7ec272417540f3d7b633e0cb18d1d246f04b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-24 19:11:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4660e63f3094c7c4d9d60f0f3ca2c7906baf8cc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 20:37:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6128214dc3c4ac18cc8a349f76bc783\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92621b9fa678419eb5f99de9eda61303 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0bb76e3406fc47a98ed4147404e156ad : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b022a5acf60e493fa527d1fdd75ec6b2 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a9d0a5005ce45f2a2dc8276e70500dc : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95ed534c24c342f89d41ba6c5c09a33c : SUCCESS in 2m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf149bfc25754c6b8825e0eae80dff5a : SUCCESS in 3m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdbd60a221664ab59768a8b06bd80917 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/359cb73a633749d698463eed293dc30a : SUCCESS in 10m 36s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/20afcc10fbc645f48350556a14eda7f1 : SUCCESS in 8m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1686233c2184d13b5de304e7b22c4f7 : SUCCESS in 7m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/89ea6a51a692441a91525cc3d6a9f600 : SUCCESS in 11m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7099405e243942a5bb9dc9ff795d3da6 : SUCCESS in 15m 02s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c432c3b5f8134c3889c4246f0133b6c0 : FAILURE in 9m 59s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cff262967214473fa7efbb679eedf6f3 : SUCCESS in 47m 30s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2aced3c30e714834bbe902b7c157ac1c : SUCCESS in 55m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ca9272f17d14ef5a611c9fdd3a6d651 : SUCCESS in 45m 45s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/514d0ba127c3414cb3ec31147fa984d1 : SUCCESS in 45m 42s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/975448e69ea84091b7c55f4d573f90e4 : SUCCESS in 1h 24m 54s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a486f0ae2f546189752be2db3da130e : SUCCESS in 1h 24m 37s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/002f6bea291e4222bb08b57514d98324 : SUCCESS in 10m 08s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/923294b6d67c47ed889bc73d2f5607ff : SUCCESS in 9m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e76626a67b3d4c9a809128f108ce8f0b : SUCCESS in 8m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ca9237d31f44fa38c09bbd5cc7b0caa : SUCCESS in 8m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2ac0a46085a4480faa7c38c3bc67e032 : SUCCESS in 15m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8f609841c184521843ae525fe88bcc7 : SUCCESS in 13m 27s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c1f941cc3e4a4dc490e6ec4fad753198 : SUCCESS in 8m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b2ee862568f4729ae894eb761ad6d11 : SUCCESS in 13m 22s","accounts_in_message":[],"_revision_number":4},{"id":"8a050fe3a74fcebed58a1857cd2904a5ab146100","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-25 12:27:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5411eaf70e6f942dccfd8bbc4452daba6a1a5c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 12:45:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/436d481787ac4c63a75184c7f2f7cad8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3461c065e394f47b7a3e03b271ac3b9 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e49039fa81c74e82abfc112b8aec22cd : FAILURE in 2m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/803fc2543f6e4bd2a30f56945a4609aa : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f0004d4a5b45919260776c85fe6164 : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a111e384657a4e318b4a27612c2d8993 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c29a2125a5624a619f7051b405a457e2 : FAILURE in 7m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/822746ddcadc4776922aa9c37f4a5930 : FAILURE in 8m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37dac2f6255640af9562069108ad5031 : FAILURE in 5m 26s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/dd85e10f79e844d9a37eb530d3079ba2 : FAILURE in 8m 57s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e31b076dae234f00901822132987f9a2 : FAILURE in 7m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/810fca97b6344875a0f27a08664d440d : FAILURE in 5m 44s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8740e29c01cc449b908767bfb5063574 : FAILURE in 7m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6accb8707dac408ebcdf7597c9e4430d : FAILURE in 7m 36s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5dab0af44d2e416cade6aa4d02fd15d7 : FAILURE in 6m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ee62a938598d42bf992bc5daf11aaf69 : FAILURE in 6m 24s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f13a7b9ba49b4a0fa24e4ffc10094b55 : FAILURE in 6m 17s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6c14dffd5354140ad80aedfcaee881f : FAILURE in 6m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a167afcfcb594d53b89ab33e6b48762c : FAILURE in 6m 50s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12fcd9d6d71c4cb99954107ba9169e4f : FAILURE in 6m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/36d95fb63da64ec990e55f83904e306e : FAILURE in 6m 42s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a65f7f32c176485c93cfb907a7e77172 : FAILURE in 5m 55s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c859986cbfe43288c0a79dd15c9f848 : FAILURE in 7m 45s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/07d4d9b7d5a74666af0936b7b92ad7b7 : FAILURE in 7m 14s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b5ec56e174a141eea580d72000d6a5db : FAILURE in 7m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ecbd507203e48d7bae7736a3bb0cfef : FAILURE in 6m 43s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/352247aa341544919be53bfd9792dda0 : FAILURE in 6m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93538bfd5ffb4c63bd502433d9ba1063 : FAILURE in 8m 52s","accounts_in_message":[],"_revision_number":4},{"id":"abb1e80a3e25ae0790ebaf4e0debd9c40740cdb2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-25 14:10:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50e2011ce0bffb8f756a6b17d54d4be04d39c5a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 14:27:27.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/590c016882f34b7f96b0cf1dcb969ebf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c18c99cbbc64af79078ca726f02c56c : SUCCESS in 4m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95a62fb205ea4cb59abf0465ac8a4a4c : FAILURE in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50c60833ac67496cbc01001ac5e0b9f1 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ad8ba1a3404497986339d0354b093e : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/946aea276170460faf5caf546db9bfa9 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e767dd3bee9425aaae038b447b0e849 : FAILURE in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d85b39a9369443cb63d93fc5e5446f6 : FAILURE in 4m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ddc3945d02f47bc9f3152d5cb76a11d : FAILURE in 5m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/adef0fa5dff145e499f3fe9826d6e6e5 : FAILURE in 7m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c8c169296ce14f4bba4d9c74315bdc60 : FAILURE in 4m 35s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/98e8962f81b14753a81b8c6131fadb38 : FAILURE in 7m 29s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a373ab8e13e540c88f51cd4d57ca465f : FAILURE in 7m 14s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a45b67828bfc49099484f2a08189752a : FAILURE in 6m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/49272da13ae343a1bd1802f86c795b29 : FAILURE in 5m 29s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8359e8de22944a98830f632c0b61b813 : FAILURE in 6m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/96d61e4712774be4956cc3d24a0645ad : FAILURE in 6m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/116cee5c47814b949f4d33d5543dc1c5 : FAILURE in 7m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/efc21002764e49379c3d0febbb21a917 : FAILURE in 7m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01f2c79e59974ca9a6accafe594fd422 : FAILURE in 7m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/944b088391f24133a09658fdf6c16322 : FAILURE in 6m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e5af25ed65614d82bb779d84d93e3e57 : FAILURE in 6m 23s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/693175a9fa9b4402ab6f95788f679fbe : FAILURE in 7m 08s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/430a6cd2a3ac4e00b04d513437404b31 : FAILURE in 6m 51s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/151c8f6adf8b43df978da7ac7557f77d : FAILURE in 7m 12s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef8aa144b3754df9b2bcf51cd780d282 : FAILURE in 6m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2c4bc2ccf0af4a3b8fddbf73b802c80d : FAILURE in 6m 58s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f39757ad116e404385cb463685ffbd14 : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":4},{"id":"6b74e38ebf211fcab68432e355e2072f7b5dea5d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-25 19:37:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"157d86d16fffc8d6728fab5d0d1bb0bcd2612f38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 19:56:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0f55b307a484ec8b0f4c2b93d540287\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84bee010b8f348c38ecf5c8ed98c9859 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a3b3e576314349939084cc445d904b54 : FAILURE in 4m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e92e9558485241c0ae53be03d88516b8 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e538f51e571f4580a95c9a67e7c7d41b : SUCCESS in 2m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e49cffc2027345babcefadf60dd22f9d : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f9ed1c3f6e5640fd8ba602b8352a50dd : FAILURE in 6m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e39d1045aa2f4c6bae9ddee2e0d45bf2 : FAILURE in 4m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2975f8060ee425a87e4d3891b643493 : FAILURE in 5m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ee8a976eb323419cbd9d06da55326ea8 : FAILURE in 4m 56s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0971570bb0994c1bb4254995dac3cd7e : FAILURE in 6m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/e8c5705580a84728aa195386ee799d26 : FAILURE in 6m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14532541b377428eb53c924ff1b7dbf9 : FAILURE in 6m 35s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/038730e29a16457f91602032d6ffda51 : FAILURE in 4m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/3ceab2713a934d628c9ce92d05de30be : FAILURE in 6m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/32cf402cd0694d479f89b9c97e0003d5 : FAILURE in 6m 34s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b64b98c307324cb39dfbbd247b602ffc : FAILURE in 6m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0cbe0e16d740437195970e0a2539603b : FAILURE in 6m 36s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/64faa59c18c94cf89d53d0a316b40854 : FAILURE in 6m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7cbda00cbd674bbb846533ee183b175b : FAILURE in 6m 42s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5d70ff0bd6ea442fbc93761076bc4416 : FAILURE in 5m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9769cb8f23534345b7b00ee5b85f1c99 : FAILURE in 6m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b055ed349b97449293aceb74f1c93390 : FAILURE in 6m 46s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/648a2f12d1f04a22be564b1c087e4fe5 : FAILURE in 6m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c849810e077b4cb69c407d5641b8ce97 : FAILURE in 5m 03s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93213fcf774b4b6db07292b2fb97c1c7 : FAILURE in 7m 42s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/20c9061ba8f14886b0686a4c8ce8bd50 : FAILURE in 6m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d0479b9d7894978985cd589d7544b0f : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":4},{"id":"1a78cb70a45fc3956c121762efe9caf900930fab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 14:27:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd3160396125e2908203040122e49a833d5bf374","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-26 16:00:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"65b1f3f2c8ec5a351a856b4ea221323388c894f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:00:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feae08e857734951988b03bfbc390760\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f28a89ac5540c88bd139e79aba9bca : SUCCESS in 4m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f50056191c58490bbd26baf447335d78 : SUCCESS in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a45ed398f1924ba0852914a1ad828068 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbeae09e3efe47f5bbca3021602145fe : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ee1aeac11474463979e455256a9dacf : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31f9ae12d6cf4a99b01c70bcecf9a185 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e85ac5eff7aa4af4a80013eafd4857e9 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65ddbf9a7d7442e982d5030d7e94cb47 : SUCCESS in 10m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8a8fa340e37f49818d508f8010239e04 : SUCCESS in 9m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a06fa9bca3d24a408d6f3687998413ab : SUCCESS in 12m 03s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0205830413774809ab6e8f8b6e6e1c0c : SUCCESS in 11m 43s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f276405b375a45bfa8caa61b515f90bf : SUCCESS in 15m 30s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/06ea6d3d210345e0af163d8d3c1385ec : SUCCESS in 49m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/7ad1ef717c814631a9100d615ffba90d : SUCCESS in 47m 09s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/a7d34d69409f4fe98d52d9ef1de96ef1 : SUCCESS in 55m 51s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0fd8f240b6b4ae1af2ce50a3410b89a : SUCCESS in 47m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0caffa8b74c041689faaca75183ef737 : SUCCESS in 45m 30s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d1c918741c6d492bb1cdd0fbfab974e5 : SUCCESS in 1h 28m 53s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49b5eeeb7cea403d8190124121ca702b : SUCCESS in 1h 27m 45s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/70000085d37c4a27a7c89c1dcf9aa043 : SUCCESS in 22m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/440deb75fb894b6fb8fc73de4e982b9c : SUCCESS in 10m 54s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8070b77d67a941f09cb2597596e0aeb9 : SUCCESS in 16m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f37a081b3fe14a469b91e17767605b7a : SUCCESS in 12m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c95a487705fd456f9d6763a63aba8db7 : SUCCESS in 36m 02s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69190a85bade4c159444769fbbe30aa1 : SUCCESS in 30m 01s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5a892106acbe48cba25d962d33f7dbd9 : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0e78c74055f475eb188189e8819ff53 : SUCCESS in 15m 18s","accounts_in_message":[],"_revision_number":4},{"id":"ff2f1fe8f098a10bef96713ba3917cbaa11a3a17","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-26 16:02:43.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d81fc4a2422b28cba2c0cb1f16fb7add77332bb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 16:03:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3ecd59ede36382301a299bd494f05e355f87c8de","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-26 17:42:12.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4035d18d1c0ac4b1a4c3ad48790b4f8f82507541","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 18:06:29.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4e82f3562ef4b0e85d2b2a3a1e2e235\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d86f215317849478426bf33a28113f6 : SUCCESS in 10m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a30892d15c6349e0a78461fa677512ca : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee0bb3cab494ab382415683eb1ff2bf : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0a654f0450ed4396843a32c455c24ff6 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7d7a1bae4c9428682213bec1060f4a7 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d043fb6f8be434d91e91eed9799b81c : SUCCESS in 14m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/112304fd37c64fa4bbc036cf825b334f : SUCCESS in 11m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a6bca610c4b245b2bf6c311cd156a584 : SUCCESS in 12m 42s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/539562a212b340ce9ff659253848c7eb : SUCCESS in 14m 13s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2c13250910464e7495bba55a7b6139aa : SUCCESS in 25m 04s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bba2855160e640e2bd8355aca5acfb92 : SUCCESS in 15m 33s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f349b39d71664fe895aab669a66e8f1e : SUCCESS in 49m 19s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/5e6700b2f53848b987965e7ab34dcc05 : SUCCESS in 48m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/13d6ab34dafa499f9c4fa26f987a9ccc : SUCCESS in 56m 03s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19fbe82e5d22463582641eb2a13203ef : SUCCESS in 46m 15s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c3fa938d34b841a3ada9794d541a37ac : SUCCESS in 45m 44s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7dff05ab43be465692d35e016efb4c6a : SUCCESS in 1h 27m 49s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/41d578a1a6bf4ad882a6e2fb86287689 : FAILURE in 3m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd9a6789b03f4c9e8fb449325e9a67e0 : SUCCESS in 1h 22m 43s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2f1949bb902c468ab32ee50acb1f60fa : SUCCESS in 1h 24m 11s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/84365fab94ce4f82b3ed0f9723e4afed : SUCCESS in 9m 33s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/73a21c4d5dac4d7fadda94c5c3033ec1 : SUCCESS in 16m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41a824bf89174535a4ddfe9d779261d6 : SUCCESS in 9m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4d47d4268901411aa941a97381afe063 : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/35c78be02d05494cb5934048b98044eb : SUCCESS in 18m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ac6d493adc10478b8656c0cfa155fb0c : SUCCESS in 27m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aca71abc476c4e1a95dd11b62d21acc6 : SUCCESS in 33m 06s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8c3201b5def4840a5a893abd4f3309e : SUCCESS in 25m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/73bc9dd8a99a41a8824008336fe91b08 : SUCCESS in 15m 21s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47e1cfba1ceb43718089c550f8432ca7 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":4},{"id":"e447fa782a079d2068cc4911fee8f52322b1f5fc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 18:37:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"83d7fee477d6e527b96451c57592546204550eaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 20:06:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b5699b832f443e8b479a3bf022169a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0847a9c7f5940ffbc5971f5402bb696 : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce6a84a762fb45ea90bbab222eda5a17 : SUCCESS in 6m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7159e8299cf847ca920654357a517dac : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701e375e0a4e4f0ebde54d89f70042aa : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ba1827a05f48415b86243ad06fd5df29 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/13707763a5404ffa81ec1a3b2f3bdcd9 : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc9896d0d79e48f2b2f5ff4556cbf048 : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5728fb52af654595845c522d5e2711aa : SUCCESS in 11m 25s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/88cad18a57f04581955e683747488292 : SUCCESS in 9m 53s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e4483f10ff94d65b7c52814d43c082c : SUCCESS in 13m 43s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/eebdb3e2dc6e40a68f4272a6ce8fd0fb : SUCCESS in 19m 40s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c81692f1b5f4b0a87665a04e9fc0683 : SUCCESS in 16m 18s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8cbc06848be0427ab5fcd809915a4d99 : SUCCESS in 48m 04s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b68c2733057844b2b1af26b9546e0e6d : SUCCESS in 46m 51s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f5831f3b52f846fab5427ef09767bfa2 : SUCCESS in 57m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb50de63ede54e57b00bd58d1b4826b4 : SUCCESS in 46m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/106e8b21402b49ecac36f9373fe30ce0 : SUCCESS in 44m 45s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ad66b040e37444aa9a844d45ce0fbf34 : SUCCESS in 1h 28m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8698182d7e3d4e24accd7e3777993e3d : SUCCESS in 1h 25m 50s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eec84f6426d74d23a46e76ccf2d9c5e2 : SUCCESS in 1h 17m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a02dbcdef7924a119eafa20f429c7703 : SUCCESS in 1h 21m 00s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/005fa5f8f15e48cfbff7392398d82a2c : SUCCESS in 17m 39s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9862c12909d547feb5c6cc6bd53bcc2f : SUCCESS in 9m 25s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ead41faffe442778061621420548daa : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a225919d7cd24c3699d9df5db4a1bdd0 : SUCCESS in 14m 54s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6e1bacdfbc594e9f805735f76485b489 : SUCCESS in 23m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/432caea01fc34e8e9673430e28804d5b : SUCCESS in 34m 56s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4c4ccc8276841d3a260a01a4fbc1179 : SUCCESS in 34m 39s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f8082d5b3b834b2192d8fe10539d3c63 : SUCCESS in 27m 35s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8ea5769fc6a0450a93a122a6248286b9 : SUCCESS in 19m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c25e4f52b96f41bfb48c2c9837b79a4b : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":4},{"id":"c7e5172db085940c4759f670460c64ea246545b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 20:07:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"19e69a1869114c306ad8c1c3f943a5f05941126c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:33:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16f5b04067164e81944b7fa957882868\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e195cde8e94bbfbd06a61e749ffb4d : SUCCESS in 9m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ed4cc52809c44789a054d187b7ebe6a1 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf3b622e263435bb3b382ada9a78d0b : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb2e9d6fecfa414da5e2426936fd0548 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/900597ac06c6449fbb1bcf615d330dde : SUCCESS in 2m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38d83eff310242368ded124ac5baa66b : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/084586df00284df8b87323fb774da076 : SUCCESS in 11m 41s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/745bdbdfcc9b4c2ebca9b9cd9dbd80f3 : SUCCESS in 8m 11s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd396d9d0fc24d4e9d52c08b8fdc63ac : SUCCESS in 6m 51s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0d69031133664bb98b70b3539efccb88 : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9068fa3e06046c59faff63c3dc9d419 : SUCCESS in 19m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/47b559f46df44874ac9626c67736855c : SUCCESS in 46m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6d38b58d648e40ee8a1a8542fcd3f87f : SUCCESS in 47m 13s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/86cacc7958d94c02b56da56d73edaa93 : SUCCESS in 55m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8b3350a480344383bfaef72a8574787c : SUCCESS in 43m 51s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/21d7894050d0483a8267b7132a377bb9 : SUCCESS in 45m 24s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a803acd151844e6bab1a8a8055ca0372 : SUCCESS in 1h 24m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/497deaebc1db493ebff7562d79357bfc : SUCCESS in 1h 22m 44s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76bdfdc4f13841edaf9bf721be1dc2ea : SUCCESS in 1h 20m 06s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6d04e6f47a0d42af8ce96c0ee6a7348e : SUCCESS in 1h 21m 33s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5666537a294b493d9c5411cdd6781aec : SUCCESS in 24m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/81eb673b06ce4d99901468c0ad26b84f : SUCCESS in 10m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aa24e9ef7ec4cf6a9e89f16cbd246a1 : SUCCESS in 19m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ff9f547bc0742b7b3e2bf710c9721bb : SUCCESS in 15m 25s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7eb282cfabca46d2b5b8b39dd3d39761 : SUCCESS in 35m 08s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e3022362e0fe410c8a42e5b1f9934375 : SUCCESS in 36m 01s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6187355a0b374f58af35ef9e8c37fe4b : SUCCESS in 29m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e82577fb45a4a838c6d31727d42b6ee : SUCCESS in 27m 08s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/52d0dade32d7439da96f4bcb6b080c77 : SUCCESS in 18m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b27c2aa28c73493ba92e5366d8c03bb7 : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a7b0c2e4248b59a67d50734b30d7b5faa44669cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:33:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cf9f874949c94b2bec0404f2d0404cbadbb71380","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 21:35:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da88100841124470b3a21c3cb462f0d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355e390a65104a7fb8b047771a794e5a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"32214e2b6b14fa834e8d210763b14bae7678e771","revisions":{"ec294e2a89df9e02267dabe14e90e103bbd55e3e":{"kind":"REWORK","_number":1,"created":"2026-01-23 17:09:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/90/974490/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/974490/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/974490/1"}}},"commit":{"parents":[{"commit":"7ad864048f10f0366d088440aec885307711957d","subject":"Merge \"CI: Set previous_release for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ad864048f10f0366d088440aec885307711957d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:08:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:08:09.000000000","tz":60},"subject":"CI: Make Tenks nodes compatible with c9s IPA","message":"CI: Make Tenks nodes compatible with c9s IPA\n\nIncrease the Tenks nodes to 3 GiB ram and force `cpu_mode: maximum` so\nthe CentOS Stream IPA no longer errors with \"Fatal glibc error: CPU does\nnot support x86-64-v2\", like in kolla-ansible [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/965244\n\nChange-Id: If0352355db140b5c96776a9380af9afa933b9af3\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec294e2a89df9e02267dabe14e90e103bbd55e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec294e2a89df9e02267dabe14e90e103bbd55e3e"}]},"branch":"refs/heads/master"},"44983df575a42a811663cd3bd83aea8956d37245":{"kind":"REWORK","_number":2,"created":"2026-01-23 17:31:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/90/974490/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/974490/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/974490/2"}}},"commit":{"parents":[{"commit":"7ad864048f10f0366d088440aec885307711957d","subject":"Merge \"CI: Set previous_release for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ad864048f10f0366d088440aec885307711957d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:31:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 17:31:14.000000000","tz":60},"subject":"CI: Make Tenks nodes compatible with c9s IPA","message":"CI: Make Tenks nodes compatible with c9s IPA\n\nIncrease the Tenks nodes to 3 GiB ram and force `cpu_mode: maximum` so\nthe CentOS Stream IPA no longer errors with \"Fatal glibc error: CPU does\nnot support x86-64-v2\", like in kolla-ansible [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/965244\n\nChange-Id: If0352355db140b5c96776a9380af9afa933b9af3\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44983df575a42a811663cd3bd83aea8956d37245"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44983df575a42a811663cd3bd83aea8956d37245"}]},"branch":"refs/heads/master"},"81d7f0df4beb189fabc3f7cfb9a94d80ef36f505":{"kind":"REWORK","_number":3,"created":"2026-01-23 20:47:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/90/974490/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/974490/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/974490/3"}}},"commit":{"parents":[{"commit":"7ad864048f10f0366d088440aec885307711957d","subject":"Merge \"CI: Set previous_release for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ad864048f10f0366d088440aec885307711957d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 20:46:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-23 20:46:45.000000000","tz":60},"subject":"CI: Make Tenks nodes compatible with c9s IPA","message":"CI: Make Tenks nodes compatible with c9s IPA\n\nIncrease the Tenks nodes to 3 GiB ram and force `cpu_mode: maximum` so\nthe CentOS Stream IPA no longer errors with \"Fatal glibc error: CPU does\nnot support x86-64-v2\", like in kolla-ansible [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/965244\n\nChange-Id: If0352355db140b5c96776a9380af9afa933b9af3\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d7f0df4beb189fabc3f7cfb9a94d80ef36f505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81d7f0df4beb189fabc3f7cfb9a94d80ef36f505"}]},"branch":"refs/heads/master"},"32214e2b6b14fa834e8d210763b14bae7678e771":{"kind":"REWORK","_number":4,"created":"2026-01-24 19:11:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/90/974490/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/974490/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/974490/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/974490/4"}}},"commit":{"parents":[{"commit":"7ad864048f10f0366d088440aec885307711957d","subject":"Merge \"CI: Set previous_release for 2026.1 Gazpacho\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ad864048f10f0366d088440aec885307711957d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-24 19:11:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-24 19:11:08.000000000","tz":60},"subject":"CI: Make Tenks nodes compatible with c9s IPA","message":"CI: Make Tenks nodes compatible with c9s IPA\n\nIncrease the Tenks nodes to 3 GiB ram and force `cpu_mode: maximum` so\nthe CentOS Stream IPA no longer errors with \"Fatal glibc error: CPU does\nnot support x86-64-v2\", like in kolla-ansible [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/965244\n\nChange-Id: If0352355db140b5c96776a9380af9afa933b9af3\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32214e2b6b14fa834e8d210763b14bae7678e771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32214e2b6b14fa834e8d210763b14bae7678e771"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
