)]}'
{"id":"openstack%2Fkayobe~974778","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~If0352355db140b5c96776a9380af9afa933b9af3","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-27 12:38:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0352355db140b5c96776a9380af9afa933b9af3","subject":"CI: Make Tenks nodes compatible with c9s IPA","status":"MERGED","created":"2026-01-26 21:46:16.000000000","updated":"2026-01-27 12:40:41.000000000","submitted":"2026-01-27 12:38:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974778","cherry_pick_of_change":974490,"cherry_pick_of_patch_set":4,"meta_rev_id":"42278586261834928e9b1cbd719c21a24f426aef","_number":974778,"virtual_id_number":974778,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 12:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-27 11:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-26 23:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-27 09:32:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-01-27 11:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 23:18:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-26 23:23:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-27 09:32:21.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-27 11:00:56.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"0a2d634f9c6fcf7e4e5ed55a38c7ea686f29e224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 21:46:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9cc39c11736a7a6d07de24356798e0e57cf44312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 23:18:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c193980d3d544d5faaecd1d9d75defcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f60c7a3453e4c7088292239f133107a : SUCCESS in 4m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/112b3131b5c94b9b91364c3100b07957 : SUCCESS in 6m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5c100d6748846b09b147b5404ad8145 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f8e42e88104b8a939c9f17a2fdc499 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76f650f28e054556b362a6d60a46bed9 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb36c7d8789245a5968aac4dd098eadb : SUCCESS in 6m 48s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/795aae3578d049d4b675d60e5908d0ed : SUCCESS in 3m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd515a27fbe8438a997e7a214cb37638 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7994226da0e74f8d8a9251c61b45ec65 : SUCCESS in 11m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a3fc3a59907642c1bdd4cae359a7ad89 : SUCCESS in 8m 44s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/620ae2d8e3ce467cbda0ae0dde599594 : SUCCESS in 17m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/f61058d8085f44f1a667495491442808 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f4b882ab8067444bae9de547d138ad43 : SUCCESS in 14m 59s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f95d05049bd14f6cb667ed3e0a69984f : SUCCESS in 56m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/655d6721f7f448c2b7e87502329cdb14 : SUCCESS in 58m 54s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/69d6f489eca5443ca9134eb355e6d1de : SUCCESS in 1h 00m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/736073ce002e423da4bee07b9957603f : SUCCESS in 55m 55s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d932ea8de25d4496953b430d7dc79fbd : SUCCESS in 57m 29s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1fb01e86de184478ac1e61af0a360fda : SUCCESS in 1h 29m 25s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ddffd50b6b4d4187ba44bc3cfe3927ca : SUCCESS in 1h 23m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d2b55fbd07aa4d5ea9ea51e838709bda : SUCCESS in 9m 38s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/22ead2923c414a8bbbb80ab1ea9e13f8 : SUCCESS in 9m 26s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df467c7036af4a459b84ef3aad94f71d : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/00b2de52012347249778361d39fa5c8d : SUCCESS in 18m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6c81621a30dd4f6085203116ea0c035b : SUCCESS in 24m 24s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/243e63f6a0b6400ea71aa6306b1f49a7 : SUCCESS in 27m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/62a8ad0033a3499f9619be1323469521 : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d97ee94f43b4966964f8c07b81043e9 : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"aea9aeecdfed04793afe56bc42539acf05968d4d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-26 23:23:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e6899c070c0aafdaaa16be852f2ab5e1d172a914","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-27 09:32:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa2caec25aebf648de374e1808f739297efded5a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 11:00:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f274fd4498f8b50c1d46a400af36101a17826f58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:01:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4782b2f49346f87192062815eb3f33542cbdbeb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:38:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b54a920edb3403fa26019b84e69989a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5633b6b5f64c48c2aa15aef3681389e8 : SUCCESS in 6m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7ba548d5f5345ef83507bc0efd1e2d6 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d9a54c0bd04a959ce0a878d9b034cd : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e639dc7845f47b5a4851cb99380e4a2 : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8b9519705ce4761b4b06497201ffc02 : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/608174321d984ce2abac4902060a58a7 : SUCCESS in 5m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e555d1717bdf4d0081cab42fe4a2ec0f : SUCCESS in 11m 12s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ebb179221d04cb0aebae31ee317d8ac : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4262feecafb4ff49b274435815b203d : SUCCESS in 13m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/fb5c4ae27698493581081aa3e57c837c : SUCCESS in 11m 55s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cabf6ac1f0da44efafe015102e65b0fa : SUCCESS in 17m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c93bc0baf4884da39a5a7490171a880e : SUCCESS in 51m 41s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9018333c307f4c308a7b3a07aca1776a : SUCCESS in 57m 35s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/22ad44fe9d724c85af75edfd4ff63dab : SUCCESS in 58m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ed418023c4840a699bf1f1ef174f5a7 : SUCCESS in 50m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f8024510030a4bab9ef3bdb2fb15eae8 : SUCCESS in 55m 52s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a81624a098cb454e81d9b4986c79e5f3 : SUCCESS in 1h 36m 46s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73a2f033560a4d18a2e01765b06fbcbc : SUCCESS in 1h 17m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e41065e75ab244a687dcbbe5bdb1a3e6 : SUCCESS in 11m 40s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/d69549c006d34662af2bf9cb057a5c8f : SUCCESS in 16m 30s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/acc8ed8b322c456c825fdb3c9e0af271 : SUCCESS in 20m 37s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7938f500993b47af98f62a7dc5efe137 : SUCCESS in 19m 19s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d5fa3c86d2f84aaeabd67b92eff4e580 : SUCCESS in 29m 31s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a115f64b5aa843a48431cb82927592ad : SUCCESS in 33m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/82c445c066cc4559a5234cf2a1df6281 : SUCCESS in 10m 20s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dfe53215bac24ce388933500dd52ba75 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c2223ee760a2664399229721f1fa2679193b4891","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:38:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42278586261834928e9b1cbd719c21a24f426aef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5714257282a49b689d199a1974ebe94\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c1de348d474455891c2c1815d7d28b : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ae230ece61d348e05c49079b7ac5c0bff1f7849","revisions":{"3ae230ece61d348e05c49079b7ac5c0bff1f7849":{"kind":"REWORK","_number":1,"created":"2026-01-26 21:46:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/974778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/974778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/974778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/974778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/974778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/974778/1"}}},"commit":{"parents":[{"commit":"2cf8ac7ca52ca1d4b5935d135b51ddb071784445","subject":"Add infra-vms to ntp group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cf8ac7ca52ca1d4b5935d135b51ddb071784445"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-24 19:11:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-26 21:46:16.000000000","tz":0},"subject":"CI: Make Tenks nodes compatible with c9s IPA","message":"CI: Make Tenks nodes compatible with c9s IPA\n\nIncrease the Tenks nodes to 3 GiB ram and force `cpu_mode: maximum` so\nthe CentOS Stream IPA no longer errors with \"Fatal glibc error: CPU does\nnot support x86-64-v2\", like in kolla-ansible [1].\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/965244\n\nChange-Id: If0352355db140b5c96776a9380af9afa933b9af3\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 32214e2b6b14fa834e8d210763b14bae7678e771)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ae230ece61d348e05c49079b7ac5c0bff1f7849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ae230ece61d348e05c49079b7ac5c0bff1f7849"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
