)]}'
{"id":"openstack%2Fkayobe~974783","triplet_id":"openstack%2Fkayobe~stable%2F2025.2~I3971d05f3c09a6841f6666a18f2d2b44b226411d","project":"openstack/kayobe","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-27 12:38:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3971d05f3c09a6841f6666a18f2d2b44b226411d","subject":"Support swap configuration for all host groups","status":"MERGED","created":"2026-01-26 22:19:29.000000000","updated":"2026-01-27 12:41:37.000000000","submitted":"2026-01-27 12:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974783","cherry_pick_of_change":974044,"cherry_pick_of_patch_set":4,"meta_rev_id":"e489ed78b32350d54e89a8523f49db1ae7452915","_number":974783,"virtual_id_number":974783,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 12:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-27 11:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-26 23:24:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-27 09:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-01-27 11:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 23:24:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-27 00:30:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 09:30:33.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-27 11:03:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"cd6e4c40d3704b8130c349780207813c2dc0d156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 22:19:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d0cb76941042c8247e259d54d369fe51dc26fccd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-26 23:24:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cde046c34e012c9a8e9daa791853234e71a8caa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 00:30:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24b3c3ae14cc46bb940f5b5f4e36590b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4145212fc7e0468f89ef11e54f19f6a9 : SUCCESS in 9m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e756751b10af4dab8c9fc03beaab3363 : SUCCESS in 6m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1271905b00234b52ac4bad777149916f : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8fb507aa64c44f2bcbf16d9c3ed041d : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9787c38ba3b4577abeca282b53b3e34 : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/627d315a0fea46b7ab202b7d2c4425aa : SUCCESS in 6m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c365c694a09d47588d42113e179d1561 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b14dd04b8cfa472289bdc93dd6d28f8d : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a08796ac34f84a868c456c82eba046c6 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa7db625e0bb45ccbfc29ac811b38326 : SUCCESS in 6m 04s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b1a3ca4bcd524e05837b3fd24c56925a : SUCCESS in 8m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/114d8037b05d43468ac6d4858b708b89 : SUCCESS in 12m 38s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7f9731aa823c4f9fa49e4122ea7815c1 : SUCCESS in 11m 28s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6c3f30cc95949ca83c33f112acb728c : SUCCESS in 15m 42s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2fedc0a181e54acca51ace72cbc645af : SUCCESS in 37m 26s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/b4b8ddcc484a4f3ab821235a86f783f6 : SUCCESS in 38m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/5a12a68790e9453085aed2e1561579b5 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff27b9c5f56843cc905138623c41d1a5 : SUCCESS in 1h 19m 29s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fea66daf7ea945eca8433b1eeb5c9b37 : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70bc25cdb9e5451c8e5d5c8a74a2a503 : SUCCESS in 2h 00m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f05e53a6c57f425fadbde20cf142fe0e : SUCCESS in 1h 54m 23s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/3d424d7ed8634b34a3f526696106e0c6 : SUCCESS in 15m 16s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4fc37e40ff9645dfb40152eb8135311f : SUCCESS in 16m 19s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/718ba1a28515438f855c573973736b41 : SUCCESS in 17m 15s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/232e3ce58264481aafe29f5cfaaf47d0 : SUCCESS in 8m 28s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e89f5022c9e4464daa7da7790e1ca531 : SUCCESS in 17m 35s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcbd6808f7e1421aa4a387bd06d024ce : SUCCESS in 28m 53s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/99379e8a7f8a447b9bcf169e6549859b : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/519999bb3eb3408fbf7514856c5a83e2 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f20311fca964baaf544e72b16f2462547bd8013b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-27 09:30:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb2a549392c7fd8367efc069006253abe91b224c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 11:03:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b887184739fec1e0d238ec9be2a8a84bb5b9d62c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:03:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e22280fb3e170b72986c24e2613775dc8a702660","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:38:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175f32ae39cf4d1b90ecbe844bc1fc6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cc709943de74fe9b310485021902aa5 : SUCCESS in 8m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/897efc247cc84433aa5c6a4f49028f3d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30e68ca10a94560b07819c785704a9d : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9cfe01e77799408495fa91d7b9133382 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73b5aa69f16c4354a730957029cb6395 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/febe0a569b204287a0c464a1ebf0fe0b : SUCCESS in 4m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc4ec06d841644788f0c64830f0ac216 : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b94c1cd3fe64cd7ac73b13ee495130d : SUCCESS in 6m 05s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f43bd7c3e9d0461ca59795d0663d664f : SUCCESS in 15m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5eef4c03f383465aa10673f364a14b71 : SUCCESS in 7m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/32eeebdaae964dd497fca3062e8e31f3 : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/02dd1786faec4c00a98efd35ebf39537 : SUCCESS in 15m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6554cc21f2574550b6b74e1fa547b5c7 : SUCCESS in 51m 52s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/c51ebd6e53be479f913c1a59224d3834 : SUCCESS in 52m 33s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/ef49df3f990a4022b60e117b47796323 : SUCCESS in 55m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/addfbce081044685a78982f413f253fd : SUCCESS in 51m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/04169396ebbe42b7b2cef8bfe01747fd : SUCCESS in 52m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/56b59ee9c4d745848f4983ed50bb4801 : SUCCESS in 1h 20m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ae3539a861e4b798c1ed54aa9fcce73 : SUCCESS in 1h 28m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/10f99fbab3f447d69ecfb6e967855a70 : SUCCESS in 14m 37s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/9d7dc5db98b04dc5a65d9963e1c3d985 : SUCCESS in 9m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/44849942a6674a10868cbd560180085a : SUCCESS in 16m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/77ded85881604b14892c58edab8eae31 : SUCCESS in 15m 18s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/12dcfd220ad346698077ce16327b8ea2 : SUCCESS in 23m 13s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06bc95404e1943e5a129392016cb4f86 : SUCCESS in 26m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/35c77b31b45945b19dfcac915b9355ac : SUCCESS in 16m 19s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49b44e19efcb4e408e0fcb78f1cc56aa : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":1},{"id":"56fd35a5a1c65ba4f8f30ce6ddcd01e0621d0c1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:38:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e489ed78b32350d54e89a8523f49db1ae7452915","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:41:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bda70bc3e3940429315dd30ef2777c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e19d202bfb4544b1f50602b142c78b : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d068d404c44846c9a822aa579910ccce : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0364c5612a00345a13d8e022289a3409f9203c63","revisions":{"0364c5612a00345a13d8e022289a3409f9203c63":{"kind":"REWORK","_number":1,"created":"2026-01-26 22:19:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/974783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/974783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/974783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/974783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/974783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/974783/1"}}},"commit":{"parents":[{"commit":"2cf8ac7ca52ca1d4b5935d135b51ddb071784445","subject":"Add infra-vms to ntp group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cf8ac7ca52ca1d4b5935d135b51ddb071784445"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 11:56:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-26 22:19:29.000000000","tz":0},"subject":"Support swap configuration for all host groups","message":"Support swap configuration for all host groups\n\nAn earlier change [1] added support for swap configuration, but the swap\nplaybook was only imported in ansible/overcloud-host-configure.yml.\n\nThis completes swap support for the seed hypervisor, seed and infra VM\nhosts and adds missing variables in etc/kayobe/*.yml files.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/851195\n\nCloses-Bug: #2138800\nChange-Id: I3971d05f3c09a6841f6666a18f2d2b44b226411d\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 6b4028754321b3e20b45f93b6fba159985e850af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0364c5612a00345a13d8e022289a3409f9203c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0364c5612a00345a13d8e022289a3409f9203c63"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
