)]}'
{"id":"openstack%2Fkayobe~974784","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~I3971d05f3c09a6841f6666a18f2d2b44b226411d","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-27 13:48:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3971d05f3c09a6841f6666a18f2d2b44b226411d","subject":"Support swap configuration for all host groups","status":"MERGED","created":"2026-01-26 22:20:05.000000000","updated":"2026-01-27 13:49:54.000000000","submitted":"2026-01-27 13:48:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974784","cherry_pick_of_change":974044,"cherry_pick_of_patch_set":4,"meta_rev_id":"3a70541cb94fe7781c43aeff2b74eb9610563bd1","_number":974784,"virtual_id_number":974784,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 13:48:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-27 11:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-26 23:24:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-27 09:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-01-27 11:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 23:24:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-27 00:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 09:31:14.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-27 11:03:24.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"b5e979dc3a8d802272a43729497589393304720f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 22:20:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"806642ec26a6e447c9d4d4f0df71832ddaa1f47d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-26 23:24:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e8976b348a67320cf46f93fc38558b65f5733f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 00:50:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/877c04cb91044da1a73cce17c0794260\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/014a56523f96494e825fbb6a23ace062 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf35618e4b54070aec70a737f312f98 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c40bf9a2619749c6a0ce2b82884c41b5 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f24ebd5df394ddf8b5768b1ef8c4a13 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58f1832a6d5448189ea313e4f0eb12b8 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82967e7bc36241b9a9921f53a2990677 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7c1dfdeac401494a85d32968eda7fdb4 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/658560fbe69144b4b9878f8d31d6651d : SUCCESS in 12m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ef6e78cb757463f81e158c33254d9d8 : SUCCESS in 18m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a34dff2dfec49c09133d57d1ee09666 : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06571298243a48f2bc7d547c9788dc37 : SUCCESS in 6m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/0e3d531ae4454faa8fb27627e2244796 : SUCCESS in 10m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5cdb117743ea4653b873f437014cfc32 : SUCCESS in 20m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/066900c9f676400b8a71dd94808687a8 : SUCCESS in 17m 46s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/6fbf338b915f45f7909502786df29db6 : SUCCESS in 55m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cd19f7264abe4a8c81389d17f4128bac : SUCCESS in 35m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6e47e3c18d6845c9afc5559322fb2043 : SUCCESS in 1h 02m 47s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/eb9f776c3bdf4fa8877e7744a1709883 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c9efbc5ce13540c7a039c41fc7b5506c : SUCCESS in 39m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6894cbee7bd649a68fef5461c8455937 : SUCCESS in 36m 38s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0d32d4f69df4037be7d52aa2c8de194 : SUCCESS in 47m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/66e42469616b406485d28c2e73f78742 : SUCCESS in 30m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/46c5f3b2ac5a465fad80eaa2cdce82f8 : SUCCESS in 2h 14m 41s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/e21967fdcb05406ab197aac8391caec2 : SUCCESS in 2h 20m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62004d471f71476ca8eaee86e1b8f44e : SUCCESS in 2h 17m 36s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7abe305ed2d443d0bfefadcd06012cc7 : SUCCESS in 1h 15m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/89d8ea414eb1420fae579cc1b8c08c56 : SUCCESS in 8m 59s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/f16cd0f871cc428db9acc969eb6af49b : SUCCESS in 8m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4fc78dfaf8db42f0a70ec6e79014daa8 : SUCCESS in 10m 59s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/b3bce7968a704b148e8cfdea9869a024 : SUCCESS in 9m 15s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e05b92190abd4e6c930133864e4db471 : SUCCESS in 15m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/622add0791864f86ac6a63254c212629 : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8148f13b4e804c7c9ffd1b83342c49c4 : SUCCESS in 16m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/9b096d308a05460f99b1e398b06757a8 : SUCCESS in 34m 04s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb4000a417264aed8a466d558a197e86 : SUCCESS in 27m 36s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/55120570b8024218981f0f8b3d6124de : SUCCESS in 13m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/edf9eb63a3b349ef999854e66f49789b : SUCCESS in 15m 55s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4cb0b58076a64a56a9e8b285f5f17c5c : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f385ac218ccbdda12cd42aa2f244368d89d814f3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-27 09:31:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"21d41d41d6e1490a9a793656c73d30e171af9c9a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 11:03:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27c711a792f3492413ae37acc24504659709e37d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:04:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8a13c83e373d171c27be73d36fe484864a0f4a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:48:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fd02f356ddf4744a22530eb9d55332d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f33274d9b62f41c49a7c88e52508396b : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b4f0cd97ff744dc8ca00315475f1621 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2c381172d0495aa9eb900ee812683b : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba5017318c18414b92f23e5cd5c7db38 : SUCCESS in 2m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85e815b557cb46e38d1208f5b99156b5 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42f0ec1132614ef1be1ad3e8f0a3c5db : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6b16d7c524c43f58644b26bc4c9a25c : SUCCESS in 5m 45s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d28f8b7f4fbf4978a937f001a9c980de : SUCCESS in 12m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2f99c3eb988b4deca5b99f314c457aa0 : SUCCESS in 13m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/21acdcaf53ec47afb32b348bdf60ad2e : SUCCESS in 11m 17s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/01f412bfa5724445b289c0f3b0b05e23 : SUCCESS in 11m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3db72d92de724c2db9cb32c9f45f2f06 : SUCCESS in 18m 56s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/985aa66ee3f348f88c578453b4032a63 : SUCCESS in 13m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/97f169c509b54331ba70b45b06426224 : SUCCESS in 32m 21s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9487cf8df3904c42a241464734a19ca6 : SUCCESS in 34m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d586ac07873d49d9a51d8ab7c689edd3 : SUCCESS in 1h 13m 02s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/de51cb3caae8481793ae08565051a265 : SUCCESS in 1h 14m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/8a0b134a2eea4e5f9641621dcdb1aef7 : SUCCESS in 40m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/18c6bc88945443849a103ee5ce4f6f52 : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b23b84e9a75f4727afd9c22b16e99a24 : SUCCESS in 54m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d57e877533f4be694676d93aa4fc39f : SUCCESS in 1h 06m 34s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/562d95d2f2644ceca87a5e6a3402fed4 : SUCCESS in 2h 24m 03s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/a231c69eb2f94cbcbacd2d039dba3bf0 : SUCCESS in 2h 36m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6517e1fc9f904af7b4d806eb9451c36b : SUCCESS in 2h 02m 48s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40abbcc4c7a04d5c940952442ea5edc8 : SUCCESS in 1h 56m 18s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/98c5e0ce71a349dab72b7b973ef00b7a : SUCCESS in 21m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/e884041971454fa9aff4562551839c29 : SUCCESS in 16m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2497aa769cd14c23bc3a60dd388b1393 : SUCCESS in 20m 59s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/c8702973a8494f1aa9367fad255e9e1a : SUCCESS in 21m 52s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52cdeca741e8406b9330157f09a55f75 : SUCCESS in 18m 55s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/01aee14572fd4f009d5f757664ba3bae : SUCCESS in 13m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/39f1355bf68b46ee9282fd2057298a94 : SUCCESS in 34m 57s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/d4cb3843bbd643eb94322d932c84f602 : SUCCESS in 38m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/236b080e760b4d67ae0e41508fb59d3c : SUCCESS in 32m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8926bbb1185445c58c9d6d33eea0994e : SUCCESS in 15m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9420d84937c94c8db60a921c2111084d : SUCCESS in 9m 14s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81dc31e8a843461ba1ae33c6316d75c5 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"416519d53eeda6a225102ba25275586d6f4d542c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:48:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3a70541cb94fe7781c43aeff2b74eb9610563bd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:49:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e749514e5943e2b2be10dcc28781de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8957773ec44132b13c5f057f978020 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b60340cee67a40e5bb7291467108dfb3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6f76976563dd0a89497f8b4cd15113579ba07ad","revisions":{"e6f76976563dd0a89497f8b4cd15113579ba07ad":{"kind":"REWORK","_number":1,"created":"2026-01-26 22:20:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/974784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/974784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/974784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/974784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/974784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/974784/1"}}},"commit":{"parents":[{"commit":"010aff4d2e75dc0aff921658d9114149e1d69457","subject":"Add infra-vms to ntp group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/010aff4d2e75dc0aff921658d9114149e1d69457"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 11:56:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-26 22:20:05.000000000","tz":0},"subject":"Support swap configuration for all host groups","message":"Support swap configuration for all host groups\n\nAn earlier change [1] added support for swap configuration, but the swap\nplaybook was only imported in ansible/overcloud-host-configure.yml.\n\nThis completes swap support for the seed hypervisor, seed and infra VM\nhosts and adds missing variables in etc/kayobe/*.yml files.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/851195\n\nCloses-Bug: #2138800\nChange-Id: I3971d05f3c09a6841f6666a18f2d2b44b226411d\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 6b4028754321b3e20b45f93b6fba159985e850af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6f76976563dd0a89497f8b4cd15113579ba07ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6f76976563dd0a89497f8b4cd15113579ba07ad"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
