)]}'
{"id":"openstack%2Fkayobe~974786","triplet_id":"openstack%2Fkayobe~stable%2F2024.2~I3971d05f3c09a6841f6666a18f2d2b44b226411d","project":"openstack/kayobe","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-27 14:48:11.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I3971d05f3c09a6841f6666a18f2d2b44b226411d","subject":"Support swap configuration for all host groups","status":"MERGED","created":"2026-01-26 22:21:33.000000000","updated":"2026-01-27 18:21:16.000000000","submitted":"2026-01-27 18:20:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974786","meta_rev_id":"32d6a4f38f2b72036db9757c864904482ce72168","_number":974786,"virtual_id_number":974786,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 18:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-27 11:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-26 23:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-27 09:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-01-27 11:03:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 23:24:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-27 00:31:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 09:31:35.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-01-27 11:03:30.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"f956a1e36d8e0b20b1b984d62968a503ad36f7c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-26 22:21:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6763e413b8e6e0e622a1b21f40b5c3fbd9d19c7f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-26 23:24:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6ea893c29ec7153ccabf6b221b079ff934f78457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 00:31:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5adef061aa34ead82375a5a101c6322\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72714c128158432eb2b51f4b4f60e226 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbd66738a5e4bad85ab7ec04eb83e8d : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4edcbbd953f4b2ea076266513b40589 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fabcde19e25b4d818662cfcbdec33440 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895079e3944d4926b7cb5cc6c738f5cb : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61fe160246f74b22813a5b8aa13d199a : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/974f1f1c02e7421b87a24d906a8fdd1c : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9789981c920a4e05a5bb41e3961b0f56 : SUCCESS in 5m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc5ba35f44a44ac78855900e289b5eca : SUCCESS in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bccba3fa2a7340329a83e4846532fe42 : SUCCESS in 59m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/df3777bdf368494a9b83c10c11c60773 : FAILURE in 7m 53s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b68b2bd9c279413fa7fb64655c8b5319 : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5180b2c30de74cc6a4926ab5aee75e49 : SUCCESS in 1h 15m 46s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d361004e54964c2db31f75b2d8999f20 : SUCCESS in 25m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f9b32157a9bb4be7a377f8e80fe54ec3 : FAILURE in 9m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/274f9bfe73214ec8983cd8f299493c53 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8342e21b67f544b6952f585241dd95e2 : SUCCESS in 1h 58m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbd141eec9d34e71aa3cbcd1c96ba48c : SUCCESS in 1h 31m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/408190f189b546718827a778fc049619 : SUCCESS in 17m 21s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81bd73bb575d45638069f252944f950e : SUCCESS in 19m 25s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d98e9a03bb054ba38cc5058d71a8073b : SUCCESS in 34m 02s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9360de113b4644a098d9fa928e9b9093 : SUCCESS in 33m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8883d714b436440b8e250b36e698b6ad : SUCCESS in 18m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e701519a59a54a048d73bc24a2bdd775 : SUCCESS in 16m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb5bed3b57494727936df583c5925863 : SUCCESS in 18m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebee9d04d52145bb9bb9d0224457634d : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2112d8e6f692b373a8454f1ce7cc6ab277cc8824","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-27 09:31:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e801ac66d8645631486c84a09c8b6dbb536029da","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 11:03:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"13b3abbfd7c1c620c67e676d5720371476d42d2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:04:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a77e5cd455393ad081952c70849d9c283a4faef6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:48:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbc59b75d47e4c6d9f9ef8ccba2d542a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08aef5da33954ac2882aee399a2f897b : SUCCESS in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce81b29213ef4b5492272b8fc96e0005 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c37834506a248bfaea980cc5ead8ba7 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47acc4d81e6f4a9096135e393c7816f0 : SUCCESS in 2m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f378e668290042d8aa1e4beedcb11d5c : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4bea8fc2982c47b89d0d85bae6dd185d : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/00a9d8d4b91f4b87a8ae84224e9c1417 : SUCCESS in 11m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/99cb78eabd104caea5e01b426c780657 : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/32bfdcd416554175a44ee996d9b43865 : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/775ec5a551e2473999d3dec5779cb679 : SUCCESS in 45m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/df7d9253cf56454a8f5109dec7ae81e2 : SUCCESS in 19m 22s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/291a5e4802ab4cf4b68dfa63b86cd2a3 : SUCCESS in 14m 07s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aedb54ad36334ad7ba60214dc68e29cc : SUCCESS in 1h 08m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7fd159df771451d8c868603581ff7c7 : SUCCESS in 1h 54m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/27fe74435a3d4c30b5ac9fd0146988bc : SUCCESS in 18m 06s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3a73dcb79ab4f008cb29430b3431574 : SUCCESS in 23m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3a5783981e1942f29201bec1a8950c69 : SUCCESS in 25m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/704e09daf50547c3ad14c8872ff60fd2 : SUCCESS in 19m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bb3697e6247644859bb2958138b0767a : SUCCESS in 16m 56s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7c121f96b7f4f638709290edd724bb6 : FAILURE in 14m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ed8bb52670d9454c882617aaeda3d0b0 : SUCCESS in 15m 54s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c9b69dae6a184940ae4e8445a4737408 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":1},{"id":"835662f0c38b2d4a88bb173a1f7fe0f408234eba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-27 14:48:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41cbaa794f72bcb0cdf359d1b3d141f245c3452d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:58:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebead18a8d34be99bde97d3f0addf3c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a78939c9f3db49d58f76727022960031 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2b7cb97b3a49b8873cb15eeff7269c : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9120cc9242be4235be03d22defcdd62c : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50e640f7a95d4d30988ae6e43f1ad9e2 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5616e22d1dc0490db478859b6734646f : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/976aab0c923d43649f0f7fba3ff29b94 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a53928eeaf0403091473da976464930 : SUCCESS in 3m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07e6fbf724ac48ada0b7f976e1ab6d98 : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cc6ef21dfed045d2a1fb8e2629abcc8b : SUCCESS in 11m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cdc00bf914a5425e9aabd0dcc29634f0 : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d3f03c098b2e439fa39ea8c8ea40fda4 : FAILURE in 9m 41s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52414dadf0af4899bf111c6cb5537c16 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fd6267daa1484c9c84d6cd913bf62125 : SUCCESS in 43m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7476ceac7d5c41688f8b1e48ce17a182 : SUCCESS in 22m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e22e944de41d4e588b98583fc56b69b7 : FAILURE in 10m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82011fcb2758456baae763230370207e : SUCCESS in 10m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9f18198b80dc48daa9e9ec5faf144a34 : SUCCESS in 2h 01m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c00e94b53c74f4a88458aff29fa1034 : SUCCESS in 1h 48m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/13257022e3f04dd5a306112bd07996a5 : SUCCESS in 27m 03s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c33cbaa6d8b247e5a506eb0fdc13d940 : SUCCESS in 21m 41s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/13d88b4580774c0ba681ff7e1db9d136 : SUCCESS in 39m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cda617732f6342f6810c1f46fab9f1b6 : SUCCESS in 30m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f94d998e9bfb4d95a6581ea31c6239cd : SUCCESS in 20m 51s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d75a54be9a784211b1e8e153fcff6672 : SUCCESS in 17m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cb622f5e013e42b68132072aa6d085ce : SUCCESS in 15m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b582ef9df2fc41a28259ca22d422e4ee : SUCCESS in 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"72886e2003def28680eaa2d924befa328222ccfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 16:58:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01f444e8ffaf035bf865d7b87da513011b723f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:20:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd135d76f33b41a982f0b1d0a89789a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c959bc3410b475eb81e61a6a2658188 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c64d9e45c9a4a6ba2740108ade662bd : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0479e63d4e18484aa87bc159586ef0b8 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d4a5be803cd4b4dbc5d618e93b00984 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e0ac16bfa45464fb6f00c320c349a70 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1970514e65844cfa9251b7122195b5e0 : SUCCESS in 7m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/993f47d45e0a4ad684b42ebf95ba5cdd : SUCCESS in 9m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cb560ed1a2e74b09be1fb1957a555fa1 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9fa69fff77894fcfb43cc6c240ff61f4 : SUCCESS in 33m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f7f7365dab154c33befbfbd440ff854e : SUCCESS in 1h 16m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/61904a3fd17a45589178dec2b461b7eb : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b6e04bb0401424dadd011d4c5e381c5 : SUCCESS in 11m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6478ddbe24a046e7b3ec2303428ed6dc : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac44971c4bcc48c691943e080b2be684 : SUCCESS in 54m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bcfeae1960074b1fb3e9c47b339bf299 : SUCCESS in 24m 37s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99b1cbc279e244a0a264009df089fdf1 : SUCCESS in 15m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c623b630c1fe4f4aab7a19a81be42d2d : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/545f6558951343bc8bbdf56ee4ce47ba : SUCCESS in 31m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/17fe6adf52f74f3c9e6e2013cef9391e : SUCCESS in 17m 04s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/663051db120d4fe4abb6d88d3f3c6348 : SUCCESS in 13m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1b1a9cdc9b474ace8a146eaa06c83e59 : SUCCESS in 19m 16s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a096a2224ecc4f1d8b3dbcf4137d48a2 : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4d90b22652f11c25cf5117fc67bfd50bbef2b793","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:20:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"32d6a4f38f2b72036db9757c864904482ce72168","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:21:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6571707d46d3495ea5a99c300cbce2c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1abb73a5e0644df290486a1a15df81bf : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea61fd4402424ff4964097e3a45a6b81 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa27bc08d4849bbed106f7bcb27ccc1bd4e594b9","revisions":{"fa27bc08d4849bbed106f7bcb27ccc1bd4e594b9":{"kind":"REWORK","_number":1,"created":"2026-01-26 22:21:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/974786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/974786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/974786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/974786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/974786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/974786/1"}}},"commit":{"parents":[{"commit":"755beae3b1702d5ecdbfdceca468495ea20cfb54","subject":"Add infra-vms to ntp group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/755beae3b1702d5ecdbfdceca468495ea20cfb54"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-21 11:56:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-26 22:21:13.000000000","tz":60},"subject":"Support swap configuration for all host groups","message":"Support swap configuration for all host groups\n\nAn earlier change [1] added support for swap configuration, but the swap\nplaybook was only imported in ansible/overcloud-host-configure.yml.\n\nThis completes swap support for the seed hypervisor, seed and infra VM\nhosts and adds missing variables in etc/kayobe/*.yml files.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/851195\n\nCloses-Bug: #2138800\nChange-Id: I3971d05f3c09a6841f6666a18f2d2b44b226411d\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n(cherry picked from commit 6b4028754321b3e20b45f93b6fba159985e850af)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa27bc08d4849bbed106f7bcb27ccc1bd4e594b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa27bc08d4849bbed106f7bcb27ccc1bd4e594b9"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
