)]}'
{"id":"openstack%2Fkayobe~974833","triplet_id":"openstack%2Fkayobe~master~I65d60fe49bbc7c0fb7472d065040d7f446258705","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-01-28 12:50:36.000000000","reason":"Change was submitted"},"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2026-01-28 12:50:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I65d60fe49bbc7c0fb7472d065040d7f446258705","subject":"Changed the way bashate is executed to exclude local dev artifacts.","status":"MERGED","created":"2026-01-27 11:36:36.000000000","updated":"2026-01-28 12:53:17.000000000","submitted":"2026-01-28 12:50:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974833","meta_rev_id":"3a93f04732c078290d2a62f206079a707b510cf2","_number":974833,"virtual_id_number":974833,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 12:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-28 10:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-28 10:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-28 10:32:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-28 10:32:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 11:55:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 12:52:14.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-01-27 13:26:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-28 10:32:51.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"e95cd48c47036bff31be05663740287d2ea4aa61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-01-27 11:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0297a24dccf7687e86410759f5fedd6545b270f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-01-27 11:39:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e88ecc5b776d5ee0d910b8b6b8b9421adbb10ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:55:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d346a997ee0e4bafb2db186785323acc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3d0e56568642659f81991fbdd93008 : SUCCESS in 9m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4fed5b9cd9c4570ba457c9928476776 : SUCCESS in 4m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07d012ff80744ed79a89f9268b30f7e5 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2512ef456b124e2aa3e0b6f51c450a4d : SUCCESS in 5m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/562ce18db9fc4e7795fee2d0c8d53cbe : SUCCESS in 2m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e2d75a8ba9943dcac4af291ba75573e : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84398e2d8e3341ae8c4dcd552d108714 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b115d8e80c14e4e9e98f97cf19215d7 : SUCCESS in 10m 29s","accounts_in_message":[],"_revision_number":2},{"id":"275f051450812dbc04bee897cdf5eaa2e247a6d9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 12:52:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0095a3b227597203798783246785c7ec85ec692","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-27 13:26:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ccd18ccd36e694c41c89bcf23d4bb6200a07acd3","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2026-01-27 14:39:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bd07258e60052a9ca9fbdfebf44a7520557bbe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-27 18:03:38.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9a2bfa3665122206545c543de50db8fc30063020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 18:24:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3d82d4612ab4b848c497bfd7a8160ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903aef00d4f74aa2b4ce3e97f56b3d61 : SUCCESS in 9m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5f360dc1da5e424299b2dc4fb8b0addc : FAILURE in 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a18a8f416a6479aa7c427556dcd717b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc46c53c79748e2bda5c1fb5f57c991 : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f0ca1dc97df49a391fad250e8d0e3fe : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24be0343c43c40f68f82a52c8bc9d205 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee24d9fb9e554a38a80751f21c1f8a06 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f8ca2f86af5452fab6d7e864beaaad4 : SUCCESS in 10m 55s\n\nWarning:\n  Comments left for invalid file ansible/provision-net.yml","accounts_in_message":[],"_revision_number":3},{"id":"f8d812f2a4b6cf81f98c77305a4bb4e0c9359f97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 10:04:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7da8df707e74674023c1607e89e9b60288ff3609","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 10:05:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"eb429e445403176c14b6685190f04c2777955a53","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 10:06:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"84c3229dc3e323273fcdbfbc759edba0020fc748","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-01-28 10:32:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"847e828bcf87168d5316276d33eb5e44925b5227","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:35:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a29558fa04564c59a2cf479ac1ed4735\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7a1925d6824cb0b3992a51c3682774 : SUCCESS in 9m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9afa135e3a3d48fd9493c2eefe5f5c49 : SUCCESS in 7m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1994040f4318422597ff99714669d51f : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaf5087a8a31423b82b017447880fe51 : SUCCESS in 2m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75cd07e544c24755aca55e700487872c : SUCCESS in 2m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aee21131385a4cb799cb3ebca6d3de6f : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/baa2fe87872949099c933643c72ab02b : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0cbd9185b480437ea826e807a963d43a : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":5},{"id":"d42afd5829e6c912cc0c44fbdd18c453dae791f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:35:58.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a4d4ed24376fea20ece035e35d79737cf2cce675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:50:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d179d9dade048899df7713d7f839e87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b04f0f9b2e9405c8d27454e8975f272 : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9b5dc1e2cfbe41b6a0b668647110585f : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9320838fb1b54197aee0d1213826aade : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b452fa38a554d18ad0a2aa01f56e291 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9bbb84a4d0845a6b046974476361102 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd764556a17748449d624d3eb7dd5230 : SUCCESS in 12m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da2b895bb0fe482aa9768427512dbc67 : SUCCESS in 10m 43s","accounts_in_message":[],"_revision_number":5},{"id":"4b747b182e4d8479613dd2ab238f0a255f458a1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:50:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3a93f04732c078290d2a62f206079a707b510cf2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:53:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e24c7375d35d439e930a50c074fa0804\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae5954f5ca94863b542aa5511c6d2a6 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7b11017c732c8aa233bb72ad71fc646a3e806f2c","revisions":{"fb2430083cb1960063b69f0bc01c3fcc3b546b0e":{"kind":"REWORK","_number":1,"created":"2026-01-27 11:36:36.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/33/974833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/974833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/974833/1"}}},"commit":{"parents":[{"commit":"0788c6ded5ccac152859bb16b1866c6c269bcba4","subject":"CI: Add missing job to experimental queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0788c6ded5ccac152859bb16b1866c6c269bcba4"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-27 11:28:35.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-27 11:36:24.000000000","tz":60},"subject":"Changed the way bashate is executed to exclude local dev artifacts.","message":"Changed the way bashate is executed to exclude local dev artifacts.\n\nCloses-bug: https://bugs.launchpad.net/kayobe/+bug/2139142\nChange-Id: I65d60fe49bbc7c0fb7472d065040d7f446258705\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb2430083cb1960063b69f0bc01c3fcc3b546b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb2430083cb1960063b69f0bc01c3fcc3b546b0e"}]},"branch":"refs/heads/master"},"856bbb2f0f611f4c31a410672b223b0039b4ec2a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-01-27 11:39:05.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/33/974833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/974833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/974833/2"}}},"commit":{"parents":[{"commit":"0788c6ded5ccac152859bb16b1866c6c269bcba4","subject":"CI: Add missing job to experimental queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0788c6ded5ccac152859bb16b1866c6c269bcba4"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-27 11:28:35.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-27 11:38:51.000000000","tz":60},"subject":"Changed the way bashate is executed to exclude local dev artifacts.","message":"Changed the way bashate is executed to exclude local dev artifacts.\n\nCloses-bug: #2139142\nChange-Id: I65d60fe49bbc7c0fb7472d065040d7f446258705\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/856bbb2f0f611f4c31a410672b223b0039b4ec2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/856bbb2f0f611f4c31a410672b223b0039b4ec2a"}]},"branch":"refs/heads/master"},"bf342c1163df411af78e48615d0d28266cac9e02":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-27 18:03:38.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/974833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/974833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/974833/3"}}},"commit":{"parents":[{"commit":"b7236b8ac3d8578f9320ab646076fb436fbfdc44","subject":"Adds option to disable registration of networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7236b8ac3d8578f9320ab646076fb436fbfdc44"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-27 11:28:35.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-27 18:02:36.000000000","tz":0},"subject":"Changed the way bashate is executed to exclude local dev artifacts.","message":"Changed the way bashate is executed to exclude local dev artifacts.\n\nCloses-bug: #2139142\nChange-Id: I65d60fe49bbc7c0fb7472d065040d7f446258705\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf342c1163df411af78e48615d0d28266cac9e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf342c1163df411af78e48615d0d28266cac9e02"}]},"branch":"refs/heads/master"},"4771a9128d66c13f49e1510a5c5f45d4a7fc91d2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-01-28 10:04:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/974833/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/974833/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/974833/4"}}},"commit":{"parents":[{"commit":"9b03310327630fbf6f9e64d65364502fbd212d1c","subject":"Adds option to disable registration of networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b03310327630fbf6f9e64d65364502fbd212d1c"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-27 11:28:35.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-28 10:04:37.000000000","tz":0},"subject":"Changed the way bashate is executed to exclude local dev artifacts.","message":"Changed the way bashate is executed to exclude local dev artifacts.\n\nCloses-bug: #2139142\nChange-Id: I65d60fe49bbc7c0fb7472d065040d7f446258705\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4771a9128d66c13f49e1510a5c5f45d4a7fc91d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4771a9128d66c13f49e1510a5c5f45d4a7fc91d2"}]},"branch":"refs/heads/master","description":"Rebase"},"7b11017c732c8aa233bb72ad71fc646a3e806f2c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-28 10:05:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/974833/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/974833/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/974833/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/974833/5"}}},"commit":{"parents":[{"commit":"b622c92b37b16148505821f8eeec8f9ac6d78422","subject":"Merge \"Add support for url sources in kolla_sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b622c92b37b16148505821f8eeec8f9ac6d78422"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2026-01-27 11:28:35.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-28 10:05:10.000000000","tz":0},"subject":"Changed the way bashate is executed to exclude local dev artifacts.","message":"Changed the way bashate is executed to exclude local dev artifacts.\n\nCloses-bug: #2139142\nChange-Id: I65d60fe49bbc7c0fb7472d065040d7f446258705\nSigned-off-by: Grzegorz Koper \u003cgrzegorzk@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b11017c732c8aa233bb72ad71fc646a3e806f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b11017c732c8aa233bb72ad71fc646a3e806f2c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
