)]}'
{"id":"openstack%2Fkayobe~974959","triplet_id":"openstack%2Fkayobe~master~I9395aa664a24f31d90e2d09f4191f904b437c840","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-01-28 09:59:57.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-01-28 09:19:08.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-01-28 12:50:30.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-01-28 12:50:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9395aa664a24f31d90e2d09f4191f904b437c840","subject":"CI: Enable baremetal inspection testing","status":"MERGED","created":"2026-01-27 21:59:31.000000000","updated":"2026-01-28 12:52:29.000000000","submitted":"2026-01-28 12:50:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974959","meta_rev_id":"3d0b56804aab7508e06ee70f428f62f082c16e26","_number":974959,"virtual_id_number":974959,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 12:50:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-01-28 09:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-01-28 09:19:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-01-28 09:59:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 22:10:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-01-27 23:52:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 08:18:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-01-28 08:18:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"b8ce0ee81f303c3336767c4ca0ddad6a9fc5c7d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-01-27 21:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20c3867a02ae6b691627ea6f355e5154995f95b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 23:52:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1db3af38717e40559a08e9b3ceb12405\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780b66998d154046833162d2844a64aa : SUCCESS in 9m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/922552862c4b4c388156172500014d70 : SUCCESS in 7m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaa65c129d4e4481a0fbeacca69297be : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780a4c0a383a473eaf02e18537c2f626 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e569f96678f74075a975277d3ff7175c : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c9627a418f44e328be37cafa22cee04 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6626ea0c56264cd8925f251f8a78eb7d : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f51dfe99f01949968b49bf88c4364e48 : SUCCESS in 13m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/bbc28e8b49e9468b971a1dbf366c1be1 : SUCCESS in 18m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed8404c15b6147dd9233c209b9f2d7af : SUCCESS in 12m 28s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/5d69cfcd599342fbb762545b9523feca : SUCCESS in 17m 24s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6991839a51414ac980e6624990204e84 : SUCCESS in 15m 34s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/118fe7081a46465fb8c56fb1e64af525 : SUCCESS in 54m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/881e41fbdf0c405c958dfd66533a5925 : SUCCESS in 1h 14m 18s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f6ac4bf05fae410c8407ff97c8718ab9 : SUCCESS in 58m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b5c30e28af5448cc889f70bdf99c403a : SUCCESS in 59m 44s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fe7e1af958d449dbbe32d2818a49bb3d : SUCCESS in 1h 15m 14s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2a2727d901cc49bea09bb391a4ca138c : SUCCESS in 1h 51m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/4e7f8b38af284c3f92ab47c03d812559 : SUCCESS in 1h 48m 12s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60d81b7bf55240fd95378e3e2f88e7e2 : SUCCESS in 1h 24m 31s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8fef1420d0a47f1b9cceab7a55e37c0 : SUCCESS in 1h 51m 20s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e9c1512e21ed4cf294d7a20da2d378df : SUCCESS in 9m 55s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/0efca922607d44378dbff91bcf702f73 : SUCCESS in 10m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/849152c175094edaa27bf465614b1cba : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d9e4182cdc8d477cb4e714c50ca01f7c : SUCCESS in 15m 13s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b3cff72f902b43049b63da535ce4e0c3 : SUCCESS in 15m 30s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8513cebd01324752bcfda1aea218c5af : SUCCESS in 36m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c615ece3a1254630a15877f98f667b53 : SUCCESS in 28m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c3b76dc38604d08b56a79e9df62da0f : SUCCESS in 39m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/70b9eeac19d14978abd90081d292c09f : SUCCESS in 15m 05s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/464a944135944660970e081f98a97be7 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0de147bfc8c63a165ba8d89e298034c2ba42b72c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-01-28 09:19:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e8073a5add17ece69dccc44ed256c66062d36033","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-28 09:59:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80f8e30c557b166fa37da01017951dcd56f86035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:00:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"981557b9afdf5a1bd55231fa4f9a8b85c876d5bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:50:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8000a9375664aecbaf31acd48e555a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6afa9fdaf91b4fb791d24ef53158100e : SUCCESS in 4m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b9fa9a68d9cd4ddfbef0c5dd28a0332a : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83d303f56db249aaaba1348dead0be8b : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b34a9d13aeca4acaa5a27e0d30164b3c : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee22b08b27e04b16a6c6c591e60792eb : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e3bc0b2c078445da8c133566a5bfdfc : SUCCESS in 14m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a99eaa29fb704f17866f5bbc3548dd29 : SUCCESS in 6m 20s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d6dd5514c0c64b0c856767bf285829e1 : SUCCESS in 15m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75c05b4063ca4f978d51c6e7b5ae5a68 : SUCCESS in 12m 02s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/82dabd2c87ac4703b05d52d5ce858665 : SUCCESS in 17m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1c1c1dcc20a94c44bc7647fc7164876e : SUCCESS in 15m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8b980b8a2b9349f3b231173ebfde2951 : SUCCESS in 55m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/709c09567c2147108b4cd7abd2501ef6 : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2622eac1a33043b59f8758f07d3ce1a2 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33317d4ea2d144c48b68e73944361706 : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/531cb47eb13b4d0d872a664977a51034 : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/900cee31171949a89675872f3c190302 : SUCCESS in 1h 34m 12s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a2f159b1851b494d94f411a7d155ba4c : SUCCESS in 1h 28m 57s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1da1b877fd544edf82ab5eeac2c4f355 : SUCCESS in 1h 31m 23s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c9a6e898a904f398ccf6a55f6e527cf : SUCCESS in 1h 30m 03s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/749b4b1127aa46f4b95d9ec181b9f253 : SUCCESS in 23m 25s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/8d43918af76d42ba8443a8841059dd77 : SUCCESS in 15m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05dde8ecbef04869aa2b9f14c6586672 : SUCCESS in 20m 04s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/47a18be7578542d996bf46e70641b0a3 : SUCCESS in 19m 57s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/20ca15061c60432ba936597a6b1343e0 : SUCCESS in 33m 27s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ae999528032b4be3a43d976be2be31e3 : SUCCESS in 30m 07s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68f48d7aa7194fc29530fdee33e56678 : SUCCESS in 29m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92fc94b1a7e84798811980107eced716 : SUCCESS in 27m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/950fc3dfccdd4392a592d257532a689b : SUCCESS in 15m 02s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0af393eed3274426babb29842bb02b03 : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0596aff5cc6c4bd408afae2353424505b7c5489c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:50:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d0b56804aab7508e06ee70f428f62f082c16e26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:52:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed27997a661641e4a9341a3a34a585ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f8c4e36a0a433e95d7f5f50c31a6d7 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16f7e2e14be59d32bd2ad39175ac92e8ac824d0c","revisions":{"16f7e2e14be59d32bd2ad39175ac92e8ac824d0c":{"kind":"REWORK","_number":1,"created":"2026-01-27 21:59:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/974959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/974959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/974959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/974959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/974959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/974959/1"}}},"commit":{"parents":[{"commit":"b622c92b37b16148505821f8eeec8f9ac6d78422","subject":"Merge \"Add support for url sources in kolla_sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b622c92b37b16148505821f8eeec8f9ac6d78422"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 21:59:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-01-27 21:59:14.000000000","tz":60},"subject":"CI: Enable baremetal inspection testing","message":"CI: Enable baremetal inspection testing\n\nChange-Id: I9395aa664a24f31d90e2d09f4191f904b437c840\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16f7e2e14be59d32bd2ad39175ac92e8ac824d0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16f7e2e14be59d32bd2ad39175ac92e8ac824d0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
