)]}'
{"id":"openstack%2Fkayobe~975186","triplet_id":"openstack%2Fkayobe~master~I88e0f12e838070196d8f3ffa4ec95464ab323632","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-03-17 11:19:29.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-16 09:33:37.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-03-17 09:26:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I88e0f12e838070196d8f3ffa4ec95464ab323632","subject":"[firewalld] Fix broken conditional","status":"MERGED","created":"2026-01-29 17:41:11.000000000","updated":"2026-03-17 11:20:31.000000000","submitted":"2026-03-17 11:19:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975186","meta_rev_id":"91eab2d23df52b75eb9e3466ef12d71e81f3c1ca","_number":975186,"virtual_id_number":975186,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-17 11:19:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2026-03-16 09:33:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-17 09:26:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2026-03-16 09:04:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-17 09:26:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 19:23:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 19:48:23.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-04 11:12:09.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"CC"},{"updated":"2026-03-16 09:04:05.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-03-17 09:26:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"9ba6e363f56b5be09d0bddd3ec6bae97a745e374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-29 17:41:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a501681acb2795f13c51024d6c24b876052c60c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-29 17:42:25.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1de00c0c57f39fc96bcdf15a9168410824ec6148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-29 17:49:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ee97b62c06ef5482eb4d8d5895550e95f387f315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 19:23:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5ee14f6419a4eb5b1bc568d6f6bd1d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2091e8871d40e4ac725167855d0ef9 : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f267cd0d519446ad994af17d63e4e276 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09950a1ef8fb451baab35f68a90b2a1b : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c8e95b4ddc24e7ca20904496eff5404 : SUCCESS in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/716df97e97e8454f9b40311dd83758e4 : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d41b827d259c4160b1b27408927cc955 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27c6175657054719873275e3a1d3d1b3 : SUCCESS in 15m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2087f2ff12f04885b82f09f00aefedc1 : SUCCESS in 11m 18s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f379473f86f6409fb3157b0d17cae34f : SUCCESS in 18m 58s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4f332dfb60745c18dcbabd9fdff86ca : SUCCESS in 6m 40s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3555f5e506ab4eb0adb4dac370c0b461 : SUCCESS in 12m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3085caeab2b1418c9ae5fe2972186b63 : SUCCESS in 17m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/a4d58fe12bba48469f365edac4b7386c : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d40f00f4ace545e1a93228b0837b8911 : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/f0e6e3d1184149e19065aae9a47348be : SUCCESS in 1h 01m 53s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38d42a76cd534b5588e4a2473bc131c4 : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fc85ba1a5f8f499686f0f7d79da1417c : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cbc466449c1e40d78b9a7eb226122261 : SUCCESS in 1h 30m 35s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/145af8b9daac4f07985abe3f75b9cb50 : FAILURE in 6m 23s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/220d3b3a0a51474fa26bb5d3392fc2e0 : SUCCESS in 1h 32m 32s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7505eafb9e442d3be3bc30787695124 : SUCCESS in 1h 30m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ad7cb16dcee9489782c99e1a752f9dc5 : SUCCESS in 9m 09s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a3ecfa009dd142d69413b2890b8e948b : SUCCESS in 9m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a147b05c84b4cfba4fa3e811da5d5b1 : SUCCESS in 13m 34s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f911cd273fa248edad6bb971c05832e8 : SUCCESS in 19m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/c90c38bc09274d4693eead0d8a13ff7c : SUCCESS in 19m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dc5e8baaa5c5475c835f2d87a50aa807 : SUCCESS in 28m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1b88cad339434578ae83068f3823f49d : SUCCESS in 15m 34s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59481a3762b348cfb0fabe463efbb62c : SUCCESS in 33m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d9ba30c9a09f4a9eab6cf727e9c38866 : SUCCESS in 16m 22s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5defd3db9dc4405fba998d7bbbc4a6b9 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"9e60b70c6c302c325fe14764fd6100207434364e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-30 11:26:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"836e704ae304dee7a089fcc1c9460a964d6db24f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-01-30 11:34:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b4e2555c79aa37464dc6e23df1ac61ef1da19f41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 13:06:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220a8e30699e48d3b983547c46e951b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd875ee27f648a188ece3bec31d5942 : SUCCESS in 8m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d25cae358a1d4c6abbb5ed92a1a373b3 : SUCCESS in 4m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a73a63f232b34355b04965d9fb836a85 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc34e368f036467ab80b0fbd1e9e16ec : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5d5612eacc0c4ce7a4e4db2f4fa8ede0 : SUCCESS in 3m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/330074d544ab4da59b0b132498cf3f32 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/878dca399ed44f9ea6f91c28445a985a : SUCCESS in 12m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb7dd6eeacba495aa2ee5a40dcb1a934 : SUCCESS in 9m 02s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/17d106a2cf47468b8243feb7d3b56e5a : SUCCESS in 9m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de48ad7f40bf4bfab7003d3bbaab6a1b : SUCCESS in 12m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1ec447cd412b4a579eaef407536535b6 : SUCCESS in 11m 42s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3dd363983f7044f9a0f4775e1c52af08 : SUCCESS in 14m 36s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/9ce031d4684d440593ab77e926b465d2 : SUCCESS in 57m 53s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8569985e441a424383a8d5fe1c00932d : SUCCESS in 55m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/04828596ee924d088a6d797807dd0366 : SUCCESS in 57m 45s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fdfd9b227f047a3b393b30d3feece3e : SUCCESS in 54m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/850cfd7cd67441e6818f24a3dcb271dc : SUCCESS in 56m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f667b60eb0e94b97aa79556a8ee3803c : SUCCESS in 1h 30m 00s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/95cc5f99088c4c42b19716abdbba3780 : SUCCESS in 1h 25m 21s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca8f5b77761547209a1bd67ff0a62e63 : SUCCESS in 1h 22m 12s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2690f18450c4953bab4483d16dcf770 : SUCCESS in 1h 24m 39s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/09eabfbfef7d4f79bfdf7c68de43c77e : SUCCESS in 11m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4e549b46d8f94954818ca8872dcf80bf : SUCCESS in 15m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ece0444a1faa4815b97e383e720e17aa : SUCCESS in 9m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/31cbb78188be4fd99202fd35ade3fabc : SUCCESS in 19m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cc992830935f4697b8456c5ca6956b47 : SUCCESS in 15m 10s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/ec89be8d9fdb463f950af70fda6f24a7 : SUCCESS in 36m 15s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ff5a37fd93614034b3323b63aec808a8 : SUCCESS in 30m 05s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2cd09c015a514a92af3d0b7bdbc2dde4 : SUCCESS in 26m 39s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/905ec3f224e649e9888995d26da9864b : SUCCESS in 10m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c992e0173e114510a5bf3ebcde0584ee : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":5},{"id":"1c37197a83090ffd4d32dc4c3833b4a6cad99dcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-04 07:29:34.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9e0569326ece9c495240d66d12c01147b223da81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 09:19:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa32a6a26412420e8b4790a9c0b9b66c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/801fc631125f4cff81332f75f28f6103 : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c18b2ff24b742e3975f8641f758653e : SUCCESS in 5m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/227a15d6d6aa4518b48a024515954476 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea823d93dff4e90825a7d530ddaeb00 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/158e71e2bf584182b8d966ee94b2bfbc : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/442aff6713244cce97f50e3fc811d496 : SUCCESS in 6m 48s (non-voting)\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/798c227867ea4f2fa355b490ce602ddf : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77a8b9e9a0ab4bba9a45b3d6dba4978a : SUCCESS in 13m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e92749e84e6a45328a8a99b5a5768f02 : SUCCESS in 6m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/63efc1c35d634e9181a813f2c5be1c1e : SUCCESS in 8m 59s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d14f397e71cf43c7b69ed64edd823fc5 : SUCCESS in 12m 05s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/2ea29cb2bbac4c13947ea35c4a9b85fd : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ea897a6dc4d456a82f38d12a5e9d4f7 : SUCCESS in 15m 08s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/705785a9f50e45418f770a3b2e008050 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ec83d22348aa4884bec958701d8854a8 : SUCCESS in 58m 59s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/21649d73ee604dc7abc42448d1021fe5 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ace3e09030e435595ad8220c8f41c5b : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0db62cbd63784c0aaed680e00984b563 : SUCCESS in 56m 03s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cee6cde4a1e34e84aa1718681cc6acbc : SUCCESS in 1h 35m 55s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/fa5b0428a82a47808a590650c3232748 : SUCCESS in 1h 35m 09s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/15ab706a50014575b86631d88d55d430 : SUCCESS in 1h 35m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57e3d0f4ff37410db031e38740540057 : SUCCESS in 1h 49m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/eb4562d5f5ff4404bc566839d2a70ea3 : SUCCESS in 14m 48s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b01e9fab4ac644a3b71eb9c1c46c7057 : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":6},{"id":"92728470defb406c04284c33c285dff202a9b505","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-03-16 09:04:05.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"dd8140881ae3f940b78b583b9fb596b23737d1e8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-16 09:33:37.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"a4136086d3f6c2ac92a37ebe5eb897fdc7843a2a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-03-17 09:26:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2aed02c3637f38bf74aa2ca4470622171411a877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 09:26:38.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8479c20b4f7a9a03fae34a10a31bd2e80ef678df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 11:19:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d028c78a17c4321a949c446b3b3a346\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7da2b184d6437ab797e74b98c43689 : SUCCESS in 6m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39f1c61393db477481c83e6060f0b3b2 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e160d5910b441cdb1d70f83744727c2 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a617bc6935042718cb75ea76b6c31a4 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69fd8e92197146f58bd2854e983bb8bc : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b211f7dc9cf4119a55a3ebe25fe12f2 : SUCCESS in 6m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/977791c3e99c448b93a3eea19f42e246 : SUCCESS in 6m 23s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ca594f28dd784b2eab2edc1a5202ec92 : SUCCESS in 18m 28s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/93998ee4dd694f9884536a116eb09012 : SUCCESS in 12m 16s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a3e3051efaee48809065525d52c015d6 : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49d4c9939cd5454a8b8135665c1a7d72 : SUCCESS in 18m 43s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/e39a562f8fda407aa8dc67f5798ecdda : SUCCESS in 57m 38s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/bdca23eb720a4a539ab5698914a5f49d : SUCCESS in 1h 08m 42s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/0e6824c768774373a511d6e74861ef46 : SUCCESS in 1h 01m 32s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6c064cb513184ea084d3905e0c460490 : SUCCESS in 58m 30s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/b176eff1a5924dceb1dd86492f3bf972 : SUCCESS in 58m 55s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/2546f40ef1544ed0b465a59f3ee9fdf2 : SUCCESS in 1h 52m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/253d1505b0354c7b88f8de5c64d725b2 : SUCCESS in 1h 50m 34s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9ed9ff02782f44f19f639f7f3e983c21 : SUCCESS in 1h 36m 51s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2e503e7c32f04d5fbfa62d02932111b7 : SUCCESS in 1h 38m 36s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e966c2a8bb494249a03332a4a958502e : SUCCESS in 11m 20s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cfc42aea2afe4723b00ded5894c62ea8 : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aedb8bc3e6734827bf6fec38e0c89093 : SUCCESS in 17m 57s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c365318ac3c947c9b83f5d3986a3003e : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e6ba786f95c445a5a743c05bcac594e5 : SUCCESS in 36m 03s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1684f03cb15b41c4b327def898fccd52 : SUCCESS in 32m 38s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31d76d777c81425b8e55cdb4c484e47b : SUCCESS in 25m 11s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/498df996c9a54df9af5d75c15cbfafa6 : SUCCESS in 19m 58s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/8976c4af0c1a4b14b4a32e2f2840a139 : SUCCESS in 10m 43s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3be6c1f62c544b7af1a91630b3df397 : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":6},{"id":"8b5d044cf0101086c9536842e3509f51bd1ce1bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 11:19:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"91eab2d23df52b75eb9e3466ef12d71e81f3c1ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 11:20:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c0c0f2a54c455cbbb6d561fe6cf302\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a0959f2dbe49b9a9511a82826c4d65 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"190b79b5d02a67be5353b5e908f6f3aa909aac49","revisions":{"1dcd45385be620ddc297af62f0fd1e3a03cd58bd":{"kind":"REWORK","_number":1,"created":"2026-01-29 17:41:11.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/975186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/975186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/975186/1"}}},"commit":{"parents":[{"commit":"785f36cadc88826b7dfb7a90cc4458ea87db377e","subject":"Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/785f36cadc88826b7dfb7a90cc4458ea87db377e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:36:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:40:57.000000000","tz":0},"subject":"[firewalld] Fix broken conditional","message":"[firewalld] Fix broken conditional\n\nBroken conditionals are currently allowed because the `ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\n[DEPRECATION WARNING]: Conditional result (True) was derived from value of type \u0027str\u0027 at \u0027/home/zuul/src/opendev.org/openstack/kayobe-config-dev/etc/kayobe/zz-30-overrides.yml:47:25\u0027. Conditionals must have a boolean result. This feature will be removed from ansible-core version 2.23.\nOrigin: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/firewalld/tasks/enabled.yml:50:9\n\n48   become: true\n49   loop: \"{{ network_interfaces }}\"\n50   when: item | net_zone\n           ^ column 9\n\nTrivialFix\n\nChange-Id: I88e0f12e838070196d8f3ffa4ec95464ab323632\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dcd45385be620ddc297af62f0fd1e3a03cd58bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dcd45385be620ddc297af62f0fd1e3a03cd58bd"}]},"branch":"refs/heads/master"},"68e4b254bfdded39a4693e8640fc2e58307ba79a":{"kind":"NO_CHANGE","_number":2,"created":"2026-01-29 17:42:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/975186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/975186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/975186/2"}}},"commit":{"parents":[{"commit":"b556941be61034c8dffa8632739331bc26db8733","subject":"[networkd] Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b556941be61034c8dffa8632739331bc26db8733"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:36:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:42:18.000000000","tz":0},"subject":"[firewalld] Fix broken conditional","message":"[firewalld] Fix broken conditional\n\nBroken conditionals are currently allowed because the `ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\n[DEPRECATION WARNING]: Conditional result (True) was derived from value of type \u0027str\u0027 at \u0027/home/zuul/src/opendev.org/openstack/kayobe-config-dev/etc/kayobe/zz-30-overrides.yml:47:25\u0027. Conditionals must have a boolean result. This feature will be removed from ansible-core version 2.23.\nOrigin: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/firewalld/tasks/enabled.yml:50:9\n\n48   become: true\n49   loop: \"{{ network_interfaces }}\"\n50   when: item | net_zone\n           ^ column 9\n\nTrivialFix\n\nChange-Id: I88e0f12e838070196d8f3ffa4ec95464ab323632\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68e4b254bfdded39a4693e8640fc2e58307ba79a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68e4b254bfdded39a4693e8640fc2e58307ba79a"}]},"branch":"refs/heads/master"},"20bc4b9da2c609f6492fd1f44e5455b935cc1a07":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-29 17:49:33.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/975186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/975186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/975186/3"}}},"commit":{"parents":[{"commit":"df2249e8c1579806d3276ea92d427d3c9bbb9fef","subject":"[networkd] Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df2249e8c1579806d3276ea92d427d3c9bbb9fef"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:36:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:49:27.000000000","tz":0},"subject":"[firewalld] Fix broken conditional","message":"[firewalld] Fix broken conditional\n\nBroken conditionals are currently allowed because the `ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\n[DEPRECATION WARNING]: Conditional result (True) was derived from value of type \u0027str\u0027 at \u0027/home/zuul/src/opendev.org/openstack/kayobe-config-dev/etc/kayobe/zz-30-overrides.yml:47:25\u0027. Conditionals must have a boolean result. This feature will be removed from ansible-core version 2.23.\nOrigin: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/firewalld/tasks/enabled.yml:50:9\n\n48   become: true\n49   loop: \"{{ network_interfaces }}\"\n50   when: item | net_zone\n           ^ column 9\n\nTrivialFix\n\nChange-Id: I88e0f12e838070196d8f3ffa4ec95464ab323632\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20bc4b9da2c609f6492fd1f44e5455b935cc1a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20bc4b9da2c609f6492fd1f44e5455b935cc1a07"}]},"branch":"refs/heads/master"},"6ecae62731adc3c78333c47bab5320ef8078d639":{"kind":"REWORK","_number":4,"created":"2026-01-30 11:26:22.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/975186/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/975186/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/975186/4"}}},"commit":{"parents":[{"commit":"77d80be617a6f48943263b6e347d6be4b1328b22","subject":"[networkd] Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77d80be617a6f48943263b6e347d6be4b1328b22"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:36:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-30 11:25:14.000000000","tz":0},"subject":"[firewalld] Fix broken conditional","message":"[firewalld] Fix broken conditional\n\nBroken conditionals are currently allowed because the `ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\n[DEPRECATION WARNING]: Conditional result (True) was derived from value of type \u0027str\u0027 at \u0027/home/zuul/src/opendev.org/openstack/kayobe-config-dev/etc/kayobe/zz-30-overrides.yml:47:25\u0027. Conditionals must have a boolean result. This feature will be removed from ansible-core version 2.23.\nOrigin: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/firewalld/tasks/enabled.yml:50:9\n\n48   become: true\n49   loop: \"{{ network_interfaces }}\"\n50   when: item | net_zone\n           ^ column 9\n\nTrivialFix\n\nChange-Id: I88e0f12e838070196d8f3ffa4ec95464ab323632\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ecae62731adc3c78333c47bab5320ef8078d639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ecae62731adc3c78333c47bab5320ef8078d639"}]},"branch":"refs/heads/master"},"6db74c59e7d2c7b71e54ff998cebe50e5c2fab34":{"kind":"REWORK","_number":5,"created":"2026-01-30 11:34:46.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/86/975186/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/975186/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/975186/5"}}},"commit":{"parents":[{"commit":"b6c43d9274fc4ac65870fda39193da325ea9420b","subject":"[networkd] Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b6c43d9274fc4ac65870fda39193da325ea9420b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:36:21.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-30 11:33:07.000000000","tz":0},"subject":"[firewalld] Fix broken conditional","message":"[firewalld] Fix broken conditional\n\nBroken conditionals are currently allowed because the `ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\n[DEPRECATION WARNING]: Conditional result (True) was derived from value of type \u0027str\u0027 at \u0027/home/zuul/src/opendev.org/openstack/kayobe-config-dev/etc/kayobe/zz-30-overrides.yml:47:25\u0027. Conditionals must have a boolean result. This feature will be removed from ansible-core version 2.23.\nOrigin: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/firewalld/tasks/enabled.yml:50:9\n\n48   become: true\n49   loop: \"{{ network_interfaces }}\"\n50   when: item | net_zone\n           ^ column 9\n\nTrivialFix\n\nChange-Id: I88e0f12e838070196d8f3ffa4ec95464ab323632\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6db74c59e7d2c7b71e54ff998cebe50e5c2fab34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6db74c59e7d2c7b71e54ff998cebe50e5c2fab34"}]},"branch":"refs/heads/master"},"190b79b5d02a67be5353b5e908f6f3aa909aac49":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2026-03-04 07:29:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/975186/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/975186/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/975186/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/975186/6"}}},"commit":{"parents":[{"commit":"a46f6484f8b361c322533bfe60b6cfe12aeca85a","subject":"[networkd] Fix broken conditional","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a46f6484f8b361c322533bfe60b6cfe12aeca85a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-01-29 17:36:21.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-03-04 07:28:38.000000000","tz":60},"subject":"[firewalld] Fix broken conditional","message":"[firewalld] Fix broken conditional\n\nBroken conditionals are currently allowed because the\n`ALLOW_BROKEN_CONDITIONALS` configuration option is enabled.\n\n    [DEPRECATION WARNING]: Conditional result (True) was derived from\n    value of type \u0027str\u0027 at\n    \u0027/home/zuul/src/opendev.org/openstack/kayobe-config-dev/etc/kayobe/zz-30-overrides.yml:47:25\u0027.\n    Conditionals must have a boolean result. This feature will be removed\n    from ansible-core version 2.23.\n    Origin: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/firewalld/tasks/enabled.yml:50:9\n\n    48   become: true\n    49   loop: \"{{ network_interfaces }}\"\n    50   when: item | net_zone\n               ^ column 9\n\nTrivialFix\n\nChange-Id: I88e0f12e838070196d8f3ffa4ec95464ab323632\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/190b79b5d02a67be5353b5e908f6f3aa909aac49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/190b79b5d02a67be5353b5e908f6f3aa909aac49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
