)]}'
{"id":"openstack%2Fkayobe~975379","triplet_id":"openstack%2Fkayobe~master~Ida4b38b98ab1140d44941355bc7ff991b0b5aa47","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-02 10:17:45.000000000","reason":"removed on reply"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-02-02 15:28:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida4b38b98ab1140d44941355bc7ff991b0b5aa47","subject":"Bump stackhpc.linux collection to v1.5.1","status":"MERGED","created":"2026-02-02 09:57:08.000000000","updated":"2026-02-02 15:29:33.000000000","submitted":"2026-02-02 15:28:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975379","meta_rev_id":"172cd80f592ac4cbd9d24cccc315981d5ad84e58","_number":975379,"virtual_id_number":975379,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-02 15:28:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-02-02 10:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-02-02 12:21:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-02-02 10:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-02 10:17:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2026-02-02 10:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2026-02-02 10:14:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 10:04:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2026-02-02 10:17:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-02-02 10:29:40.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-02-02 10:32:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-02-02 11:52:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-02 12:21:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"fa0813098323cc56521ae409399ff0c2b44d078e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 09:57:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c90b76d052c43f9dbccb6bf0addf58fcf251650d","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 09:57:24.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"c5e8ef0ab763d18e019ce78e8c7ecd1526af5d66","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-02 10:04:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da9c0676438a74e383768a67449e6737125c1302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 10:10:32.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":2},{"id":"360ea7fdd9577070cff3922aad17e7a8f85a3bf1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 10:10:40.000000000","message":"Patch Set 2: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6503cf37d02a9993686914d9d157b900ddb5b9f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-02 10:11:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"503f6a032ae15f87447d6df5699e708176c146e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 10:14:52.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fbc84d2a5e8eaeb1e24a73bb1e75316ccd8019c1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 10:14:57.000000000","message":"Patch Set 2: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79c7dcac9e7ea81a5ced375e49aaee07f78bbcf3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-02 10:17:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"daf3f15f1b0181a05ae613a78ecf9cbddb62aa91","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-02-02 10:29:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ca44d2d645343b3afa79cb7e41ab5bec8732db6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-02 10:32:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3418a52e029f2d0fdbb189f4b443d85cfaf230ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 11:52:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/730e9c6a19b44e6a836b43a4abf6f76f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48821d02968746e49b845b139020b1f6 : SUCCESS in 8m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d5ee7c9d5b674c8c92f3247f5f151b1d : SUCCESS in 6m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfea43c3fb6b4dc0925acfef8194052c : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3777c062c2436ab8a434b91cced179 : SUCCESS in 5m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d38d9000f62a46c19bce11c2cb0b3b9c : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3771fafd91a46318181ab9fccda05bf : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4f6850aa8414b16bea761019c090a22 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a71e8437f912494a9dacadf028daf181 : POST_FAILURE in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9ce8fcec0b04d9186f1dd25e802ea5d : SUCCESS in 11m 15s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ad9b50a2bdf24aaeaccc81fff19ac841 : SUCCESS in 13m 52s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a1d4a6670f64174a80849e8b28d34ab : SUCCESS in 13m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3d71a9bfb0964453895ba3b96c215538 : SUCCESS in 24m 25s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b92393e9efc4d4d8aede3dbf13ba2ea : SUCCESS in 18m 17s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b3c74301b88142da9b890c7cc0bcc558 : SUCCESS in 56m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a5073c65d12e47ce92959ff45bf5ddc7 : SUCCESS in 54m 24s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/86835f3b6091424e8d7493884983c06c : SUCCESS in 55m 13s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5408563e59ae4f1a84e1183561f040ba : SUCCESS in 55m 02s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c850c240f19d4e94a3e74fc9a3aef4ae : SUCCESS in 55m 40s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fb0671ce95024991820b9c488dd60ada : SUCCESS in 1h 32m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e7c9694b30214f8b831026592973944d : SUCCESS in 1h 25m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2c4e1d3d9234bb99ccf792167608673 : SUCCESS in 1h 22m 42s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0538c36a5c149a0a2a730a5549ebf17 : SUCCESS in 1h 30m 48s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/5898a28c578d45a69ed999c38af29051 : SUCCESS in 9m 47s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/20810b5bb740405480275e95de3392cb : SUCCESS in 16m 07s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33016172bae8403880585ba6975eb038 : SUCCESS in 17m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a38a254ea1f54138ba73657a222efa5a : SUCCESS in 15m 32s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/1e649672ae4e4dd8934d8e983ab17d74 : SUCCESS in 26m 02s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/22721f24f12742f28c33a861838b5c93 : SUCCESS in 27m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b958c13894204f6d80bbfa168e74a3dd : SUCCESS in 27m 37s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c945ff03458b419580e76537d2e2a41e : SUCCESS in 32m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6ab3d1e1d75c487587933a88c1a95cef : SUCCESS in 8m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bef9950b3c847e2ac4f236254c29bcd : SUCCESS in 15m 30s","accounts_in_message":[],"_revision_number":3},{"id":"cd74e9929f8334786e61b8161a30ed15f2319665","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-02 12:21:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4789e27bc893afed584049fd69402e13c6f584d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 13:53:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed106cf65b5f4f29a0e23cf085a36d46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81a71541a6d4313a363d4ab35751a82 : SUCCESS in 4m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7053b994a7fa43b88e1d9082339f6242 : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/378afd1d1fef480d87c9c95bac85882a : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d3406ef96543519579099de5ae8c4e : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e53b91b77204c119e1baa476bceb6a6 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06b8ed2bbb70468bbd2b4bf5779e4e4d : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/132a72d132624525b1bfbe868b31140f : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/02bec26b09e149789877be5cff0a2e3b : SUCCESS in 5m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c37be06774dd48208d740f533543b5b0 : SUCCESS in 8m 44s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/104daeac17474639b24ec73b24367e63 : SUCCESS in 13m 00s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5885d078a3bc4f6ca10c8b96dfc9e873 : SUCCESS in 14m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a8fbff1ff83c4f16b67833f104a93d22 : SUCCESS in 16m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ba0652477644e829a3ba0a11e7e0779 : SUCCESS in 14m 29s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/22abbd9d362d4986a79fc7431461dcfc : SUCCESS in 53m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fab69a861b1d454f86f6a2e85222c7f4 : SUCCESS in 55m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e8b60b401d56453886e43c140d635904 : SUCCESS in 58m 48s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d26e7c9209f4e4b893ed6a9eb0a6a3a : SUCCESS in 56m 01s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e6515d2e09b64554b0951d2d9e4eb3c6 : SUCCESS in 52m 16s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/6f48b9b0c66c4c84987c4693e543938b : SUCCESS in 1h 27m 25s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/81fc6650a5014ca7bb186dae6d390cba : SUCCESS in 1h 25m 04s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfc3c40f96154131a29f7870a4882fa6 : SUCCESS in 1h 21m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6823180b2dff4443931885fcb2af3083 : SUCCESS in 1h 28m 28s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/c94887e3871045aabd5f84d50fe88ec8 : SUCCESS in 9m 28s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/caeba40a2ce74beebd3f9b62591193aa : SUCCESS in 22m 00s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ee0c18c34b84f5da8d5b23752a4fc98 : SUCCESS in 8m 30s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7ee88a0a20084dc580be214c17cddfc9 : SUCCESS in 16m 39s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/8c45beadcee44b268b701555779dbec0 : SUCCESS in 16m 21s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1d9a766ca36e4f5ab8cc5c6ca03c5381 : SUCCESS in 16m 33s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a144411a6ac449db940a9095caaad3b : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7cf42f25cb349a7a9afaccba5e7c1f3 : SUCCESS in 46m 47s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/90fee3dd68e94d4d91028c4b1200811c : SUCCESS in 18m 24s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/045c22fa9a0e4297a3e10b9a883426f5 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":3},{"id":"314c3ab39366d575d163f312580bc6f57d272250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 13:53:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"49334ce823c222ad1d67109308c9153ab30f4e11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 15:28:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e445fc6da841bb8000fa011d94b403\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2af04cd2ba284b51bc0f79b1f8c4f9df : SUCCESS in 9m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56c62097adcc4d3aaced80d2f529cfff : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab004bc07cc54085928dc69722cd03d8 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86a33f9ef7ae420a99179140907436fe : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a019b606f87c4916840014e8896faaf0 : SUCCESS in 1m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b43440584212471e84d1669bc313259f : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/efd4ec81511a48029592de958d52e271 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bdaef9ae418947c4beb9e266ad643b87 : SUCCESS in 11m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0cadaa6ad376483fbd5c289830face08 : SUCCESS in 15m 38s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eadc0339e3bd44c094fd76e6f907cc60 : SUCCESS in 13m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/c98da6ac100f4af086b64058b2743fc0 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e06268495dc4689b86bdb6bceae6845 : SUCCESS in 15m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/273a1f1e6100483a9a182f9230f8421f : SUCCESS in 56m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a897aa984d204eed9c6c8fe96ea40439 : SUCCESS in 56m 03s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/96f388bd62c24edb853e4417df66a98e : SUCCESS in 56m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47cb73724e4845479d27ff60af216636 : SUCCESS in 53m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/83258b3c0d754f759f1031c39be96c05 : SUCCESS in 54m 02s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/b8d8bb8b14b948049ab398042904f951 : SUCCESS in 1h 31m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3fc9643547f3475c91fc5016e3886cf9 : SUCCESS in 1h 25m 53s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b36c9385fcd149d49c1ebd23fc8f85fc : SUCCESS in 1h 22m 26s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b77c5b949dcd46179274c1ec9c5fae48 : SUCCESS in 1h 30m 05s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/ca56ffefecd74344aaf54c9bede58bc9 : SUCCESS in 22m 56s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2f783796f8e84dd1993ea19f2dadd81e : SUCCESS in 13m 45s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1c8e5c59aa246e084059f241eb3c532 : SUCCESS in 17m 33s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a2e84a80f1804cbda6f2692d64f4e08f : SUCCESS in 16m 09s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/0d2f471ff96d432ca7eb0549ea38758e : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/86a6bf906eb3453aa824bd5e21e59a81 : SUCCESS in 32m 22s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e63cbb01b904cfc856da0c41ce93076 : SUCCESS in 28m 19s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/648ab989f7a04e28a0fcfe2fb1db5829 : SUCCESS in 27m 26s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5933cc99c9394e80bb7fda966e6850dd : SUCCESS in 8m 44s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d87b99f7428e46488abec74f2d6ae2be : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":3},{"id":"4656068db48bce975e7d2fd8f2d854169cd28f29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 15:28:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"172cd80f592ac4cbd9d24cccc315981d5ad84e58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 15:29:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f6a95aa3671496caa379873c1a73bf9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29ff7e82d2d47539cccbb3b4c457617 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93e076aaf6349029fbeb4b81a0b5da7 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0115b6ea5efa9a8ee0583b50f948ee1758960a67","revisions":{"afbce56677f0e6b1bd05c3ccb5ea50c1859750a5":{"kind":"REWORK","_number":1,"created":"2026-02-02 09:57:08.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/79/975379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/975379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/975379/1"}}},"commit":{"parents":[{"commit":"9f1a265be77b80858b3e65d7f69c1f5aa08ca4d3","subject":"Merge \"Remove undocumented playbook for creating external network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f1a265be77b80858b3e65d7f69c1f5aa08ca4d3"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 09:53:30.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 09:57:02.000000000","tz":0},"subject":"Increase stackhpc.linux collection to v1.5.1","message":"Increase stackhpc.linux collection to v1.5.1\n\nFixes race conditions in setup of vGPU SRIOV devices.\nAdds support for configuring MIG devices without creating vGPUs.\n\nChange-Id: Ida4b38b98ab1140d44941355bc7ff991b0b5aa47\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afbce56677f0e6b1bd05c3ccb5ea50c1859750a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/afbce56677f0e6b1bd05c3ccb5ea50c1859750a5"}]},"branch":"refs/heads/master"},"32f22a2ca558bddcd53ac45e62f0a7d0112bd4db":{"kind":"REWORK","_number":2,"created":"2026-02-02 10:10:32.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/79/975379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/975379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/975379/2"}}},"commit":{"parents":[{"commit":"9f1a265be77b80858b3e65d7f69c1f5aa08ca4d3","subject":"Merge \"Remove undocumented playbook for creating external network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f1a265be77b80858b3e65d7f69c1f5aa08ca4d3"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 09:53:30.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 10:10:27.000000000","tz":0},"subject":"Bump stackhpc.linux collection to v1.5.1","message":"Bump stackhpc.linux collection to v1.5.1\n\nFixes race conditions in setup of vGPU SRIOV devices.\nAdds support for configuring MIG devices without creating vGPUs.\n\nChange-Id: Ida4b38b98ab1140d44941355bc7ff991b0b5aa47\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32f22a2ca558bddcd53ac45e62f0a7d0112bd4db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32f22a2ca558bddcd53ac45e62f0a7d0112bd4db"}]},"branch":"refs/heads/master"},"0115b6ea5efa9a8ee0583b50f948ee1758960a67":{"kind":"REWORK","_number":3,"created":"2026-02-02 10:14:52.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/79/975379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/975379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/975379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/975379/3"}}},"commit":{"parents":[{"commit":"9f1a265be77b80858b3e65d7f69c1f5aa08ca4d3","subject":"Merge \"Remove undocumented playbook for creating external network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f1a265be77b80858b3e65d7f69c1f5aa08ca4d3"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 09:53:30.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 10:14:42.000000000","tz":0},"subject":"Bump stackhpc.linux collection to v1.5.1","message":"Bump stackhpc.linux collection to v1.5.1\n\nFixes race conditions in setup of vGPU SR-IOV devices.\nAdds support for configuring MIG devices without creating vGPUs.\n\nChange-Id: Ida4b38b98ab1140d44941355bc7ff991b0b5aa47\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0115b6ea5efa9a8ee0583b50f948ee1758960a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0115b6ea5efa9a8ee0583b50f948ee1758960a67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
