)]}'
{"id":"openstack%2Fkayobe~975410","triplet_id":"openstack%2Fkayobe~stable%2F2025.1~Ida4b38b98ab1140d44941355bc7ff991b0b5aa47","project":"openstack/kayobe","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-02-02 20:48:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida4b38b98ab1140d44941355bc7ff991b0b5aa47","subject":"Bump stackhpc.linux collection to v1.5.1","status":"MERGED","created":"2026-02-02 15:29:09.000000000","updated":"2026-02-02 20:49:23.000000000","submitted":"2026-02-02 20:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975410","cherry_pick_of_change":975379,"cherry_pick_of_patch_set":3,"meta_rev_id":"f4466e0e95d864b0171479a8a2230ecef37199af","_number":975410,"virtual_id_number":975410,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-02 20:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-02-02 16:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-02-02 15:36:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2026-02-02 15:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-02-02 16:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 15:36:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-02 15:52:55.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-02-02 16:03:48.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-02-02 18:11:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a52f42d5d5d99699cde3bb58996ac52fd1e7d46d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-02 15:29:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d7fee79760447837bad004124ae1988b9a8d97ee","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-02 15:36:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d4023b1d3ab7fcdda9161f215acd63e86a5322e1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-02-02 15:52:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bcc191c0f9f1d6b37656a16ff5249bda017641e4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-02 16:03:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e634c5216dd7b019efbebf3b46ec9ac248d8d06f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:11:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f842da94f8624d6d9c7f6b891a5be899\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b496d675fccc4aeaa7c733f72a5f8ac6 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12b98b90712e4ae7bd11beda878dc9fd : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d745fb0fee943d2a0033fb72f8f2d85 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f1c57f560814df18e7efdd9fe88ae00 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c1dfa099aa4455d8f29e6ceb9ef27c2 : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f64909d000c42b6b0166685e1049c88 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/05db16a5c2d1454fb1d9dbda6ee35ffd : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6bbe79527bb94db5a98e4389eaa57d09 : SUCCESS in 12m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/d3f6dfb41a4c4052a7c9efade5715d14 : SUCCESS in 13m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c4bcbd883f84217a1cf063562f71cca : SUCCESS in 14m 32s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aa4271a959a245a29ce8afa6b858a072 : SUCCESS in 12m 33s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3fd68e14fde74c039ffb8437f54cd198 : SUCCESS in 9m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/eacc7f4daefc448cb6bdaec72625197b : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/450ee5166bd14308b9e97001527ca367 : SUCCESS in 13m 54s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8afeb63163784435832b4e06a2d237ee : SUCCESS in 50m 57s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/71255a6e43a64ab1a0048b83aebc62bc : SUCCESS in 57m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dd7dc94bdb6f4911958f7be5e0b62011 : SUCCESS in 1h 09m 16s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/f2d002dbfcc74c49a10f00eb96e54eae : SUCCESS in 33m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3c5450bfb9af450da4343139b8c3ebdf : SUCCESS in 35m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/37b4cba6bcd342688dc9a6c63caf46a0 : SUCCESS in 1h 06m 31s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb3a3e3f82a34e1b9ed77db73265047d : SUCCESS in 57m 57s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ccbb6202cddb4a488dd18a588eb03c9c : SUCCESS in 29m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ce0a2af53fcc41d49ad93b2de027efbc : SUCCESS in 2h 34m 12s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/8b30a4332bcc4ae0a9fe6af9f6889eb7 : SUCCESS in 2h 33m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7595e5dde0574087b53bb1f69f64dea0 : SUCCESS in 2h 29m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fcc45e45669e492189a0fb8bac92d04e : SUCCESS in 2h 30m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d6b3d0f36f3a44dda478e20e0edede4a : SUCCESS in 16m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2e60cf97c6c541d5bdb5be986df91a94 : SUCCESS in 20m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/86941f87f7e945e7a52797562c60794c : SUCCESS in 18m 18s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/161b1582b52043adb4fb20be58f5cd22 : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cf0c251f9bcf46dfa0980a2fabc5846f : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6003fd8fdbfe430ca0ef2f3de3abd86c : SUCCESS in 14m 45s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/51423ce8f96c4fefac127bffb8f4fe72 : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/771490a740aa4b8ea971949d84d7da09 : SUCCESS in 45m 07s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2032cb38b874fbf9275a2232b3c3538 : SUCCESS in 31m 57s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/1a2868291c2b48a19abfa578a6768489 : SUCCESS in 8m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/075221bc2db94da0a06f644dff02f2e9 : SUCCESS in 7m 32s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35a90fd129254168899e92899ea60704 : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1ca98ccfe5ff6e6e41dff1a34245996d94890990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:11:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab6bc3c58323b6eebf1d5c667a419ff7b894bafe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 20:48:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0945be1af9fe474491cc7d2a527961fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee2464277154fd89ad15f15b187194c : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c8697672f5c4821bd536af8087a4b84 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a286e2522b3a41c38893d9c77a625d21 : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f1b15770cff4a979e4e6ced53db6155 : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d776256594f4568b45d6e9cd325a0d1 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/310c12b5e7e04b989ca0b33db51f8443 : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5884b302ce6b4722aa2720dd91695e7c : SUCCESS in 8m 34s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3da1b8553a3c44949e992a997e2ad5d4 : SUCCESS in 13m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/abdaaf5bdb7d40b193c2e9dfc8257bc5 : SUCCESS in 8m 19s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/671324e0dffa40b7ae91581e698c6163 : SUCCESS in 12m 14s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7e7d0305087a4b73bf0c9cf486411ffc : SUCCESS in 11m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6c49c0973385423f8589bdc8bf6fefdd : SUCCESS in 10m 19s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99a6880d73e741d189f048c50ba0a8af : SUCCESS in 16m 24s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/8d8401ec4681440dabf81d460c75e2f2 : SUCCESS in 52m 55s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/d43077c3f59e4b4bb251db58f32d0bca : SUCCESS in 32m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ef94f365e43f4937bbf833468d4ca7dc : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-rocky9-podman https://zuul.opendev.org/t/openstack/build/a4cea054bc714842bcaa498d6893f408 : SUCCESS in 1h 04m 31s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/2e848d439ffe4679b7167ecb893ae10a : SUCCESS in 37m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/62b94913c1be40838a3ae9fc784a278e : SUCCESS in 1h 13m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eafc2e14101042c8bdde448bafb6ec22 : SUCCESS in 30m 13s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f62dfac1a8c7430189fe9ebc899462ec : SUCCESS in 33m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/762febdb38e74ffb9f4b55e34b504e58 : SUCCESS in 2h 31m 48s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/ae931a0cd79d488fb9e11a667605f70b : SUCCESS in 2h 28m 08s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ded76b0200648798e5a2bd81177e22c : SUCCESS in 1h 42m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74a8ba2c3dcc40dfaeefb864bf91267b : SUCCESS in 2h 12m 55s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/1f0b58a136f74f449f0baca86d94a43c : SUCCESS in 17m 10s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/09f5986c41964e8888e1bdd0606d1495 : SUCCESS in 21m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b71a00b1e75b4666adc1ca17ba7776b0 : SUCCESS in 9m 11s\n- kayobe-seed-rocky9-podman https://zuul.opendev.org/t/openstack/build/bcc6262b359b43c8933e008182c8a5ff : SUCCESS in 14m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/13265419335f41c2883676a22c980b50 : SUCCESS in 16m 03s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8d52fda056dc44c48ad4d784cd95b6a3 : SUCCESS in 14m 07s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/14524a6a703e4e0bb97ab02d1a90320e : SUCCESS in 31m 58s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/00bb466eb3604e978abfbaa314ea75f0 : SUCCESS in 42m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94362056ac724db3b7331820952c811d : SUCCESS in 21m 32s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/2376460fec194c27bde57a941ce8c31b : SUCCESS in 18m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6575b47d66ff45f58db07645e80a82ac : SUCCESS in 16m 03s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c497708e2c17416eb42ae49a4a62a075 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"19d897b4fac52b292f3cb6085b4b2959e6101568","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 20:48:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f4466e0e95d864b0171479a8a2230ecef37199af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 20:49:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccfc605f80fa45a1909bba056e9d7554\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5f5331d6b54f7199556a133ec1472b : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/286da429ae6d4842a5af68dbf372f893 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d850105571080d48b81693574c2e2b4d4d0843db","revisions":{"d850105571080d48b81693574c2e2b4d4d0843db":{"kind":"REWORK","_number":1,"created":"2026-02-02 15:29:09.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/10/975410/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/975410/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/975410/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/975410/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/975410/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/975410/1"}}},"commit":{"parents":[{"commit":"e6f76976563dd0a89497f8b4cd15113579ba07ad","subject":"Support swap configuration for all host groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6f76976563dd0a89497f8b4cd15113579ba07ad"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 09:53:30.000000000","tz":0},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2026-02-02 15:29:09.000000000","tz":0},"subject":"Bump stackhpc.linux collection to v1.5.1","message":"Bump stackhpc.linux collection to v1.5.1\n\nFixes race conditions in setup of vGPU SR-IOV devices.\nAdds support for configuring MIG devices without creating vGPUs.\n\nChange-Id: Ida4b38b98ab1140d44941355bc7ff991b0b5aa47\nSigned-off-by: Matt Crees \u003cmattc@stackhpc.com\u003e\n(cherry picked from commit 0115b6ea5efa9a8ee0583b50f948ee1758960a67)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d850105571080d48b81693574c2e2b4d4d0843db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d850105571080d48b81693574c2e2b4d4d0843db"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
