)]}'
{"id":"openstack%2Fkayobe~975963","triplet_id":"openstack%2Fkayobe~master~Ic5094dbc2223a46638582be1159e37d404c6d5fb","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-02-09 14:54:49.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2026-02-09 09:38:26.000000000","reason":"\u003cGERRIT_ACCOUNT_17669\u003e replied on the change","reason_account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-09 11:35:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-02-09 09:59:11.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"Ic5094dbc2223a46638582be1159e37d404c6d5fb","subject":"Remove compatibility with Ansible 11","status":"MERGED","created":"2026-02-06 21:14:23.000000000","updated":"2026-02-09 14:56:41.000000000","submitted":"2026-02-09 14:54:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975963","meta_rev_id":"86ab9b711c22ec304601cb3d7d66fd8940bf58a4","_number":975963,"virtual_id_number":975963,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-09 14:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2026-02-09 09:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2026-02-07 18:18:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-09 09:59:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-09 09:59:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-06 22:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-07 14:03:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-02-07 14:03:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-02-07 14:03:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2026-02-07 18:18:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"5424b06196b9ad1c4d4fc89371b520996d1be98b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-06 21:14:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"583c99a801415e89727915e586377537e1001302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-06 21:14:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e59950a12c997eeb9b8453b660b81283399dba13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 22:53:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b59ee3237ec40df80a779c20c6dac63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18aea01cfd2249f6bf30ff0752095b63 : SUCCESS in 8m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2e1e9b23aaf44b19af6d7039ebd8a4d : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaa36103647d43b084b60c57e64c31a4 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d826a92a9dc4171ae320540bddf5d7e : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e5554fe21774f2c8fbfbdff296fa679 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b7f44d0278a4228a1c43c4184b459be : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2afbdb3099b4d4ba570e10bb1518fab : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d3b2a2ef2fbc4ab1a8600fc108e1d79e : SUCCESS in 12m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e06913f73bd4449c97d452a8816722fd : SUCCESS in 18m 36s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c29d1b1820534c4fa0c4f50eb6e59c9b : SUCCESS in 13m 44s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/15dce1ef99374960b500772fb7d6a66c : SUCCESS in 12m 17s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1aede973157a423097d94e351643fe82 : SUCCESS in 15m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f3af4e233a044a9a940349a2eeb4f8df : SUCCESS in 59m 48s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cb94af529d204dd1a250c7528042ef6f : SUCCESS in 59m 15s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/173a9f4d35fc462c9caa5debf73ebd5f : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f952d330e7c9434e99a007837817275f : SUCCESS in 56m 59s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ad69c09870c4b7890912dbab7c0c363 : SUCCESS in 54m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3e0d6f33e37844f3bd2119d9cd656712 : SUCCESS in 1h 36m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/de157251277249ae80924491936bd29d : SUCCESS in 1h 32m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/31d02bacf59b496d8799f06ca727e849 : SUCCESS in 1h 23m 00s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d2785126971140e38547a3530dc8c4ee : SUCCESS in 1h 31m 08s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/a12d80cb68c24943bbe42db7f52a4357 : SUCCESS in 24m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c09245a4d0654b0bbc5cf3e5aba993d8 : SUCCESS in 21m 20s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2181a51e1e9a410294aca03920b9701c : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5d291dd2b9b14cc0bd33131e7b620c1c : SUCCESS in 17m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/a974071427ff461d8c5fcf06e69076fc : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1e212b6eb07a4826a7912c5536e4ae44 : SUCCESS in 39m 23s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a410845aa15441a2bde70f2dba241bb2 : SUCCESS in 29m 26s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29177d685e1848d4a3d4ac5f6cb570e4 : SUCCESS in 31m 00s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/abf0ef5e1712469996f0d027a85fde32 : SUCCESS in 10m 25s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52227912840e441a879fc1bd89029c64 : SUCCESS in 15m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b8f253787fdfa55ff811e1225f161a2c5f43404d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-07 18:18:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"94000d8b1d1a77c5b8b9be65e61ae6b9040da035","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2026-02-09 09:38:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d72314046d6c973926e635884dd9664bbef02c3","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-09 09:59:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"60a8e1d96f35dabf632bcb923e805fb46ad764f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:59:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c51b1c9ff63570428852614c90f55081622c6071","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 11:34:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eed8cb9cd6ff48f891c4b69d58ea9b54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4225afa91141caab1fa455d7d96607 : SUCCESS in 8m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cfc2d498122543009e76fe7f88bf5236 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cec2e2c0b9347bba0728b55f88c9560 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/694e2d67fff54e33823ad2e916d0417c : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff9001b7d1eb4e1cb6404882ee82a89f : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12295c9714f848eb992c83704270af26 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/895bb5f326734d51907deb4923a27737 : SUCCESS in 13m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b9a045d0c41d402bac02faed91e65977 : SUCCESS in 13m 05s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/54cbeb1f80ed4f47bd492d129dd9f70f : SUCCESS in 15m 04s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7de0ec853c1148f79262b4b839e2e77e : SUCCESS in 13m 01s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/68158b7a25fc4acd9a0277a4a3730f6d : SUCCESS in 13m 52s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2b8f9e37787d485ba2e1208dad692cf9 : FAILURE in 41m 09s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/cbeb3226e81b46229c8461e116e76f90 : SUCCESS in 54m 57s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6dde15c2ad7141a9a5173af195c20d6f : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/08dd353ccf8e4bb495784f2723c07335 : SUCCESS in 56m 24s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/eb7587b39ef2446da63f44a5af91a6e7 : SUCCESS in 54m 51s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/bae154b69a74434996466ce314465f59 : SUCCESS in 1h 31m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/8b1262b8f94c4b13a931291953c087b5 : SUCCESS in 1h 25m 34s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e9dfac91988a4732bde1a10f61bc22d7 : SUCCESS in 1h 25m 28s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69e18464900e4db6bb7c77363f168f4d : SUCCESS in 1h 33m 04s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/d6d8cf5b9054454a8dd683165e39f26e : SUCCESS in 16m 13s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c91aec245ea042d5b08ce5d505b5b687 : SUCCESS in 21m 40s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26bd2fc02fd54b81863ef97c023b3c90 : SUCCESS in 9m 58s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/aed6e6b617dc4ad9916f7ec95ea83156 : SUCCESS in 16m 10s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/70b09aef95c64c9da007452a50d295c3 : SUCCESS in 32m 47s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/5565767f256d45ceac0721221b5755d7 : SUCCESS in 37m 16s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/89e46788f1fa4efdba0a4185db32aafc : SUCCESS in 26m 50s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0503302d653243809679516a824b83ed : SUCCESS in 24m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f4c7c5b10d7746d7b1d1d5fd080ecdb2 : SUCCESS in 18m 45s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae371462a314473dbcef8ff11b08854a : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":2},{"id":"e4d8d9a49d263aa3593b518e2f6b2d298a27cae1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 11:35:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca25552df0f6436459c28fa166729832e8ac9f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 13:11:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17694374ed7043f3a31afce419fcad8b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13e2a767978b4defa985c3e03adb5187 : SUCCESS in 8m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/63985db199af4f7f8ce78910917ce965 : SUCCESS in 5m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89e061f7e9534f34ba8e3fbccfded5d5 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a9e8c8018441f19734c924c3865610 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4c899e54151944c480934c2c3752a7b2 : SUCCESS in 2m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24c0a548305e456c92fb633a4cedf7e1 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b93d6a1b3d44e9b8921ae7bb48f889d : SUCCESS in 5m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22fc66e0acb94a64bb33d213d74eae4a : SUCCESS in 13m 01s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5ecbd0313e07419c96087a5d01fb2df5 : SUCCESS in 15m 13s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6a89ae17c80a43f2a40dbe401aa9bd9d : SUCCESS in 11m 25s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6d46db98bc82470b8c05b33e2df72b5b : SUCCESS in 23m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71e78b803cb044c7ae35cfa6ddf0172a : SUCCESS in 18m 48s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5dec9a9e9e6540158bc36a7ea0219f08 : SUCCESS in 57m 10s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/94abbe1c956242349b49360c89e52dfe : SUCCESS in 54m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/cd46a1f3b7c049719d7b9ab002c5094a : SUCCESS in 57m 09s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e1f345d4666472a935948a05d4a28d7 : SUCCESS in 54m 08s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cee49e5dfce14d80a43489907680dde3 : SUCCESS in 54m 18s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/76714e164afb4200b586661d5c2401be : SUCCESS in 1h 34m 11s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d992a0e9c9d34f6685bce5494b7f7d40 : SUCCESS in 1h 22m 59s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cca50782896c495d9c7fe01325d28f59 : SUCCESS in 1h 26m 39s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6abc0f14543442bbacc748f634ecd19 : SUCCESS in 1h 35m 02s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/8216a51cf38d41ed99eeb154cbae72c4 : SUCCESS in 9m 14s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/5a046b5663db4d26a875b5b7123cddeb : SUCCESS in 9m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fe84c325289a446b98835d6d0701782d : SUCCESS in 16m 31s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/89d5a22a88084f628c68415ac4c4e1d4 : SUCCESS in 18m 03s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f430e54cc86c4f6c9dcc91708a1447a7 : SUCCESS in 32m 38s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/d091aa4afd26471a8ba035b7878ed406 : SUCCESS in 16m 54s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29ce16d237884862894f69a92e3c504d : SUCCESS in 32m 15s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26c8013c77304c44a68c77df8ae929ee : SUCCESS in 36m 51s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3ddbde5bb5bd4870b6ea57bf2bb5518c : SUCCESS in 18m 11s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/128d31129e6f48c5827ed77957a1e335 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":2},{"id":"49be928919dc99aec86e762081ccc77e34b1ab7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 13:12:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a95b62b43562a21c619c5a00edae712d6edc449d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 14:54:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b361e21658874fdfa9ddf0c20f0d3f20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452cf51d02d647908aefceacfd722c35 : SUCCESS in 8m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4794464d05e740c2adf47d24f87bec91 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f61cd6edbac4a629dd29dc9d3b955b7 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7efba6cae6a0477da992a13dbe8dbd57 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4ecc8fbe7514eefaeda0b6a99050c29 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a488c24137d14990af69fc5e9f4d5ad4 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b4bcf35a7164c8aa0b326580ffb2b82 : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f5ecb3a40f534b84a6034994162dc3a5 : SUCCESS in 12m 43s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d32738237564e848230591df7b95bc6 : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/3f8029840f504cb2bdc57c7d0de167c3 : SUCCESS in 12m 16s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6920562269264efeb66dde6292f9e3c8 : SUCCESS in 15m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/fb0310042c964c97aee18ef69e989107 : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/998d7cf6b3ae4a81a60c3d84ac5cbddf : SUCCESS in 1h 01m 58s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/29cf8d30380d42d5ac74804466f66430 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/09220a5bd74b4336b2500ccee561ea03 : SUCCESS in 1h 04m 00s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5f2326d319dc4aeba0c56ba91b1b028c : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/47e83cd3ec5842198ad4176219f2922f : SUCCESS in 1h 39m 28s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e672346e82bb4f468a342ff7e8f921df : SUCCESS in 1h 35m 38s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0e551001d9b64063af611ba02b514f04 : SUCCESS in 1h 33m 30s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dcc106acf31344d083e287c93e46f6a2 : SUCCESS in 1h 37m 10s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/03e3d65427ef4b9cbcc7a9673fb009c3 : SUCCESS in 10m 57s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/2d48adcd5de8431e8d58bddf33e83fed : SUCCESS in 10m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e71ce93eaab41fbba9622478be9bdb6 : SUCCESS in 16m 05s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/718a50f81fad40b480e4f38befb3978c : SUCCESS in 9m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/66b8795684fd4e95ad4e30f2b05ad321 : SUCCESS in 17m 22s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/e5a36572f1724053b0203f8013b5195f : SUCCESS in 18m 00s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2053078275a34deb85f2bf60ed2623ae : SUCCESS in 29m 46s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/726c4d3efd284e0e8b6c4ebd95757362 : SUCCESS in 26m 31s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/7eb49f6cd2794954bdaed586c098dcd0 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1fe659f41bf4a6dbc48c8c80401973b : SUCCESS in 13m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7fa9cbdd42daf099754c516a8c69fe2e734154f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 14:54:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"86ab9b711c22ec304601cb3d7d66fd8940bf58a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 14:56:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a941821f24c144f9be2397d5569f02a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3799a5462294cf28086c7b7598b936a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06c025e7f4688c5f59b035948978b12d0f9f862b","revisions":{"aea71126ebcdec87e8e34444e986201dac2e8179":{"kind":"REWORK","_number":1,"created":"2026-02-06 21:14:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/975963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/975963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/975963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/975963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/975963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/975963/1"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-06 21:14:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-06 21:14:18.000000000","tz":60},"subject":"Remove compatibility with Ansible 11","message":"Remove compatibility with Ansible 11\n\nChange-Id: Ic5094dbc2223a46638582be1159e37d404c6d5fb\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aea71126ebcdec87e8e34444e986201dac2e8179"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aea71126ebcdec87e8e34444e986201dac2e8179"}]},"branch":"refs/heads/master"},"06c025e7f4688c5f59b035948978b12d0f9f862b":{"kind":"REWORK","_number":2,"created":"2026-02-06 21:14:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/975963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/975963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/975963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/975963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/975963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/975963/2"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-06 21:14:47.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-06 21:14:47.000000000","tz":60},"subject":"Remove compatibility with Ansible 11","message":"Remove compatibility with Ansible 11\n\nChange-Id: Ic5094dbc2223a46638582be1159e37d404c6d5fb\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
