)]}'
{"id":"openstack%2Fkayobe~976011","triplet_id":"openstack%2Fkayobe~master~Ibab43c633cf718664e144470e944ef24425695fe","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-17 19:32:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibab43c633cf718664e144470e944ef24425695fe","subject":"Split Python installation from user bootstrap","status":"MERGED","created":"2026-02-07 21:47:09.000000000","updated":"2026-02-17 19:33:35.000000000","submitted":"2026-02-17 19:32:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976011","meta_rev_id":"2f526a2b76a9d698e4dd397d7e040254b4607f3c","_number":976011,"virtual_id_number":976011,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 19:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-02-11 15:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-02-09 09:12:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-11 08:05:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":2,"date":"2026-02-17 17:52:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":1,"date":"2026-02-17 17:52:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-02-09 09:12:29.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2026-02-11 15:16:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-08 00:02:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-09 09:12:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-11 08:05:35.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-02-11 15:15:47.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-02-17 17:52:06.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"4c86aab581a46d0c4cf1dbe96dd5d55a3941d33a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-07 21:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4106e9a5d89dde964b1c8204df507c4c547036a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-08 00:02:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f302367426fe41548ba01faaf24d9a20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c49138e8924dae8fa66b1e98828941 : SUCCESS in 9m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e9c176fbd0d14b25aeac7a907640b6f5 : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ad4281330d405186a5ddef2ebba8ca : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bdac8251bc4c59959f5e0b791bfb9c : SUCCESS in 6m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05cd4d4337094c7c8d4d320abd5b48c6 : SUCCESS in 2m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e822eae1512841d1a796a895639fa44c : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd55795786524594ae82e4b6dc5a7d40 : SUCCESS in 5m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16159347760645139fb2fead4cc15207 : SUCCESS in 9m 30s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c19885a1cba34c37b199c241a684ee94 : SUCCESS in 8m 20s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34bd49d4bcfe488fa0a5cf1f278a4a87 : SUCCESS in 11m 30s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/90eafb062a9d492bb38c0cda23cd128d : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95abc10dc077421c847569b247478714 : SUCCESS in 19m 56s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/12b0cd73ad2e40c79d3c2c13e656dfcb : SUCCESS in 35m 34s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/004702629c1b439da20ff290efd71fa5 : SUCCESS in 54m 43s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/682be736968d4013af2fa0addbde8bbe : SUCCESS in 39m 15s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06f9a7d41fa34461989350bb66cdce14 : SUCCESS in 56m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7dfb6218be2f4cc394773360a2c6ecda : SUCCESS in 55m 27s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5037a47b555644e9837825ac48cc5dd7 : SUCCESS in 1h 27m 10s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2faa29b8e88b42a79ccfe17d78ae6a6a : SUCCESS in 1h 25m 18s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/26e0a0b433eb468f8cd64b857822ea1f : SUCCESS in 1h 22m 44s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b35d7f23c1644f80b11e1caca8b53913 : SUCCESS in 2h 10m 34s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0697d736da804094b1227a6e2a691c4a : SUCCESS in 16m 45s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c1970c8b82124b61a1bcc10cd3a6dd33 : SUCCESS in 16m 32s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76e9e2daef374a88855086f22816c053 : SUCCESS in 15m 29s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c249f75b572a44179a60e9ca8fba7158 : SUCCESS in 18m 29s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/3c35e9dbbe444147a5e85c0f9bf46605 : SUCCESS in 35m 13s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/98d4b36978c543ddaaefa8b577e04cc3 : SUCCESS in 27m 36s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8dcecdfeddfe43809ea1ac2504756ba2 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e3a574b67b504ab99b446f466ffaa22d : SUCCESS in 30m 45s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/73d95457ce5547a4bf0dd145cfe7fb5f : SUCCESS in 18m 00s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0320a3241f24bf0a1efb32f5f1f1231 : SUCCESS in 14m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a20dd8d2862c433581ae6940b970858ecd160faf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-09 09:12:29.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"19cacd23270fc047867f8ecba1ee405e414a6916","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-11 08:05:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5ae25c5f45072ded65f5e8f92acddbf0ee5fc6e5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 15:15:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c2d98cff7b28404394a614e6b983ff954de52bf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 15:16:23.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"906bff9f9ebb31649594c4d550f0886b19a70db1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-17 17:52:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c38102a9814a8375536f350a245e8a5ec8bd0797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 18:00:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86cc0d7c6d8b6e951b6efd517f066061268abc97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:32:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec189dd80284de9ba27f032cf6cb5c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db522fd4e9a047d99b27074e85bac4d8 : SUCCESS in 5m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/29e4574606e64851b229c3c0ef09b04d : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79a44d88b244c1ab23a0d7f240e91d1 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b88f09d2f9754a79a5a56fe23f1143e5 : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/74de872011a7458784b11a33ff3ebd3c : SUCCESS in 3m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a77ff5b106734aa682c3f03b2cc5eb41 : SUCCESS in 6m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8528202bcb4433d9fcae292e885e564 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/42991457974b4ed9ab562e54117d3256 : SUCCESS in 8m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77606d38582d45049ebf776303ead2f3 : SUCCESS in 11m 26s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8b781088ebc24e39836ffa5edf18acbf : SUCCESS in 20m 11s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0c7a78b150a4a39a876b7aad8807924 : SUCCESS in 16m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/5dd06a2bdd534eb68a392db2355c9b5a : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/ff92ac33624d43859b415f9dc4d4faef : SUCCESS in 54m 12s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/3d95f270b1824f80a82578d58fe16118 : SUCCESS in 1h 00m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d86feb9bc1f3472e8d191e397b690c84 : SUCCESS in 54m 53s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/28397ef2c5fc4ceaa0cbb41d5f2ac0f1 : SUCCESS in 54m 54s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/166fd9c800634d7496dc4ed0f413c8dd : SUCCESS in 1h 28m 43s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/b3edd0fa94eb49238a245714848f515e : SUCCESS in 1h 25m 54s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1eb2665bf8ec4ec4b5b874e7f8885b67 : SUCCESS in 1h 21m 49s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/251c8303e0194c4082705d9c73319862 : SUCCESS in 1h 31m 27s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e31281a1c2e744d88bf444925e211a3f : SUCCESS in 16m 32s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/4635ee685a6a460686d28c66ab366e57 : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0352c0a0876442b899b2081805d76754 : SUCCESS in 11m 21s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/48b81c003d66421f80baf4ae49b381a2 : SUCCESS in 9m 49s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/23c4654391de42a997ef2611cb720a2d : SUCCESS in 18m 11s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6dbeffd715204a51951d1c0f696eebca : SUCCESS in 19m 02s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b128ecb59f84b418b8a3cfc32cbb7e0 : SUCCESS in 26m 09s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc1a8151c0304d8587d6052887c1e230 : SUCCESS in 16m 04s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/748e1360bbae48c9bf079e875ae50a9a : SUCCESS in 14m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00412d2166e84798a4f8c399bf673c7c : SUCCESS in 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fa72e256a85ad14e6b94925b44fc45ddee32a9ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:32:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2f526a2b76a9d698e4dd397d7e040254b4607f3c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 19:33:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4df496f7044cf492eba76af6be2ac5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254ffc7196a14ebf9859792a7b6179d9 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"981d63926c421a5a39bc2cf16394a5b668b94aba","revisions":{"981d63926c421a5a39bc2cf16394a5b668b94aba":{"kind":"REWORK","_number":1,"created":"2026-02-07 21:47:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/976011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/976011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/976011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/976011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/976011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/976011/1"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-07 21:43:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-07 21:43:41.000000000","tz":60},"subject":"Split Python installation from user bootstrap","message":"Split Python installation from user bootstrap\n\nThe check for the absence of the Python interpreter cannot run with\nansible_python_interpreter set to /usr/bin/python3.\n\nChange-Id: Ibab43c633cf718664e144470e944ef24425695fe\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981d63926c421a5a39bc2cf16394a5b668b94aba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981d63926c421a5a39bc2cf16394a5b668b94aba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
