)]}'
{"id":"openstack%2Fkayobe~976059","triplet_id":"openstack%2Fkayobe~master~I3c793f1f63d5d220efd967dcea17beff4f861644","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-02-13 17:30:44.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-02-16 14:51:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c793f1f63d5d220efd967dcea17beff4f861644","subject":"Bump all Ansible collections and roles","status":"MERGED","created":"2026-02-08 17:59:24.000000000","updated":"2026-02-16 14:52:55.000000000","submitted":"2026-02-16 14:51:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"976059","meta_rev_id":"159169e380ebb221be00e1fd0cc1d198a717359c","_number":976059,"virtual_id_number":976059,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 14:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2026-02-13 17:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2026-02-13 18:01:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":2,"date":"2026-02-16 13:12:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":1,"date":"2026-02-16 13:12:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-08 21:07:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-11 08:44:30.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-02-11 15:19:26.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2026-02-13 17:30:44.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2026-02-13 18:01:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2026-02-16 13:12:13.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"}],"messages":[{"id":"3b46f257208611bca9c861eab19e80d4676cece1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-08 17:59:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3a3d4a759d2439f9ba3d06cf3ff19966a1cdd3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-08 18:05:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33684d05f4d56e0c8ed3f797f0918f69e9d0f104","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-08 18:14:48.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e828afdb84bfdff83e0f9944c682e4c74ac953d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-08 21:07:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4824d65dd59e415998355e75cfe6facf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2c5447b1da4439a7cdb5f762efad2e : SUCCESS in 9m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/35e40d5fba9f47c6bc37f7d51354e0b4 : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0974ad786724e23a5d7af4efd5a5bd2 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6d5b638db9402aadf2f485d0fb2769 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a25872abb1de4f6f8eb2adf6da7e8c0b : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a5af67a130f4b938b674944c2764168 : SUCCESS in 3m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16978bc904a347dcba6ce4a372610b94 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7f5ed7f684e4ef39e23b239b002d134 : SUCCESS in 10m 31s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/3a50b0be2b444459a735a99320294249 : FAILURE in 14m 30s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e7858b643bec4b59b79dee8deb0c6ceb : FAILURE in 9m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/b475b6bb5d2e473880a16d993f3973f1 : FAILURE in 7m 07s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3cf00caace345cd8d311c824fecee58 : FAILURE in 10m 55s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/efaa439923974e4394c4a66870657e13 : SUCCESS in 58m 06s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/8b2daca276924baabb1e3b8937bc06ec : SUCCESS in 58m 19s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/c5db8963d1844c3897b7454aeb38e52c : FAILURE in 15m 23s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5b192d02a624fa185ae5b897927461c : SUCCESS in 56m 46s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1ff4218cbaa44b1f9cb50f60efdbff0e : SUCCESS in 1h 12m 45s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/f0e7bed464e74740a6f046c204efcd50 : SUCCESS in 1h 33m 54s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/9d4d7a7cb5ce4229a24758b64b454392 : SUCCESS in 1h 27m 28s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/25e72257c24243c3bd499f8913577c3e : SUCCESS in 1h 59m 52s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3027ee73d55d4d3ca2faf440b98a2797 : SUCCESS in 2h 55m 17s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/e1ceb7ebccf748be8e7bc6cf32fee4f6 : SUCCESS in 9m 36s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/a5604c3c20574b40a016881d1e44f7fa : SUCCESS in 20m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f19f442beb9843e381e328263d7091fc : SUCCESS in 12m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3d214e1a9a3a4c3496a352612287e5e5 : SUCCESS in 8m 16s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/efa14c248470433e8cf1bec888f5ea3e : SUCCESS in 26m 35s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/3c3cd595a731497cb56e5468a95bd7db : SUCCESS in 15m 51s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f9de4ee3196499486d805dfb2845e19 : SUCCESS in 34m 49s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0fb994b7e8714e029f1e7c1ea7c49a3a : SUCCESS in 34m 06s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/f95e84b98e8144c8b342ab0082947592 : FAILURE in 10m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0345e92cfce4424bbdd6303beebdfa25 : FAILURE in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9323590d4aec6026a29b4970566111a74570054e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 09:00:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cbaf0c1e44860a011b3de6e4b63816adfacbc70a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 09:20:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fe3679bef385bbf5b7522a9a82a95b2c3a7599e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 09:45:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6490e70d2fcf4ed2a8131b1bd5e69acc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9371e35887d1467c9175c8d8cfd297fb : SUCCESS in 9m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4a95eb83e26d40d9a65d52c5fa837f3e : SUCCESS in 6m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de5625ba38e34089b2a65da677400bab : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9edf3a53557b4e5b976f9452324c3122 : SUCCESS in 3m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0bf81661509548819e7a9cb526f75a6e : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/acbf8d1043a54a74b8abca3321793e7e : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a39387521e104c9faef7acc2ae474cf3 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5174078e4a554f15831cb78e1cccb603 : SUCCESS in 11m 23s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/53a29efc12314a0a9d40a3d8ddbe7536 : FAILURE in 7m 09s","accounts_in_message":[],"_revision_number":4},{"id":"f1481b318114a3029c20215954133fc9b6df2297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 10:33:23.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1d9f6d4ca6065166c2812db0ea448b47074694d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 10:51:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa4fbf8681494e88a0f0fccd37caa160\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ab520423b8044b5a920792b323f4178 : SUCCESS in 9m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/902ab49e834f4e18a68c4ecf70ab1577 : SUCCESS in 6m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64743acfda1144ca84212da4d0ab6acc : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82e56b0bdbb4ae796643cc7574dbd15 : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1fe9400d9794a3d9894927b2231da18 : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73915cf1ba4b4216afdd880d6eea1e02 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6dfeddfc36824aa6aff6b8dbdca1be6f : SUCCESS in 12m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b24edbf17d244714848e87d39132ae1a : SUCCESS in 10m 36s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/04ae94cfcf5c4f299a1fed7221e38d70 : FAILURE in 12m 56s","accounts_in_message":[],"_revision_number":5},{"id":"608be7b4b9d1af4a2e5d90f3314dbd98aab6b116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 10:59:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0107f46a5d57e99b878dc63128d01fdf8c200af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 11:17:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c79b9ea87b140688cb47a8e60b7a5e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e09d8fbf4946aaaef25224fb44b7ba : SUCCESS in 9m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/790cd04a5b23474eb75ab2534ce4da5b : SUCCESS in 6m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ff226165ffc427aa254d832132cb64b : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7bff9a8ee914b6988c59962047a3916 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bc8bacfcb4e744afa981f85b54f77727 : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b3a408eee824c798e25ef316a02e237 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae55541e95b8466ab4cb557f1202f569 : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/761a334ab1534ad2a0618efb823953da : SUCCESS in 13m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/8c7dfce9e1ef4f8c8632adf523175298 : FAILURE in 9m 46s","accounts_in_message":[],"_revision_number":6},{"id":"7b89977fbb9feba5543286bc1f056d9b4d4d3c4d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 11:32:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"93ed8c0aab01eb9f2c0258a7507b8f2ffffaa369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 11:45:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6b7ce830fcf426c81535920daa1104a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b53a76fe964b138121c4f647b8c617 : SUCCESS in 5m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0b0c8e379516477da3a62e0d07442f81 : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c21bf89e33c470c91d39f1b930998b0 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63cd150caaf644229e85f2d5865195d6 : SUCCESS in 3m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8624176f07449f6b71b3c43fa92cf96 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/03b026f654554d07af3c9f31d773ae20 : SUCCESS in 3m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b18027a55e624f7890f21d8aa7609a83 : SUCCESS in 10m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c2ade7a833a40929d91edef9f54d433 : SUCCESS in 9m 01s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/9e991a1edf414b4fbf3221cd57d2e56e : SUCCESS in 11m 28s","accounts_in_message":[],"_revision_number":6},{"id":"1f789a6f227aa95dbfa19cf8194d5fb51e1d901e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 12:03:47.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"08fb30e2714a54e94b783cbb5d58fe88935fc92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 12:22:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9057d7cd33a049e481e76e0e0519265f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53cd7b4fa8584d30bde52f5530dbaf15 : SUCCESS in 9m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a80afc5d5c6414ebf438e3a37565456 : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1791315f503b448eb7d18661f7d1a495 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23819edb4860468b968192d7e8dbc0d8 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f520190dc0e46479afbc5276f59640f : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0677e859efdf4614979ff6fcdb6882c3 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e92b9138588c4d24976a3d9ee20d8797 : SUCCESS in 13m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c8acf38ec2a54043b488645bd9bf40e5 : SUCCESS in 6m 57s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/df9801c7eeeb446cbdf6bc126da49eb9 : SUCCESS in 16m 23s","accounts_in_message":[],"_revision_number":7},{"id":"0a49197006033c1fc40333566d0026efb4201499","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 12:30:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c5c775e771e53e9beb97d3b28095a63393b3ef4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 13:00:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/500ac9a967054688ab4481eede1492df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c727a98a0fb4c9495ec0ff0b587b760 : SUCCESS in 8m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/217c2b8fc622402f8a6a2fe8c455c7eb : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/349cee2c66bf472f9d5c01766e5b091e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3272a2d426c249a7807ba88a8b1a0109 : SUCCESS in 2m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/389387263f1b4cf3a72e8af2d92a1de5 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1280af1baa3d4c5b9623345cd1a83e26 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17536ed0faf94f1aba2464c55ce22349 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0023fb19dfc04cb09021e8972ce353ce : SUCCESS in 5m 59s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/1d3841d4bc2649b08f0e35eb3fc65d65 : SUCCESS in 23m 25s","accounts_in_message":[],"_revision_number":7},{"id":"c51dd2e5bf5c0f4c0ff5409a3c1f38b9c2eaa9b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-09 14:03:37.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"be13ee8a556e560962f483b83aa2b9a8ae8c8d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-09 15:45:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/509fa260d57d4622b98091cc65757f70\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e291bdc69c840d6a77f72f680576f9b : SUCCESS in 4m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/73dedf4b5f8c47e8ac7bfd0ff4fae807 : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75ae955700ae46c29fa14aaed138e7fc : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abd612a7e77420fb292377a6a73f012 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05b2d4917e09411b8f901132ebc2815b : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/311d3a42fdda4c52b90b86468b6bfa8a : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0f541f8a467462bb04b7d43e07386ce : SUCCESS in 6m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c093027c8dd3409298a22c923c3576f8 : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/79b39807470e4a1781527fee62c1a286 : SUCCESS in 8m 17s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f908e690b8944011b4940b47f44b8fce : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6eb08ccffd724cba99f5d51062ed718c : SUCCESS in 16m 27s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/746b10ce3125462491acee62292f971f : SUCCESS in 15m 12s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/f1e07eb2d33541eca61bbeba732fd973 : SUCCESS in 57m 16s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/fdd7e19a4fc24499995812c2af943ab2 : SUCCESS in 57m 17s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/e1f21307fbd84d649faeb5855204e708 : SUCCESS in 58m 36s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a1a851e4a80452bbce1b999263980e1 : SUCCESS in 56m 19s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e82a93a9d7e4586aa8d3ee95f7e79e6 : SUCCESS in 57m 17s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/71c8c5e00a81462eb49ecd5aee5fae9e : SUCCESS in 1h 31m 26s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a76f13c4aeb74cbf8b887252b8c498b0 : SUCCESS in 1h 25m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd92dabf2ad44d449d8cc39a0dd8fee9 : SUCCESS in 1h 25m 02s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e53304213d324bcd9b004bd9825f1f47 : SUCCESS in 1h 38m 52s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/321eca436a8240b4b8555d82038e0d22 : SUCCESS in 11m 00s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/51f2c748d06e4dad97917246a1e4612a : SUCCESS in 9m 24s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dfb712bd1054e50a371d456b5c44921 : SUCCESS in 13m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/749a08894b7b4d958ec6798fd2fa41dc : SUCCESS in 15m 00s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/66aceaf643bb4227b5ca311c4ba18280 : SUCCESS in 15m 57s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/faa924cd9c4841a0b1f8828e1a59696a : SUCCESS in 35m 18s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/42b48df051d6400c94d1eccc6d4a8eef : SUCCESS in 31m 48s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1f40a2fbb05473f80f7d79feb9f3c2c : SUCCESS in 16m 13s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b4ef27ffd0e74dfd8fcf058e63f6256e : SUCCESS in 10m 12s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9a8e0eb83bf843c6beebd45ad14ac32b : SUCCESS in 15m 14s","accounts_in_message":[],"_revision_number":8},{"id":"ec83ac149adc965f8cd3afb28a2110b87f2227b5","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-11 08:44:30.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"9da9213bdde4cfd24484a90ff815354fb61e591b","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-11 08:44:42.000000000","message":"Patch Set 8: -Code-Review","accounts_in_message":[],"_revision_number":8},{"id":"bd67a25fb996bdf3d7f7658927e4379007755b07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 15:19:03.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6dc265dc4227ff84a22cc9375667bd97fd83db7d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 15:19:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"969a2cfc82ed35db05cd700b15cf6a7e6bd26b40","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-11 15:44:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"aa3b69a5d394004672f895c72f1c6fccc14800d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 17:00:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa85e1baea82435995868ac0986851f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aae0d9041264d5fb98b3c70b3615ad1 : SUCCESS in 9m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/017b54d0da3c4f6aa31171970498bbfa : SUCCESS in 6m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d3e8df316c42279f3906d0457a19e8 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca776b8c0a644161a2ea2cf472d986dc : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d326d544b53e4ba88306e63910fc5522 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa74c0a0e8834571b35c923a2d6b7b37 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bd5b642a2714eb4a5e1c459bf5883a9 : SUCCESS in 13m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ff40c968c74476386975aec2023bd5f : SUCCESS in 12m 28s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/28f043f994d9461ea2e617b332d1bd64 : SUCCESS in 14m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fafbbec3ea742328a0a5170286450dc : SUCCESS in 11m 08s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/bbb230f872d243b3b594f3f565bbaa79 : FAILURE in 8m 47s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9e8182076ef4f4ca99a8cefad4a6b2f : SUCCESS in 17m 04s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/1a7dbaa94af4496991730c538a74b6f3 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a40d1dc546e54d14abbd48cbbf1aaa59 : SUCCESS in 56m 27s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/6a1ff1afbe1045a99b7add6d9780aa8f : SUCCESS in 1h 03m 46s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4ff9276a5eaa4c339cce95e2bdfa192e : SUCCESS in 1h 03m 07s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a5001dea4eff44b7908f372d3f07a0d2 : SUCCESS in 1h 06m 01s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/324dffa707994006990cfcc32ac35a58 : SUCCESS in 1h 39m 01s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/60c365ad2c074ec5a3c21deea6cc0c05 : SUCCESS in 1h 38m 51s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1799e74ac5cd4268b20b36e584d6094b : SUCCESS in 1h 31m 33s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a8886361761425d9df0fd96da7358e2 : SUCCESS in 1h 39m 35s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/0f11ef5003044d1a989ea19a04db0bc7 : SUCCESS in 16m 51s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/207abeba69634e49b345d7ae0aaa47e3 : SUCCESS in 11m 05s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b84a7e32005844a4bbe8cd4e73671f8a : SUCCESS in 19m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/95bde0058ede49189bb64f8c312160a3 : SUCCESS in 14m 02s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/5d1953ac77074aa3ac1bced9b1ca2d83 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/53dedb0e505c4727b3bc1a242dbb32de : SUCCESS in 25m 28s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a0b8c7f5c1784faf84f20184912097b6 : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/00d739ab032b45fbb9ed552d3d145fa2 : SUCCESS in 30m 41s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/44cda261d8284d2c9b2a021d8abe59b3 : SUCCESS in 10m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e021671540c5488ebdf8a3077d7d440f : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":9},{"id":"08625ba351ccf7abdf6ed2a859127906fd27c182","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-11 17:39:24.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"80f6c2a8bd7dda35867aa8da6bda523213687883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 17:57:36.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"2205f2d113c6e2f3c468d53f2ff07c672b01042d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 17:58:44.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f1b605adadf3e5c88c8b8d990674f1416c74cbb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-11 18:18:43.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"8947c851ce1dc87b3690a8c1a415d8e83a26447c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-11 19:51:04.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d13c0f250e55470a93800e838015dcba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420e1a15e475491c9b3aa16631c61df5 : SUCCESS in 6m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/61fb03db36ba45e4800a87b2f5cddad3 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a426106007564acabfa370070612f69f : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9353e8cb04814e30aa2d9e55a621a997 : SUCCESS in 3m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59c09d7dc6164a8da2ed2970f388a96b : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6b4b57732e346edbd2ec6158b35a87c : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6737b38c8f1a49339b0ee329fa4f24c5 : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d70249f07b7348a3a04f834afbc50ac7 : SUCCESS in 11m 10s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/b094fd0e20a34181a282cc3efa618d66 : SUCCESS in 15m 47s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b0b672c80894b3597eb6bb02be60dca : SUCCESS in 13m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/461c4c9dcbca4333b6d7895c4cd070ac : SUCCESS in 12m 15s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d43750de87244ff9b66517bf93f5a1ca : SUCCESS in 15m 22s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/2b95d2e7145742dda66cba8f9ffda608 : SUCCESS in 56m 18s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/9cf6b80c4ddb44bfa77084201eeb81e2 : SUCCESS in 58m 47s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d896de4f7e0547d3b8d5b3b0f51e463f : SUCCESS in 56m 56s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36f6d3c4097a4024804341052f36a491 : SUCCESS in 55m 37s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/c34d7f5727ac4a1aa5ce6d6fe5055785 : SUCCESS in 55m 39s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/7dac31bbcb6342ad95df4d231b373c01 : SUCCESS in 1h 31m 23s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/2ce7a21c6d144104b286d0a6b254c33d : SUCCESS in 1h 26m 55s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae8c02406d9b45fcb393069ff11f1ba1 : SUCCESS in 1h 23m 19s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9102d697b50e4377a0800f8191ec144c : SUCCESS in 1h 28m 06s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/fc23b969a3fc4b3a837163da30ca6afe : SUCCESS in 11m 21s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/1eb1f7f1f7bb422386f0a523698623a5 : SUCCESS in 11m 28s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c64fce136d2d45c787af15df3ba08114 : SUCCESS in 9m 32s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0757549c0cb2491e9c80acacbad4eea6 : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/59f29f68238e42b088390ce9159d6453 : SUCCESS in 17m 33s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/43997eac48cf465db31d11722f60a463 : SUCCESS in 20m 25s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6b5cd749005f45dd92747eade698f594 : SUCCESS in 27m 47s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ec1552b07e9481ab1d30afa4571db45 : SUCCESS in 15m 19s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/0d5e302184c1441ab8e71e745424d059 : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/925f09ead3304791b52a5463e1f3eaea : SUCCESS in 15m 55s","accounts_in_message":[],"_revision_number":12},{"id":"5671631597480627b88ece041c542fab11009b5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-12 09:35:51.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"4b0883eadbc2d0c70164aee492ad015f25a0e19c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 10:10:26.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6a8b4a62c8484147a548878e11b0e1b597949d47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 10:10:58.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"498350938b22715afc3eeb746125622ca60ded17","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 11:15:05.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6b19482d3e2068b30222f138d84bdcb979c4245e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 11:46:30.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f090572aa1049798f21a141e08469ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b1266b4715431d9532279fd77af39a : SUCCESS in 4m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b77d1a0456fa4e4aaa99e75ff6f883da : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2118410a1eeb45f899b084c271b705bd : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/185dcd7b068d4c7eb72ce86d1625720f : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ef7a5d536ba478c8a23c0d84619e40f : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9926bf680a4f44dea812477e3ef88ec9 : SUCCESS in 3m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c806b3f3939e4fa690da1f9627670b12 : SUCCESS in 17m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/499d7e43ebf046178728d3e2a742ce6d : SUCCESS in 12m 06s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/a893c3838e6141eaa29b62cc0435b0d4 : SUCCESS in 18m 21s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fd4f0611b304a3082557c2bf2861a55 : SUCCESS in 14m 52s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/a490a2a8046145298fc627bdb8d90f15 : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/305090792b9243b095dc8e9a2badf353 : SUCCESS in 16m 40s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c9236e5f718a400f9ddbb99d9384886f : SUCCESS in 53m 30s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a4be11f2b7a64474a73bb039789eb4ed : SUCCESS in 54m 22s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/b69303e73baa4bae874be157b471a1df : FAILURE in 9m 26s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d954cfeb2ee64b7d9146b0b0e77a949b : SUCCESS in 57m 23s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8084cdd93ca94dcf9779764f4cffe141 : SUCCESS in 58m 07s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/d6e34105b3754902a636643da6f0a076 : SUCCESS in 1h 32m 34s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a446a756baff48e283653774166dd317 : SUCCESS in 1h 23m 39s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49172f85c8f54fe3adec9522bba41137 : SUCCESS in 1h 24m 09s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ae05b13d29614d5a98163927362d52e6 : SUCCESS in 1h 34m 24s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/950f4c47e51942d78e8e87a5813430de : SUCCESS in 10m 02s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/427abfd38cb14a488abdcccabf9a7b22 : SUCCESS in 23m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9e8ca4200ad40f2aa216bd447472e11 : SUCCESS in 16m 47s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a75b2ae07061403d988a8a2f9012e5f3 : SUCCESS in 8m 24s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/06085706a6f14c42a4b5ceb371022309 : SUCCESS in 33m 06s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/6159c69aaf25410898d6849f4900891a : SUCCESS in 18m 12s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75e45b29a2f847dab359a491024a0dc0 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59ce1a92784a47db94935e3634426a40 : SUCCESS in 29m 16s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/88bfea2e6a3a4f40abf814b91d9cbb75 : SUCCESS in 17m 38s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8cdae65173eb4c53aed5b585c810f9ac : SUCCESS in 15m 16s","accounts_in_message":[],"_revision_number":14},{"id":"6e345ab119c022efdf57bcf8c49995f945614055","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 12:14:23.000000000","message":"Patch Set 14:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3444f14a5dfb48108da7004827effc7f\n\n- kayobe-infra-vm-centos10s https://zuul.opendev.org/t/openstack/build/4175f6eda9cc4fadad73d38a21b03505 : SUCCESS in 12m 41s (non-voting)\n- kayobe-infra-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/389bae8a729e467e83c805614e22429c : SUCCESS in 10m 22s (non-voting)\n- kayobe-infra-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/e42e5748d7c54ad4a8c995398f295e9f : SUCCESS in 10m 55s\n- kayobe-infra-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/a1b1476f99654d818e6626e27ddde920 : SUCCESS in 20m 01s (non-voting)\n- kayobe-overcloud-centos10s https://zuul.opendev.org/t/openstack/build/38679ec7c0874d08a684da636f2172ab : SUCCESS in 51m 36s (non-voting)\n- kayobe-overcloud-host-configure-centos10s https://zuul.opendev.org/t/openstack/build/af80327f6d024b79adc2fe8833113a66 : SUCCESS in 16m 46s (non-voting)\n- kayobe-overcloud-tls-centos10s https://zuul.opendev.org/t/openstack/build/530fc23da5d143caa99f813e817bcdc7 : SUCCESS in 51m 41s (non-voting)\n- kayobe-seed-centos10s https://zuul.opendev.org/t/openstack/build/5bff0a0e7826474b83efa92b2ace1399 : SUCCESS in 13m 47s (non-voting)\n- kayobe-seed-images-centos10s https://zuul.opendev.org/t/openstack/build/1d63ef310dd04314962338173a233a48 : FAILURE in 9m 30s (non-voting)\n- kayobe-seed-images-rocky10 https://zuul.opendev.org/t/openstack/build/b6218aecd06449ce8ce6bfe62a250a32 : FAILURE in 9m 31s\n- kayobe-seed-images-rocky10-podman https://zuul.opendev.org/t/openstack/build/e272e3916dbe4caa8b0d95f249bd218c : FAILURE in 14m 14s\n- kayobe-seed-images-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1053e85c65ae4c1bad08fbfd6bda3753 : SUCCESS in 53m 57s\n- kayobe-seed-images-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/74860a219049416a9142ce1c420708ec : FAILURE in 16m 44s\n- kayobe-seed-vm-centos10s https://zuul.opendev.org/t/openstack/build/bf4fc89980dc4c5aab54565418b17f19 : SUCCESS in 8m 54s (non-voting)\n- kayobe-seed-vm-centos10s-cloud-image https://zuul.opendev.org/t/openstack/build/465d335b4afe4a2db68690deaac33f8c : SUCCESS in 10m 48s (non-voting)\n- kayobe-seed-vm-centos10s-q35 https://zuul.opendev.org/t/openstack/build/106c501cc90245ca856bededf1f32ca6 : SUCCESS in 13m 07s (non-voting)\n- kayobe-seed-vm-rocky10-cloud-image https://zuul.opendev.org/t/openstack/build/7ff0d66a811d43a1ac60c6c24c59499c : SUCCESS in 11m 50s\n- kayobe-seed-vm-rocky10-q35 https://zuul.opendev.org/t/openstack/build/ae71e0cc95984da5bcfc8542683c6555 : SUCCESS in 21m 38s\n- kayobe-seed-vm-ubuntu-noble-cloud-image https://zuul.opendev.org/t/openstack/build/00471878132740649b3c23626844a3e6 : SUCCESS in 22m 03s (non-voting)\n- kayobe-seed-vm-ubuntu-noble-q35 https://zuul.opendev.org/t/openstack/build/7f7906bb975942ef9a945a73624c6e7c : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":14},{"id":"bd5a9686c43e7f3f1f033e8a32cf14a14afa1ca7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 14:46:42.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2d30bc50ea387e4d9fa49ebb1c440d08200a276f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 16:35:39.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1520389a8d5645a0a58571da427a9767\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e454b38d6a6437393a78937cc4c05fd : SUCCESS in 9m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ffaaa515d188427e9252882e2a38b578 : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccc54ad8621b4733838b49ac56980b04 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeacf25713814a62af37d27d3e883dd6 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f43688eb76104f9bb60418a7e0e528b1 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/afb2be12beae429db4e982f7009491e1 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fc592a5315c451188ef14bb90f4bcc7 : SUCCESS in 15m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94ffc6ecdd78477f90c01e97d1b18bb2 : SUCCESS in 12m 00s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/e00949ad1336447eb248c90a252b6823 : SUCCESS in 16m 45s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ca17b5d9369442718cc3ad2d2a57b9a0 : SUCCESS in 8m 24s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/905e9de718924908a30dbd08a9dfaa58 : SUCCESS in 25m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85a44f2c9a714c5da61fb8acb9a0de94 : SUCCESS in 19m 38s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/26b34247c8034febb3acd54d9018a53a : FAILURE in 8m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6c170ca7d20342c592b35350bce016b2 : SUCCESS in 57m 11s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/85f1440091a74259877b7e9592959bef : SUCCESS in 58m 00s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d87f28c9f9740f09a336e48e7cca087 : SUCCESS in 58m 43s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/87fab77aee974deba65329e570ed0dad : SUCCESS in 55m 45s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/e7c18ea913a64b4ba972f04775cc1346 : SUCCESS in 1h 32m 14s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/1a74bd3c8292467bb645d967b715d238 : SUCCESS in 1h 32m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/751cae8644284e409d674558433c2cce : SUCCESS in 1h 32m 01s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/55924310e4ef49b897fdf2b64fa376e9 : SUCCESS in 1h 42m 53s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/7a99d63ef9df4b789113a646b7dd698a : SUCCESS in 11m 29s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/cc4eb060e4b644c1a668db97a8c7581f : SUCCESS in 23m 17s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f856640729a541ae96613339433040f2 : SUCCESS in 16m 12s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/16d9220a99d74f779fdadf3c2f14ef5d : SUCCESS in 16m 01s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/631fc5f367e3489099e3ffab7799aa63 : SUCCESS in 17m 32s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/22c0ef52ccd0462fb811624158d459c3 : SUCCESS in 19m 09s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60c7acb7571849028d79da4b1a1361b9 : SUCCESS in 19m 10s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/20703e482fa247edb1764c9a12db4fc6 : SUCCESS in 31m 33s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/160a9c0bdc9a489487bddc2ca6bf842f : SUCCESS in 18m 46s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e58e07fa83254c1da37022850dd97870 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":14},{"id":"1b37485bd7e2354292317eafb330086282b901a2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 18:08:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0a87f77794dad797119731f389acb1bd0741d6b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2026-02-12 18:09:02.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"787625375e77ceb7b0dbf36ea67dc287d5ace281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 19:54:27.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7047ccc928524843ba02a0cb84f43050\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b47f920b3c4839b5c0a5cd24ee5163 : SUCCESS in 4m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0951f3d425644cbbbf575b0991d4f748 : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc713d9321d8434793e42ac6ff1a3852 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4c52ed739f343f898519b328a165739 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cf758832fa7428cb93a316281a0d349 : SUCCESS in 2m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe8b97cd34d14338b2dbd23f51dff49f : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22f00d69e463423faada5cf5c0c83a66 : SUCCESS in 9m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9d2233530b974585b5e9d9499336bb32 : SUCCESS in 13m 17s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/c7878bc7e9d14cffb3c5494136fe14fd : SUCCESS in 14m 46s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7bfed3ef59d64e4c9cded1bcd4a8e7f9 : SUCCESS in 14m 12s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/7dda8ef3e7cf401eba17d0b944eac059 : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/750b1b41484b4fd78fd534ee51e6b65b : SUCCESS in 14m 20s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/ab716f3e74f84281a5ffdefd036ddabb : SUCCESS in 1h 01m 37s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/55f9d246d471424eb3e1655cd0cf02a8 : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/d661b6b5193141b98973e0a49cbe934b : SUCCESS in 1h 05m 04s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/91e8472503644270ade57c953e38b60f : SUCCESS in 1h 09m 03s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1136c027c124414aa220ded988dd8b72 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9deebcbbfa7341168f43addac56b5b58 : SUCCESS in 1h 43m 37s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/0e42738b2da349428684f42625156955 : SUCCESS in 1h 34m 40s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f0929690c6aa42368ef37226d009768f : SUCCESS in 1h 29m 04s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ee9e56f2fd36411eb6258ea64ecdac46 : SUCCESS in 1h 38m 14s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/b476cd4000d64fd8a25b1aa6d3e5bcba : SUCCESS in 10m 17s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/c85311eff21844dfb6395006a1fae8f6 : SUCCESS in 9m 41s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7c8aae40ad8247579b782bb98c470bb0 : SUCCESS in 8m 23s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/70fe7c164ab04ecc816b6f0e64aaf58d : SUCCESS in 16m 40s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/cf8c15356f1549b9830081ca8fbea738 : SUCCESS in 32m 00s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/a9a7576e2b4845fd9381b347282f9803 : SUCCESS in 16m 48s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/75542e6949c94e68906d012395631edc : SUCCESS in 28m 57s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/605d9278fb7a4e8e98d5e22043ec760c : SUCCESS in 30m 21s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/021bf837084d4da6bea8b156b1b39af0 : SUCCESS in 9m 07s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/908950a41fcc4a43947904b4ef764bf3 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":14},{"id":"1855722e677f8491795367b6653f83e22752d1a3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2026-02-13 17:30:44.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"f02d040d1cdf669ed7debd2e841058d02f07d41f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-13 18:01:26.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"5bd8f1f116e169f542cd057cca7b1eca4082f7e4","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-02-16 13:12:13.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"6cf7566749942d0dae55f2663425740fd4725599","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 13:12:32.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"590f89e2c9f479e7da10315cc1710830421af350","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 14:51:57.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14473ac98f964486bf77dce0240cfc9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ba6826a8d14417a8e0832afaad1520 : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d4a0f793b8448f08a18e88e48507b00 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf52c4add0fc44ef86140041f480cf5b : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd5e484408ab40ce88c58b0580116fb8 : SUCCESS in 4m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dd1b9608a464046840efb7cb01aebaf : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1660a9adb82f47b49c7210c6532fbc1f : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/abf00c6adf2049b487e4cb7f57467ca6 : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/293f4641e5694ec7a5dbe08938b55d0a : SUCCESS in 18m 04s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57336e8b0263429286112e3b9144897e : SUCCESS in 13m 47s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/d4a233a03fb543d789d83bc3310f880d : SUCCESS in 23m 53s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10bc36f56c624cf9b97823a1d1d89e9f : SUCCESS in 8m 47s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/dff0ab2b51804f4eb1cee3d1444df16a : SUCCESS in 55m 20s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/a8c91f6a23594ba5b786e0cd7032dcf1 : SUCCESS in 54m 49s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/96cde1650c5d48caa9bf272b8d3585a1 : SUCCESS in 57m 25s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fec0acdb4cb46db815e01c0fbda7363 : SUCCESS in 55m 11s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6e8aa2f77a2942d2a9dc82bd077f94e0 : SUCCESS in 55m 13s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9e7c6af919f140ad85fdb0fb8534815e : SUCCESS in 1h 30m 41s\n- kayobe-overcloud-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/72109ecba9e7439485ecabadae5731bc : SUCCESS in 1h 34m 33s\n- kayobe-overcloud-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/759480e09a384a1ab1c9a7bf54f81193 : SUCCESS in 1h 21m 07s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95572ed7ba944f3ba4ce5e48b10b95c7 : SUCCESS in 1h 24m 31s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/548cf70725544e83b016ad4b5644c007 : SUCCESS in 14m 43s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/b7c35b653a7249a1be5afa0976efd4cc : SUCCESS in 15m 14s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/60ca27cdc43d44319f731dc9d5434810 : SUCCESS in 16m 49s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a613545f323a4f22acb2cc34139d0700 : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/fd401bcef991454e8d5301191f665d38 : SUCCESS in 14m 51s\n- kayobe-seed-upgrade-slurp-rocky10 https://zuul.opendev.org/t/openstack/build/dc8e1986880946f3a5cdcd2cc9315f20 : SUCCESS in 17m 14s\n- kayobe-seed-upgrade-slurp-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cd7f852225e434185e7ed0924489a4f : SUCCESS in 15m 28s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/46f318d4b85f4bf28d376f1c11ce06bb : SUCCESS in 27m 11s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/6fb68ca7573349978f537577a1ada3f9 : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/04a849e814544ceab2b75f5474a1b562 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":14},{"id":"54372023ad27a817b2e1c8b5829ca95e0f93995f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 14:51:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"159169e380ebb221be00e1fd0cc1d198a717359c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 14:52:55.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf6b6907cda34a3c93c3c227736b3aee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ed720f490a4f72a75d9ef68b6ef589 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a5b72b0b4714ed670a4104de3439421e6f5a5d78","revisions":{"ff796b5636855d02675c9958bf124b32c73f4adf":{"kind":"REWORK","_number":1,"created":"2026-02-08 17:59:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/1"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-08 17:59:05.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-08 17:59:05.000000000","tz":60},"subject":"Bump all Ansible collections and roles","message":"Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff796b5636855d02675c9958bf124b32c73f4adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff796b5636855d02675c9958bf124b32c73f4adf"}]},"branch":"refs/heads/master"},"1c4693e908816c3ec88dda837f0ee9afc9ebb7b9":{"kind":"REWORK","_number":2,"created":"2026-02-08 18:05:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/2"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-08 18:05:23.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-08 18:05:23.000000000","tz":60},"subject":"Bump all Ansible collections and roles","message":"Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c4693e908816c3ec88dda837f0ee9afc9ebb7b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1c4693e908816c3ec88dda837f0ee9afc9ebb7b9"}]},"branch":"refs/heads/master"},"46ca926d8bf7c4fa9cb05fd727a57ddf87f7d753":{"kind":"REWORK","_number":3,"created":"2026-02-09 09:00:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/3"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 08:59:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 08:59:52.000000000","tz":60},"subject":"Bump all Ansible collections and roles","message":"Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46ca926d8bf7c4fa9cb05fd727a57ddf87f7d753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46ca926d8bf7c4fa9cb05fd727a57ddf87f7d753"}]},"branch":"refs/heads/master"},"925f351ee08e296c3cd1f8f75e9bfdd952b3ffb9":{"kind":"REWORK","_number":4,"created":"2026-02-09 09:20:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/4"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 08:59:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 09:20:32.000000000","tz":60},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/925f351ee08e296c3cd1f8f75e9bfdd952b3ffb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/925f351ee08e296c3cd1f8f75e9bfdd952b3ffb9"}]},"branch":"refs/heads/master"},"3a7999be82ccb2c2e721ff8a9b9f13bdabb379fa":{"kind":"REWORK","_number":5,"created":"2026-02-09 10:33:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/5"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 10:33:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 10:33:15.000000000","tz":60},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7999be82ccb2c2e721ff8a9b9f13bdabb379fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3a7999be82ccb2c2e721ff8a9b9f13bdabb379fa"}]},"branch":"refs/heads/master"},"d1e8188ca8144f55b3278686769f1ea70f81c979":{"kind":"REWORK","_number":6,"created":"2026-02-09 10:59:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/6"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 10:59:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 10:59:04.000000000","tz":60},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1e8188ca8144f55b3278686769f1ea70f81c979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d1e8188ca8144f55b3278686769f1ea70f81c979"}]},"branch":"refs/heads/master"},"e0942ef25be1d64e90e08c490b7666ef9c5b1b9e":{"kind":"REWORK","_number":7,"created":"2026-02-09 12:03:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/7"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 12:03:40.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 12:03:40.000000000","tz":60},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0942ef25be1d64e90e08c490b7666ef9c5b1b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0942ef25be1d64e90e08c490b7666ef9c5b1b9e"}]},"branch":"refs/heads/master"},"47d9a69a9d913bcf0ce614906b3413ff69016177":{"kind":"REWORK","_number":8,"created":"2026-02-09 14:03:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/8"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 14:03:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 14:03:29.000000000","tz":60},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47d9a69a9d913bcf0ce614906b3413ff69016177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47d9a69a9d913bcf0ce614906b3413ff69016177"}]},"branch":"refs/heads/master"},"742ca4b117129a1f9240f43ff4f67a0b6850878b":{"kind":"REWORK","_number":9,"created":"2026-02-11 15:19:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/59/976059/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/9"}}},"commit":{"parents":[{"commit":"6142c9fcd7c0a565bf0edda36a5a471047c936c4","subject":"Merge \"CI: Add bandit testing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6142c9fcd7c0a565bf0edda36a5a471047c936c4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-09 14:03:29.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-11 15:19:02.000000000","tz":0},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/742ca4b117129a1f9240f43ff4f67a0b6850878b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/742ca4b117129a1f9240f43ff4f67a0b6850878b"}]},"branch":"refs/heads/master"},"0c8785eb33ab2c39476ca2e0c25552fdeab399a5":{"kind":"REWORK","_number":10,"created":"2026-02-11 17:39:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/10"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-11 17:39:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-11 17:39:13.000000000","tz":60},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c8785eb33ab2c39476ca2e0c25552fdeab399a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c8785eb33ab2c39476ca2e0c25552fdeab399a5"}]},"branch":"refs/heads/master"},"936ce2d167fa54466c8acb1105d4eee65cad7ef2":{"kind":"REWORK","_number":11,"created":"2026-02-11 17:57:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/59/976059/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/11"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-11 17:39:08.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-11 17:57:36.000000000","tz":0},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/936ce2d167fa54466c8acb1105d4eee65cad7ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/936ce2d167fa54466c8acb1105d4eee65cad7ef2"}]},"branch":"refs/heads/master"},"b17fc96965bed50b7afc510540bd9a95634134ce":{"kind":"REWORK","_number":12,"created":"2026-02-11 18:18:43.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/59/976059/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/12"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-11 17:39:08.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-11 18:18:43.000000000","tz":0},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b17fc96965bed50b7afc510540bd9a95634134ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b17fc96965bed50b7afc510540bd9a95634134ce"}]},"branch":"refs/heads/master"},"1abd3a1298716453fd17f388dfb3acdf92067f03":{"kind":"REWORK","_number":13,"created":"2026-02-12 09:35:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/59/976059/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/13"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-11 17:39:08.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2026-02-12 09:35:51.000000000","tz":0},"subject":"[WIP] Bump all Ansible collections and roles","message":"[WIP] Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1abd3a1298716453fd17f388dfb3acdf92067f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1abd3a1298716453fd17f388dfb3acdf92067f03"}]},"branch":"refs/heads/master"},"a5b72b0b4714ed670a4104de3439421e6f5a5d78":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2026-02-12 10:10:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/976059/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/976059/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/976059/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/976059/14"}}},"commit":{"parents":[{"commit":"06c025e7f4688c5f59b035948978b12d0f9f862b","subject":"Remove compatibility with Ansible 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06c025e7f4688c5f59b035948978b12d0f9f862b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-12 10:08:32.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2026-02-12 10:08:32.000000000","tz":60},"subject":"Bump all Ansible collections and roles","message":"Bump all Ansible collections and roles\n\nChange-Id: I3c793f1f63d5d220efd967dcea17beff4f861644\nSigned-off-by: Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5b72b0b4714ed670a4104de3439421e6f5a5d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5b72b0b4714ed670a4104de3439421e6f5a5d78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
